)]}'
{"id":"openstack%2Fmistral~605633","triplet_id":"openstack%2Fmistral~master~Ic98e2db02abd8483591756d73e06784cc2e9cbe3","project":"openstack/mistral","branch":"master","topic":"bug/1789680","hashtags":[],"change_id":"Ic98e2db02abd8483591756d73e06784cc2e9cbe3","subject":"Cleanup transport along RPC clients","status":"MERGED","created":"2018-09-27 09:49:04.000000000","updated":"2018-09-29 01:06:30.000000000","submitted":"2018-09-28 23:06:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"605633-1538175989533-fbf4e33a","meta_rev_id":"07fa04b040a97ab6962f8c1b6bf638511336cebc","_number":605633,"virtual_id_number":605633,"owner":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},{"value":2,"date":"2018-09-28 23:06:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11085,"name":"Toure Dunnon","email":"toure@redhat.com","username":"Toure"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":2,"date":"2018-09-28 20:01:15.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2018-09-27 14:48:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2018-09-27 17:26:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-09-27 13:36:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11085,"name":"Toure Dunnon","email":"toure@redhat.com","username":"Toure"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":1,"date":"2018-09-28 20:01:10.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11085,"name":"Toure Dunnon","email":"toure@redhat.com","username":"Toure"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":11085,"name":"Toure Dunnon","email":"toure@redhat.com","username":"Toure"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-27 13:36:20.000000000","updated_by":{"_account_id":11085,"name":"Toure Dunnon","email":"toure@redhat.com","username":"Toure"},"reviewer":{"_account_id":11085,"name":"Toure Dunnon","email":"toure@redhat.com","username":"Toure"},"state":"REVIEWER"},{"updated":"2018-09-27 14:48:42.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2018-09-27 17:26:10.000000000","updated_by":{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},"reviewer":{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},"state":"REVIEWER"},{"updated":"2018-09-28 20:01:10.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2018-09-28 23:06:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f077eca6c1ffe6d4e21f338dad4c3e0ea00ed72f","author":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"date":"2018-09-27 09:49:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ded0591d8c161a4e0ec9b606a5b25776d046b99","author":{"_account_id":11085,"name":"Toure Dunnon","email":"toure@redhat.com","username":"Toure"},"date":"2018-09-27 13:36:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"548d3e4651090798fc579ad8aa321aa4a7561f80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-27 14:39:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/33/605633/1/check/openstack-tox-pep8/df322b0/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/33/605633/1/check/openstack-tox-py27/e665595/ : SUCCESS in 9m 41s\n- openstack-tox-py35 http://logs.openstack.org/33/605633/1/check/openstack-tox-py35/121ffe4/ : SUCCESS in 9m 20s\n- openstack-tox-py36 http://logs.openstack.org/33/605633/1/check/openstack-tox-py36/84a4f29/ : SUCCESS in 9m 29s\n- openstack-tox-docs http://logs.openstack.org/33/605633/1/check/openstack-tox-docs/e8c60a0/html/ : SUCCESS in 6m 41s\n- openstack-tox-cover http://logs.openstack.org/33/605633/1/check/openstack-tox-cover/c6019ba/cover/ : SUCCESS in 10m 15s (non-voting)\n- mistral-devstack http://logs.openstack.org/33/605633/1/check/mistral-devstack/a601c24/ : SUCCESS in 45m 36s\n- mistral-devstack-non-apache http://logs.openstack.org/33/605633/1/check/mistral-devstack-non-apache/3f012ac/ : SUCCESS in 38m 57s\n- mistral-devstack-kombu http://logs.openstack.org/33/605633/1/check/mistral-devstack-kombu/5322b30/ : SUCCESS in 38m 09s\n- mistral-tox-unit-mysql http://logs.openstack.org/33/605633/1/check/mistral-tox-unit-mysql/f403673/ : SUCCESS in 30m 24s\n- mistral-tox-unit-postgresql http://logs.openstack.org/33/605633/1/check/mistral-tox-unit-postgresql/3fffe41/ : SUCCESS in 27m 47s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/33/605633/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/e2185ed/ : SUCCESS in 2h 17m 41s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/33/605633/1/check/tripleo-ci-centos-7-undercloud-containers/2136154/ : SUCCESS in 1h 41m 19s\n- mistral-rally-task http://logs.openstack.org/33/605633/1/check/mistral-rally-task/901e9ad/ : SUCCESS in 1h 36m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/33/605633/1/check/openstack-tox-lower-constraints/036e745/ : SUCCESS in 11m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5d1677e94660fb30f6d5a5a8b7fa5c7a6b4236c0","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2018-09-27 14:48:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cbac1569178274178dd708b605b2ff3c8c520ba2","author":{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},"date":"2018-09-27 17:26:10.000000000","message":"Patch Set 1: Code-Review+1\n\nI have a high degree of confidence in this fix.","accounts_in_message":[],"_revision_number":1},{"id":"87d11eecf0f9564c85d2d8d0ef4d62e22e3be759","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2018-09-28 20:01:10.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d7ab485fe90d11fd2c1469992ae90e75f8fdf173","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2018-09-28 20:01:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3837709cd596a592a97ea2c5ee0103bcf2c9f332","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-28 20:01:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"657c869536796971f687e5e445c283c7f308503b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-28 23:06:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/33/605633/1/gate/openstack-tox-pep8/d417381/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/33/605633/1/gate/openstack-tox-py27/f9b447e/ : SUCCESS in 8m 33s\n- openstack-tox-py35 http://logs.openstack.org/33/605633/1/gate/openstack-tox-py35/9df517f/ : SUCCESS in 11m 10s\n- openstack-tox-py36 http://logs.openstack.org/33/605633/1/gate/openstack-tox-py36/c0fe08f/ : SUCCESS in 9m 15s\n- openstack-tox-docs http://logs.openstack.org/33/605633/1/gate/openstack-tox-docs/b735089/html/ : SUCCESS in 7m 11s\n- mistral-devstack http://logs.openstack.org/33/605633/1/gate/mistral-devstack/332ad21/ : SUCCESS in 36m 36s\n- mistral-devstack-non-apache http://logs.openstack.org/33/605633/1/gate/mistral-devstack-non-apache/90a31f3/ : SUCCESS in 41m 12s\n- mistral-tox-unit-mysql http://logs.openstack.org/33/605633/1/gate/mistral-tox-unit-mysql/ce18706/ : SUCCESS in 31m 03s\n- mistral-tox-unit-postgresql http://logs.openstack.org/33/605633/1/gate/mistral-tox-unit-postgresql/983d625/ : SUCCESS in 29m 10s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/33/605633/1/gate/tripleo-ci-centos-7-scenario003-multinode-oooq-container/611ce3a/ : SUCCESS in 2h 45m 03s\n- mistral-devstack-kombu http://logs.openstack.org/33/605633/1/gate/mistral-devstack-kombu/b8d8bfc/ : SUCCESS in 47m 42s\n- mistral-rally-task http://logs.openstack.org/33/605633/1/gate/mistral-rally-task/d39e773/ : SUCCESS in 1h 35m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/33/605633/1/gate/openstack-tox-lower-constraints/7629d32/ : SUCCESS in 10m 52s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/33/605633/1/gate/tripleo-ci-centos-7-undercloud-containers/89da542/ : SUCCESS in 1h 48m 48s","accounts_in_message":[],"_revision_number":1},{"id":"dc4ccfa54f8eff2de7bee00c6ec05a9bfb03ca22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-28 23:06:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a7ce58935cff9c8c90b251300c2f266c766fd5de","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-09-29 01:05:54.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 4137f549800624d2d64da3fcf331d3ea441f53cc","accounts_in_message":[],"_revision_number":1},{"id":"1f7db9f1bc5447063fba4a0e20da043f7e36c053","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-09-29 01:06:06.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 204e96e8fc9c093f7468387bbda996d257a03be2","accounts_in_message":[],"_revision_number":1},{"id":"def5b01e6c029f5ce08c2a3a5cc1be875a69a5ff","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-09-29 01:06:30.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/pike as commit 3ccdf730c3c0c4426d8a0f3590b358f598b2df8b","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c005a7926060b6898b536962d2e88ea922f1a32","revisions":{"5c005a7926060b6898b536962d2e88ea922f1a32":{"kind":"REWORK","_number":1,"created":"2018-09-27 09:49:04.000000000","uploader":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"ref":"refs/changes/33/605633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/33/605633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/605633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/605633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/605633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/33/605633/1"}}},"commit":{"parents":[{"commit":"8720a2711b736b87ff4bf96239a52e278b0b6fa2","subject":"Merge \"Fix how Mistral calculates workflow output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8720a2711b736b87ff4bf96239a52e278b0b6fa2"}]}],"author":{"name":"Thomas Herve","email":"therve@redhat.com","date":"2018-09-27 09:45:42.000000000","tz":120},"committer":{"name":"Thomas Herve","email":"therve@redhat.com","date":"2018-09-27 09:45:42.000000000","tz":120},"subject":"Cleanup transport along RPC clients","message":"Cleanup transport along RPC clients\n\nThis fixes a bad weird condition in the API server related to\ncron-triggers and SIGHUP. The parent API server creates a RPC connection\nwhen creating workflows from cron. If a SIGUP signal happens after, the\nchild inherits the connection, but it\u0027s non-functional.\n\nChange-Id: Ic98e2db02abd8483591756d73e06784cc2e9cbe3\nCloses-Bug: #1789680\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5c005a7926060b6898b536962d2e88ea922f1a32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5c005a7926060b6898b536962d2e88ea922f1a32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
