)]}'
{"id":"openstack%2Fmistral~607960","triplet_id":"openstack%2Fmistral~master~I9cabb519a2c229b5269ae56c339160913322648a","project":"openstack/mistral","branch":"master","topic":"bug/1779071","hashtags":[],"change_id":"I9cabb519a2c229b5269ae56c339160913322648a","subject":"Fix state change propagation in workflows","status":"ABANDONED","created":"2018-10-04 13:35:05.000000000","updated":"2026-03-28 19:37:16.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"33b27a69118abd6e80332cac10eb535e538cc84b","_number":607960,"virtual_id_number":607960,"owner":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-01-30 12:33:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-09 10:43:02.000000000","updated_by":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"reviewer":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"state":"REVIEWER"},{"updated":"2020-01-30 12:33:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f70d41e576e7015666bfcd27b925d82c7fca17ab","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2018-10-04 13:35:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a7ebda6bf36aee12da8bf4b0aec905f1f9f646d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-04 17:39:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/60/607960/1/check/openstack-tox-pep8/7981733/ : SUCCESS in 6m 03s\n- openstack-tox-py27 http://logs.openstack.org/60/607960/1/check/openstack-tox-py27/7abfd0d/ : FAILURE in 8m 50s\n- openstack-tox-py35 http://logs.openstack.org/60/607960/1/check/openstack-tox-py35/0f3c156/ : FAILURE in 9m 21s\n- openstack-tox-py36 http://logs.openstack.org/60/607960/1/check/openstack-tox-py36/a7bacd3/ : FAILURE in 8m 51s\n- openstack-tox-docs http://logs.openstack.org/60/607960/1/check/openstack-tox-docs/2bae5bf/html/ : SUCCESS in 4m 52s\n- openstack-tox-cover http://logs.openstack.org/60/607960/1/check/openstack-tox-cover/bad90f7/ : FAILURE in 9m 32s (non-voting)\n- mistral-devstack http://logs.openstack.org/60/607960/1/check/mistral-devstack/1658cb3/ : SUCCESS in 39m 49s\n- mistral-devstack-non-apache http://logs.openstack.org/60/607960/1/check/mistral-devstack-non-apache/2f5779c/ : SUCCESS in 39m 53s\n- mistral-devstack-kombu http://logs.openstack.org/60/607960/1/check/mistral-devstack-kombu/9799c2c/ : SUCCESS in 41m 58s\n- mistral-tox-unit-mysql http://logs.openstack.org/60/607960/1/check/mistral-tox-unit-mysql/6205b33/ : FAILURE in 30m 03s\n- mistral-tox-unit-postgresql http://logs.openstack.org/60/607960/1/check/mistral-tox-unit-postgresql/20b55a8/ : FAILURE in 28m 24s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/60/607960/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/654649e/ : FAILURE in 2h 15m 04s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/60/607960/1/check/tripleo-ci-centos-7-undercloud-containers/4c04376/ : SUCCESS in 1h 53m 32s\n- mistral-rally-task http://logs.openstack.org/60/607960/1/check/mistral-rally-task/9bfa323/ : SUCCESS in 1h 46m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/607960/1/check/openstack-tox-lower-constraints/703fe12/ : FAILURE in 10m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d316f5a94fcd4d869c6cc45a28a9708b1d7136fb","author":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"date":"2018-10-09 09:59:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"45c6cb2509a405f2cd5a596819e78244c43798ef","author":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"date":"2018-10-09 10:43:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"08ffc460726af2c47c1d80f7515e9405cf365bf1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-09 12:59:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/60/607960/2/check/openstack-tox-pep8/7302b9b/ : SUCCESS in 6m 03s\n- openstack-tox-py27 http://logs.openstack.org/60/607960/2/check/openstack-tox-py27/37642b8/ : FAILURE in 9m 22s\n- openstack-tox-py35 http://logs.openstack.org/60/607960/2/check/openstack-tox-py35/94088aa/ : FAILURE in 10m 00s\n- openstack-tox-py36 http://logs.openstack.org/60/607960/2/check/openstack-tox-py36/37c5a78/ : FAILURE in 9m 59s\n- openstack-tox-docs http://logs.openstack.org/60/607960/2/check/openstack-tox-docs/51e9817/html/ : SUCCESS in 4m 46s\n- openstack-tox-cover http://logs.openstack.org/60/607960/2/check/openstack-tox-cover/48a3b63/ : FAILURE in 9m 42s (non-voting)\n- mistral-devstack http://logs.openstack.org/60/607960/2/check/mistral-devstack/b0fe972/ : SUCCESS in 37m 08s\n- mistral-devstack-non-apache http://logs.openstack.org/60/607960/2/check/mistral-devstack-non-apache/5e1455b/ : SUCCESS in 40m 31s\n- mistral-devstack-kombu http://logs.openstack.org/60/607960/2/check/mistral-devstack-kombu/5fc5b57/ : SUCCESS in 44m 12s\n- mistral-tox-unit-mysql http://logs.openstack.org/60/607960/2/check/mistral-tox-unit-mysql/7c53393/ : FAILURE in 29m 48s\n- mistral-tox-unit-postgresql http://logs.openstack.org/60/607960/2/check/mistral-tox-unit-postgresql/c58b08b/ : FAILURE in 28m 11s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/60/607960/2/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/3297532/ : SUCCESS in 2h 31m 11s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/60/607960/2/check/tripleo-ci-centos-7-undercloud-containers/5bd8a87/ : FAILURE in 1h 47m 51s\n- mistral-rally-task http://logs.openstack.org/60/607960/2/check/mistral-rally-task/1ca0257/ : SUCCESS in 1h 30m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/607960/2/check/openstack-tox-lower-constraints/ab2e355/ : FAILURE in 11m 11s","accounts_in_message":[],"_revision_number":2},{"id":"62d6daf96fdacbcb62ee367199cdea688aa187dc","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2018-10-18 10:28:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84a3f847bd0f3cb1676f9218dd95518ad8947eb7","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2018-10-18 11:00:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8c042f3dda55838acbd2446e14680c3685907104","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-18 13:42:19.000000000","message":"Patch Set 3: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/60/607960/3/check/openstack-tox-pep8/8962563/ : FAILURE in 6m 52s\n- openstack-tox-py27 http://logs.openstack.org/60/607960/3/check/openstack-tox-py27/5d8844f/ : FAILURE in 9m 23s\n- openstack-tox-py35 http://logs.openstack.org/60/607960/3/check/openstack-tox-py35/06b75db/ : FAILURE in 9m 42s\n- openstack-tox-py36 http://logs.openstack.org/60/607960/3/check/openstack-tox-py36/0619698/ : FAILURE in 10m 28s\n- openstack-tox-docs http://logs.openstack.org/60/607960/3/check/openstack-tox-docs/71fb089/html/ : SUCCESS in 5m 25s\n- openstack-tox-cover http://logs.openstack.org/60/607960/3/check/openstack-tox-cover/baee2be/ : FAILURE in 11m 24s (non-voting)\n- mistral-devstack http://logs.openstack.org/60/607960/3/check/mistral-devstack/e0ba667/ : SUCCESS in 40m 17s\n- mistral-devstack-non-apache http://logs.openstack.org/60/607960/3/check/mistral-devstack-non-apache/a41d059/ : SUCCESS in 1h 07m 45s\n- mistral-devstack-kombu http://logs.openstack.org/60/607960/3/check/mistral-devstack-kombu/6d13873/ : SUCCESS in 37m 50s\n- mistral-tox-unit-mysql http://logs.openstack.org/60/607960/3/check/mistral-tox-unit-mysql/d04c93e/ : FAILURE in 30m 36s\n- mistral-tox-unit-postgresql http://logs.openstack.org/60/607960/3/check/mistral-tox-unit-postgresql/9a6557a/ : FAILURE in 31m 47s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/60/607960/3/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/052b7f3/ : SUCCESS in 2h 24m 28s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/60/607960/3/check/tripleo-ci-centos-7-undercloud-containers/cef85b6/ : SUCCESS in 1h 49m 40s\n- mistral-rally-task http://logs.openstack.org/60/607960/3/check/mistral-rally-task/ff5b0ec/ : FAILURE in 1h 55m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/60/607960/3/check/openstack-tox-lower-constraints/beea711/ : FAILURE in 11m 56s","accounts_in_message":[],"_revision_number":3},{"id":"2c24e2b243ea252acb62b8250381bc4f52d5c666","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-01-30 09:30:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7568918f5ae4e7e6bf44283c09c4fcb2720524e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:33:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de442bec6b6f4168a1eccdc7ac78851e : FAILURE in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c1b969f14504b04b7b310345d11281a : FAILURE in 8m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9accfae3a57e48dfa04db8969e900b8d : FAILURE in 9m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/916c0acf565142579b9529b66d09e192 : FAILURE in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9861659a2f7943cfacaa3ac9729b50c9 : SUCCESS in 8m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ce54841e74649efb3e534771a27409b : FAILURE in 8m 13s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c39a7ecb7e79466dbc9b8eeeb29a605f : SUCCESS in 49m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8af3ab96ad348df9f5b5acd8893d671 : SUCCESS in 44m 04s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67962b50ca204e25a4c9914b506ed3e5 : SUCCESS in 41m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a8d8a35a56ca40aabf2c6ca3b1feaf04 : SUCCESS in 41m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6f280d32ee424f0daefd31428bccf791 : SUCCESS in 40m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f42da084a11740879582ade43cf863f1 : FAILURE in 29m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e937cd333d6a45b095c850acee582a7e : FAILURE in 26m 48s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d6413bd986db4697929b7b3a311b80ad : SUCCESS in 2h 58m 38s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2acd13c1e17c47118d61b50993a86ea2 : SUCCESS in 56m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f77b4356bce49b1bc828ccbc89823b3 : FAILURE in 9m 51s","accounts_in_message":[],"_revision_number":4},{"id":"33b27a69118abd6e80332cac10eb535e538cc84b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-03-28 19:37:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e5f55a545314b5daea759b2926099f2fc502a1d6","revisions":{"91517c15e64afeb55b2213f7a636247ad6117143":{"kind":"REWORK","_number":1,"created":"2018-10-04 13:35:05.000000000","uploader":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"ref":"refs/changes/60/607960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/607960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/607960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/607960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/607960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/607960/1"}}},"commit":{"parents":[{"commit":"f0b49a82b4c4d00961f051f69470325c8c481999","subject":"Add a release note for Ic98e2db02abd8483591756d73e06784cc2e9cbe3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f0b49a82b4c4d00961f051f69470325c8c481999"}]}],"author":{"name":"Andras Kovi","email":"akovi@nokia.com","date":"2018-10-04 08:42:00.000000000","tz":120},"committer":{"name":"Andras Kovi","email":"akovi@nokia.com","date":"2018-10-04 13:31:17.000000000","tz":120},"subject":"Fix state change propagation in workflows","message":"Fix state change propagation in workflows\n\nThe workflow state may be updated concurrently from many sources,\nthus, if the state update (guarded by state conditions) does\nnot change the state in the DB, then the current state must be\nobtained from the DB and should be used to propagate the changes.\n\nCloses-Bug##1779071\n\nChange-Id: I9cabb519a2c229b5269ae56c339160913322648a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/91517c15e64afeb55b2213f7a636247ad6117143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/91517c15e64afeb55b2213f7a636247ad6117143"}]},"branch":"refs/heads/master"},"aaf6e74924298f9b5bbcc954ca7714fef94a99ef":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-10-09 09:59:48.000000000","uploader":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"ref":"refs/changes/60/607960/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/607960/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/607960/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/607960/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/607960/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/607960/2"}}},"commit":{"parents":[{"commit":"f0b49a82b4c4d00961f051f69470325c8c481999","subject":"Add a release note for Ic98e2db02abd8483591756d73e06784cc2e9cbe3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f0b49a82b4c4d00961f051f69470325c8c481999"}]}],"author":{"name":"Andras Kovi","email":"akovi@nokia.com","date":"2018-10-04 08:42:00.000000000","tz":120},"committer":{"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","date":"2018-10-09 09:59:37.000000000","tz":0},"subject":"Fix state change propagation in workflows","message":"Fix state change propagation in workflows\n\nThe workflow state may be updated concurrently from many sources,\nthus, if the state update (guarded by state conditions) does\nnot change the state in the DB, then the current state must be\nobtained from the DB and should be used to propagate the changes.\n\nCloses-Bug: #1779071\nChange-Id: I9cabb519a2c229b5269ae56c339160913322648a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aaf6e74924298f9b5bbcc954ca7714fef94a99ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aaf6e74924298f9b5bbcc954ca7714fef94a99ef"}]},"branch":"refs/heads/master"},"cabac805580b06326fbc847209170a31acec591f":{"kind":"REWORK","_number":3,"created":"2018-10-18 11:00:05.000000000","uploader":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"ref":"refs/changes/60/607960/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/607960/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/607960/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/607960/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/607960/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/607960/3"}}},"commit":{"parents":[{"commit":"9be7e928d683ff00aaa6617c5ba57456fe6a02cc","subject":"Remove remaining references to the rpc_backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9be7e928d683ff00aaa6617c5ba57456fe6a02cc"}]}],"author":{"name":"Andras Kovi","email":"akovi@nokia.com","date":"2018-10-04 08:42:00.000000000","tz":120},"committer":{"name":"Andras Kovi","email":"akovi@nokia.com","date":"2018-10-18 10:58:27.000000000","tz":120},"subject":"Fix state change propagation in workflows","message":"Fix state change propagation in workflows\n\nThe workflow state may be updated concurrently from many sources,\nthus, if the state update (guarded by state conditions) does\nnot change the state in the DB, then the current state must be\nobtained from the DB and should be used to propagate the changes.\n\nCloses-Bug: #1779071\nChange-Id: I9cabb519a2c229b5269ae56c339160913322648a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cabac805580b06326fbc847209170a31acec591f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cabac805580b06326fbc847209170a31acec591f"}]},"branch":"refs/heads/master"},"e5f55a545314b5daea759b2926099f2fc502a1d6":{"kind":"REWORK","_number":4,"created":"2020-01-30 09:30:39.000000000","uploader":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"ref":"refs/changes/60/607960/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/607960/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/607960/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/607960/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/607960/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/607960/4"}}},"commit":{"parents":[{"commit":"680a9e2e0791ddd1acf8968be641903a12b993f0","subject":"Merge \"wrapped the value of parameters in inputs(in wf get API) with \"\".\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/680a9e2e0791ddd1acf8968be641903a12b993f0"}]}],"author":{"name":"Andras Kovi","email":"akovi@nokia.com","date":"2018-10-04 08:42:00.000000000","tz":120},"committer":{"name":"Andras Kovi","email":"akovi@nokia.com","date":"2020-01-30 09:30:27.000000000","tz":60},"subject":"Fix state change propagation in workflows","message":"Fix state change propagation in workflows\n\nThe workflow state may be updated concurrently from many sources,\nthus, if the state update (guarded by state conditions) does\nnot change the state in the DB, then the current state must be\nobtained from the DB and should be used to propagate the changes.\n\nCloses-Bug: #1779071\nChange-Id: I9cabb519a2c229b5269ae56c339160913322648a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e5f55a545314b5daea759b2926099f2fc502a1d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e5f55a545314b5daea759b2926099f2fc502a1d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
