)]}'
{"id":"openstack%2Fmistral~607974","triplet_id":"openstack%2Fmistral~master~Ida3b502b25887ec9a7b51c4d6497699cc9466f05","project":"openstack/mistral","branch":"master","topic":"ironic_action","hashtags":[],"change_id":"Ida3b502b25887ec9a7b51c4d6497699cc9466f05","subject":"Use SessionClient for Ironic actions","status":"MERGED","created":"2018-10-04 14:35:13.000000000","updated":"2019-07-17 09:34:04.000000000","submitted":"2019-07-17 09:34:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"607974-1563356044719-b2961d89","meta_rev_id":"8e07345a6fc7dd36a4e57ab766f68bad1e4b1d21","_number":607974,"virtual_id_number":607974,"owner":{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-07-17 09:34:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-07-17 07:39:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-07-17 07:39:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-08 05:28:07.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2018-10-09 08:53:51.000000000","updated_by":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"reviewer":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"state":"REVIEWER"},{"updated":"2018-11-19 15:52:36.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-01-26 16:21:01.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2019-05-13 14:27:20.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-07-04 12:26:31.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-07-17 07:39:40.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-07-17 09:34:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab070ec1b56d05e46b25fc50927923efd05fd2a8","author":{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},"date":"2018-10-04 14:35:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99ae8d163d5036d87cd438ac6fd796380006cd1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-04 19:10:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/1/check/openstack-tox-pep8/8116053/ : SUCCESS in 6m 06s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/1/check/openstack-tox-py27/e0b3bb6/ : SUCCESS in 10m 01s\n- openstack-tox-py35 http://logs.openstack.org/74/607974/1/check/openstack-tox-py35/2b2ddcd/ : SUCCESS in 10m 11s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/1/check/openstack-tox-py36/252c74a/ : SUCCESS in 9m 30s\n- openstack-tox-docs http://logs.openstack.org/74/607974/1/check/openstack-tox-docs/086963b/html/ : SUCCESS in 5m 16s\n- openstack-tox-cover http://logs.openstack.org/74/607974/1/check/openstack-tox-cover/adddd61/cover/ : SUCCESS in 10m 16s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/1/check/mistral-devstack/8964c38/ : SUCCESS in 39m 30s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/1/check/mistral-devstack-non-apache/9f64b59/ : SUCCESS in 38m 27s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/1/check/mistral-devstack-kombu/b35fce1/ : POST_FAILURE in 39m 01s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/1/check/mistral-tox-unit-mysql/0c737be/ : SUCCESS in 31m 08s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/1/check/mistral-tox-unit-postgresql/e9a1edc/ : SUCCESS in 29m 05s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/74/607974/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/d8a8180/ : FAILURE in 2h 19m 57s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/607974/1/check/tripleo-ci-centos-7-undercloud-containers/7d870d2/ : SUCCESS in 1h 36m 54s\n- mistral-rally-task http://logs.openstack.org/74/607974/1/check/mistral-rally-task/85bae01/ : FAILURE in 1h 58m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/1/check/openstack-tox-lower-constraints/091131a/ : FAILURE in 10m 59s","accounts_in_message":[],"_revision_number":1},{"id":"514268d2c17eb4cecf384060d952add05da38f91","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2018-10-08 05:28:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e1fc40262a813e614c6411e1ecfd439f733dae27","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2018-10-08 05:28:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2251a6b0edeabfb6cf8b3bfaf0d6c709b6bc6410","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-08 08:21:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/1/check/openstack-tox-pep8/af04e01/ : SUCCESS in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/1/check/openstack-tox-py27/c0d4f6b/ : SUCCESS in 8m 46s\n- openstack-tox-py35 http://logs.openstack.org/74/607974/1/check/openstack-tox-py35/4e31dfd/ : SUCCESS in 10m 08s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/1/check/openstack-tox-py36/5c21968/ : SUCCESS in 10m 09s\n- openstack-tox-docs http://logs.openstack.org/74/607974/1/check/openstack-tox-docs/bb36256/html/ : SUCCESS in 4m 32s\n- openstack-tox-cover http://logs.openstack.org/74/607974/1/check/openstack-tox-cover/9811cbe/cover/ : SUCCESS in 10m 18s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/1/check/mistral-devstack/0311364/ : SUCCESS in 42m 33s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/1/check/mistral-devstack-non-apache/2f071a9/ : SUCCESS in 42m 52s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/1/check/mistral-devstack-kombu/5e7ba52/ : SUCCESS in 38m 54s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/1/check/mistral-tox-unit-mysql/01894ed/ : SUCCESS in 31m 26s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/1/check/mistral-tox-unit-postgresql/7b84eac/ : SUCCESS in 27m 59s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/74/607974/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/30d0dac/ : SUCCESS in 2h 52m 33s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/607974/1/check/tripleo-ci-centos-7-undercloud-containers/fb059c0/ : SUCCESS in 1h 40m 09s\n- mistral-rally-task http://logs.openstack.org/74/607974/1/check/mistral-rally-task/b65babd/ : SUCCESS in 1h 37m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/1/check/openstack-tox-lower-constraints/708b862/ : FAILURE in 10m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7074a94c24159bcb646ba2a2c2cc7f505ad0adc0","author":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"date":"2018-10-09 08:53:51.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a88ab20a48e63bc0df70f18539b1cfcebcdba9ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-09 11:45:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/1/check/openstack-tox-pep8/fe60121/ : SUCCESS in 5m 42s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/1/check/openstack-tox-py27/a67b21d/ : SUCCESS in 11m 46s\n- openstack-tox-py35 http://logs.openstack.org/74/607974/1/check/openstack-tox-py35/49f760e/ : SUCCESS in 9m 52s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/1/check/openstack-tox-py36/5bb080b/ : SUCCESS in 10m 41s\n- openstack-tox-docs http://logs.openstack.org/74/607974/1/check/openstack-tox-docs/294c62b/html/ : SUCCESS in 5m 19s\n- openstack-tox-cover http://logs.openstack.org/74/607974/1/check/openstack-tox-cover/52ab6f8/cover/ : SUCCESS in 9m 54s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/1/check/mistral-devstack/62c002c/ : SUCCESS in 46m 52s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/1/check/mistral-devstack-non-apache/c6891cc/ : SUCCESS in 41m 23s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/1/check/mistral-devstack-kombu/070473c/ : SUCCESS in 40m 04s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/1/check/mistral-tox-unit-mysql/36907d8/ : SUCCESS in 31m 05s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/1/check/mistral-tox-unit-postgresql/a880063/ : SUCCESS in 27m 57s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/74/607974/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/38c60b3/ : SUCCESS in 2h 48m 29s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/607974/1/check/tripleo-ci-centos-7-undercloud-containers/bc2e899/ : FAILURE in 1h 06m 00s\n- mistral-rally-task http://logs.openstack.org/74/607974/1/check/mistral-rally-task/cb48550/ : SUCCESS in 1h 39m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/1/check/openstack-tox-lower-constraints/952944a/ : FAILURE in 11m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f75f05c8726d7ea9f3a1ba265241c7469fc334ae","author":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"date":"2018-10-15 16:31:09.000000000","message":"Patch Set 1: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"328e59d14d89966d34707646173db052c8a645b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-15 22:09:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/1/check/openstack-tox-pep8/98d293d/ : SUCCESS in 5m 50s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/1/check/openstack-tox-py27/1fcc0c6/ : SUCCESS in 8m 42s\n- openstack-tox-py35 http://logs.openstack.org/74/607974/1/check/openstack-tox-py35/ff0a150/ : SUCCESS in 11m 22s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/1/check/openstack-tox-py36/4b2cb92/ : SUCCESS in 8m 18s\n- openstack-tox-docs http://logs.openstack.org/74/607974/1/check/openstack-tox-docs/ef33133/html/ : SUCCESS in 7m 20s\n- openstack-tox-cover http://logs.openstack.org/74/607974/1/check/openstack-tox-cover/b113732/cover/ : SUCCESS in 11m 37s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/1/check/mistral-devstack/dbdcb89/ : SUCCESS in 41m 17s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/1/check/mistral-devstack-non-apache/2a70729/ : SUCCESS in 49m 43s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/1/check/mistral-devstack-kombu/d98ee05/ : SUCCESS in 42m 16s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/1/check/mistral-tox-unit-mysql/ed5f5b3/ : SUCCESS in 33m 34s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/1/check/mistral-tox-unit-postgresql/b88ba7b/ : SUCCESS in 30m 47s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/74/607974/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/056da91/ : SUCCESS in 3h 18m 21s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/607974/1/check/tripleo-ci-centos-7-undercloud-containers/beb4480/ : SUCCESS in 1h 45m 38s\n- mistral-rally-task http://logs.openstack.org/74/607974/1/check/mistral-rally-task/247ffa0/ : SUCCESS in 1h 31m 59s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/1/check/openstack-tox-lower-constraints/dfc9c49/ : FAILURE in 12m 36s","accounts_in_message":[],"_revision_number":1},{"id":"fe2d353f1707cafd015117fd8fba2905121cee53","author":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"date":"2018-10-15 22:52:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b5be9ec0f63b29029313152e2e40b33c2bcd31ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-16 03:18:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/1/check/openstack-tox-pep8/366a1f5/ : SUCCESS in 6m 07s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/1/check/openstack-tox-py27/cf00992/ : SUCCESS in 9m 48s\n- openstack-tox-py35 http://logs.openstack.org/74/607974/1/check/openstack-tox-py35/7a80313/ : SUCCESS in 8m 47s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/1/check/openstack-tox-py36/44d20b3/ : SUCCESS in 10m 27s\n- openstack-tox-docs http://logs.openstack.org/74/607974/1/check/openstack-tox-docs/85fbe34/html/ : SUCCESS in 5m 47s\n- openstack-tox-cover http://logs.openstack.org/74/607974/1/check/openstack-tox-cover/1da51ad/cover/ : SUCCESS in 10m 47s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/1/check/mistral-devstack/4c9fbc6/ : SUCCESS in 38m 01s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/1/check/mistral-devstack-non-apache/1bc0627/ : SUCCESS in 38m 45s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/1/check/mistral-devstack-kombu/f0aeef5/ : SUCCESS in 39m 48s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/1/check/mistral-tox-unit-mysql/989e959/ : SUCCESS in 30m 37s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/1/check/mistral-tox-unit-postgresql/206931d/ : SUCCESS in 33m 32s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/74/607974/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/3b8e56a/ : SUCCESS in 2h 28m 05s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/607974/1/check/tripleo-ci-centos-7-undercloud-containers/f51f15c/ : SUCCESS in 1h 40m 15s\n- mistral-rally-task http://logs.openstack.org/74/607974/1/check/mistral-rally-task/cbcffe8/ : TIMED_OUT in 2h 31m 01s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/1/check/openstack-tox-lower-constraints/d98516f/ : FAILURE in 11m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4ec416dfc7fc3ba840118c072847ff5281f862fa","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-10-25 09:19:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"53b645d5646c6037b067bef440c8a0aa41561400","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-10-25 09:19:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"06013631bd41a549d10fb78dba38a3d117ff5879","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-25 13:31:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/2/check/openstack-tox-pep8/2cd4775/ : SUCCESS in 5m 52s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/2/check/openstack-tox-py27/fec15d3/ : FAILURE in 10m 42s\n- openstack-tox-py35 http://logs.openstack.org/74/607974/2/check/openstack-tox-py35/047203a/ : FAILURE in 8m 30s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/2/check/openstack-tox-py36/5a2ae6c/ : FAILURE in 11m 32s\n- openstack-tox-docs http://logs.openstack.org/74/607974/2/check/openstack-tox-docs/8e29e1a/html/ : SUCCESS in 6m 12s\n- openstack-tox-cover http://logs.openstack.org/74/607974/2/check/openstack-tox-cover/29fb4db/ : FAILURE in 9m 57s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/2/check/mistral-devstack/1848f55/ : SUCCESS in 44m 26s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/2/check/mistral-devstack-non-apache/9f817c6/ : SUCCESS in 39m 35s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/2/check/mistral-devstack-kombu/ad74c95/ : SUCCESS in 44m 11s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/2/check/mistral-tox-unit-mysql/703ccf9/ : FAILURE in 31m 27s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/2/check/mistral-tox-unit-postgresql/e9af540/ : FAILURE in 32m 09s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/74/607974/2/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/e8fb894/ : TIMED_OUT in 3h 20m 32s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/607974/2/check/tripleo-ci-centos-7-undercloud-containers/112e253/ : SUCCESS in 1h 53m 06s\n- mistral-rally-task http://logs.openstack.org/74/607974/2/check/mistral-rally-task/427ccdc/ : FAILURE in 1h 25m 11s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/2/check/openstack-tox-lower-constraints/9305132/ : FAILURE in 10m 57s","accounts_in_message":[],"_revision_number":2},{"id":"a49cfc016d8ae44557dda07baef7224298e194b0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-11-13 09:44:07.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"491ac921648c0d2707492d7bc489c62eb1b5c8c4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-11-13 09:44:10.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"53253c7f0168687aa5bbad50ac3d391aa88accab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-13 12:45:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/3/check/openstack-tox-pep8/5ec5965/ : SUCCESS in 9m 06s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/3/check/openstack-tox-py27/d15fef2/ : SUCCESS in 7m 52s\n- openstack-tox-py35 http://logs.openstack.org/74/607974/3/check/openstack-tox-py35/9527fcc/ : SUCCESS in 9m 02s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/3/check/openstack-tox-py36/c7effe2/ : SUCCESS in 10m 55s\n- openstack-tox-docs http://logs.openstack.org/74/607974/3/check/openstack-tox-docs/766206a/html/ : SUCCESS in 4m 40s\n- openstack-tox-cover http://logs.openstack.org/74/607974/3/check/openstack-tox-cover/d31201b/cover/ : SUCCESS in 11m 20s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/3/check/mistral-devstack/3bfd318/ : SUCCESS in 42m 16s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/3/check/mistral-devstack-non-apache/61b2d97/ : SUCCESS in 38m 05s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/3/check/mistral-devstack-kombu/5679e1b/ : SUCCESS in 42m 06s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/3/check/mistral-tox-unit-mysql/b2b115f/ : SUCCESS in 26m 21s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/3/check/mistral-tox-unit-postgresql/81b90e8/ : SUCCESS in 25m 23s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/74/607974/3/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/d756cab/ : SUCCESS in 2h 59m 35s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/607974/3/check/tripleo-ci-centos-7-undercloud-containers/daf9db6/ : SUCCESS in 1h 40m 20s\n- mistral-rally-task http://logs.openstack.org/74/607974/3/check/mistral-rally-task/37936ef/ : SUCCESS in 1h 00m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/3/check/openstack-tox-lower-constraints/8ec0156/ : FAILURE in 9m 38s","accounts_in_message":[],"_revision_number":3},{"id":"144bbe96d854bf3fdbe8341de3c90fc07d44e38f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-11-19 08:16:59.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"52a6975e8c14f75c810fd382ee727b1f015b9c9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-19 11:12:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/4/check/openstack-tox-pep8/7b533a4/ : SUCCESS in 6m 09s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/4/check/openstack-tox-py27/8be1eed/ : SUCCESS in 8m 29s\n- openstack-tox-py35 http://logs.openstack.org/74/607974/4/check/openstack-tox-py35/089ce97/ : SUCCESS in 8m 42s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/4/check/openstack-tox-py36/5012dc6/ : SUCCESS in 8m 28s\n- openstack-tox-docs http://logs.openstack.org/74/607974/4/check/openstack-tox-docs/a3d7496/html/ : SUCCESS in 6m 00s\n- openstack-tox-cover http://logs.openstack.org/74/607974/4/check/openstack-tox-cover/5cbadb1/cover/ : SUCCESS in 9m 59s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/4/check/mistral-devstack/84cceb6/ : SUCCESS in 43m 33s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/4/check/mistral-devstack-non-apache/3548e1a/ : SUCCESS in 45m 31s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/4/check/mistral-devstack-kombu/a8dba59/ : SUCCESS in 43m 04s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/4/check/mistral-tox-unit-mysql/05299a8/ : SUCCESS in 30m 03s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/4/check/mistral-tox-unit-postgresql/316f6c0/ : SUCCESS in 25m 48s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/74/607974/4/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/f797f74/ : SUCCESS in 2h 48m 45s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/607974/4/check/tripleo-ci-centos-7-undercloud-containers/6a18bbb/ : FAILURE in 2h 53m 40s\n- mistral-rally-task http://logs.openstack.org/74/607974/4/check/mistral-rally-task/dcce52f/ : SUCCESS in 58m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/4/check/openstack-tox-lower-constraints/c9dbe01/ : FAILURE in 10m 08s","accounts_in_message":[],"_revision_number":4},{"id":"9dcc946885b10d068b1034e8c325f3b7b475bf8f","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2018-11-19 11:17:11.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"750359f65ba460b4664be350798022ad83c12fc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-19 14:01:42.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/4/check/openstack-tox-pep8/a3002c7/ : SUCCESS in 5m 45s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/4/check/openstack-tox-py27/9236a97/ : SUCCESS in 8m 53s\n- openstack-tox-py35 http://logs.openstack.org/74/607974/4/check/openstack-tox-py35/4c22214/ : SUCCESS in 8m 55s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/4/check/openstack-tox-py36/9396f53/ : SUCCESS in 8m 48s\n- openstack-tox-docs http://logs.openstack.org/74/607974/4/check/openstack-tox-docs/aacf10a/html/ : SUCCESS in 4m 20s\n- openstack-tox-cover http://logs.openstack.org/74/607974/4/check/openstack-tox-cover/581c0c3/cover/ : SUCCESS in 9m 01s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/4/check/mistral-devstack/3596fc7/ : SUCCESS in 49m 12s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/4/check/mistral-devstack-non-apache/e58d3ec/ : SUCCESS in 40m 19s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/4/check/mistral-devstack-kombu/6e6fe25/ : SUCCESS in 43m 50s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/4/check/mistral-tox-unit-mysql/ed080ca/ : SUCCESS in 27m 37s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/4/check/mistral-tox-unit-postgresql/51ca8dc/ : SUCCESS in 25m 33s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/74/607974/4/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/6128bfb/ : SUCCESS in 2h 42m 09s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/607974/4/check/tripleo-ci-centos-7-undercloud-containers/c1fb39f/ : SUCCESS in 1h 36m 05s\n- mistral-rally-task http://logs.openstack.org/74/607974/4/check/mistral-rally-task/aab8bb9/ : SUCCESS in 1h 08m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/4/check/openstack-tox-lower-constraints/9d3f665/ : FAILURE in 9m 57s","accounts_in_message":[],"_revision_number":4},{"id":"f0a9c10f8a96ebc627bc22f16f48557c2bddaf34","author":{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},"date":"2018-11-19 14:06:54.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"422783dc7fbfe26b201e227193c28e7a19600626","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2018-11-19 15:52:36.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"246295312a6f9c802a761f649079289557884ee9","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2018-11-19 15:52:47.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"6b3353b293c8a6825609a0db449fc47917d1bcda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-19 17:42:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/5/check/openstack-tox-pep8/243f8e9/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/5/check/openstack-tox-py27/4a09b84/ : SUCCESS in 8m 32s\n- openstack-tox-py35 http://logs.openstack.org/74/607974/5/check/openstack-tox-py35/eaa9ac2/ : SUCCESS in 8m 33s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/5/check/openstack-tox-py36/abefca7/ : SUCCESS in 11m 07s\n- openstack-tox-docs http://logs.openstack.org/74/607974/5/check/openstack-tox-docs/b99e09e/html/ : SUCCESS in 5m 06s\n- openstack-tox-cover http://logs.openstack.org/74/607974/5/check/openstack-tox-cover/9fec1fa/cover/ : SUCCESS in 9m 11s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/5/check/mistral-devstack/438bcf7/ : SUCCESS in 41m 34s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/5/check/mistral-devstack-non-apache/b249d22/ : SUCCESS in 36m 01s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/5/check/mistral-devstack-kombu/ff2d55c/ : FAILURE in 1h 09m 29s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/5/check/mistral-tox-unit-mysql/2dacc57/ : SUCCESS in 26m 08s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/5/check/mistral-tox-unit-postgresql/f6a9432/ : SUCCESS in 24m 29s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/74/607974/5/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/e31897a/ : SUCCESS in 2h 25m 04s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/607974/5/check/tripleo-ci-centos-7-undercloud-containers/4d10a64/ : TIMED_OUT in 3h 32m 20s\n- mistral-rally-task http://logs.openstack.org/74/607974/5/check/mistral-rally-task/37b06ee/ : SUCCESS in 1h 00m 30s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/5/check/openstack-tox-lower-constraints/1e8d63a/ : FAILURE in 9m 52s","accounts_in_message":[],"_revision_number":5},{"id":"3ec0d700f286a8790478c13e46ee9701bd08af94","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-11-27 04:34:08.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"b50bbffa99528439ad17e227c1374bc524a4a2f4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-11-27 04:34:15.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"634be9f198202dd2e91ae44c98ac5529d8101d8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 06:49:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/6/check/openstack-tox-pep8/986a82e/ : SUCCESS in 10m 55s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/6/check/openstack-tox-py27/1e95237/ : SUCCESS in 7m 44s\n- openstack-tox-py35 http://logs.openstack.org/74/607974/6/check/openstack-tox-py35/a475f59/ : SUCCESS in 9m 15s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/6/check/openstack-tox-py36/09fb4f7/ : SUCCESS in 8m 28s\n- openstack-tox-docs http://logs.openstack.org/74/607974/6/check/openstack-tox-docs/7453b28/html/ : SUCCESS in 5m 53s\n- openstack-tox-cover http://logs.openstack.org/74/607974/6/check/openstack-tox-cover/bab3b9d/cover/ : SUCCESS in 8m 57s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/6/check/mistral-devstack/191e825/ : SUCCESS in 55m 31s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/6/check/mistral-devstack-non-apache/71e37f9/ : SUCCESS in 40m 20s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/6/check/mistral-devstack-kombu/5da53aa/ : SUCCESS in 43m 13s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/6/check/mistral-tox-unit-mysql/23d1f75/ : SUCCESS in 27m 07s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/6/check/mistral-tox-unit-postgresql/57e93b6/ : SUCCESS in 26m 02s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/74/607974/6/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/b205c8c/ : SUCCESS in 2h 14m 08s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/607974/6/check/tripleo-ci-centos-7-undercloud-containers/45b336d/ : SUCCESS in 1h 47m 27s\n- mistral-rally-task http://logs.openstack.org/74/607974/6/check/mistral-rally-task/fd4d232/ : SUCCESS in 59m 47s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/6/check/openstack-tox-lower-constraints/22df423/ : FAILURE in 10m 42s","accounts_in_message":[],"_revision_number":6},{"id":"0d9f9a073e0b4efc059943c05dfbc87b5d2a5421","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-01-26 16:21:01.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"fc3b2ac0d647258c3e19dc211d954904534f59d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-26 18:35:53.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/6/check/openstack-tox-pep8/5e4f82e/ : SUCCESS in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/6/check/openstack-tox-py27/8979730/ : SUCCESS in 9m 38s\n- openstack-tox-py35 http://logs.openstack.org/74/607974/6/check/openstack-tox-py35/974f138/ : SUCCESS in 10m 03s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/6/check/openstack-tox-py36/0114c54/ : SUCCESS in 8m 18s\n- openstack-tox-docs http://logs.openstack.org/74/607974/6/check/openstack-tox-docs/4d0baff/html/ : SUCCESS in 5m 03s\n- openstack-tox-cover http://logs.openstack.org/74/607974/6/check/openstack-tox-cover/dc41a8b/cover/ : SUCCESS in 9m 36s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/6/check/mistral-devstack/3ef71c4/ : SUCCESS in 41m 06s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/6/check/mistral-devstack-non-apache/0aae761/ : SUCCESS in 35m 27s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/6/check/mistral-devstack-kombu/9774526/ : SUCCESS in 48m 48s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/6/check/mistral-tox-unit-mysql/a3ac2ce/ : SUCCESS in 27m 01s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/6/check/mistral-tox-unit-postgresql/9a56357/ : SUCCESS in 27m 06s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/74/607974/6/check/tripleo-ci-centos-7-scenario003-standalone/d002306/ : SUCCESS in 2h 09m 17s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/607974/6/check/tripleo-ci-centos-7-undercloud-containers/953f863/ : SUCCESS in 2h 02m 14s\n- mistral-rally-task http://logs.openstack.org/74/607974/6/check/mistral-rally-task/679ce4b/ : FAILURE in 1h 22m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/6/check/openstack-tox-lower-constraints/c3fd224/ : FAILURE in 9m 43s","accounts_in_message":[],"_revision_number":6},{"id":"687912816112ec49dce9d985eb0fbf12c1af2cca","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-02-01 08:56:28.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"5d4089acb9f703b8f7c7f5b3ec64409875f24f4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-01 10:55:40.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/7/check/openstack-tox-pep8/30236dd/ : SUCCESS in 5m 42s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/7/check/openstack-tox-py27/1ac9487/ : SUCCESS in 8m 59s\n- openstack-tox-py35 http://logs.openstack.org/74/607974/7/check/openstack-tox-py35/32b776c/ : SUCCESS in 8m 49s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/7/check/openstack-tox-py36/6d95abf/ : SUCCESS in 9m 05s\n- openstack-tox-docs http://logs.openstack.org/74/607974/7/check/openstack-tox-docs/cce2a46/html/ : SUCCESS in 5m 03s\n- openstack-tox-cover http://logs.openstack.org/74/607974/7/check/openstack-tox-cover/611cedb/cover/ : SUCCESS in 9m 26s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/7/check/mistral-devstack/77c2a1d/ : SUCCESS in 46m 12s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/7/check/mistral-devstack-non-apache/2df134e/ : SUCCESS in 48m 22s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/7/check/mistral-devstack-kombu/92f6983/ : SUCCESS in 48m 18s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/7/check/mistral-tox-unit-mysql/9ff2f41/ : SUCCESS in 28m 06s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/7/check/mistral-tox-unit-postgresql/4c53d52/ : SUCCESS in 26m 50s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/74/607974/7/check/tripleo-ci-centos-7-scenario003-standalone/a153d97/ : SUCCESS in 1h 43m 09s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/607974/7/check/tripleo-ci-centos-7-undercloud-containers/5ad8f8b/ : SUCCESS in 1h 53m 13s\n- mistral-rally-task http://logs.openstack.org/74/607974/7/check/mistral-rally-task/9442157/ : SUCCESS in 1h 06m 18s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/7/check/openstack-tox-lower-constraints/808626a/ : FAILURE in 10m 41s","accounts_in_message":[],"_revision_number":7},{"id":"b0ce386260ab1a03939b0940b74065ddae952ac4","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-05-09 09:51:27.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"3f870ea4c7318fbed23297bb08249849e8ab9d75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 11:57:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/8/check/openstack-tox-pep8/3fd1fc2/ : SUCCESS in 5m 21s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/8/check/openstack-tox-py27/4096558/ : SUCCESS in 8m 47s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/8/check/openstack-tox-py36/8286fb0/ : SUCCESS in 8m 13s\n- openstack-tox-py37 http://logs.openstack.org/74/607974/8/check/openstack-tox-py37/4351c26/ : SUCCESS in 8m 47s\n- openstack-tox-docs http://logs.openstack.org/74/607974/8/check/openstack-tox-docs/d564331/html/ : SUCCESS in 4m 35s\n- openstack-tox-cover http://logs.openstack.org/74/607974/8/check/openstack-tox-cover/2bcbcf2/cover/ : SUCCESS in 8m 50s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/8/check/mistral-devstack/f986fa8/ : SUCCESS in 44m 19s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/8/check/mistral-devstack-non-apache/ee8b296/ : SUCCESS in 46m 35s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/8/check/mistral-devstack-kombu/7a33a1d/ : SUCCESS in 43m 00s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/8/check/mistral-tox-unit-mysql/8674a0d/ : FAILURE in 27m 20s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/8/check/mistral-tox-unit-postgresql/527b31e/ : SUCCESS in 26m 35s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/74/607974/8/check/tripleo-ci-centos-7-scenario003-standalone/de268f7/ : SUCCESS in 1h 14m 27s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/607974/8/check/tripleo-ci-centos-7-containers-multinode/2d42407/ : SUCCESS in 1h 59m 37s\n- mistral-rally-task http://logs.openstack.org/74/607974/8/check/mistral-rally-task/49c4b11/ : SUCCESS in 1h 03m 52s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/8/check/openstack-tox-lower-constraints/8faaf6b/ : FAILURE in 11m 07s","accounts_in_message":[],"_revision_number":8},{"id":"7f23d81806f02681acf07859d3c92463c8c32f5e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-13 02:27:28.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"58a68abfcc0e9167b0b4b7070053be57833b1ed5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 04:45:16.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/8/check/openstack-tox-pep8/833787c/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/8/check/openstack-tox-py27/a43d6f8/ : SUCCESS in 10m 03s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/8/check/openstack-tox-py36/70e8de5/ : SUCCESS in 8m 21s\n- openstack-tox-py37 http://logs.openstack.org/74/607974/8/check/openstack-tox-py37/9aba72d/ : SUCCESS in 9m 28s\n- openstack-tox-docs http://logs.openstack.org/74/607974/8/check/openstack-tox-docs/0630256/html/ : SUCCESS in 4m 36s\n- openstack-tox-cover http://logs.openstack.org/74/607974/8/check/openstack-tox-cover/de2fe39/cover/ : SUCCESS in 9m 36s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/8/check/mistral-devstack/4507423/ : SUCCESS in 43m 25s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/8/check/mistral-devstack-non-apache/e10df3e/ : SUCCESS in 42m 20s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/8/check/mistral-devstack-kombu/a5eaa4d/ : SUCCESS in 34m 52s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/8/check/mistral-tox-unit-mysql/d16f330/ : SUCCESS in 29m 12s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/8/check/mistral-tox-unit-postgresql/529adf3/ : SUCCESS in 26m 02s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/74/607974/8/check/tripleo-ci-centos-7-scenario003-standalone/3307518/ : SUCCESS in 1h 15m 36s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/607974/8/check/tripleo-ci-centos-7-containers-multinode/1a59772/ : SUCCESS in 1h 57m 34s\n- mistral-rally-task http://logs.openstack.org/74/607974/8/check/mistral-rally-task/97b1d70/ : SUCCESS in 56m 01s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/8/check/openstack-tox-lower-constraints/eaf4c32/ : FAILURE in 10m 18s","accounts_in_message":[],"_revision_number":8},{"id":"c5e8e2d2375b8f7f7a581ba30f7b1c96d23d0b98","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-05-13 14:27:20.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"0fa93b9565fff726bb88ccf0fa597905d9350b98","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-14 09:24:56.000000000","message":"Patch Set 8:\n\nShould we bump a version of Ironic client in low constraints?","accounts_in_message":[],"_revision_number":8},{"id":"7513b2393f1b96fbf791587ef238810dbdc6946f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-20 07:52:01.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"08acdf8b855b8f0851f6d77adfcce6080f0c27b0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-20 07:52:19.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"0fd1807b11eee3b471306524ad96e575c652fdbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 10:33:37.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/9/check/openstack-tox-pep8/bace336/ : SUCCESS in 7m 03s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/9/check/openstack-tox-py27/9fc2d95/ : SUCCESS in 11m 01s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/9/check/openstack-tox-py36/4920792/ : SUCCESS in 9m 12s\n- openstack-tox-py37 http://logs.openstack.org/74/607974/9/check/openstack-tox-py37/9ce6973/ : SUCCESS in 12m 34s\n- openstack-tox-docs http://logs.openstack.org/74/607974/9/check/openstack-tox-docs/160ca31/html/ : SUCCESS in 5m 24s\n- openstack-tox-cover http://logs.openstack.org/74/607974/9/check/openstack-tox-cover/e431ba7/cover/ : SUCCESS in 10m 41s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/9/check/mistral-devstack/67e093c/ : SUCCESS in 40m 52s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/9/check/mistral-devstack-non-apache/e73d5c8/ : SUCCESS in 44m 52s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/9/check/mistral-devstack-kombu/918ac9d/ : SUCCESS in 48m 04s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/9/check/mistral-tox-unit-mysql/9a68637/ : SUCCESS in 28m 16s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/9/check/mistral-tox-unit-postgresql/25bd662/ : SUCCESS in 27m 22s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/74/607974/9/check/tripleo-ci-centos-7-scenario003-standalone/e5f297a/ : SUCCESS in 1h 26m 46s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/607974/9/check/tripleo-ci-centos-7-containers-multinode/496d571/ : SUCCESS in 2h 11m 18s\n- mistral-rally-task http://logs.openstack.org/74/607974/9/check/mistral-rally-task/c9d0c2c/ : SUCCESS in 1h 36m 19s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/9/check/openstack-tox-lower-constraints/d9599b6/ : FAILURE in 11m 21s","accounts_in_message":[],"_revision_number":9},{"id":"9b157d04ef7c21f589d934ccc913aea34af0c648","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-06-21 04:10:42.000000000","message":"Patch Set 9: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"2005461b4bdd45cbc8e3196d122b58e1bf3a15d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 06:26:01.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/9/check/openstack-tox-pep8/bb680f1/ : SUCCESS in 7m 27s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/9/check/openstack-tox-py27/99316a5/ : SUCCESS in 10m 04s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/9/check/openstack-tox-py36/c00809d/ : SUCCESS in 8m 59s\n- openstack-tox-py37 http://logs.openstack.org/74/607974/9/check/openstack-tox-py37/e40101b/ : SUCCESS in 9m 19s\n- openstack-tox-docs http://logs.openstack.org/74/607974/9/check/openstack-tox-docs/04feb5e/html/ : SUCCESS in 5m 05s\n- openstack-tox-cover http://logs.openstack.org/74/607974/9/check/openstack-tox-cover/324ecf3/cover/ : SUCCESS in 10m 56s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/9/check/mistral-devstack/06e3612/ : SUCCESS in 38m 37s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/9/check/mistral-devstack-non-apache/252ec5a/ : SUCCESS in 47m 00s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/9/check/mistral-devstack-kombu/d4e7b1e/ : SUCCESS in 36m 23s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/9/check/mistral-tox-unit-mysql/d4da9bc/ : SUCCESS in 27m 31s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/9/check/mistral-tox-unit-postgresql/eef246a/ : SUCCESS in 35m 32s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/74/607974/9/check/tripleo-ci-centos-7-scenario003-standalone/64b0f96/ : SUCCESS in 1h 10m 57s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/607974/9/check/tripleo-ci-centos-7-containers-multinode/587457a/ : SUCCESS in 2h 13m 26s\n- mistral-rally-task http://logs.openstack.org/74/607974/9/check/mistral-rally-task/6aadbc1/ : SUCCESS in 58m 06s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/9/check/openstack-tox-lower-constraints/d2b3a37/ : FAILURE in 25m 23s","accounts_in_message":[],"_revision_number":9},{"id":"99a23dbf17c79825c956fea36b7e7b4e57af6dbf","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-21 06:43:06.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"d6acacff7a7f4d3f7bbc5feaf51e26b1f6218fb0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-21 06:43:13.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"31ccfd11373c0e0b9d2620140e353972705bc309","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 09:13:50.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/10/check/openstack-tox-pep8/f491805/ : SUCCESS in 13m 15s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/10/check/openstack-tox-py27/4372c00/ : SUCCESS in 8m 59s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/10/check/openstack-tox-py36/a74cb0e/ : SUCCESS in 11m 59s\n- openstack-tox-py37 http://logs.openstack.org/74/607974/10/check/openstack-tox-py37/9768c03/ : SUCCESS in 17m 19s\n- openstack-tox-docs http://logs.openstack.org/74/607974/10/check/openstack-tox-docs/fd70909/html/ : SUCCESS in 13m 03s\n- openstack-tox-cover http://logs.openstack.org/74/607974/10/check/openstack-tox-cover/d350e0d/cover/ : SUCCESS in 11m 08s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/10/check/mistral-devstack/a454a96/ : SUCCESS in 39m 30s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/10/check/mistral-devstack-non-apache/3738bd6/ : SUCCESS in 43m 32s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/10/check/mistral-devstack-kombu/b67710f/ : SUCCESS in 46m 56s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/10/check/mistral-tox-unit-mysql/7c0b1cc/ : SUCCESS in 27m 59s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/10/check/mistral-tox-unit-postgresql/c2659d5/ : SUCCESS in 26m 58s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/74/607974/10/check/tripleo-ci-centos-7-scenario003-standalone/ca9f9e3/ : SUCCESS in 1h 12m 10s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/607974/10/check/tripleo-ci-centos-7-containers-multinode/933b26d/ : SUCCESS in 2h 03m 17s\n- mistral-rally-task http://logs.openstack.org/74/607974/10/check/mistral-rally-task/1a36ca2/ : SUCCESS in 1h 37m 19s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/10/check/openstack-tox-lower-constraints/ba17b31/ : FAILURE in 12m 04s","accounts_in_message":[],"_revision_number":10},{"id":"a6a46f590348d7ad8634cefe02197f2d9ceff598","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-04 12:26:31.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"8679b953490375a895f35a95b917af98a403fd73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 14:50:49.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/10/check/openstack-tox-pep8/93278b8/ : SUCCESS in 5m 52s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/10/check/openstack-tox-py27/8b6b947/ : SUCCESS in 8m 08s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/10/check/openstack-tox-py36/c640b7a/ : SUCCESS in 7m 50s\n- openstack-tox-py37 http://logs.openstack.org/74/607974/10/check/openstack-tox-py37/8dda21d/ : SUCCESS in 9m 32s\n- openstack-tox-docs http://logs.openstack.org/74/607974/10/check/openstack-tox-docs/3025109/html/ : SUCCESS in 4m 13s\n- openstack-tox-cover http://logs.openstack.org/74/607974/10/check/openstack-tox-cover/fff8014/cover/ : SUCCESS in 8m 08s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/10/check/mistral-devstack/a375172/ : SUCCESS in 51m 36s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/10/check/mistral-devstack-non-apache/2da382f/ : SUCCESS in 41m 20s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/10/check/mistral-devstack-kombu/b71af73/ : SUCCESS in 46m 51s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/10/check/mistral-tox-unit-mysql/ba9f1e6/ : SUCCESS in 27m 47s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/10/check/mistral-tox-unit-postgresql/6a2027b/ : SUCCESS in 25m 39s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/74/607974/10/check/tripleo-ci-centos-7-scenario003-standalone/91626e6/ : SUCCESS in 1h 17m 34s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/607974/10/check/tripleo-ci-centos-7-containers-multinode/5b36763/ : SUCCESS in 2h 09m 34s\n- mistral-rally-task http://logs.openstack.org/74/607974/10/check/mistral-rally-task/2a69df9/ : SUCCESS in 57m 23s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/10/check/openstack-tox-lower-constraints/d5274a3/ : FAILURE in 9m 53s","accounts_in_message":[],"_revision_number":10},{"id":"9f4a447fd9e54f4ad8a3395ed465cd83468c7bac","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-07 06:10:10.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"26bd472e3a002f28696694645b58853ecf432cd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-07 08:34:37.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/11/check/openstack-tox-pep8/1526ea1/ : SUCCESS in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/11/check/openstack-tox-py27/00a02db/ : SUCCESS in 8m 06s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/11/check/openstack-tox-py36/354cb9b/ : SUCCESS in 8m 42s\n- openstack-tox-py37 http://logs.openstack.org/74/607974/11/check/openstack-tox-py37/7227970/ : SUCCESS in 8m 37s\n- openstack-tox-docs http://logs.openstack.org/74/607974/11/check/openstack-tox-docs/3142681/html/ : SUCCESS in 4m 09s\n- requirements-check http://logs.openstack.org/74/607974/11/check/requirements-check/a95a8de/ : SUCCESS in 2m 52s\n- openstack-tox-cover http://logs.openstack.org/74/607974/11/check/openstack-tox-cover/571bb4e/cover/ : SUCCESS in 9m 52s (non-voting)\n- mistral-devstack http://logs.openstack.org/74/607974/11/check/mistral-devstack/0b3b58d/ : SUCCESS in 48m 35s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/11/check/mistral-devstack-non-apache/3801693/ : SUCCESS in 49m 21s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/11/check/mistral-devstack-kombu/b9162f6/ : SUCCESS in 52m 50s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/11/check/mistral-tox-unit-mysql/bdc00cc/ : SUCCESS in 28m 25s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/11/check/mistral-tox-unit-postgresql/11db8bf/ : SUCCESS in 28m 35s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/74/607974/11/check/tripleo-ci-centos-7-scenario003-standalone/eae74c6/ : SUCCESS in 1h 15m 52s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/607974/11/check/tripleo-ci-centos-7-containers-multinode/e57dd63/ : SUCCESS in 2h 16m 28s\n- mistral-rally-task http://logs.openstack.org/74/607974/11/check/mistral-rally-task/6c69bcd/ : SUCCESS in 52m 46s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/11/check/openstack-tox-lower-constraints/09ef3ac/ : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":11},{"id":"31ce5226fa1c86672e8c62ad824ad770ad95b9a2","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-17 07:39:40.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"1fdf2f76158a34050db9ae162021a3fc6e4f1c60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 07:39:52.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"554670c17dda36bc72b5b99f72ef201f9ed39120","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 09:34:04.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/74/607974/11/gate/openstack-tox-pep8/09e13b2/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/74/607974/11/gate/openstack-tox-py27/097058f/ : SUCCESS in 8m 08s\n- openstack-tox-py36 http://logs.openstack.org/74/607974/11/gate/openstack-tox-py36/9285561/ : SUCCESS in 8m 01s\n- openstack-tox-py37 http://logs.openstack.org/74/607974/11/gate/openstack-tox-py37/58f7cd4/ : SUCCESS in 8m 15s\n- openstack-tox-docs http://logs.openstack.org/74/607974/11/gate/openstack-tox-docs/f2aa1d2/html/ : SUCCESS in 4m 06s\n- requirements-check http://logs.openstack.org/74/607974/11/gate/requirements-check/a1a50ad/ : SUCCESS in 2m 32s\n- mistral-devstack http://logs.openstack.org/74/607974/11/gate/mistral-devstack/82d3507/ : SUCCESS in 48m 32s\n- mistral-devstack-non-apache http://logs.openstack.org/74/607974/11/gate/mistral-devstack-non-apache/8bc01da/ : SUCCESS in 45m 20s\n- mistral-tox-unit-mysql http://logs.openstack.org/74/607974/11/gate/mistral-tox-unit-mysql/3fbd1d1/ : SUCCESS in 27m 35s\n- mistral-tox-unit-postgresql http://logs.openstack.org/74/607974/11/gate/mistral-tox-unit-postgresql/4034ae5/ : SUCCESS in 26m 03s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/74/607974/11/gate/tripleo-ci-centos-7-scenario003-standalone/9644483/ : SUCCESS in 1h 25m 59s\n- mistral-devstack-kombu http://logs.openstack.org/74/607974/11/gate/mistral-devstack-kombu/3cf23cc/ : SUCCESS in 57m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/607974/11/gate/openstack-tox-lower-constraints/8bb28c2/ : SUCCESS in 9m 38s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/607974/11/gate/tripleo-ci-centos-7-undercloud-containers/31f44ea/ : SUCCESS in 1h 47m 52s","accounts_in_message":[],"_revision_number":11},{"id":"155d1a229b9584d2c0f02fd7f357a94bfe04d663","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 09:34:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"b6ebcb3bd080bf1617dbe32f102e09ce42aa7921","revisions":{"2b70995cb2261e01b5347eb6767e60b567abd1bc":{"kind":"REWORK","_number":1,"created":"2018-10-04 14:35:13.000000000","uploader":{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},"ref":"refs/changes/74/607974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/607974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/607974/1"}}},"commit":{"parents":[{"commit":"f0b49a82b4c4d00961f051f69470325c8c481999","subject":"Add a release note for Ic98e2db02abd8483591756d73e06784cc2e9cbe3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f0b49a82b4c4d00961f051f69470325c8c481999"}]}],"author":{"name":"Brad P. Crochet","email":"brad@redhat.com","date":"2018-10-04 14:33:44.000000000","tz":-240},"committer":{"name":"Brad P. Crochet","email":"brad@redhat.com","date":"2018-10-04 14:33:44.000000000","tz":-240},"subject":"Use SessionClient for Ironic actions","message":"Use SessionClient for Ironic actions\n\nThe HTTPClient for Ironic is deprecated and will be removed in Stein.\nUse the SessionClient for Ironic actions instead. Also uses\nendpoint_override param instead of endpoint, as that is also\ndeprecated.\n\nChange-Id: Ida3b502b25887ec9a7b51c4d6497699cc9466f05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2b70995cb2261e01b5347eb6767e60b567abd1bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2b70995cb2261e01b5347eb6767e60b567abd1bc"}]},"branch":"refs/heads/master"},"4e2fe11c91ae76983b1c8e5cb283531ca2ec78e6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-10-25 09:19:20.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/74/607974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/607974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/607974/2"}}},"commit":{"parents":[{"commit":"eb402e40ffbe9074d5a01b01ad7a844a75f2f46e","subject":"Merge \"Improve join by removing periodic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eb402e40ffbe9074d5a01b01ad7a844a75f2f46e"}]}],"author":{"name":"Brad P. Crochet","email":"brad@redhat.com","date":"2018-10-04 14:33:44.000000000","tz":-240},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2018-10-25 09:19:20.000000000","tz":0},"subject":"Use SessionClient for Ironic actions","message":"Use SessionClient for Ironic actions\n\nThe HTTPClient for Ironic is deprecated and will be removed in Stein.\nUse the SessionClient for Ironic actions instead. Also uses\nendpoint_override param instead of endpoint, as that is also\ndeprecated.\n\nChange-Id: Ida3b502b25887ec9a7b51c4d6497699cc9466f05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e2fe11c91ae76983b1c8e5cb283531ca2ec78e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e2fe11c91ae76983b1c8e5cb283531ca2ec78e6"}]},"branch":"refs/heads/master"},"edc8a86b9975dee03f9c70531d5b41916d796d96":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-11-13 09:44:07.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/74/607974/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/607974/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/607974/3"}}},"commit":{"parents":[{"commit":"4b83cef3af8586463898ed165db4866564dec48c","subject":"Merge \"Divide yaml input to save it into definitions separately.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4b83cef3af8586463898ed165db4866564dec48c"}]}],"author":{"name":"Brad P. Crochet","email":"brad@redhat.com","date":"2018-10-04 14:33:44.000000000","tz":-240},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2018-11-13 09:44:07.000000000","tz":0},"subject":"Use SessionClient for Ironic actions","message":"Use SessionClient for Ironic actions\n\nThe HTTPClient for Ironic is deprecated and will be removed in Stein.\nUse the SessionClient for Ironic actions instead. Also uses\nendpoint_override param instead of endpoint, as that is also\ndeprecated.\n\nChange-Id: Ida3b502b25887ec9a7b51c4d6497699cc9466f05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/edc8a86b9975dee03f9c70531d5b41916d796d96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/edc8a86b9975dee03f9c70531d5b41916d796d96"}]},"branch":"refs/heads/master"},"be64afeb7c6bc1781e18042d440ae6368af21b0b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2018-11-19 08:16:59.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/74/607974/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/607974/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/607974/4"}}},"commit":{"parents":[{"commit":"c9e08a883946540bf245dd372406642e3aa50808","subject":"Fix \"join\" when the last indirect inbound task failed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c9e08a883946540bf245dd372406642e3aa50808"}]}],"author":{"name":"Brad P. Crochet","email":"brad@redhat.com","date":"2018-10-04 14:33:44.000000000","tz":-240},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2018-11-19 08:16:59.000000000","tz":0},"subject":"Use SessionClient for Ironic actions","message":"Use SessionClient for Ironic actions\n\nThe HTTPClient for Ironic is deprecated and will be removed in Stein.\nUse the SessionClient for Ironic actions instead. Also uses\nendpoint_override param instead of endpoint, as that is also\ndeprecated.\n\nChange-Id: Ida3b502b25887ec9a7b51c4d6497699cc9466f05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/be64afeb7c6bc1781e18042d440ae6368af21b0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/be64afeb7c6bc1781e18042d440ae6368af21b0b"}]},"branch":"refs/heads/master"},"cffde94232ff3db3d45fcf4f1808418425574e3e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2018-11-19 14:06:54.000000000","uploader":{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},"ref":"refs/changes/74/607974/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/607974/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/607974/5"}}},"commit":{"parents":[{"commit":"916393eca80589bd7f89f5f4d1015d81be6701ff","subject":"Merge \"Fix error workbook example\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/916393eca80589bd7f89f5f4d1015d81be6701ff"}]}],"author":{"name":"Brad P. Crochet","email":"brad@redhat.com","date":"2018-10-04 14:33:44.000000000","tz":-240},"committer":{"name":"Brad P. Crochet","email":"brad@redhat.com","date":"2018-11-19 14:06:26.000000000","tz":-300},"subject":"Use SessionClient for Ironic actions","message":"Use SessionClient for Ironic actions\n\nThe HTTPClient for Ironic is deprecated and will be removed in Stein.\nUse the SessionClient for Ironic actions instead. Also uses\nendpoint_override param instead of endpoint, as that is also\ndeprecated.\n\nChange-Id: Ida3b502b25887ec9a7b51c4d6497699cc9466f05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cffde94232ff3db3d45fcf4f1808418425574e3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cffde94232ff3db3d45fcf4f1808418425574e3e"}]},"branch":"refs/heads/master"},"9101de37573a7c7869114df3eb3b2e596df62871":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2018-11-27 04:34:08.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/74/607974/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/607974/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/607974/6"}}},"commit":{"parents":[{"commit":"3153f1ca584450bdd1b983b8ebc13ef5cb809c0e","subject":"Set admin security context in the action execution checker thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3153f1ca584450bdd1b983b8ebc13ef5cb809c0e"}]}],"author":{"name":"Brad P. Crochet","email":"brad@redhat.com","date":"2018-10-04 14:33:44.000000000","tz":-240},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2018-11-27 04:34:08.000000000","tz":0},"subject":"Use SessionClient for Ironic actions","message":"Use SessionClient for Ironic actions\n\nThe HTTPClient for Ironic is deprecated and will be removed in Stein.\nUse the SessionClient for Ironic actions instead. Also uses\nendpoint_override param instead of endpoint, as that is also\ndeprecated.\n\nChange-Id: Ida3b502b25887ec9a7b51c4d6497699cc9466f05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9101de37573a7c7869114df3eb3b2e596df62871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9101de37573a7c7869114df3eb3b2e596df62871"}]},"branch":"refs/heads/master"},"14ac40cd01943d6753504725004c2b45206ffa73":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-02-01 08:56:28.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/74/607974/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/607974/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/607974/7"}}},"commit":{"parents":[{"commit":"81af1b48384d04400a3b988aca7d4b1fa9596af5","subject":"Process all task batches in wf output evaluation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/81af1b48384d04400a3b988aca7d4b1fa9596af5"}]}],"author":{"name":"Brad P. Crochet","email":"brad@redhat.com","date":"2018-10-04 14:33:44.000000000","tz":-240},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-02-01 08:56:28.000000000","tz":0},"subject":"Use SessionClient for Ironic actions","message":"Use SessionClient for Ironic actions\n\nThe HTTPClient for Ironic is deprecated and will be removed in Stein.\nUse the SessionClient for Ironic actions instead. Also uses\nendpoint_override param instead of endpoint, as that is also\ndeprecated.\n\nChange-Id: Ida3b502b25887ec9a7b51c4d6497699cc9466f05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14ac40cd01943d6753504725004c2b45206ffa73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14ac40cd01943d6753504725004c2b45206ffa73"}]},"branch":"refs/heads/master"},"a6ceba02593997de4474439f89863c491c07600a":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-05-09 09:51:27.000000000","uploader":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"ref":"refs/changes/74/607974/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/607974/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/607974/8"}}},"commit":{"parents":[{"commit":"8e59d27ba1fef4e495cca29677aba1c5532ba0d4","subject":"Merge \"Adding root_execution_id to the jinja executor function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8e59d27ba1fef4e495cca29677aba1c5532ba0d4"}]}],"author":{"name":"Brad P. Crochet","email":"brad@redhat.com","date":"2018-10-04 14:33:44.000000000","tz":-240},"committer":{"name":"dharmendra kushwaha","email":"dharmendra.kushwaha@india.nec.com","date":"2019-05-09 09:51:27.000000000","tz":0},"subject":"Use SessionClient for Ironic actions","message":"Use SessionClient for Ironic actions\n\nThe HTTPClient for Ironic is deprecated and will be removed in Stein.\nUse the SessionClient for Ironic actions instead. Also uses\nendpoint_override param instead of endpoint, as that is also\ndeprecated.\n\nChange-Id: Ida3b502b25887ec9a7b51c4d6497699cc9466f05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a6ceba02593997de4474439f89863c491c07600a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a6ceba02593997de4474439f89863c491c07600a"}]},"branch":"refs/heads/master"},"389d330bea135f86771600c462b6a957365b251a":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-06-20 07:52:01.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/74/607974/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/607974/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/607974/9"}}},"commit":{"parents":[{"commit":"74b2fffec299c2c96d94248296fd368843a77434","subject":"Merge \"Make more JSON fields in execution objects lazy-loaded\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/74b2fffec299c2c96d94248296fd368843a77434"}]}],"author":{"name":"Brad P. Crochet","email":"brad@redhat.com","date":"2018-10-04 14:33:44.000000000","tz":-240},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-06-20 07:52:01.000000000","tz":0},"subject":"Use SessionClient for Ironic actions","message":"Use SessionClient for Ironic actions\n\nThe HTTPClient for Ironic is deprecated and will be removed in Stein.\nUse the SessionClient for Ironic actions instead. Also uses\nendpoint_override param instead of endpoint, as that is also\ndeprecated.\n\nChange-Id: Ida3b502b25887ec9a7b51c4d6497699cc9466f05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/389d330bea135f86771600c462b6a957365b251a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/389d330bea135f86771600c462b6a957365b251a"}]},"branch":"refs/heads/master"},"1d53c3c7e5cd8b128bca96d301bbd06de79cdf2b":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-06-21 06:43:06.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/74/607974/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/607974/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/607974/10"}}},"commit":{"parents":[{"commit":"f75e7198c61da0547ab6ab7e96c797ac5c6657da","subject":"Merge \"Fix workflow execution cascade delete error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f75e7198c61da0547ab6ab7e96c797ac5c6657da"}]}],"author":{"name":"Brad P. Crochet","email":"brad@redhat.com","date":"2018-10-04 14:33:44.000000000","tz":-240},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-06-21 06:43:06.000000000","tz":0},"subject":"Use SessionClient for Ironic actions","message":"Use SessionClient for Ironic actions\n\nThe HTTPClient for Ironic is deprecated and will be removed in Stein.\nUse the SessionClient for Ironic actions instead. Also uses\nendpoint_override param instead of endpoint, as that is also\ndeprecated.\n\nChange-Id: Ida3b502b25887ec9a7b51c4d6497699cc9466f05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1d53c3c7e5cd8b128bca96d301bbd06de79cdf2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1d53c3c7e5cd8b128bca96d301bbd06de79cdf2b"}]},"branch":"refs/heads/master"},"b6ebcb3bd080bf1617dbe32f102e09ce42aa7921":{"kind":"REWORK","_number":11,"created":"2019-07-07 06:10:10.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/74/607974/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/607974/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/607974/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/607974/11"}}},"commit":{"parents":[{"commit":"f75e7198c61da0547ab6ab7e96c797ac5c6657da","subject":"Merge \"Fix workflow execution cascade delete error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f75e7198c61da0547ab6ab7e96c797ac5c6657da"}]}],"author":{"name":"Brad P. Crochet","email":"brad@redhat.com","date":"2018-10-04 14:33:44.000000000","tz":-240},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-07 06:10:05.000000000","tz":0},"subject":"Use SessionClient for Ironic actions","message":"Use SessionClient for Ironic actions\n\nThe HTTPClient for Ironic is deprecated and will be removed in Stein.\nUse the SessionClient for Ironic actions instead. Also uses\nendpoint_override param instead of endpoint, as that is also\ndeprecated.\n\nChange-Id: Ida3b502b25887ec9a7b51c4d6497699cc9466f05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b6ebcb3bd080bf1617dbe32f102e09ce42aa7921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b6ebcb3bd080bf1617dbe32f102e09ce42aa7921"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
