)]}'
{"id":"openstack%2Fmistral~621470","triplet_id":"openstack%2Fmistral~master~I0459206be5cc390853b9c69e8c5002568d1efa60","project":"openstack/mistral","branch":"master","topic":"openstackclient-version","hashtags":[],"change_id":"I0459206be5cc390853b9c69e8c5002568d1efa60","subject":"Support OpenStack services dynamic versions","status":"MERGED","created":"2018-12-03 04:45:07.000000000","updated":"2019-09-04 14:09:01.000000000","submitted":"2019-09-04 14:06:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"621470-1567606005773-233c4fa7","meta_rev_id":"58900235f161e4d968b1e4a50e17c8b2bfe1180d","_number":621470,"virtual_id_number":621470,"owner":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},{"value":0,"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-09-04 14:06:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19741,"name":"Hieu LE","email":"hieulq2@viettel.com.vn","username":"hieulq"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"recommended":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-09-03 01:25:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},{"value":1,"date":"2019-09-03 01:25:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2019-09-03 01:25:34.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},{"value":0,"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},{"value":2,"date":"2019-09-03 01:25:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19741,"name":"Hieu LE","email":"hieulq2@viettel.com.vn","username":"hieulq"},{"value":1,"date":"2019-09-03 01:25:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},{"value":0,"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},{"value":1,"date":"2019-09-04 03:57:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19741,"name":"Hieu LE","email":"hieulq2@viettel.com.vn","username":"hieulq"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":19741,"name":"Hieu LE","email":"hieulq2@viettel.com.vn","username":"hieulq"},{"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-07 01:42:04.000000000","updated_by":{"_account_id":19741,"name":"Hieu LE","email":"hieulq2@viettel.com.vn","username":"hieulq"},"reviewer":{"_account_id":19741,"name":"Hieu LE","email":"hieulq2@viettel.com.vn","username":"hieulq"},"state":"REVIEWER"},{"updated":"2019-05-07 01:48:17.000000000","updated_by":{"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},"reviewer":{"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},"state":"REVIEWER"},{"updated":"2019-05-13 01:11:02.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-06-06 01:55:12.000000000","updated_by":{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},"reviewer":{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},"state":"REVIEWER"},{"updated":"2019-08-29 12:00:47.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-08-29 12:06:58.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-09-02 07:15:29.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-09-04 03:57:43.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-09-04 14:06:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"edb8cba7166b50f07f896cb9ff826dead99e740e","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-12-03 04:45:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d8d665f18fd3d27d574608a5cf2d3aac0d45f84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-03 07:04:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/70/621470/1/check/openstack-tox-pep8/7abf464/ : SUCCESS in 5m 15s\n- openstack-tox-py27 http://logs.openstack.org/70/621470/1/check/openstack-tox-py27/1420bd1/ : SUCCESS in 8m 07s\n- openstack-tox-py35 http://logs.openstack.org/70/621470/1/check/openstack-tox-py35/20f3053/ : SUCCESS in 8m 39s\n- openstack-tox-py36 http://logs.openstack.org/70/621470/1/check/openstack-tox-py36/1593fe8/ : SUCCESS in 12m 19s\n- openstack-tox-docs http://logs.openstack.org/70/621470/1/check/openstack-tox-docs/f331c33/html/ : SUCCESS in 4m 35s\n- openstack-tox-cover http://logs.openstack.org/70/621470/1/check/openstack-tox-cover/9d2d970/cover/ : SUCCESS in 10m 04s (non-voting)\n- mistral-devstack http://logs.openstack.org/70/621470/1/check/mistral-devstack/24d418a/ : FAILURE in 46m 40s\n- mistral-devstack-non-apache http://logs.openstack.org/70/621470/1/check/mistral-devstack-non-apache/6940ef0/ : FAILURE in 41m 54s\n- mistral-devstack-kombu http://logs.openstack.org/70/621470/1/check/mistral-devstack-kombu/cf157c0/ : FAILURE in 34m 27s\n- mistral-tox-unit-mysql http://logs.openstack.org/70/621470/1/check/mistral-tox-unit-mysql/c361678/ : SUCCESS in 29m 39s\n- mistral-tox-unit-postgresql http://logs.openstack.org/70/621470/1/check/mistral-tox-unit-postgresql/df99abe/ : SUCCESS in 29m 03s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/70/621470/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/bba1588/ : SUCCESS in 2h 17m 46s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/621470/1/check/tripleo-ci-centos-7-undercloud-containers/b873c96/ : FAILURE in 1h 50m 41s\n- mistral-rally-task http://logs.openstack.org/70/621470/1/check/mistral-rally-task/31cb0fc/ : SUCCESS in 1h 06m 29s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/70/621470/1/check/openstack-tox-lower-constraints/6a1048f/ : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d5ced5c17bcdf66135b442e483e96767c1ce1fee","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2018-12-03 07:56:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1c6d341f73d220fe9694f8ae0fabdb74b5d42d88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-03 11:09:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/70/621470/1/check/openstack-tox-pep8/3fbab10/ : SUCCESS in 6m 48s\n- openstack-tox-py27 http://logs.openstack.org/70/621470/1/check/openstack-tox-py27/0c2faee/ : SUCCESS in 10m 41s\n- openstack-tox-py35 http://logs.openstack.org/70/621470/1/check/openstack-tox-py35/49ed7cb/ : SUCCESS in 8m 11s\n- openstack-tox-py36 http://logs.openstack.org/70/621470/1/check/openstack-tox-py36/a38ae9c/ : SUCCESS in 9m 20s\n- openstack-tox-docs http://logs.openstack.org/70/621470/1/check/openstack-tox-docs/8690ebe/html/ : SUCCESS in 6m 15s\n- openstack-tox-cover http://logs.openstack.org/70/621470/1/check/openstack-tox-cover/703a146/cover/ : SUCCESS in 12m 39s (non-voting)\n- mistral-devstack http://logs.openstack.org/70/621470/1/check/mistral-devstack/00d9af6/ : FAILURE in 45m 02s\n- mistral-devstack-non-apache http://logs.openstack.org/70/621470/1/check/mistral-devstack-non-apache/1f01014/ : FAILURE in 47m 59s\n- mistral-devstack-kombu http://logs.openstack.org/70/621470/1/check/mistral-devstack-kombu/bf46751/ : FAILURE in 52m 13s\n- mistral-tox-unit-mysql http://logs.openstack.org/70/621470/1/check/mistral-tox-unit-mysql/6b12bf6/ : SUCCESS in 28m 11s\n- mistral-tox-unit-postgresql http://logs.openstack.org/70/621470/1/check/mistral-tox-unit-postgresql/9ea3806/ : SUCCESS in 27m 51s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/70/621470/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/1a8d9a9/ : SUCCESS in 3h 10m 50s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/621470/1/check/tripleo-ci-centos-7-undercloud-containers/b3262df/ : FAILURE in 1h 42m 38s\n- mistral-rally-task http://logs.openstack.org/70/621470/1/check/mistral-rally-task/599f6eb/ : SUCCESS in 1h 02m 03s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/70/621470/1/check/openstack-tox-lower-constraints/5e6e2ad/ : SUCCESS in 9m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3356fb70ab5c7ca3bd16ceaa5dd52d37e38d3e1d","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2019-05-07 01:05:40.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"31ba2e425962394c9a17fb00699126326aca71ab","author":{"_account_id":19741,"name":"Hieu LE","email":"hieulq2@viettel.com.vn","username":"hieulq"},"date":"2019-05-07 01:42:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"489b63d488950c63c00210f43c4e72df47ca9d91","author":{"_account_id":19930,"name":"Dai Dang Van","email":"daidv@omzcloud.vn","username":"daidv"},"date":"2019-05-07 01:48:17.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":2},{"id":"9cc4494ae756ea57ed793c1eaec1e151be23c2ec","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2019-05-07 01:54:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8106777348c01083e7c69fe035116acc83ea475f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 03:23:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/70/621470/2/check/openstack-tox-pep8/8ae0f44/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/70/621470/2/check/openstack-tox-py27/8089549/ : SUCCESS in 8m 17s\n- openstack-tox-py36 http://logs.openstack.org/70/621470/2/check/openstack-tox-py36/d662e8c/ : SUCCESS in 8m 31s\n- openstack-tox-py37 http://logs.openstack.org/70/621470/2/check/openstack-tox-py37/94110d4/ : SUCCESS in 9m 16s\n- openstack-tox-docs http://logs.openstack.org/70/621470/2/check/openstack-tox-docs/9f7312e/html/ : SUCCESS in 4m 59s\n- openstack-tox-cover http://logs.openstack.org/70/621470/2/check/openstack-tox-cover/3a0490e/cover/ : SUCCESS in 10m 10s (non-voting)\n- mistral-devstack http://logs.openstack.org/70/621470/2/check/mistral-devstack/d4fc6a7/ : FAILURE in 47m 01s\n- mistral-devstack-non-apache http://logs.openstack.org/70/621470/2/check/mistral-devstack-non-apache/f14315e/ : FAILURE in 47m 42s\n- mistral-devstack-kombu http://logs.openstack.org/70/621470/2/check/mistral-devstack-kombu/5ee4b59/ : FAILURE in 47m 50s\n- mistral-tox-unit-mysql http://logs.openstack.org/70/621470/2/check/mistral-tox-unit-mysql/c2e1a7f/ : FAILURE in 29m 04s\n- mistral-tox-unit-postgresql http://logs.openstack.org/70/621470/2/check/mistral-tox-unit-postgresql/abde1ba/ : SUCCESS in 29m 34s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/70/621470/2/check/tripleo-ci-centos-7-scenario003-standalone/55cc016/ : SUCCESS in 1h 28m 12s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/621470/2/check/tripleo-ci-centos-7-containers-multinode/32fdb09/ : SUCCESS in 2h 12m 34s\n- mistral-rally-task http://logs.openstack.org/70/621470/2/check/mistral-rally-task/f06d701/ : FAILURE in 1h 08m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/70/621470/2/check/openstack-tox-lower-constraints/00f6f64/ : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":2},{"id":"4ddf51d9ace8c2ae7a633285ac9e241ca79ef94a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-07 06:50:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c38837d93bb8dfca6432b895960bedd0b78e2fe3","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2019-05-07 08:22:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7a2e307fda9397cde7050d22bc211ab49ce0a67e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 10:40:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/70/621470/3/check/openstack-tox-pep8/6570e57/ : SUCCESS in 7m 09s\n- openstack-tox-py27 http://logs.openstack.org/70/621470/3/check/openstack-tox-py27/82c0e9a/ : SUCCESS in 11m 28s\n- openstack-tox-py36 http://logs.openstack.org/70/621470/3/check/openstack-tox-py36/f2c819e/ : SUCCESS in 10m 20s\n- openstack-tox-py37 http://logs.openstack.org/70/621470/3/check/openstack-tox-py37/8abb5e5/ : SUCCESS in 12m 26s\n- openstack-tox-docs http://logs.openstack.org/70/621470/3/check/openstack-tox-docs/f7a998f/html/ : SUCCESS in 6m 11s\n- openstack-tox-cover http://logs.openstack.org/70/621470/3/check/openstack-tox-cover/dca9c81/cover/ : SUCCESS in 9m 54s (non-voting)\n- mistral-devstack http://logs.openstack.org/70/621470/3/check/mistral-devstack/56b53fa/ : SUCCESS in 48m 19s\n- mistral-devstack-non-apache http://logs.openstack.org/70/621470/3/check/mistral-devstack-non-apache/f6d1f9f/ : SUCCESS in 49m 56s\n- mistral-devstack-kombu http://logs.openstack.org/70/621470/3/check/mistral-devstack-kombu/b74c33d/ : SUCCESS in 48m 22s\n- mistral-tox-unit-mysql http://logs.openstack.org/70/621470/3/check/mistral-tox-unit-mysql/b4d4267/ : SUCCESS in 30m 27s\n- mistral-tox-unit-postgresql http://logs.openstack.org/70/621470/3/check/mistral-tox-unit-postgresql/8d3ddc3/ : SUCCESS in 30m 18s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/70/621470/3/check/tripleo-ci-centos-7-scenario003-standalone/5e64d85/ : SUCCESS in 1h 18m 49s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/621470/3/check/tripleo-ci-centos-7-containers-multinode/719840b/ : SUCCESS in 2h 10m 52s\n- mistral-rally-task http://logs.openstack.org/70/621470/3/check/mistral-rally-task/9e8e877/ : SUCCESS in 1h 05m 50s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/70/621470/3/check/openstack-tox-lower-constraints/87f4bc0/ : SUCCESS in 13m 25s","accounts_in_message":[],"_revision_number":3},{"id":"317c928ab572311bae95398645a01c72cd4c9378","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-29 04:18:38.000000000","message":"Patch Set 3:\n\nIs it possible to provide some tests? May be under mistral-tempest-plugin (with the dependency on this patch)?","accounts_in_message":[],"_revision_number":3},{"id":"f647b583628aa64eba9a5c6812f027d566227d28","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-31 07:28:51.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM but please answer my last question. Thanks.","accounts_in_message":[],"_revision_number":3},{"id":"98ebf7f1d3e0b0178950b8ec1c76acd2bf1a9cb7","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2019-05-31 08:18:10.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"3d658746d1b03319d37cffc7d42d96624d10eac2","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2019-05-31 08:19:58.000000000","message":"Patch Set 4:\n\n\u003e Is it possible to provide some tests? May be under\n \u003e mistral-tempest-plugin (with the dependency on this patch)?\n\nSure, I\u0027m working on a WIP patch now.","accounts_in_message":[],"_revision_number":4},{"id":"5d275ee51e903d041935060992298d82f0d67b4c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-31 08:47:38.000000000","message":"Patch Set 4:\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"e10df8828286c2d796c15244475c603eba4db494","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 10:26:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/70/621470/4/check/openstack-tox-pep8/38e9f3a/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/70/621470/4/check/openstack-tox-py27/4e30f53/ : SUCCESS in 11m 36s\n- openstack-tox-py36 http://logs.openstack.org/70/621470/4/check/openstack-tox-py36/dbeea1a/ : SUCCESS in 8m 47s\n- openstack-tox-py37 http://logs.openstack.org/70/621470/4/check/openstack-tox-py37/29cef2c/ : SUCCESS in 9m 44s\n- openstack-tox-docs http://logs.openstack.org/70/621470/4/check/openstack-tox-docs/7f776fd/html/ : SUCCESS in 4m 26s\n- openstack-tox-cover http://logs.openstack.org/70/621470/4/check/openstack-tox-cover/7ecf780/cover/ : SUCCESS in 8m 30s (non-voting)\n- mistral-devstack http://logs.openstack.org/70/621470/4/check/mistral-devstack/4abd321/ : FAILURE in 28m 15s\n- mistral-devstack-non-apache http://logs.openstack.org/70/621470/4/check/mistral-devstack-non-apache/539430d/ : FAILURE in 44m 00s\n- mistral-devstack-kombu http://logs.openstack.org/70/621470/4/check/mistral-devstack-kombu/05569f7/ : SUCCESS in 49m 26s\n- mistral-tox-unit-mysql http://logs.openstack.org/70/621470/4/check/mistral-tox-unit-mysql/a8db786/ : FAILURE in 30m 44s\n- mistral-tox-unit-postgresql http://logs.openstack.org/70/621470/4/check/mistral-tox-unit-postgresql/bd698fc/ : SUCCESS in 28m 28s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/70/621470/4/check/tripleo-ci-centos-7-scenario003-standalone/05c126b/ : SUCCESS in 1h 19m 17s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/621470/4/check/tripleo-ci-centos-7-containers-multinode/0bf5ec8/ : SUCCESS in 2h 02m 31s\n- mistral-rally-task http://logs.openstack.org/70/621470/4/check/mistral-rally-task/23fba1b/ : FAILURE in 1h 06m 03s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/70/621470/4/check/openstack-tox-lower-constraints/7c2c30a/ : SUCCESS in 10m 28s","accounts_in_message":[],"_revision_number":4},{"id":"b480fd22774ab739e0035a435d3dbd60cd5aa7ed","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2019-06-04 06:53:34.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"63c8a93ad2ab018a9a9a46de15392fcc944e6976","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 09:28:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/70/621470/5/check/openstack-tox-pep8/afcc94c/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/70/621470/5/check/openstack-tox-py27/b40ba70/ : SUCCESS in 9m 28s\n- openstack-tox-py36 http://logs.openstack.org/70/621470/5/check/openstack-tox-py36/519a68d/ : SUCCESS in 8m 01s\n- openstack-tox-py37 http://logs.openstack.org/70/621470/5/check/openstack-tox-py37/f18bf7d/ : SUCCESS in 11m 06s\n- openstack-tox-docs http://logs.openstack.org/70/621470/5/check/openstack-tox-docs/bae403f/html/ : SUCCESS in 5m 06s\n- openstack-tox-cover http://logs.openstack.org/70/621470/5/check/openstack-tox-cover/920b404/cover/ : SUCCESS in 10m 42s (non-voting)\n- mistral-devstack http://logs.openstack.org/70/621470/5/check/mistral-devstack/2aaac1c/ : RETRY_LIMIT in 3m 32s\n- mistral-devstack-non-apache http://logs.openstack.org/70/621470/5/check/mistral-devstack-non-apache/c161caf/ : SUCCESS in 47m 37s\n- mistral-devstack-kombu http://logs.openstack.org/70/621470/5/check/mistral-devstack-kombu/cc458d6/ : SUCCESS in 45m 39s\n- mistral-tox-unit-mysql http://logs.openstack.org/70/621470/5/check/mistral-tox-unit-mysql/57ce4bd/ : SUCCESS in 28m 59s\n- mistral-tox-unit-postgresql http://logs.openstack.org/70/621470/5/check/mistral-tox-unit-postgresql/839b0c1/ : SUCCESS in 29m 57s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/70/621470/5/check/tripleo-ci-centos-7-scenario003-standalone/b3f416a/ : SUCCESS in 1h 16m 51s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/621470/5/check/tripleo-ci-centos-7-containers-multinode/576a0c4/ : SUCCESS in 2h 01m 41s\n- mistral-rally-task http://logs.openstack.org/70/621470/5/check/mistral-rally-task/29173f0/ : SUCCESS in 58m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/70/621470/5/check/openstack-tox-lower-constraints/f7fd48f/ : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":5},{"id":"7f6a31ae33f3c8da648b519a99e1b610f2236729","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2019-06-06 01:50:46.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"0cd83c4b997f88b9c83542e06f2b99146a3bfab5","author":{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},"date":"2019-06-06 01:55:12.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"6590947207057e1143da4eb59caf5a800c0e9a2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 04:00:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/70/621470/6/check/openstack-tox-pep8/9dc0625/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/70/621470/6/check/openstack-tox-py27/18d0ca0/ : SUCCESS in 8m 23s\n- openstack-tox-py36 http://logs.openstack.org/70/621470/6/check/openstack-tox-py36/32f9a84/ : SUCCESS in 9m 12s\n- openstack-tox-py37 http://logs.openstack.org/70/621470/6/check/openstack-tox-py37/52b9eab/ : SUCCESS in 9m 09s\n- openstack-tox-docs http://logs.openstack.org/70/621470/6/check/openstack-tox-docs/ba73fdf/html/ : SUCCESS in 5m 54s\n- openstack-tox-cover http://logs.openstack.org/70/621470/6/check/openstack-tox-cover/519b55e/cover/ : SUCCESS in 10m 29s (non-voting)\n- mistral-devstack http://logs.openstack.org/70/621470/6/check/mistral-devstack/de140f7/ : SUCCESS in 42m 39s\n- mistral-devstack-non-apache http://logs.openstack.org/70/621470/6/check/mistral-devstack-non-apache/6eff11d/ : SUCCESS in 44m 32s\n- mistral-devstack-kombu http://logs.openstack.org/70/621470/6/check/mistral-devstack-kombu/a81fe36/ : SUCCESS in 41m 55s\n- mistral-tox-unit-mysql http://logs.openstack.org/70/621470/6/check/mistral-tox-unit-mysql/576841d/ : SUCCESS in 27m 45s\n- mistral-tox-unit-postgresql http://logs.openstack.org/70/621470/6/check/mistral-tox-unit-postgresql/609ca8a/ : SUCCESS in 26m 38s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/70/621470/6/check/tripleo-ci-centos-7-scenario003-standalone/8d8acca/ : SUCCESS in 1h 12m 21s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/621470/6/check/tripleo-ci-centos-7-containers-multinode/147462a/ : SUCCESS in 2h 02m 44s\n- mistral-rally-task http://logs.openstack.org/70/621470/6/check/mistral-rally-task/61fe062/ : SUCCESS in 1h 10m 04s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/70/621470/6/check/openstack-tox-lower-constraints/7e6fb4a/ : SUCCESS in 10m 16s","accounts_in_message":[],"_revision_number":6},{"id":"f00dbc058fd3696a1b3d51231289053728be86b7","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2019-07-08 08:16:31.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"7d1549b6ed3880e677aa827615f2b32a81c2cddb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 10:33:46.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/70/621470/7/check/openstack-tox-pep8/9b66732/ : SUCCESS in 6m 40s\n- openstack-tox-py27 http://logs.openstack.org/70/621470/7/check/openstack-tox-py27/dec78f2/ : SUCCESS in 8m 33s\n- openstack-tox-py36 http://logs.openstack.org/70/621470/7/check/openstack-tox-py36/2117da5/ : SUCCESS in 8m 18s\n- openstack-tox-py37 http://logs.openstack.org/70/621470/7/check/openstack-tox-py37/06f6630/ : SUCCESS in 8m 38s\n- openstack-tox-docs http://logs.openstack.org/70/621470/7/check/openstack-tox-docs/21051dc/html/ : SUCCESS in 5m 57s\n- openstack-tox-cover http://logs.openstack.org/70/621470/7/check/openstack-tox-cover/47b7c01/cover/ : SUCCESS in 10m 48s (non-voting)\n- mistral-devstack http://logs.openstack.org/70/621470/7/check/mistral-devstack/b764a3c/ : SUCCESS in 43m 48s\n- mistral-devstack-non-apache http://logs.openstack.org/70/621470/7/check/mistral-devstack-non-apache/3ba21b6/ : SUCCESS in 49m 31s\n- mistral-devstack-kombu http://logs.openstack.org/70/621470/7/check/mistral-devstack-kombu/7378509/ : SUCCESS in 49m 50s\n- mistral-tox-unit-mysql http://logs.openstack.org/70/621470/7/check/mistral-tox-unit-mysql/95b2d05/ : SUCCESS in 29m 04s\n- mistral-tox-unit-postgresql http://logs.openstack.org/70/621470/7/check/mistral-tox-unit-postgresql/8c03637/ : SUCCESS in 31m 01s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/70/621470/7/check/tripleo-ci-centos-7-scenario003-standalone/442295c/ : SUCCESS in 1h 28m 30s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/621470/7/check/tripleo-ci-centos-7-containers-multinode/00e1869/ : SUCCESS in 2h 10m 25s\n- mistral-rally-task http://logs.openstack.org/70/621470/7/check/mistral-rally-task/8000e4e/ : SUCCESS in 1h 14m 17s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/70/621470/7/check/openstack-tox-lower-constraints/b3537fb/ : SUCCESS in 13m 33s","accounts_in_message":[],"_revision_number":7},{"id":"ab954f4a99c4b363886394fa0dfccfb426f0bf2d","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-08-29 12:00:47.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"8cf975c7bac39ed6e3964d3a261fbd3aecc85f8d","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-29 12:06:58.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"93124b55a2d6d2676651d5b3b0e0c414912be38d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-02 07:07:54.000000000","message":"Patch Set 7: Code-Review+2\n\nLooks good.","accounts_in_message":[],"_revision_number":7},{"id":"bec6f33e40db145cee5ca257b384d58292071ac6","author":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"date":"2019-09-02 07:15:29.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"d059d6c887b4b0e031e1393f0f708c07d6cdd84f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 07:15:42.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"23de8082b1a0758db5d49be9f44e41367ee821a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 08:31:50.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d87aca38bb5541b59a1fa43cd008a56d : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99e53e21e5b743de9cf5fa41b80cf26c : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/382da9677d134ea9ace50ce6788330a7 : SUCCESS in 8m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a2b136ee56044d5ca6f911d54070aab5 : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b02407e3bc27489fa0139ac5dbd03c35 : SUCCESS in 5m 24s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/120bfe8edf1741d98f21e595f3f4b648 : SUCCESS in 40m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6d42d1f94767430aa9507499061de978 : SUCCESS in 41m 52s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a685a2e1e7804c1d960a062231b3ce87 : SUCCESS in 27m 45s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/294bd28e3764403da1fc096a8ced7138 : SUCCESS in 27m 09s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/279caf3f547249a08226ded5f6c137e2 : FAILURE in 15m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4d45eb4da4ab49bab5dd32287276c3c6 : SUCCESS in 44m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21e923522b5543b689c13eea5bed53b8 : SUCCESS in 9m 57s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/061eb6ebd77b454588c11f8392655511 : FAILURE in 15m 48s","accounts_in_message":[],"_revision_number":7},{"id":"d86f034fb99cbb6b6f21b9b71d226a7a8d3af279","author":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"date":"2019-09-03 01:25:34.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"a70ed988563dbd49a1b1e80b8d6cc585f1153180","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 02:42:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab66e67e3e0847ddbac950187dd602ef : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/967beb4fefbd46b6a9fd965b5dad1e52 : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf6f11f4307648f588277edb63f7f62b : SUCCESS in 9m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a1199c30f534755920ffe81a2a78b7c : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4431cf964249918d26f672b600376d : SUCCESS in 5m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/532a3303b3cb4118a4383e8137ad0f3f : SUCCESS in 9m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/26d53782cb1b4789b5afa25672deb0c3 : SUCCESS in 45m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/72c000a7895a4fe79a70a5f86297b04a : SUCCESS in 42m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9c83021b55f94d57b2043313d3f8eff3 : SUCCESS in 44m 54s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7dbb94e3625a446c9a1b29327951e0a5 : FAILURE in 27m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/25431ef149a1476e8de2ed3934a56bfb : SUCCESS in 26m 52s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/89cf7adaf1a142f0bc0d12c24aca0fd8 : FAILURE in 15m 32s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4622ecb84e0249ca8897cad3cc29f5c1 : FAILURE in 17m 17s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b3405e58bed54e7ab975327b0d163311 : SUCCESS in 1h 09m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2d539b5b3664ef292431ce93e7ba684 : SUCCESS in 9m 52s","accounts_in_message":[],"_revision_number":8},{"id":"815c6be5c974df5e6c2bd1d0a069e22ecb6d3fe3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-03 08:21:32.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"a1c4d50926965d484ace7e77924817d566c79d75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 11:31:01.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6aa978c0d8341139ca5efeabd10f699 : SUCCESS in 7m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/20be6d3f491d434ab320f5b59336b3b5 : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f694947604ca46f693c88647470f345b : SUCCESS in 9m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31468295f8814e36a97fe6415f2ab2c9 : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43a15dfd7e1749f393545e5f5557aef7 : SUCCESS in 7m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f45467ba0a24cf09c19d4dbc030f37b : SUCCESS in 9m 10s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9e9719db3fcb4122a3c7db3da3c0c6b5 : SUCCESS in 46m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/20b6b93c2f6f451a8462837955883b30 : SUCCESS in 52m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/541771ad4569411dadbfe8828b4bc51a : SUCCESS in 41m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e4205e3a1a7d41d2b26d428462dbce91 : SUCCESS in 28m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e429f9b1585745bfaef0d3478837b6fd : SUCCESS in 28m 42s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/42817986766644f4bfe947e95b13d0c6 : SUCCESS in 1h 35m 58s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0851038b60f84ebd901c3e90aea53ab0 : SUCCESS in 2h 46m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/848a56db848f453c891d077135681e88 : SUCCESS in 1h 09m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b233584d883496980eca9918e55e5f4 : SUCCESS in 10m 39s","accounts_in_message":[],"_revision_number":8},{"id":"5f721ea98293fb3233e9000e4ca3c2b1594ff892","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-04 03:57:43.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"94b6605782f3bb2b7a5ca16bc1d233b1c2b66286","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 03:57:55.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"4a16af565f120303be5f66d25e6b653e166a3bcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 05:38:55.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d352a8a054f545269b494ead9bba698f : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/057691b94b194d93b1d1717c87d61de6 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abba7e0f181849858d7e55ba663018ea : SUCCESS in 8m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d82111854044afcaed8e25fccdb4d7b : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d8ea3f15199438bafed1901e358c554 : SUCCESS in 5m 05s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2f3ee32f7c5e423d8f911c75d6622072 : SUCCESS in 37m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0aa7706076464f609ffd6ce68dba567a : SUCCESS in 41m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f3b234a42bc9489e8df23fab4a3b11f5 : SUCCESS in 27m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/82a04a6f31814f60bd9a22f334916f00 : SUCCESS in 26m 21s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/24f91ed96fb149aa88c8c5b4999555ba : SUCCESS in 1h 27m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/900ac32abd844196bcf28650c985e086 : SUCCESS in 43m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b42f6a7c757d4643bba037bf81fa7e51 : SUCCESS in 9m 44s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/155ecb0c752444daa6c8588028846457 : POST_FAILURE in 1h 35m 35s","accounts_in_message":[],"_revision_number":8},{"id":"fbe0baea7dde8f2e3570a74d0be1be212fe51109","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-04 08:31:59.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"eb4b5881551385daeebccfb6ae6dc15f20db2ca3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 12:10:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca4f53677fd4983bd29eeaf4d977129 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a92e86e9dac4db48decee2c33b359e4 : SUCCESS in 14m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d75e9d7c82dd414995ea687204f4c988 : SUCCESS in 9m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5864a51aed5e4cb786f3b0f1db9331df : SUCCESS in 11m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e379be528741729f74a5584426c74b : SUCCESS in 5m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6f4bbaf6aa9469e831769dd908c6048 : SUCCESS in 10m 00s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0048a08901474c06bc8de5c96f82c392 : SUCCESS in 44m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/23aab82c00ec4ac0877f188fff01ad1f : SUCCESS in 1h 00m 19s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ef95144ef6454db18a47138cbdba1811 : SUCCESS in 46m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/605ed80e5246445e98c62dbd154aef33 : SUCCESS in 28m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7b6c6f64a2c24686b83d68414b5e76ef : SUCCESS in 28m 06s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a714ea4a0faa442a80f8b6b3837f4cff : SUCCESS in 1h 51m 07s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3bca97d7e6094920a97f8e6e27a24c80 : SUCCESS in 2h 58m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/34745f38ea584ffb9418fa751298f708 : SUCCESS in 1h 05m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81978b5d326542d0839d1403bbc8b5d9 : SUCCESS in 11m 51s","accounts_in_message":[],"_revision_number":8},{"id":"c93b9f3097fba739e03afc43548efdf0d739f8f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 12:10:26.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"4c1245f2f5444562092f23df9a7942775a45ce27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 14:06:45.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa120e7b0b50416fa580fb058c16c6b3 : SUCCESS in 11m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95f387e9d12f4642976161cce1cc7d88 : SUCCESS in 9m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c8fcb1be7434695becca7377cbb292d : SUCCESS in 8m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58c9915f52e04ce0b7d604364a77730a : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1612b0e6f06f419098ff64a1824bb127 : SUCCESS in 4m 54s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fba80ee9feaa4ef181aa9e0d88bc033f : SUCCESS in 38m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/470aa244be704395aea4d2cbdd442bbc : SUCCESS in 46m 32s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/704e7ae34ed74f4e862c8b14569733e1 : SUCCESS in 36m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/55297f64177543e1b6953381043fa7f3 : SUCCESS in 28m 27s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a3d88c7c6b9b4e6fbe7d0e236cee6316 : SUCCESS in 1h 32m 20s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a106b24372df4ac987bdc5785f610f08 : SUCCESS in 48m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ba74a3b457c4531bc4eeac6d95dc794 : SUCCESS in 11m 17s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e02865c6b88244ae9b9a75565001a28c : SUCCESS in 1h 50m 54s","accounts_in_message":[],"_revision_number":8},{"id":"7bad2b0e63cf88d4c511e660505b31d93de9825e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 14:06:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"5e18f5c2e2b0ef44affcfc3296863e71a6670865","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 14:09:01.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/042dedfe943e4d88a7b6be7fae0c7ea3 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"6dfab10e3d885a34c32f5a4b8e0dad44de602702","revisions":{"dae819ab6daf24c6993132725b8678d2658c7fa5":{"kind":"REWORK","_number":1,"created":"2018-12-03 04:45:07.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/70/621470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/621470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/621470/1"}}},"commit":{"parents":[{"commit":"0fb96bbd2b87d10d0911314ab594dfa152d4f47f","subject":"Add loging for sending an action to executor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0fb96bbd2b87d10d0911314ab594dfa152d4f47f"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-12-03 04:43:51.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-12-03 04:47:14.000000000","tz":420},"subject":"Support OpenStack services dynamic versions","message":"Support OpenStack services dynamic versions\n\nSome OpenStack services have a discover_version method [1]\nthat returned a most recent version supported by API\nand client. Mistral should use this method rather than\nhardcode API version (manilaclient was done).\n\n[1] https://github.com/openstack/python-novaclient/blob/master/novaclient/api_versions.py#L250\n\nChange-Id: I0459206be5cc390853b9c69e8c5002568d1efa60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dae819ab6daf24c6993132725b8678d2658c7fa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dae819ab6daf24c6993132725b8678d2658c7fa5"}]},"branch":"refs/heads/master"},"d4a5623fb8fb282f4840a0dbdcb2103d2e118e28":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-07 01:05:40.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/70/621470/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/621470/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/621470/2"}}},"commit":{"parents":[{"commit":"2c7056b7095c56873772ad9bf79a7a1549a74e60","subject":"Merge \"Adds secrets_retrieve to the list of available actions in barbican\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2c7056b7095c56873772ad9bf79a7a1549a74e60"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-12-03 04:43:51.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2019-05-07 01:05:40.000000000","tz":0},"subject":"Support OpenStack services dynamic versions","message":"Support OpenStack services dynamic versions\n\nSome OpenStack services have a discover_version method [1]\nthat returned a most recent version supported by API\nand client. Mistral should use this method rather than\nhardcode API version (manilaclient was done).\n\n[1] https://github.com/openstack/python-novaclient/blob/master/novaclient/api_versions.py#L250\n\nChange-Id: I0459206be5cc390853b9c69e8c5002568d1efa60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d4a5623fb8fb282f4840a0dbdcb2103d2e118e28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d4a5623fb8fb282f4840a0dbdcb2103d2e118e28"}]},"branch":"refs/heads/master"},"c0f6cbb976a048d146cdf0721b6e556b19330ea0":{"kind":"REWORK","_number":3,"created":"2019-05-07 08:22:57.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/70/621470/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/621470/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/621470/3"}}},"commit":{"parents":[{"commit":"2c7056b7095c56873772ad9bf79a7a1549a74e60","subject":"Merge \"Adds secrets_retrieve to the list of available actions in barbican\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2c7056b7095c56873772ad9bf79a7a1549a74e60"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-12-03 04:43:51.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2019-05-07 08:21:41.000000000","tz":420},"subject":"Support OpenStack services dynamic versions","message":"Support OpenStack services dynamic versions\n\nSome OpenStack services have a discover_version method [1]\nthat returned a most recent version supported by API\nand client. Mistral should use this method rather than\nhardcode API version (manilaclient was done).\n\n[1] https://github.com/openstack/python-novaclient/blob/master/novaclient/api_versions.py#L250\n\nChange-Id: I0459206be5cc390853b9c69e8c5002568d1efa60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c0f6cbb976a048d146cdf0721b6e556b19330ea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c0f6cbb976a048d146cdf0721b6e556b19330ea0"}]},"branch":"refs/heads/master"},"87b9931c132aec0afa252bcd349fecdd5a58df88":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-31 08:18:10.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/70/621470/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/621470/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/621470/4"}}},"commit":{"parents":[{"commit":"e25da33607667767d0e34b576f1dd86d0638b854","subject":"Merge \"Fix adhoc action lookup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e25da33607667767d0e34b576f1dd86d0638b854"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-12-03 04:43:51.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2019-05-31 08:18:10.000000000","tz":0},"subject":"Support OpenStack services dynamic versions","message":"Support OpenStack services dynamic versions\n\nSome OpenStack services have a discover_version method [1]\nthat returned a most recent version supported by API\nand client. Mistral should use this method rather than\nhardcode API version (manilaclient was done).\n\n[1] https://github.com/openstack/python-novaclient/blob/master/novaclient/api_versions.py#L250\n\nChange-Id: I0459206be5cc390853b9c69e8c5002568d1efa60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87b9931c132aec0afa252bcd349fecdd5a58df88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87b9931c132aec0afa252bcd349fecdd5a58df88"}]},"branch":"refs/heads/master"},"ce23ffe38252dffa898fdd6cfe09bf3a2af49977":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-06-04 06:53:34.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/70/621470/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/621470/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/621470/5"}}},"commit":{"parents":[{"commit":"a75b07ec5f2e9a06fb3047d99a6dca3fbe8c25f6","subject":"Send task and workflow notifications out of the transaction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a75b07ec5f2e9a06fb3047d99a6dca3fbe8c25f6"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-12-03 04:43:51.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2019-06-04 06:53:34.000000000","tz":0},"subject":"Support OpenStack services dynamic versions","message":"Support OpenStack services dynamic versions\n\nSome OpenStack services have a discover_version method [1]\nthat returned a most recent version supported by API\nand client. Mistral should use this method rather than\nhardcode API version (manilaclient was done).\n\n[1] https://github.com/openstack/python-novaclient/blob/master/novaclient/api_versions.py#L250\n\nChange-Id: I0459206be5cc390853b9c69e8c5002568d1efa60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ce23ffe38252dffa898fdd6cfe09bf3a2af49977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ce23ffe38252dffa898fdd6cfe09bf3a2af49977"}]},"branch":"refs/heads/master"},"4e6c691fc92db84fbb17d2137aa04a0728fe08dd":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-06-06 01:50:46.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/70/621470/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/621470/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/621470/6"}}},"commit":{"parents":[{"commit":"55c79906848ac37c4ad4f56eca694640adbf2dd7","subject":"Merge \"Remove _get_next_clauses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55c79906848ac37c4ad4f56eca694640adbf2dd7"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-12-03 04:43:51.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2019-06-06 01:50:46.000000000","tz":0},"subject":"Support OpenStack services dynamic versions","message":"Support OpenStack services dynamic versions\n\nSome OpenStack services have a discover_version method [1]\nthat returned a most recent version supported by API\nand client. Mistral should use this method rather than\nhardcode API version (manilaclient was done).\n\n[1] https://github.com/openstack/python-novaclient/blob/master/novaclient/api_versions.py#L250\n\nChange-Id: I0459206be5cc390853b9c69e8c5002568d1efa60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e6c691fc92db84fbb17d2137aa04a0728fe08dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e6c691fc92db84fbb17d2137aa04a0728fe08dd"}]},"branch":"refs/heads/master"},"78a70b9a717596119fa8afcfd44791d1c767a69c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-07-08 08:16:31.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/70/621470/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/621470/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/621470/7"}}},"commit":{"parents":[{"commit":"6a9aa4c295b51e9b11ab4a30e5c002bcc2924609","subject":"Merge \"Initialize the lazily loaded execution \"input\" field in API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a9aa4c295b51e9b11ab4a30e5c002bcc2924609"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-12-03 04:43:51.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2019-07-08 08:16:31.000000000","tz":0},"subject":"Support OpenStack services dynamic versions","message":"Support OpenStack services dynamic versions\n\nSome OpenStack services have a discover_version method [1]\nthat returned a most recent version supported by API\nand client. Mistral should use this method rather than\nhardcode API version (manilaclient was done).\n\n[1] https://github.com/openstack/python-novaclient/blob/master/novaclient/api_versions.py#L250\n\nChange-Id: I0459206be5cc390853b9c69e8c5002568d1efa60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/78a70b9a717596119fa8afcfd44791d1c767a69c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/78a70b9a717596119fa8afcfd44791d1c767a69c"}]},"branch":"refs/heads/master"},"6dfab10e3d885a34c32f5a4b8e0dad44de602702":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-09-03 01:25:34.000000000","uploader":{"_account_id":22406,"name":"Kien Nguyen","email":"kiennt2609@gmail.com","username":"kiennt"},"ref":"refs/changes/70/621470/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/621470/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/621470/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/621470/8"}}},"commit":{"parents":[{"commit":"2cdcb5415ead23a129d2e6698833ce10e6a6cf4a","subject":"Merge \"Improve new scheduler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2cdcb5415ead23a129d2e6698833ce10e6a6cf4a"}]}],"author":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2018-12-03 04:43:51.000000000","tz":420},"committer":{"name":"Kien Nguyen","email":"kiennt65@viettel.com.vn","date":"2019-09-03 01:25:34.000000000","tz":0},"subject":"Support OpenStack services dynamic versions","message":"Support OpenStack services dynamic versions\n\nSome OpenStack services have a discover_version method [1]\nthat returned a most recent version supported by API\nand client. Mistral should use this method rather than\nhardcode API version (manilaclient was done).\n\n[1] https://github.com/openstack/python-novaclient/blob/master/novaclient/api_versions.py#L250\n\nChange-Id: I0459206be5cc390853b9c69e8c5002568d1efa60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6dfab10e3d885a34c32f5a4b8e0dad44de602702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6dfab10e3d885a34c32f5a4b8e0dad44de602702"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
