)]}'
{"id":"openstack%2Fmistral~622339","triplet_id":"openstack%2Fmistral~master~Ide438a7f6d24c8bdc9eb2c82e935fd39a6acc2c6","project":"openstack/mistral","branch":"master","topic":"bug/1806703","hashtags":[],"change_id":"Ide438a7f6d24c8bdc9eb2c82e935fd39a6acc2c6","subject":"Adds private_key parameter in the standard ssh actions","status":"MERGED","created":"2018-12-04 14:45:10.000000000","updated":"2019-05-16 06:40:47.000000000","submitted":"2019-05-16 06:40:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"622339-1557988847556-88673ecb","meta_rev_id":"579933c8752085e956adf6ec8bd69be2b9d9c7f4","_number":622339,"virtual_id_number":622339,"owner":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-05-16 06:40:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"date":"2019-05-15 08:47:11.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2018-12-09 06:50:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-05-15 09:07:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-05-15 09:07:17.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"value":0,"date":"2019-05-15 09:07:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":1,"date":"2019-05-16 05:10:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-12-09 06:50:26.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-05-15 09:07:17.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-05-15 09:07:21.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-05-16 05:10:08.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-05-16 06:40:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"98a911959434908fdd1f5bd47040a22d86bde821","author":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"date":"2018-12-04 14:45:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7307204ef600607019655ea2ed777198d3f46b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-04 17:54:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/39/622339/1/check/openstack-tox-pep8/cf30370/ : SUCCESS in 5m 41s\n- openstack-tox-py27 http://logs.openstack.org/39/622339/1/check/openstack-tox-py27/8609fab/ : SUCCESS in 8m 58s\n- openstack-tox-py35 http://logs.openstack.org/39/622339/1/check/openstack-tox-py35/2f01e8d/ : FAILURE in 5m 28s\n- openstack-tox-py36 http://logs.openstack.org/39/622339/1/check/openstack-tox-py36/e07b15f/ : FAILURE in 5m 31s\n- openstack-tox-docs http://logs.openstack.org/39/622339/1/check/openstack-tox-docs/35971db/html/ : SUCCESS in 5m 17s\n- build-openstack-releasenotes http://logs.openstack.org/39/622339/1/check/build-openstack-releasenotes/81bb465/html/ : SUCCESS in 5m 14s\n- openstack-tox-cover http://logs.openstack.org/39/622339/1/check/openstack-tox-cover/93fc8e6/ : FAILURE in 5m 45s (non-voting)\n- mistral-devstack http://logs.openstack.org/39/622339/1/check/mistral-devstack/7fd4173/ : SUCCESS in 40m 54s\n- mistral-devstack-non-apache http://logs.openstack.org/39/622339/1/check/mistral-devstack-non-apache/6a6dda1/ : SUCCESS in 39m 53s\n- mistral-devstack-kombu http://logs.openstack.org/39/622339/1/check/mistral-devstack-kombu/1ba4831/ : SUCCESS in 37m 27s\n- mistral-tox-unit-mysql http://logs.openstack.org/39/622339/1/check/mistral-tox-unit-mysql/6f00cd4/ : TIMED_OUT in 1h 00m 58s\n- mistral-tox-unit-postgresql http://logs.openstack.org/39/622339/1/check/mistral-tox-unit-postgresql/de07a58/ : TIMED_OUT in 1h 01m 04s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/39/622339/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/f37d760/ : SUCCESS in 2h 44m 59s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/39/622339/1/check/tripleo-ci-centos-7-undercloud-containers/20154fb/ : SUCCESS in 1h 48m 59s\n- mistral-rally-task http://logs.openstack.org/39/622339/1/check/mistral-rally-task/c00af7a/ : SUCCESS in 1h 12m 15s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/39/622339/1/check/openstack-tox-lower-constraints/2816866/ : FAILURE in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e6136246511e94abbf275239c0b8a76e3d0e8fcf","author":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"date":"2018-12-05 07:11:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2a5329a5db7c4dbb75103f782f59445440e76cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-05 09:35:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/39/622339/2/check/openstack-tox-pep8/3516a58/ : SUCCESS in 6m 29s\n- openstack-tox-py27 http://logs.openstack.org/39/622339/2/check/openstack-tox-py27/e30e5a0/ : SUCCESS in 10m 37s\n- openstack-tox-py35 http://logs.openstack.org/39/622339/2/check/openstack-tox-py35/60682bf/ : SUCCESS in 11m 54s\n- openstack-tox-py36 http://logs.openstack.org/39/622339/2/check/openstack-tox-py36/0afcd76/ : SUCCESS in 10m 23s\n- openstack-tox-docs http://logs.openstack.org/39/622339/2/check/openstack-tox-docs/98f7ae0/html/ : SUCCESS in 5m 31s\n- build-openstack-releasenotes http://logs.openstack.org/39/622339/2/check/build-openstack-releasenotes/f6dc980/html/ : SUCCESS in 5m 22s\n- openstack-tox-cover http://logs.openstack.org/39/622339/2/check/openstack-tox-cover/287b84a/cover/ : SUCCESS in 18m 38s (non-voting)\n- mistral-devstack http://logs.openstack.org/39/622339/2/check/mistral-devstack/2e7aad7/ : SUCCESS in 40m 01s\n- mistral-devstack-non-apache http://logs.openstack.org/39/622339/2/check/mistral-devstack-non-apache/0ee0f95/ : SUCCESS in 45m 31s\n- mistral-devstack-kombu http://logs.openstack.org/39/622339/2/check/mistral-devstack-kombu/e2f6519/ : SUCCESS in 48m 44s\n- mistral-tox-unit-mysql http://logs.openstack.org/39/622339/2/check/mistral-tox-unit-mysql/ede0eb9/ : SUCCESS in 46m 23s\n- mistral-tox-unit-postgresql http://logs.openstack.org/39/622339/2/check/mistral-tox-unit-postgresql/9572581/ : SUCCESS in 28m 15s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/39/622339/2/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/6dd4b12/ : SUCCESS in 2h 15m 00s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/39/622339/2/check/tripleo-ci-centos-7-undercloud-containers/82a87c7/ : SUCCESS in 2h 03m 27s\n- mistral-rally-task http://logs.openstack.org/39/622339/2/check/mistral-rally-task/490f088/ : SUCCESS in 1h 17m 18s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/39/622339/2/check/openstack-tox-lower-constraints/28b2f5f/ : SUCCESS in 11m 37s","accounts_in_message":[],"_revision_number":2},{"id":"38beae7ed78332274b9cc1627aeba13abe39be4a","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2018-12-09 06:50:26.000000000","message":"Patch Set 2: Code-Review+1\n\nlooks good","accounts_in_message":[],"_revision_number":2},{"id":"54bc1f655a01bf7a46de988acc0f8128b0f74b28","author":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"date":"2018-12-09 14:59:11.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"739f2bddf2589ae904828a6cbb52e2eb06c072e0","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2018-12-12 12:26:39.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\nWhy -1?\nCould you please mention it or put it in WIP.","accounts_in_message":[],"_revision_number":2},{"id":"64d529036a3d30d9ae1178392b210d435aeea235","author":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"date":"2018-12-12 13:26:08.000000000","message":"Patch Set 2: Workflow-1\n\nThe idea of the patch is to provide the private key as a parameter, I\u0027ve got some issues while getting the secret from barbican. I don\u0027t know if this related to this patch or related to the workflow so I put -1 until I debug it further","accounts_in_message":[],"_revision_number":2},{"id":"8981a0b0a518d045be59312fb434897084c94e4b","author":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"date":"2019-05-15 08:47:11.000000000","message":"Patch Set 2: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"44f70277557dd64bcc0a6e0be2df3ac498fda8f4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-15 09:07:03.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"8aa0d273f08df219ad0717da79d7ae8e39077cae","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-16 05:10:08.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ea000fbd9c1249375df1bb1660d644607514c95f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 05:10:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5810af1b2c730a044dfe0ce67f708a5553b3e47a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:40:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"af50753f19c75a7767b4e6879228dc0628960ad1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:40:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/39/622339/2/gate/openstack-tox-pep8/36ec8a8/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/39/622339/2/gate/openstack-tox-py27/18cb67b/ : SUCCESS in 8m 35s\n- openstack-tox-py36 http://logs.openstack.org/39/622339/2/gate/openstack-tox-py36/6e2c119/ : SUCCESS in 9m 18s\n- openstack-tox-py37 http://logs.openstack.org/39/622339/2/gate/openstack-tox-py37/973dc83/ : SUCCESS in 8m 57s\n- openstack-tox-docs http://logs.openstack.org/39/622339/2/gate/openstack-tox-docs/2104dbe/html/ : SUCCESS in 7m 03s\n- build-openstack-releasenotes http://logs.openstack.org/39/622339/2/gate/build-openstack-releasenotes/ddd49fd/html/ : SUCCESS in 6m 33s\n- mistral-devstack http://logs.openstack.org/39/622339/2/gate/mistral-devstack/d939d6e/ : SUCCESS in 35m 21s\n- mistral-devstack-non-apache http://logs.openstack.org/39/622339/2/gate/mistral-devstack-non-apache/fadf1af/ : SUCCESS in 38m 59s\n- mistral-tox-unit-mysql http://logs.openstack.org/39/622339/2/gate/mistral-tox-unit-mysql/868d15f/ : SUCCESS in 28m 51s\n- mistral-tox-unit-postgresql http://logs.openstack.org/39/622339/2/gate/mistral-tox-unit-postgresql/7bba6a9/ : SUCCESS in 26m 19s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/622339/2/gate/tripleo-ci-centos-7-scenario003-standalone/46bc087/ : SUCCESS in 1h 21m 42s\n- mistral-devstack-kombu http://logs.openstack.org/39/622339/2/gate/mistral-devstack-kombu/91c514b/ : SUCCESS in 38m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/39/622339/2/gate/openstack-tox-lower-constraints/d8627a6/ : SUCCESS in 10m 32s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/39/622339/2/gate/tripleo-ci-centos-7-undercloud-containers/51507d7/ : SUCCESS in 1h 22m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9bbc1d41d0a9cf8d4605a552e3fceb25c7dc4341","revisions":{"f5a36dad19a2efcefa545e7644b587fb01b6d47b":{"kind":"REWORK","_number":1,"created":"2018-12-04 14:45:10.000000000","uploader":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"ref":"refs/changes/39/622339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/39/622339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/39/622339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/39/622339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/39/622339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/39/622339/1"}}},"commit":{"parents":[{"commit":"0fb96bbd2b87d10d0911314ab594dfa152d4f47f","subject":"Add loging for sending an action to executor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0fb96bbd2b87d10d0911314ab594dfa152d4f47f"}]}],"author":{"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","date":"2018-12-04 14:43:26.000000000","tz":60},"committer":{"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","date":"2018-12-04 14:43:26.000000000","tz":60},"subject":"Adds private_key parameter in the standard ssh actions","message":"Adds private_key parameter in the standard ssh actions\n\nThis method allows to specify a private key and avoids its storage\nin the filesystem of the executors. This can be used later in\ncombination of a secrets_retrieve to use keys stored in barbican.\n\nChange-Id: Ide438a7f6d24c8bdc9eb2c82e935fd39a6acc2c6\nCloses-Bug: #1806703\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f5a36dad19a2efcefa545e7644b587fb01b6d47b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f5a36dad19a2efcefa545e7644b587fb01b6d47b"}]},"branch":"refs/heads/master"},"9bbc1d41d0a9cf8d4605a552e3fceb25c7dc4341":{"kind":"REWORK","_number":2,"created":"2018-12-05 07:11:15.000000000","uploader":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"ref":"refs/changes/39/622339/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/39/622339/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/39/622339/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/39/622339/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/39/622339/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/39/622339/2"}}},"commit":{"parents":[{"commit":"0fb96bbd2b87d10d0911314ab594dfa152d4f47f","subject":"Add loging for sending an action to executor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0fb96bbd2b87d10d0911314ab594dfa152d4f47f"}]}],"author":{"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","date":"2018-12-04 14:43:26.000000000","tz":60},"committer":{"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","date":"2018-12-05 07:10:57.000000000","tz":60},"subject":"Adds private_key parameter in the standard ssh actions","message":"Adds private_key parameter in the standard ssh actions\n\nThis method allows to specify a private key and avoids its storage\nin the filesystem of the executors. This can be used later in\ncombination of a secrets_retrieve to use keys stored in barbican.\n\nChange-Id: Ide438a7f6d24c8bdc9eb2c82e935fd39a6acc2c6\nCloses-Bug: #1806703\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9bbc1d41d0a9cf8d4605a552e3fceb25c7dc4341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9bbc1d41d0a9cf8d4605a552e3fceb25c7dc4341"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
