)]}'
{"id":"openstack%2Fmistral~636583","triplet_id":"openstack%2Fmistral~master~I888008ac8f9c12bd92a9bd72bd7c276a11827847","project":"openstack/mistral","branch":"master","topic":"bug/1815515","hashtags":[],"change_id":"I888008ac8f9c12bd92a9bd72bd7c276a11827847","subject":"Disable key sorting in workflow definition","status":"MERGED","created":"2019-02-13 10:08:23.000000000","updated":"2020-01-23 09:15:04.000000000","submitted":"2020-01-23 09:13:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"636583-1579770787169-5aeb02ef","meta_rev_id":"bfdd18968d442cdf3c89ff52b129e115c1fbbbfa","_number":636583,"virtual_id_number":636583,"owner":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2020-01-22 11:45:35.000000000","_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-23 09:13:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":1,"date":"2020-01-22 11:32:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2020-01-22 08:26:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-21 06:51:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2020-01-22 08:26:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-21 06:51:12.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-01-22 08:26:41.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-01-22 11:32:13.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-01-23 09:13:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26ff4dcf32c66da67ed75d5404f3db1a218c2c8f","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-02-13 10:08:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb9160a592a44ea3db96ae7f9e95785474788dda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-13 11:24:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/83/636583/1/check/openstack-tox-pep8/88d99f6/ : FAILURE in 6m 28s\n- openstack-tox-py27 http://logs.openstack.org/83/636583/1/check/openstack-tox-py27/041e1ab/ : SUCCESS in 9m 22s\n- openstack-tox-py35 http://logs.openstack.org/83/636583/1/check/openstack-tox-py35/366cd48/ : SUCCESS in 9m 28s\n- openstack-tox-py36 http://logs.openstack.org/83/636583/1/check/openstack-tox-py36/8ce1ceb/ : SUCCESS in 10m 35s\n- openstack-tox-docs http://logs.openstack.org/83/636583/1/check/openstack-tox-docs/44eb554/html/ : SUCCESS in 7m 00s\n- requirements-check http://logs.openstack.org/83/636583/1/check/requirements-check/2b37dcf/ : FAILURE in 4m 01s\n- openstack-tox-cover http://logs.openstack.org/83/636583/1/check/openstack-tox-cover/75d0c8e/cover/ : SUCCESS in 12m 00s (non-voting)\n- mistral-devstack http://logs.openstack.org/83/636583/1/check/mistral-devstack/843ce52/ : SUCCESS in 39m 47s\n- mistral-devstack-non-apache http://logs.openstack.org/83/636583/1/check/mistral-devstack-non-apache/6ce2aa6/ : SUCCESS in 46m 18s\n- mistral-devstack-kombu http://logs.openstack.org/83/636583/1/check/mistral-devstack-kombu/79a258b/ : SUCCESS in 42m 39s\n- mistral-tox-unit-mysql http://logs.openstack.org/83/636583/1/check/mistral-tox-unit-mysql/c62c16c/ : FAILURE in 31m 01s\n- mistral-tox-unit-postgresql http://logs.openstack.org/83/636583/1/check/mistral-tox-unit-postgresql/9767f8a/ : FAILURE in 26m 37s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/83/636583/1/check/tripleo-ci-centos-7-scenario003-standalone/8bbb8e9/ : FAILURE in 27m 44s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/83/636583/1/check/tripleo-ci-centos-7-undercloud-containers/f5c5872/ : FAILURE in 27m 06s\n- mistral-rally-task http://logs.openstack.org/83/636583/1/check/mistral-rally-task/c361746/ : SUCCESS in 1h 05m 47s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/83/636583/1/check/openstack-tox-lower-constraints/d8255c8/ : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"33d674d7642682ab26ac811de4f23ead17666c9f","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-02-13 12:15:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5bd1ec5f5d6e5e60b30f1e38d2dcdae90e165fc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-13 13:29:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/83/636583/2/check/openstack-tox-pep8/d8d2930/ : SUCCESS in 7m 49s\n- openstack-tox-py27 http://logs.openstack.org/83/636583/2/check/openstack-tox-py27/d3cf1fd/ : SUCCESS in 8m 54s\n- openstack-tox-py35 http://logs.openstack.org/83/636583/2/check/openstack-tox-py35/ea07593/ : SUCCESS in 10m 43s\n- openstack-tox-py36 http://logs.openstack.org/83/636583/2/check/openstack-tox-py36/6deb960/ : SUCCESS in 10m 56s\n- openstack-tox-docs http://logs.openstack.org/83/636583/2/check/openstack-tox-docs/a6da3b1/html/ : SUCCESS in 7m 16s\n- requirements-check http://logs.openstack.org/83/636583/2/check/requirements-check/3abff43/ : FAILURE in 5m 43s\n- openstack-tox-cover http://logs.openstack.org/83/636583/2/check/openstack-tox-cover/01b6af4/cover/ : SUCCESS in 11m 40s (non-voting)\n- mistral-devstack http://logs.openstack.org/83/636583/2/check/mistral-devstack/ddc4466/ : SUCCESS in 35m 35s\n- mistral-devstack-non-apache http://logs.openstack.org/83/636583/2/check/mistral-devstack-non-apache/15c728c/ : SUCCESS in 49m 16s\n- mistral-devstack-kombu http://logs.openstack.org/83/636583/2/check/mistral-devstack-kombu/edb3495/ : SUCCESS in 44m 58s\n- mistral-tox-unit-mysql http://logs.openstack.org/83/636583/2/check/mistral-tox-unit-mysql/afb21d1/ : FAILURE in 28m 49s\n- mistral-tox-unit-postgresql http://logs.openstack.org/83/636583/2/check/mistral-tox-unit-postgresql/aaf2c5f/ : SUCCESS in 27m 55s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/83/636583/2/check/tripleo-ci-centos-7-scenario003-standalone/84a960d/ : FAILURE in 31m 48s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/83/636583/2/check/tripleo-ci-centos-7-undercloud-containers/eab486a/ : FAILURE in 25m 05s\n- mistral-rally-task http://logs.openstack.org/83/636583/2/check/mistral-rally-task/84931f7/ : SUCCESS in 1h 08m 12s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/83/636583/2/check/openstack-tox-lower-constraints/2445146/ : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":2},{"id":"7a6e2c8b86c78caf87b8fcab6fcbe3b5704f975c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-02-14 04:46:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5c24cb2b05190f5127f800b3047f8190dac446c9","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-16 22:18:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cbffd31b572828406b7bebd8800562b893cf6c99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 01:08:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e133b57800da4bb0a9f73a54130968c7 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d4f57d0436149e8bc8d7bdd7fa2c1ad : SUCCESS in 8m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29f2ccb1b3ea40d6bddb72bc0c59cb79 : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9436539c192b4d579a802765204fd922 : SUCCESS in 7m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1135aaea0f604e4487abaa81856110cd : FAILURE in 3m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df2c4cbe452b476088d7c41ee3c5b8db : SUCCESS in 10m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/910517af05a04f08abb4867239ab09fb : SUCCESS in 44m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/450d51c3db6c4a288840dd66f7dadf49 : SUCCESS in 45m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b244f4e6d194652b28afa55fa13fad9 : SUCCESS in 53m 12s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/401417445c824c818d8ce87d4f0f1c69 : SUCCESS in 45m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ad48dddbec8d48ef882397eee140d01a : SUCCESS in 43m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cd765dd1e31e4fa7bac4de4af09fafbb : SUCCESS in 30m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1874225fef2a4b78bb65c18ff2bb0e84 : SUCCESS in 27m 39s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8570f50ebc8e4c1faa0bb7949a92fa3f : SUCCESS in 2h 45m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f4491ab4b5854c39a26df5d06bc52ca1 : SUCCESS in 59m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/491a0021a5a74e738e08f9e9644ae206 : SUCCESS in 12m 13s","accounts_in_message":[],"_revision_number":3},{"id":"47c8d28e10367338d0212e01390bd3bce5be0c3b","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-17 08:29:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"68c7feb20f190899f7d14a25c4193cbd79a3becb","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-17 11:29:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"13fa82c8007e5a73aee3697cf1171e07fc1d912f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 14:28:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e66082a8d9e640bb934ad3fb6e685cf2 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d0b32dfe79348c1b38ea2851dc21c6a : SUCCESS in 11m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/785388ef5f774f6d9d2cd49a41f54642 : SUCCESS in 12m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7b243a386ff464dbde4afb479064d09 : SUCCESS in 8m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c3d3f00459ca43aba751a372bbd306e6 : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71823ff574644741bb5be31dd6317893 : SUCCESS in 10m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/bf6645198948431b84c0bc995c055581 : SUCCESS in 42m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61df34cb832b499aa0c20644ceca5a36 : SUCCESS in 43m 59s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e30f55b814bb42e38c14e525b2e8c82e : SUCCESS in 39m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cd00aaa4cd854e799131223115d1ab77 : SUCCESS in 42m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6af0c7d0b584428393213befe752fedf : SUCCESS in 44m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ccf0649f729243c89785a9e020219956 : SUCCESS in 29m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3fa5c834209e46b8b3dd6462d99f207c : SUCCESS in 27m 24s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ad3cada7f0c04a74a5afba9886aed411 : SUCCESS in 2h 55m 05s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ac960641a64b47598f351835eac450d9 : SUCCESS in 57m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d80c4d10b68a495189e73d38eed0a5d3 : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":5},{"id":"b1ff9b819ce08faf00ff2dacb22e64f03c1acf16","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-01-19 11:07:26.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1595c7b47920e1cb1eba7979ff76ec8432891d02","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-19 11:20:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9f1be75393ad9b752293a8e1b640984e11cbbce3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-20 06:52:53.000000000","message":"Patch Set 5: Code-Review+2\n\nI\u0027m OK with the patch. However, if you address Eyal\u0027s comment it\u0027d be even better.","accounts_in_message":[],"_revision_number":5},{"id":"b2ed1d02cf8d83df2ab849f51450f0696a8ca249","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-20 21:40:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f742d3a6148e902a588cede5543ac60f5e5763f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 00:24:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/646b69bab69c475cb19464d3bb4a6d13 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55b6c94618cb4650a76c70dcd6b6b2c6 : SUCCESS in 9m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8eed4f492c204be5b8d0a3b8f271998a : SUCCESS in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1d22c7924fb4b1ea64247c20f1b0b8b : SUCCESS in 8m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ce85caa43ea3462893f952ca6b721ce7 : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6c6559916f14c1bb4520c2530ad915f : SUCCESS in 9m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1f3affdb7c5942e889aebd70dcd95a22 : SUCCESS in 45m 46s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abee3f7551804b6487aedf2afe4f1a24 : SUCCESS in 42m 46s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/121d895dd35d494e9426c3d115ac6d40 : SUCCESS in 49m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1c21d18345ea4f28ab637330ab0c69a2 : SUCCESS in 49m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6cfbf807510a4934a1c26ce0f18704fc : SUCCESS in 46m 22s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ad52c8dc7ea14108ac3bc5f5ece76905 : SUCCESS in 28m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/86eca18f509f4c12906215e4bced4c28 : SUCCESS in 28m 39s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/adf8cd1707ca453480b1a93d7bf1ec30 : SUCCESS in 2h 39m 36s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1f10d1f1a9af498491aaa5033678d86a : SUCCESS in 1h 01m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2d97b2fcef74dcdaf29ddbbed110425 : SUCCESS in 9m 18s","accounts_in_message":[],"_revision_number":6},{"id":"43af6df6676fbc56f41b982f1d961b9abd050b11","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-21 06:51:12.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"7abc07ae7776de57d7f412fe01a9a82d10ddc0ce","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-22 08:26:41.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"dcad6a96f92a5a396a0e5ee781dafa138d61ac92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 08:26:56.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"655e25592e56c0aec9b9b79e508f7377a761b89a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 10:26:56.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dd720be86a44c758fd4271fbad6631a : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ee5c486d37e42ec9fa2a7c2b86812b2 : SUCCESS in 9m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c74f6898c6ee4245b1c26680878cf3cf : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48c0f24df65c422daf60df7299d20285 : SUCCESS in 8m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0f6bb8aaa349424c9011c9a1569d2d86 : SUCCESS in 4m 09s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8879464905ec457489abbd0f9dcf22fc : SUCCESS in 46m 51s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6c3705bf5bf4610861282a640ae70f1 : SUCCESS in 46m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9583a66f4da64add930adce3915b0874 : FAILURE in 45m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/356769d5d145470293cb33aa82213ece : SUCCESS in 48m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/66fe117691ee41b5b34af870c52fb3e5 : SUCCESS in 29m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7248b4f0aa614ea2a4110111b215f14c : SUCCESS in 28m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f2e6e8084186424cab3d19f7ffff4687 : SUCCESS in 43m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f4cf8c9f9094684a8e57c6474dadf79 : SUCCESS in 10m 47s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2b1cc1298e1a4a5692a327e1befba051 : SUCCESS in 1h 54m 52s","accounts_in_message":[],"_revision_number":6},{"id":"82f4954301b936bb06bdd38639b21ac20b4a882e","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-01-22 11:32:13.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"7f82a7a6b848413c191c3e299a04592bb663e583","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-22 11:45:35.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"387805c264db02c6f4b6ea22f42de8a8dfbf50c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:03:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6274bd1bde249d78a30e2afe3065b64 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93f8337d36274109b0f76c4e1ecbcc22 : SUCCESS in 10m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ed7695779184d7d8ef052da83140d58 : SUCCESS in 10m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2731e1eff3f4106935d9ac8baf69a8f : SUCCESS in 9m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5336853a98cf4c01a79b9d7df85e327b : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87dc3c977e8843afa6a7584b86392c40 : SUCCESS in 10m 03s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b0f9313e36d34ba7b328f0e77a09e328 : SUCCESS in 41m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65adfa6ca2a049c28cb7a72e0822ac59 : SUCCESS in 41m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57a05ac67847436794b7d8062dbd82dd : SUCCESS in 40m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e03dc612a36f4ad1bf1eb3dd294b9db9 : SUCCESS in 41m 06s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f745dafe73504f47898150a406d7a633 : SUCCESS in 44m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/14d76a20c20849a1aba3b25ed8c7d2d3 : SUCCESS in 32m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/831e7a40474c494e91c86cbb14c514fc : SUCCESS in 28m 22s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/efb80ee098b742d790d898bbfafe1948 : SUCCESS in 3h 10m 17s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c81c58cbc728444c83d8b6b7f3c4e12b : SUCCESS in 1h 00m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/884c579519d84c649ef7c285e1e888b9 : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":6},{"id":"63ecd62c1cb5819e8ac30a3cb1f2a7b141a17a5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:03:16.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c1c69c5fca243fa633a6d765baf3b66f72c4ab32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 17:21:39.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1be16e8f75b64631abaa40e5c7011466 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3aad96c564e44dabdb02695932295c3 : SUCCESS in 10m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e1fd5837ded4e89b3e0896134fbcff4 : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67463ce17d7744eba35fe80653846132 : SUCCESS in 9m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/290be80157fa4a28b0c90e643f520407 : SUCCESS in 2m 53s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b8df373b96944ae9956ae14c1261e3e9 : SUCCESS in 47m 39s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6a888152d6048bc968251ac3b22ef1e : SUCCESS in 56m 47s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/034b00082d3c424ca3db995ae81217f5 : SUCCESS in 40m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4f5fbdd8b6974a329d803ece0e334253 : SUCCESS in 47m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/72aa7aad7dd24c9da9a466220cda5bf0 : FAILURE in 30m 50s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/84388bd3ab16406f853f32900e554752 : SUCCESS in 28m 50s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8a235b442b604dee999c84ad14577379 : SUCCESS in 43m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca170862cc6f498cae59a31169df6497 : SUCCESS in 12m 28s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2b282334d62f4239aac854092b2d6427 : SUCCESS in 1h 46m 07s","accounts_in_message":[],"_revision_number":6},{"id":"e6a9d7e0f9831364cfea3cbda097579b48a9e97f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-23 03:51:23.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"14eb2b5c069329e3bfff230ae9b73602f1d2bb38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 06:54:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b1e92f0bbe4907883561f9080e5629 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/702f4b02876e4100a2e6fd8ceb1e2b06 : SUCCESS in 8m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/653701223f2d4d66a908b636935c543d : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe7641df0c0b45ae9bf70b1111bba870 : SUCCESS in 7m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b852b1a8649243dcb360e96f49455234 : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edca1d1f86ff40cfa1370a5873b5f9f2 : SUCCESS in 8m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/142d081aced54f7a917a2d85ab27aa41 : SUCCESS in 38m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64fa34d6940043ef9e2b2d5163e979fc : SUCCESS in 43m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d70e7271ccf74c4591e7a16b3b1ce415 : SUCCESS in 41m 29s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6af52596e00745b695e389a4b5f8050b : SUCCESS in 42m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3582c4fc37524f47a3065927296bf7ca : SUCCESS in 42m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ea3f7c6e4f9f423499c2db2f768b8009 : SUCCESS in 29m 01s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f214660d88c2417ba8e40bee590c8623 : SUCCESS in 27m 04s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7b4651a693624a94b63db2ce989dc45e : SUCCESS in 2h 49m 53s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7b2ed6486e634becb80696c26ab6a662 : SUCCESS in 58m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/879b3899bf8341dda8d8a985361c262a : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":6},{"id":"9ee5d0807a57453df1f6273e514e242db9f68a0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 06:54:55.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9b68809da60adead7dfa11c287e276e925c87bf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 09:13:06.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c464041d90fa4843adfcb74e38fd7c65 : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2643d839f5ca44c880784338eb263597 : SUCCESS in 9m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8073355428b44c908f15a81177b37946 : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d83e1795aa604751b663e8a10fb865eb : SUCCESS in 12m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d6be1844688d4b38a10df18e660d4489 : SUCCESS in 5m 46s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/123540e4777d419a9e667fc0c46bf166 : SUCCESS in 50m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f35d2b12dadc4973a7b3b2b9e88462cd : SUCCESS in 42m 17s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/085a5292939f4c598bdb87151405df01 : SUCCESS in 54m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2d39fb5bdf8649bb98202b353cefc733 : SUCCESS in 52m 37s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1c7ea93c73b24d30b33513e537b91743 : SUCCESS in 34m 21s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4bb527f9e59d4007b07569a34c9495e6 : SUCCESS in 27m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7f261eb6855f46fa905ccf9893546412 : SUCCESS in 48m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/883d2800fd2f491086072de7cdfb96e2 : SUCCESS in 12m 22s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ae418f3967b546cda14e8e4ed4e074a4 : SUCCESS in 2h 15m 02s","accounts_in_message":[],"_revision_number":6},{"id":"d058e3cd68b974eae98e70a1d5d10a505cc766be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 09:13:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"6682037b6328cceec6a841e77c33e6496d4f80e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 09:15:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0503f91768d243dba268eaf2e8100312 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c1c21d7cddca2f307e93f42b4758e64fcdf0a630","revisions":{"fbbc00e78ba222d9b5a43832c56da4d72debac19":{"kind":"REWORK","_number":1,"created":"2019-02-13 10:08:23.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/83/636583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/83/636583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/83/636583/1"}}},"commit":{"parents":[{"commit":"58d6634702014eb3c2bf70dbe487fe5983cef8e9","subject":"Merge \"Add a workflow execution report endpoint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/58d6634702014eb3c2bf70dbe487fe5983cef8e9"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-02-11 22:43:09.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-02-11 22:43:09.000000000","tz":180},"subject":"Use yamlloader to preserve key order in dicts","message":"Use yamlloader to preserve key order in dicts\n\nBy default, yaml library doesn\u0027t guarantee the key order in load and\ndump functions. It could be fixed by custom dict_representer\nand dict_constructor, but it would be very complicated to handle\nboth 2.7 and 3.x versions, so we could use yamlloader, which provides\nthis solution OOB.\n\nChange-Id: I888008ac8f9c12bd92a9bd72bd7c276a11827847\nCloses-Bug: #1815515\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fbbc00e78ba222d9b5a43832c56da4d72debac19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fbbc00e78ba222d9b5a43832c56da4d72debac19"}]},"branch":"refs/heads/master"},"8f306d063181489c0a2d7e5d374baa08010c964f":{"kind":"REWORK","_number":2,"created":"2019-02-13 12:15:51.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/83/636583/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/83/636583/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/83/636583/2"}}},"commit":{"parents":[{"commit":"58d6634702014eb3c2bf70dbe487fe5983cef8e9","subject":"Merge \"Add a workflow execution report endpoint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/58d6634702014eb3c2bf70dbe487fe5983cef8e9"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-02-11 22:43:09.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-02-13 12:13:12.000000000","tz":180},"subject":"Use yamlloader to preserve key order in dicts","message":"Use yamlloader to preserve key order in dicts\n\nBy default, yaml library doesn\u0027t guarantee the key order in load and\ndump functions. It could be fixed by custom dict_representer\nand dict_constructor, but it would be very complicated to handle\nboth 2.7 and 3.x versions, so we could use yamlloader, which provides\nthis solution OOB.\n\nChange-Id: I888008ac8f9c12bd92a9bd72bd7c276a11827847\nCloses-Bug: #1815515\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8f306d063181489c0a2d7e5d374baa08010c964f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8f306d063181489c0a2d7e5d374baa08010c964f"}]},"branch":"refs/heads/master"},"4f624af656a33fbccd2ddfd97f41bcf3f294b426":{"kind":"REWORK","_number":3,"created":"2020-01-16 22:18:18.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/83/636583/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/83/636583/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/83/636583/3"}}},"commit":{"parents":[{"commit":"55f181f8aaf8cdc7d519c1e446d0053cd73368a6","subject":"Fix requirements remove py2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55f181f8aaf8cdc7d519c1e446d0053cd73368a6"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-02-11 22:43:09.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-16 22:18:04.000000000","tz":180},"subject":"Disable key sorting in workflow definition","message":"Disable key sorting in workflow definition\n\nUpdate to PyYAML 5.1 to be able to turn off key sorting\nin workflow definitions. We should not change user\u0027s data\nuntil user asks to.\n\nChange-Id: I888008ac8f9c12bd92a9bd72bd7c276a11827847\nCloses-Bug: #1815515\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4f624af656a33fbccd2ddfd97f41bcf3f294b426"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4f624af656a33fbccd2ddfd97f41bcf3f294b426"}]},"branch":"refs/heads/master"},"b6b21ac12dac115a51877bfcda1986e3f8ed69a5":{"kind":"REWORK","_number":4,"created":"2020-01-17 08:29:07.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/83/636583/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/83/636583/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/83/636583/4"}}},"commit":{"parents":[{"commit":"55f181f8aaf8cdc7d519c1e446d0053cd73368a6","subject":"Fix requirements remove py2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55f181f8aaf8cdc7d519c1e446d0053cd73368a6"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-02-11 22:43:09.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-17 08:28:35.000000000","tz":180},"subject":"Disable key sorting in workflow definition","message":"Disable key sorting in workflow definition\n\nUpdate to PyYAML 5.1 to be able to turn off key sorting\nin workflow definitions. We should not change user\u0027s data\nuntil user asks to.\n\nChange-Id: I888008ac8f9c12bd92a9bd72bd7c276a11827847\nCloses-Bug: #1815515\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b6b21ac12dac115a51877bfcda1986e3f8ed69a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b6b21ac12dac115a51877bfcda1986e3f8ed69a5"}]},"branch":"refs/heads/master"},"410ce0f524bd466a1d29bc870b27c02a729d6bd3":{"kind":"REWORK","_number":5,"created":"2020-01-17 11:29:45.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/83/636583/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/83/636583/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/83/636583/5"}}},"commit":{"parents":[{"commit":"4ef34609eb7e007665b59455f71efc508d883c84","subject":"Merge \"Fix keycloak authentication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4ef34609eb7e007665b59455f71efc508d883c84"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-02-11 22:43:09.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-17 11:29:25.000000000","tz":180},"subject":"Disable key sorting in workflow definition","message":"Disable key sorting in workflow definition\n\nUpdate to PyYAML 5.1 to be able to turn off key sorting\nin workflow definitions. We should not change user\u0027s data\nuntil user asks to.\n\nChange-Id: I888008ac8f9c12bd92a9bd72bd7c276a11827847\nCloses-Bug: #1815515\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/410ce0f524bd466a1d29bc870b27c02a729d6bd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/410ce0f524bd466a1d29bc870b27c02a729d6bd3"}]},"branch":"refs/heads/master"},"c1c21d7cddca2f307e93f42b4758e64fcdf0a630":{"kind":"REWORK","_number":6,"created":"2020-01-20 21:40:37.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/83/636583/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/83/636583/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/83/636583/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/83/636583/6"}}},"commit":{"parents":[{"commit":"4ef34609eb7e007665b59455f71efc508d883c84","subject":"Merge \"Fix keycloak authentication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4ef34609eb7e007665b59455f71efc508d883c84"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-02-11 22:43:09.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-20 21:40:26.000000000","tz":180},"subject":"Disable key sorting in workflow definition","message":"Disable key sorting in workflow definition\n\nUpdate to PyYAML 5.1 to be able to turn off key sorting\nin workflow definitions. We should not change user\u0027s data\nuntil user asks to.\n\nChange-Id: I888008ac8f9c12bd92a9bd72bd7c276a11827847\nCloses-Bug: #1815515\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c1c21d7cddca2f307e93f42b4758e64fcdf0a630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c1c21d7cddca2f307e93f42b4758e64fcdf0a630"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
