)]}'
{"id":"openstack%2Fmistral~652584","triplet_id":"openstack%2Fmistral~master~Ibdadb08b23738490646db676312bc87c3bf27eaa","project":"openstack/mistral","branch":"master","topic":"eyalb/py3","hashtags":[],"change_id":"Ibdadb08b23738490646db676312bc87c3bf27eaa","subject":"inspect.getargspec is deprecated on py3","status":"MERGED","created":"2019-04-15 07:35:05.000000000","updated":"2019-05-15 10:02:58.000000000","submitted":"2019-05-15 10:02:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652584-1557914578999-88673ecb","meta_rev_id":"a3c62d0518d2e2f5bcc0f5d2f51411edd3af24b1","_number":652584,"virtual_id_number":652584,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-05-15 10:02:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-15 06:19:23.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"all":[{"value":1,"date":"2019-05-02 08:28:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-05-02 17:46:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":1,"date":"2019-05-15 03:20:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-02 08:28:16.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-05-15 03:20:45.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-05-15 10:02:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c53b4b07efb4888af70e3bf696f6a7c40200fc8a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-04-15 07:35:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7ffb078c4f8212739a016e7f2f3a2bbc2db2da6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 09:44:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/84/652584/1/check/openstack-tox-pep8/a11fd09/ : SUCCESS in 6m 00s\n- openstack-tox-py27 http://logs.openstack.org/84/652584/1/check/openstack-tox-py27/46443bd/ : SUCCESS in 9m 16s\n- openstack-tox-py36 http://logs.openstack.org/84/652584/1/check/openstack-tox-py36/47ff937/ : SUCCESS in 8m 58s\n- openstack-tox-py37 http://logs.openstack.org/84/652584/1/check/openstack-tox-py37/1826cb4/ : FAILURE in 9m 11s\n- openstack-tox-docs http://logs.openstack.org/84/652584/1/check/openstack-tox-docs/085ffeb/html/ : SUCCESS in 5m 25s\n- openstack-tox-cover http://logs.openstack.org/84/652584/1/check/openstack-tox-cover/5c202d5/cover/ : SUCCESS in 9m 19s (non-voting)\n- mistral-devstack http://logs.openstack.org/84/652584/1/check/mistral-devstack/35301ea/ : SUCCESS in 46m 55s\n- mistral-devstack-non-apache http://logs.openstack.org/84/652584/1/check/mistral-devstack-non-apache/1ec75bb/ : SUCCESS in 42m 16s\n- mistral-devstack-kombu http://logs.openstack.org/84/652584/1/check/mistral-devstack-kombu/3ba9cb9/ : SUCCESS in 44m 13s\n- mistral-tox-unit-mysql http://logs.openstack.org/84/652584/1/check/mistral-tox-unit-mysql/a1cac19/ : SUCCESS in 27m 37s\n- mistral-tox-unit-postgresql http://logs.openstack.org/84/652584/1/check/mistral-tox-unit-postgresql/b494ab0/ : SUCCESS in 26m 14s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/84/652584/1/check/tripleo-ci-centos-7-scenario003-standalone/698c4ab/ : SUCCESS in 1h 08m 19s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/652584/1/check/tripleo-ci-centos-7-containers-multinode/eea75aa/ : SUCCESS in 1h 53m 25s\n- mistral-rally-task http://logs.openstack.org/84/652584/1/check/mistral-rally-task/389b137/ : SUCCESS in 1h 05m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/84/652584/1/check/openstack-tox-lower-constraints/76502e5/ : SUCCESS in 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3719670a71c50519ae74a5e0fc1aab54b932cc79","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-04-15 13:06:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e8d7bb1f9e74d34685d31bf02538cdf086367525","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 15:15:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/84/652584/1/check/openstack-tox-pep8/66dd6ff/ : SUCCESS in 6m 06s\n- openstack-tox-py27 http://logs.openstack.org/84/652584/1/check/openstack-tox-py27/9d3aa83/ : SUCCESS in 9m 53s\n- openstack-tox-py36 http://logs.openstack.org/84/652584/1/check/openstack-tox-py36/d4a179b/ : SUCCESS in 10m 27s\n- openstack-tox-py37 http://logs.openstack.org/84/652584/1/check/openstack-tox-py37/5287301/ : FAILURE in 9m 39s\n- openstack-tox-docs http://logs.openstack.org/84/652584/1/check/openstack-tox-docs/bc4729f/html/ : SUCCESS in 5m 19s\n- openstack-tox-cover http://logs.openstack.org/84/652584/1/check/openstack-tox-cover/2a1d296/cover/ : SUCCESS in 9m 35s (non-voting)\n- mistral-devstack http://logs.openstack.org/84/652584/1/check/mistral-devstack/1e96e80/ : SUCCESS in 48m 36s\n- mistral-devstack-non-apache http://logs.openstack.org/84/652584/1/check/mistral-devstack-non-apache/d5f8439/ : SUCCESS in 45m 28s\n- mistral-devstack-kombu http://logs.openstack.org/84/652584/1/check/mistral-devstack-kombu/459db73/ : SUCCESS in 45m 53s\n- mistral-tox-unit-mysql http://logs.openstack.org/84/652584/1/check/mistral-tox-unit-mysql/dd23fcc/ : SUCCESS in 29m 29s\n- mistral-tox-unit-postgresql http://logs.openstack.org/84/652584/1/check/mistral-tox-unit-postgresql/eb758d8/ : SUCCESS in 27m 56s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/84/652584/1/check/tripleo-ci-centos-7-scenario003-standalone/2b4e1d3/ : SUCCESS in 1h 05m 08s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/652584/1/check/tripleo-ci-centos-7-containers-multinode/fd7f71a/ : SUCCESS in 1h 58m 32s\n- mistral-rally-task http://logs.openstack.org/84/652584/1/check/mistral-rally-task/b05fae6/ : SUCCESS in 1h 09m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/84/652584/1/check/openstack-tox-lower-constraints/107fb76/ : SUCCESS in 11m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b72817994482d8425c18a7034644b8d0854829bb","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-04-27 10:21:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"85de880e0e64a5a5ae5b4dfa7a0e01b05ce7270d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 13:18:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/84/652584/1/check/openstack-tox-pep8/e4aa4ca/ : SUCCESS in 7m 20s\n- openstack-tox-py27 http://logs.openstack.org/84/652584/1/check/openstack-tox-py27/4a93403/ : SUCCESS in 9m 29s\n- openstack-tox-py36 http://logs.openstack.org/84/652584/1/check/openstack-tox-py36/8ea6479/ : SUCCESS in 10m 04s\n- openstack-tox-py37 http://logs.openstack.org/84/652584/1/check/openstack-tox-py37/3483b64/ : SUCCESS in 10m 14s\n- openstack-tox-docs http://logs.openstack.org/84/652584/1/check/openstack-tox-docs/0da5f2a/html/ : SUCCESS in 7m 10s\n- openstack-tox-cover http://logs.openstack.org/84/652584/1/check/openstack-tox-cover/5845aab/cover/ : SUCCESS in 8m 09s (non-voting)\n- mistral-devstack http://logs.openstack.org/84/652584/1/check/mistral-devstack/f9c865a/ : SUCCESS in 42m 41s\n- mistral-devstack-non-apache http://logs.openstack.org/84/652584/1/check/mistral-devstack-non-apache/e897395/ : SUCCESS in 47m 30s\n- mistral-devstack-kombu http://logs.openstack.org/84/652584/1/check/mistral-devstack-kombu/4c75807/ : SUCCESS in 47m 14s\n- mistral-tox-unit-mysql http://logs.openstack.org/84/652584/1/check/mistral-tox-unit-mysql/5c0a94b/ : SUCCESS in 29m 12s\n- mistral-tox-unit-postgresql http://logs.openstack.org/84/652584/1/check/mistral-tox-unit-postgresql/186cd67/ : SUCCESS in 30m 45s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/84/652584/1/check/tripleo-ci-centos-7-scenario003-standalone/b88f959/ : SUCCESS in 1h 11m 17s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/652584/1/check/tripleo-ci-centos-7-containers-multinode/9cc562d/ : SUCCESS in 2h 01m 40s\n- mistral-rally-task http://logs.openstack.org/84/652584/1/check/mistral-rally-task/895b5e2/ : SUCCESS in 1h 00m 11s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/84/652584/1/check/openstack-tox-lower-constraints/88b0369/ : SUCCESS in 11m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7eabff3e0f75bfed92a8a8ca8206738fb1cf28db","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-05-02 08:28:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"68f9f422d787d9e6330a449c76bc4e844dcaf4ef","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-02 17:46:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"83ae013038d375ffa0c2ac1e44733163d2dd8a9d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-15 03:20:45.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7ff6c396518cbd8162e8d7b021c80ed7121213f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 03:20:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b04e85192e26dc52d970a42ffecd556d3ed8f577","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 05:01:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/84/652584/1/gate/openstack-tox-pep8/6f135c6/ : SUCCESS in 6m 22s\n- openstack-tox-py27 http://logs.openstack.org/84/652584/1/gate/openstack-tox-py27/72227df/ : SUCCESS in 11m 26s\n- openstack-tox-py36 http://logs.openstack.org/84/652584/1/gate/openstack-tox-py36/8c6386a/ : SUCCESS in 8m 57s\n- openstack-tox-py37 http://logs.openstack.org/84/652584/1/gate/openstack-tox-py37/29615dc/ : SUCCESS in 10m 08s\n- openstack-tox-docs http://logs.openstack.org/84/652584/1/gate/openstack-tox-docs/563a04c/html/ : SUCCESS in 6m 09s\n- mistral-devstack http://logs.openstack.org/84/652584/1/gate/mistral-devstack/4c6013e/ : SUCCESS in 46m 43s\n- mistral-devstack-non-apache http://logs.openstack.org/84/652584/1/gate/mistral-devstack-non-apache/0617925/ : SUCCESS in 35m 46s\n- mistral-tox-unit-mysql http://logs.openstack.org/84/652584/1/gate/mistral-tox-unit-mysql/e98e5d5/ : FAILURE in 28m 29s\n- mistral-tox-unit-postgresql http://logs.openstack.org/84/652584/1/gate/mistral-tox-unit-postgresql/510b478/ : SUCCESS in 27m 19s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/84/652584/1/gate/tripleo-ci-centos-7-scenario003-standalone/7bbea09/ : SUCCESS in 1h 11m 25s\n- mistral-devstack-kombu http://logs.openstack.org/84/652584/1/gate/mistral-devstack-kombu/8ca432f/ : SUCCESS in 35m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/84/652584/1/gate/openstack-tox-lower-constraints/21d84d6/ : SUCCESS in 10m 15s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/652584/1/gate/tripleo-ci-centos-7-undercloud-containers/daa30fc/ : SUCCESS in 1h 39m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f041a5b07bc68dae5d67efe5d41f073c3a9a0b4c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-05-15 06:19:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2544352b9ddd1eceb1f27d46a7ccd81da5a5606c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-15 06:29:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ac99009c169488fe9c225a3223be51e7c51e948d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 08:31:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/84/652584/1/check/openstack-tox-pep8/3884894/ : SUCCESS in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/84/652584/1/check/openstack-tox-py27/3a06466/ : SUCCESS in 8m 11s\n- openstack-tox-py36 http://logs.openstack.org/84/652584/1/check/openstack-tox-py36/8742429/ : SUCCESS in 9m 45s\n- openstack-tox-py37 http://logs.openstack.org/84/652584/1/check/openstack-tox-py37/2665dc4/ : SUCCESS in 9m 40s\n- openstack-tox-docs http://logs.openstack.org/84/652584/1/check/openstack-tox-docs/cc17dd3/html/ : SUCCESS in 5m 09s\n- openstack-tox-cover http://logs.openstack.org/84/652584/1/check/openstack-tox-cover/e12fe37/cover/ : SUCCESS in 9m 32s (non-voting)\n- mistral-devstack http://logs.openstack.org/84/652584/1/check/mistral-devstack/02115f3/ : SUCCESS in 40m 21s\n- mistral-devstack-non-apache http://logs.openstack.org/84/652584/1/check/mistral-devstack-non-apache/a289fd5/ : SUCCESS in 46m 14s\n- mistral-devstack-kombu http://logs.openstack.org/84/652584/1/check/mistral-devstack-kombu/243fd28/ : SUCCESS in 46m 27s\n- mistral-tox-unit-mysql http://logs.openstack.org/84/652584/1/check/mistral-tox-unit-mysql/6cea2bd/ : SUCCESS in 27m 39s\n- mistral-tox-unit-postgresql http://logs.openstack.org/84/652584/1/check/mistral-tox-unit-postgresql/fef0078/ : SUCCESS in 29m 40s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/84/652584/1/check/tripleo-ci-centos-7-scenario003-standalone/ea4329a/ : SUCCESS in 1h 15m 15s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/652584/1/check/tripleo-ci-centos-7-containers-multinode/a041e7e/ : SUCCESS in 1h 59m 33s\n- mistral-rally-task http://logs.openstack.org/84/652584/1/check/mistral-rally-task/84af441/ : SUCCESS in 1h 04m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/84/652584/1/check/openstack-tox-lower-constraints/5e7fe26/ : SUCCESS in 10m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5a69ef8d2911386937a0518360324d273472aa5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 08:31:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"776a9619290f43dbf80a17d961c010ffbf92e45b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 10:02:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/84/652584/1/gate/openstack-tox-pep8/e2c7c29/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/84/652584/1/gate/openstack-tox-py27/3665c08/ : SUCCESS in 12m 09s\n- openstack-tox-py36 http://logs.openstack.org/84/652584/1/gate/openstack-tox-py36/965a0ed/ : SUCCESS in 10m 32s\n- openstack-tox-py37 http://logs.openstack.org/84/652584/1/gate/openstack-tox-py37/74f3905/ : SUCCESS in 13m 50s\n- openstack-tox-docs http://logs.openstack.org/84/652584/1/gate/openstack-tox-docs/614a3ce/html/ : SUCCESS in 5m 06s\n- mistral-devstack http://logs.openstack.org/84/652584/1/gate/mistral-devstack/9a9d564/ : SUCCESS in 43m 54s\n- mistral-devstack-non-apache http://logs.openstack.org/84/652584/1/gate/mistral-devstack-non-apache/b442fb4/ : SUCCESS in 43m 55s\n- mistral-tox-unit-mysql http://logs.openstack.org/84/652584/1/gate/mistral-tox-unit-mysql/e53ceff/ : SUCCESS in 27m 19s\n- mistral-tox-unit-postgresql http://logs.openstack.org/84/652584/1/gate/mistral-tox-unit-postgresql/c7b95b6/ : SUCCESS in 25m 55s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/84/652584/1/gate/tripleo-ci-centos-7-scenario003-standalone/8190cd6/ : SUCCESS in 1h 06m 37s\n- mistral-devstack-kombu http://logs.openstack.org/84/652584/1/gate/mistral-devstack-kombu/98bfdda/ : SUCCESS in 45m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/84/652584/1/gate/openstack-tox-lower-constraints/856115c/ : SUCCESS in 10m 10s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/652584/1/gate/tripleo-ci-centos-7-undercloud-containers/ce9b4cf/ : SUCCESS in 1h 25m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e036100693bcb4e439144c8b7a6d46619421cbee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 10:02:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59a96b73f179cbd4428b15fa5b040855f65ebe41","revisions":{"59a96b73f179cbd4428b15fa5b040855f65ebe41":{"kind":"REWORK","_number":1,"created":"2019-04-15 07:35:05.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/84/652584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/84/652584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/84/652584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/84/652584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/84/652584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/84/652584/1"}}},"commit":{"parents":[{"commit":"401eeddb2d4a9cc4f17876a8d88c6d877c95eb4a","subject":"Fix E305 codestyle errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/401eeddb2d4a9cc4f17876a8d88c6d877c95eb4a"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-04-15 07:30:20.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-04-15 07:30:20.000000000","tz":180},"subject":"inspect.getargspec is deprecated on py3","message":"inspect.getargspec is deprecated on py3\n\nuse inspect.getfullargspec instead\n\nChange-Id: Ibdadb08b23738490646db676312bc87c3bf27eaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/59a96b73f179cbd4428b15fa5b040855f65ebe41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/59a96b73f179cbd4428b15fa5b040855f65ebe41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
