)]}'
{"id":"openstack%2Fmistral~653134","triplet_id":"openstack%2Fmistral~master~I895bee811496f920b075880a6c438c53f7ecb2ca","project":"openstack/mistral","branch":"master","topic":"subworkflow-remote-run","hashtags":[],"change_id":"I895bee811496f920b075880a6c438c53f7ecb2ca","subject":"Add option to start subworkflow via RPC.","status":"MERGED","created":"2019-04-16 20:00:53.000000000","updated":"2019-04-26 08:25:16.000000000","submitted":"2019-04-26 08:25:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653134-1556267116632-36e30e48","meta_rev_id":"2bf6eb230d5bb0d7e1e351373c769e3005a0d60c","_number":653134,"virtual_id_number":653134,"owner":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-04-26 08:25:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-04-26 05:04:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-04-26 06:53:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-04-26 06:53:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 10:56:57.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-04-24 10:56:57.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-04-24 10:56:57.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-04-24 10:56:57.000000000","updated_by":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"reviewer":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"state":"REVIEWER"},{"updated":"2019-04-26 05:04:28.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-04-26 06:53:49.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-04-26 08:25:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7af3686894bc458f03d2422ad6ea757c613e7da1","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-04-16 20:00:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"124333bc29ff1baa2f107add47e522df73a32cca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 22:28:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/34/653134/1/check/openstack-tox-pep8/957ed4a/ : FAILURE in 5m 41s\n- openstack-tox-py27 http://logs.openstack.org/34/653134/1/check/openstack-tox-py27/6df6ab7/ : SUCCESS in 9m 06s\n- openstack-tox-py36 http://logs.openstack.org/34/653134/1/check/openstack-tox-py36/1b9b932/ : SUCCESS in 11m 26s\n- openstack-tox-py37 http://logs.openstack.org/34/653134/1/check/openstack-tox-py37/9cb054c/ : SUCCESS in 9m 37s\n- openstack-tox-docs http://logs.openstack.org/34/653134/1/check/openstack-tox-docs/654aa21/html/ : SUCCESS in 6m 55s\n- openstack-tox-cover http://logs.openstack.org/34/653134/1/check/openstack-tox-cover/51fef26/cover/ : SUCCESS in 10m 19s (non-voting)\n- mistral-devstack http://logs.openstack.org/34/653134/1/check/mistral-devstack/f50e5e8/ : SUCCESS in 44m 44s\n- mistral-devstack-non-apache http://logs.openstack.org/34/653134/1/check/mistral-devstack-non-apache/c627bc8/ : SUCCESS in 49m 08s\n- mistral-devstack-kombu http://logs.openstack.org/34/653134/1/check/mistral-devstack-kombu/522dfd4/ : SUCCESS in 53m 48s\n- mistral-tox-unit-mysql http://logs.openstack.org/34/653134/1/check/mistral-tox-unit-mysql/95acba7/ : FAILURE in 28m 22s\n- mistral-tox-unit-postgresql http://logs.openstack.org/34/653134/1/check/mistral-tox-unit-postgresql/64fc027/ : FAILURE in 26m 57s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/34/653134/1/check/tripleo-ci-centos-7-scenario003-standalone/0a030d0/ : SUCCESS in 1h 11m 39s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/34/653134/1/check/tripleo-ci-centos-7-containers-multinode/b0f1c61/ : SUCCESS in 2h 23m 17s\n- mistral-rally-task http://logs.openstack.org/34/653134/1/check/mistral-rally-task/87dccba/ : SUCCESS in 1h 08m 04s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/34/653134/1/check/openstack-tox-lower-constraints/19eb5df/ : SUCCESS in 10m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e854f5b9e7827a86a3997cc369af97024580f72b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-04-17 05:29:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d3916a97fc24f576e481c9f498d519e0dbc0639c","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-04-17 06:40:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b82e85c460b225f538197e30aaa39992622bb0f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 08:57:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/34/653134/2/check/openstack-tox-pep8/2117526/ : SUCCESS in 6m 03s\n- openstack-tox-py27 http://logs.openstack.org/34/653134/2/check/openstack-tox-py27/febfa67/ : SUCCESS in 14m 24s\n- openstack-tox-py36 http://logs.openstack.org/34/653134/2/check/openstack-tox-py36/42395c2/ : SUCCESS in 9m 19s\n- openstack-tox-py37 http://logs.openstack.org/34/653134/2/check/openstack-tox-py37/8498a7b/ : SUCCESS in 11m 28s\n- openstack-tox-docs http://logs.openstack.org/34/653134/2/check/openstack-tox-docs/0158caa/html/ : SUCCESS in 6m 32s\n- openstack-tox-cover http://logs.openstack.org/34/653134/2/check/openstack-tox-cover/8f511f0/cover/ : SUCCESS in 9m 42s (non-voting)\n- mistral-devstack http://logs.openstack.org/34/653134/2/check/mistral-devstack/df1e147/ : SUCCESS in 42m 57s\n- mistral-devstack-non-apache http://logs.openstack.org/34/653134/2/check/mistral-devstack-non-apache/4474c9a/ : SUCCESS in 46m 39s\n- mistral-devstack-kombu http://logs.openstack.org/34/653134/2/check/mistral-devstack-kombu/0429d6e/ : SUCCESS in 52m 52s\n- mistral-tox-unit-mysql http://logs.openstack.org/34/653134/2/check/mistral-tox-unit-mysql/749bdb0/ : FAILURE in 29m 39s\n- mistral-tox-unit-postgresql http://logs.openstack.org/34/653134/2/check/mistral-tox-unit-postgresql/1d813bc/ : SUCCESS in 29m 40s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/34/653134/2/check/tripleo-ci-centos-7-scenario003-standalone/0c44279/ : SUCCESS in 1h 14m 18s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/34/653134/2/check/tripleo-ci-centos-7-containers-multinode/5bdc62a/ : SUCCESS in 2h 11m 05s\n- mistral-rally-task http://logs.openstack.org/34/653134/2/check/mistral-rally-task/43ddaf5/ : SUCCESS in 1h 11m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/34/653134/2/check/openstack-tox-lower-constraints/26a480b/ : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":2},{"id":"4bdc738a44ae34d67900102c5d70dd3b9362253a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-04-17 09:08:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3ff94e0e9886cd829b4ab7d9ecd1591481dba62f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 11:57:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/34/653134/2/check/openstack-tox-pep8/a0167a4/ : SUCCESS in 7m 05s\n- openstack-tox-py27 http://logs.openstack.org/34/653134/2/check/openstack-tox-py27/acda794/ : SUCCESS in 10m 02s\n- openstack-tox-py36 http://logs.openstack.org/34/653134/2/check/openstack-tox-py36/55b8003/ : SUCCESS in 11m 14s\n- openstack-tox-py37 http://logs.openstack.org/34/653134/2/check/openstack-tox-py37/e07638d/ : FAILURE in 10m 11s\n- openstack-tox-docs http://logs.openstack.org/34/653134/2/check/openstack-tox-docs/bd0eff5/html/ : SUCCESS in 6m 16s\n- openstack-tox-cover http://logs.openstack.org/34/653134/2/check/openstack-tox-cover/e4587ec/cover/ : SUCCESS in 10m 50s (non-voting)\n- mistral-devstack http://logs.openstack.org/34/653134/2/check/mistral-devstack/7ed0dfe/ : SUCCESS in 43m 34s\n- mistral-devstack-non-apache http://logs.openstack.org/34/653134/2/check/mistral-devstack-non-apache/0ee7305/ : SUCCESS in 47m 58s\n- mistral-devstack-kombu http://logs.openstack.org/34/653134/2/check/mistral-devstack-kombu/dec2451/ : SUCCESS in 44m 22s\n- mistral-tox-unit-mysql http://logs.openstack.org/34/653134/2/check/mistral-tox-unit-mysql/e68d993/ : SUCCESS in 32m 26s\n- mistral-tox-unit-postgresql http://logs.openstack.org/34/653134/2/check/mistral-tox-unit-postgresql/7505f73/ : SUCCESS in 26m 52s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/34/653134/2/check/tripleo-ci-centos-7-scenario003-standalone/6475039/ : SUCCESS in 1h 17m 20s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/34/653134/2/check/tripleo-ci-centos-7-containers-multinode/8b18df0/ : SUCCESS in 2h 02m 33s\n- mistral-rally-task http://logs.openstack.org/34/653134/2/check/mistral-rally-task/b47fcf1/ : SUCCESS in 1h 04m 47s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/34/653134/2/check/openstack-tox-lower-constraints/aa98348/ : SUCCESS in 14m 19s","accounts_in_message":[],"_revision_number":2},{"id":"3352711d83f67479fb2bcc5b20e96776e7779887","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-04-17 14:40:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9f0be54fee7bd01300e6252948d2e79deea5b37f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 17:18:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/34/653134/2/check/openstack-tox-pep8/8da8de9/ : SUCCESS in 5m 35s\n- openstack-tox-py27 http://logs.openstack.org/34/653134/2/check/openstack-tox-py27/2a116a3/ : SUCCESS in 8m 47s\n- openstack-tox-py36 http://logs.openstack.org/34/653134/2/check/openstack-tox-py36/a6d871f/ : SUCCESS in 11m 09s\n- openstack-tox-py37 http://logs.openstack.org/34/653134/2/check/openstack-tox-py37/ce6924d/ : FAILURE in 10m 13s\n- openstack-tox-docs http://logs.openstack.org/34/653134/2/check/openstack-tox-docs/9745faf/html/ : SUCCESS in 6m 20s\n- openstack-tox-cover http://logs.openstack.org/34/653134/2/check/openstack-tox-cover/c77719c/cover/ : SUCCESS in 10m 39s (non-voting)\n- mistral-devstack http://logs.openstack.org/34/653134/2/check/mistral-devstack/a60eca4/ : SUCCESS in 47m 28s\n- mistral-devstack-non-apache http://logs.openstack.org/34/653134/2/check/mistral-devstack-non-apache/3dbfdea/ : SUCCESS in 50m 08s\n- mistral-devstack-kombu http://logs.openstack.org/34/653134/2/check/mistral-devstack-kombu/ca078e5/ : SUCCESS in 53m 06s\n- mistral-tox-unit-mysql http://logs.openstack.org/34/653134/2/check/mistral-tox-unit-mysql/c3b1d47/ : SUCCESS in 27m 23s\n- mistral-tox-unit-postgresql http://logs.openstack.org/34/653134/2/check/mistral-tox-unit-postgresql/2bfefb7/ : SUCCESS in 26m 18s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/34/653134/2/check/tripleo-ci-centos-7-scenario003-standalone/db2cdbf/ : SUCCESS in 1h 15m 40s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/34/653134/2/check/tripleo-ci-centos-7-containers-multinode/6b3a20d/ : SUCCESS in 2h 03m 22s\n- mistral-rally-task http://logs.openstack.org/34/653134/2/check/mistral-rally-task/347385d/ : SUCCESS in 1h 07m 27s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/34/653134/2/check/openstack-tox-lower-constraints/2b0bf12/ : SUCCESS in 12m 45s","accounts_in_message":[],"_revision_number":2},{"id":"4bcc973bee611e076f2fb48f4e90ee6b475cd509","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-04-17 22:12:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b98f63c2c5f67a35dc387ad41e916a9645397c4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 01:12:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/34/653134/2/check/openstack-tox-pep8/85db7b6/ : SUCCESS in 7m 46s\n- openstack-tox-py27 http://logs.openstack.org/34/653134/2/check/openstack-tox-py27/367797c/ : SUCCESS in 8m 57s\n- openstack-tox-py36 http://logs.openstack.org/34/653134/2/check/openstack-tox-py36/b8d4b0e/ : SUCCESS in 10m 54s\n- openstack-tox-py37 http://logs.openstack.org/34/653134/2/check/openstack-tox-py37/964522a/ : FAILURE in 10m 13s\n- openstack-tox-docs http://logs.openstack.org/34/653134/2/check/openstack-tox-docs/ce1d2d1/html/ : SUCCESS in 7m 14s\n- openstack-tox-cover http://logs.openstack.org/34/653134/2/check/openstack-tox-cover/70f72ec/cover/ : SUCCESS in 11m 35s (non-voting)\n- mistral-devstack http://logs.openstack.org/34/653134/2/check/mistral-devstack/ce2f777/ : SUCCESS in 49m 20s\n- mistral-devstack-non-apache http://logs.openstack.org/34/653134/2/check/mistral-devstack-non-apache/026e18c/ : SUCCESS in 44m 50s\n- mistral-devstack-kombu http://logs.openstack.org/34/653134/2/check/mistral-devstack-kombu/b31e483/ : SUCCESS in 44m 23s\n- mistral-tox-unit-mysql http://logs.openstack.org/34/653134/2/check/mistral-tox-unit-mysql/98d0a33/ : SUCCESS in 27m 28s\n- mistral-tox-unit-postgresql http://logs.openstack.org/34/653134/2/check/mistral-tox-unit-postgresql/b122e09/ : SUCCESS in 26m 42s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/34/653134/2/check/tripleo-ci-centos-7-scenario003-standalone/e84b505/ : SUCCESS in 1h 13m 17s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/34/653134/2/check/tripleo-ci-centos-7-containers-multinode/a91ce0f/ : SUCCESS in 2h 04m 10s\n- mistral-rally-task http://logs.openstack.org/34/653134/2/check/mistral-rally-task/cc4073a/ : SUCCESS in 1h 07m 00s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/34/653134/2/check/openstack-tox-lower-constraints/c31b364/ : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":2},{"id":"22258db6e2b007b5f4e09baa8fe3001fbd7c8830","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-04-18 18:00:07.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"52383d5f41379d39b0a2b2883fe29698d65a10de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 20:09:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/34/653134/3/check/openstack-tox-pep8/790aa19/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/34/653134/3/check/openstack-tox-py27/12b99b2/ : SUCCESS in 9m 40s\n- openstack-tox-py36 http://logs.openstack.org/34/653134/3/check/openstack-tox-py36/f5fa02b/ : SUCCESS in 8m 32s\n- openstack-tox-py37 http://logs.openstack.org/34/653134/3/check/openstack-tox-py37/42f402c/ : SUCCESS in 10m 29s\n- openstack-tox-docs http://logs.openstack.org/34/653134/3/check/openstack-tox-docs/f16c147/html/ : SUCCESS in 6m 31s\n- openstack-tox-cover http://logs.openstack.org/34/653134/3/check/openstack-tox-cover/f9f03c2/cover/ : SUCCESS in 10m 57s (non-voting)\n- mistral-devstack http://logs.openstack.org/34/653134/3/check/mistral-devstack/dd4ad36/ : SUCCESS in 39m 19s\n- mistral-devstack-non-apache http://logs.openstack.org/34/653134/3/check/mistral-devstack-non-apache/d491959/ : SUCCESS in 51m 26s\n- mistral-devstack-kombu http://logs.openstack.org/34/653134/3/check/mistral-devstack-kombu/7381dcb/ : SUCCESS in 45m 02s\n- mistral-tox-unit-mysql http://logs.openstack.org/34/653134/3/check/mistral-tox-unit-mysql/0b41254/ : FAILURE in 30m 53s\n- mistral-tox-unit-postgresql http://logs.openstack.org/34/653134/3/check/mistral-tox-unit-postgresql/c213993/ : SUCCESS in 26m 36s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/34/653134/3/check/tripleo-ci-centos-7-scenario003-standalone/ca19af6/ : SUCCESS in 1h 11m 51s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/34/653134/3/check/tripleo-ci-centos-7-containers-multinode/53d98d1/ : SUCCESS in 2h 00m 40s\n- mistral-rally-task http://logs.openstack.org/34/653134/3/check/mistral-rally-task/4969e08/ : SUCCESS in 1h 13m 57s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/34/653134/3/check/openstack-tox-lower-constraints/68a1485/ : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":3},{"id":"c6d81214fe02eb2d2cfbe5622d438485f358ad62","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-04-18 21:59:10.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"209c25a504ee60b3ca03c85d585d3594c4090773","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 04:09:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/34/653134/3/check/openstack-tox-pep8/6e28b6e/ : SUCCESS in 6m 03s\n- openstack-tox-py27 http://logs.openstack.org/34/653134/3/check/openstack-tox-py27/5a36e37/ : SUCCESS in 8m 40s\n- openstack-tox-py36 http://logs.openstack.org/34/653134/3/check/openstack-tox-py36/c7e0f50/ : SUCCESS in 14m 43s\n- openstack-tox-py37 http://logs.openstack.org/34/653134/3/check/openstack-tox-py37/0a4401a/ : SUCCESS in 16m 30s\n- openstack-tox-docs http://logs.openstack.org/34/653134/3/check/openstack-tox-docs/f1d1c9b/html/ : SUCCESS in 4m 35s\n- openstack-tox-cover http://logs.openstack.org/34/653134/3/check/openstack-tox-cover/0bd650a/cover/ : SUCCESS in 14m 53s (non-voting)\n- mistral-devstack http://logs.openstack.org/34/653134/3/check/mistral-devstack/f8088cf/ : FAILURE in 55m 02s\n- mistral-devstack-non-apache http://logs.openstack.org/34/653134/3/check/mistral-devstack-non-apache/77989cd/ : FAILURE in 52m 05s\n- mistral-devstack-kombu http://logs.openstack.org/34/653134/3/check/mistral-devstack-kombu/476dd06/ : SUCCESS in 51m 58s\n- mistral-tox-unit-mysql http://logs.openstack.org/34/653134/3/check/mistral-tox-unit-mysql/46b6758/ : SUCCESS in 28m 00s\n- mistral-tox-unit-postgresql http://logs.openstack.org/34/653134/3/check/mistral-tox-unit-postgresql/e960f1a/ : SUCCESS in 26m 58s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/34/653134/3/check/tripleo-ci-centos-7-scenario003-standalone/7d3475f/ : SUCCESS in 1h 14m 58s\n- tripleo-ci-centos-7-containers-multinode finger://ze01.openstack.org/b1f452d93be448a8bf3eab29d7cf0d20 : RETRY_LIMIT in 51m 01s\n- mistral-rally-task http://logs.openstack.org/34/653134/3/check/mistral-rally-task/fbfb62e/ : FAILURE in 1h 09m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/34/653134/3/check/openstack-tox-lower-constraints/4a6cd88/ : SUCCESS in 16m 08s","accounts_in_message":[],"_revision_number":3},{"id":"049f22013a6184e632de718715a5e25c44ef7cf3","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-04-19 09:27:52.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"39aef34330ff6976f021f73e6720dc7d8f7e6323","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 14:36:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/34/653134/3/check/openstack-tox-pep8/b67980f/ : SUCCESS in 6m 42s\n- openstack-tox-py27 http://logs.openstack.org/34/653134/3/check/openstack-tox-py27/0c97fe7/ : SUCCESS in 9m 28s\n- openstack-tox-py36 http://logs.openstack.org/34/653134/3/check/openstack-tox-py36/c136d49/ : SUCCESS in 10m 24s\n- openstack-tox-py37 http://logs.openstack.org/34/653134/3/check/openstack-tox-py37/57e1a9a/ : SUCCESS in 10m 04s\n- openstack-tox-docs http://logs.openstack.org/34/653134/3/check/openstack-tox-docs/2510a7d/html/ : SUCCESS in 6m 05s\n- openstack-tox-cover http://logs.openstack.org/34/653134/3/check/openstack-tox-cover/f5c76b9/cover/ : SUCCESS in 11m 36s (non-voting)\n- mistral-devstack http://logs.openstack.org/34/653134/3/check/mistral-devstack/2d64de2/ : SUCCESS in 48m 21s\n- mistral-devstack-non-apache http://logs.openstack.org/34/653134/3/check/mistral-devstack-non-apache/f6e7386/ : SUCCESS in 49m 16s\n- mistral-devstack-kombu http://logs.openstack.org/34/653134/3/check/mistral-devstack-kombu/4ea2e01/ : SUCCESS in 47m 36s\n- mistral-tox-unit-mysql http://logs.openstack.org/34/653134/3/check/mistral-tox-unit-mysql/17930c2/ : SUCCESS in 28m 21s\n- mistral-tox-unit-postgresql http://logs.openstack.org/34/653134/3/check/mistral-tox-unit-postgresql/35076e2/ : SUCCESS in 27m 07s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/34/653134/3/check/tripleo-ci-centos-7-scenario003-standalone/6aafbc5/ : SUCCESS in 1h 15m 22s\n- tripleo-ci-centos-7-containers-multinode finger://ze03.openstack.org/7178fbc20667410d90d4367789bfd7ab : RETRY_LIMIT in 30m 34s\n- mistral-rally-task http://logs.openstack.org/34/653134/3/check/mistral-rally-task/7327eae/ : SUCCESS in 1h 04m 47s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/34/653134/3/check/openstack-tox-lower-constraints/8d79518/ : SUCCESS in 11m 43s","accounts_in_message":[],"_revision_number":3},{"id":"51747f41aac46eb72143b52d59ef8d0488be7d3e","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-04-23 10:37:11.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d5aa935c5f5b79f44f25b1437af37004c98a545a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:56:33.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/34/653134/3/check/openstack-tox-pep8/5cb0de8/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/34/653134/3/check/openstack-tox-py27/aa08b01/ : SUCCESS in 9m 15s\n- openstack-tox-py36 http://logs.openstack.org/34/653134/3/check/openstack-tox-py36/97eedff/ : SUCCESS in 8m 42s\n- openstack-tox-py37 http://logs.openstack.org/34/653134/3/check/openstack-tox-py37/9c70524/ : SUCCESS in 9m 22s\n- openstack-tox-docs http://logs.openstack.org/34/653134/3/check/openstack-tox-docs/2886643/html/ : SUCCESS in 4m 59s\n- openstack-tox-cover http://logs.openstack.org/34/653134/3/check/openstack-tox-cover/ffae19b/cover/ : SUCCESS in 9m 20s (non-voting)\n- mistral-devstack http://logs.openstack.org/34/653134/3/check/mistral-devstack/55741e1/ : SUCCESS in 43m 12s\n- mistral-devstack-non-apache http://logs.openstack.org/34/653134/3/check/mistral-devstack-non-apache/450d0f2/ : SUCCESS in 45m 12s\n- mistral-devstack-kombu http://logs.openstack.org/34/653134/3/check/mistral-devstack-kombu/e83987b/ : SUCCESS in 44m 00s\n- mistral-tox-unit-mysql http://logs.openstack.org/34/653134/3/check/mistral-tox-unit-mysql/de60f74/ : SUCCESS in 28m 25s\n- mistral-tox-unit-postgresql http://logs.openstack.org/34/653134/3/check/mistral-tox-unit-postgresql/bf51751/ : SUCCESS in 29m 15s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/34/653134/3/check/tripleo-ci-centos-7-scenario003-standalone/07ee411/ : RETRY_LIMIT in 16m 30s\n- tripleo-ci-centos-7-containers-multinode finger://ze08.openstack.org/25e49441a22d461c84f3fc2938922e2a : RETRY_LIMIT in 18m 53s\n- mistral-rally-task http://logs.openstack.org/34/653134/3/check/mistral-rally-task/db181ef/ : RETRY_LIMIT in 16m 23s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/34/653134/3/check/openstack-tox-lower-constraints/92c3ff6/ : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":3},{"id":"9635e0b72f8d47818db749d8cf2df2f79295d1e7","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-04-24 12:32:05.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nGreat improvement. Small comments below.","accounts_in_message":[],"_revision_number":3},{"id":"a324d0a7490132d0c9b9112d503c6da13f33ffa7","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-04-25 21:29:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f0efb32046b48e890832a5be1d73af9335b42340","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-04-25 21:30:52.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a5c4c63fecc77e5e693ae963714ce62aafe6a77c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 00:24:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/34/653134/4/check/openstack-tox-pep8/4e8ec1e/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/34/653134/4/check/openstack-tox-py27/f07f80d/ : SUCCESS in 8m 27s\n- openstack-tox-py36 http://logs.openstack.org/34/653134/4/check/openstack-tox-py36/f76a9a4/ : SUCCESS in 9m 04s\n- openstack-tox-py37 http://logs.openstack.org/34/653134/4/check/openstack-tox-py37/74688c7/ : SUCCESS in 9m 21s\n- openstack-tox-docs http://logs.openstack.org/34/653134/4/check/openstack-tox-docs/3da6f3d/html/ : SUCCESS in 4m 36s\n- openstack-tox-cover http://logs.openstack.org/34/653134/4/check/openstack-tox-cover/5bbaaad/cover/ : SUCCESS in 9m 46s (non-voting)\n- mistral-devstack http://logs.openstack.org/34/653134/4/check/mistral-devstack/f90fe26/ : SUCCESS in 42m 25s\n- mistral-devstack-non-apache http://logs.openstack.org/34/653134/4/check/mistral-devstack-non-apache/0b81f08/ : SUCCESS in 45m 27s\n- mistral-devstack-kombu http://logs.openstack.org/34/653134/4/check/mistral-devstack-kombu/f04b0a9/ : SUCCESS in 44m 16s\n- mistral-tox-unit-mysql http://logs.openstack.org/34/653134/4/check/mistral-tox-unit-mysql/264042f/ : SUCCESS in 27m 26s\n- mistral-tox-unit-postgresql http://logs.openstack.org/34/653134/4/check/mistral-tox-unit-postgresql/db8a6c3/ : SUCCESS in 26m 38s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/34/653134/4/check/tripleo-ci-centos-7-scenario003-standalone/06bb877/ : SUCCESS in 1h 08m 01s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/34/653134/4/check/tripleo-ci-centos-7-containers-multinode/73e4f98/ : SUCCESS in 2h 04m 31s\n- mistral-rally-task http://logs.openstack.org/34/653134/4/check/mistral-rally-task/a58d639/ : SUCCESS in 1h 03m 19s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/34/653134/4/check/openstack-tox-lower-constraints/2e6b388/ : SUCCESS in 11m 24s","accounts_in_message":[],"_revision_number":4},{"id":"cae3e937376f0e661653d68749c4143ee7f327ba","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-04-26 05:04:28.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"38e43acc98308a2910f1871cb987bfd4a228d1ea","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-04-26 05:06:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a1ecdbb765ec2c407bef9885aac3f30811fb1912","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-04-26 06:53:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nNow looks good to me! Thanks.","accounts_in_message":[],"_revision_number":4},{"id":"a5345a3cbc429339fd0d4afa664e3d8ebc5c4ae0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 06:54:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"03e68b9e4dc827088559c342326a8b36eaf7dc03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 08:25:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"a94b1cb2ba581ea7fdb03f7f50bcc5988e566f41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 08:25:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/34/653134/4/gate/openstack-tox-pep8/b30923e/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/34/653134/4/gate/openstack-tox-py27/91366e4/ : SUCCESS in 9m 15s\n- openstack-tox-py36 http://logs.openstack.org/34/653134/4/gate/openstack-tox-py36/f6d850e/ : SUCCESS in 8m 26s\n- openstack-tox-py37 http://logs.openstack.org/34/653134/4/gate/openstack-tox-py37/8cfaeae/ : SUCCESS in 8m 21s\n- openstack-tox-docs http://logs.openstack.org/34/653134/4/gate/openstack-tox-docs/1fe0f81/html/ : SUCCESS in 5m 11s\n- mistral-devstack http://logs.openstack.org/34/653134/4/gate/mistral-devstack/7b5fefb/ : SUCCESS in 41m 19s\n- mistral-devstack-non-apache http://logs.openstack.org/34/653134/4/gate/mistral-devstack-non-apache/85e9040/ : SUCCESS in 42m 03s\n- mistral-tox-unit-mysql http://logs.openstack.org/34/653134/4/gate/mistral-tox-unit-mysql/608c70f/ : SUCCESS in 27m 20s\n- mistral-tox-unit-postgresql http://logs.openstack.org/34/653134/4/gate/mistral-tox-unit-postgresql/3645a29/ : SUCCESS in 27m 05s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/34/653134/4/gate/tripleo-ci-centos-7-scenario003-standalone/b48c129/ : SUCCESS in 1h 12m 55s\n- mistral-devstack-kombu http://logs.openstack.org/34/653134/4/gate/mistral-devstack-kombu/5009c8f/ : SUCCESS in 45m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/34/653134/4/gate/openstack-tox-lower-constraints/7f1bfbd/ : SUCCESS in 10m 19s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/34/653134/4/gate/tripleo-ci-centos-7-undercloud-containers/7aaa258/ : SUCCESS in 1h 25m 58s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"243c09d5054fd2e6ebecf36731c7758aac1c9a8d","revisions":{"50c3d30eedd4e1d980c10bcfbc7c8ec5c141d711":{"kind":"REWORK","_number":1,"created":"2019-04-16 20:00:53.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/34/653134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/653134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/653134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/653134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/653134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/653134/1"}}},"commit":{"parents":[{"commit":"2cbf543103fab2d9f6b7ab85f4bc82792840e4c3","subject":"Add reply-to to std.email","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2cbf543103fab2d9f6b7ab85f4bc82792840e4c3"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-04-16 19:57:26.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-04-16 19:57:26.000000000","tz":180},"subject":"Add option to start subworkflow via RPC.","message":"Add option to start subworkflow via RPC.\n\nIn case of several engine instances, starting subworkflows not in the\nsame instance with parent task, could significantly smooth load between\nengine instances. Furthermore, starting workflow is not always\na lightweight operation and we should try to make it as more \u0027atomic\u0027\nas possible.\n\nChange-Id: I895bee811496f920b075880a6c438c53f7ecb2ca\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/50c3d30eedd4e1d980c10bcfbc7c8ec5c141d711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/50c3d30eedd4e1d980c10bcfbc7c8ec5c141d711"}]},"branch":"refs/heads/master"},"8fcc30e5ed39e5aa50ea1052b787d64075d17b91":{"kind":"REWORK","_number":2,"created":"2019-04-17 06:40:04.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/34/653134/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/653134/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/653134/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/653134/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/653134/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/653134/2"}}},"commit":{"parents":[{"commit":"2cbf543103fab2d9f6b7ab85f4bc82792840e4c3","subject":"Add reply-to to std.email","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2cbf543103fab2d9f6b7ab85f4bc82792840e4c3"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-04-16 19:57:26.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-04-17 06:39:01.000000000","tz":180},"subject":"Add option to start subworkflow via RPC.","message":"Add option to start subworkflow via RPC.\n\nIn case of several engine instances, starting subworkflows not in the\nsame instance with parent task, could significantly smooth load between\nengine instances. Furthermore, starting workflow is not always\na lightweight operation and we should try to make it as more \u0027atomic\u0027\nas possible.\n\nChange-Id: I895bee811496f920b075880a6c438c53f7ecb2ca\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8fcc30e5ed39e5aa50ea1052b787d64075d17b91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8fcc30e5ed39e5aa50ea1052b787d64075d17b91"}]},"branch":"refs/heads/master"},"f0fce960ae56033748f18130ba769cede98124d3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-04-18 18:00:07.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/34/653134/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/653134/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/653134/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/653134/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/653134/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/653134/3"}}},"commit":{"parents":[{"commit":"6045847fe62e65edd17589e05018bf6e1300c278","subject":"Merge \"Fix an action execution controller test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6045847fe62e65edd17589e05018bf6e1300c278"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-04-16 19:57:26.000000000","tz":180},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-04-18 18:00:07.000000000","tz":0},"subject":"Add option to start subworkflow via RPC.","message":"Add option to start subworkflow via RPC.\n\nIn case of several engine instances, starting subworkflows not in the\nsame instance with parent task, could significantly smooth load between\nengine instances. Furthermore, starting workflow is not always\na lightweight operation and we should try to make it as more \u0027atomic\u0027\nas possible.\n\nChange-Id: I895bee811496f920b075880a6c438c53f7ecb2ca\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f0fce960ae56033748f18130ba769cede98124d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f0fce960ae56033748f18130ba769cede98124d3"}]},"branch":"refs/heads/master"},"243c09d5054fd2e6ebecf36731c7758aac1c9a8d":{"kind":"REWORK","_number":4,"created":"2019-04-25 21:29:23.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/34/653134/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/653134/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/653134/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/653134/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/653134/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/653134/4"}}},"commit":{"parents":[{"commit":"8e94e872591405b4132fd7dbadae61ed629169ee","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8e94e872591405b4132fd7dbadae61ed629169ee"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-04-16 19:57:26.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-04-25 21:29:00.000000000","tz":180},"subject":"Add option to start subworkflow via RPC.","message":"Add option to start subworkflow via RPC.\n\nIn case of several engine instances, starting subworkflows not in the\nsame instance with parent task, could significantly smooth load between\nengine instances. Furthermore, starting workflow is not always\na lightweight operation and we should try to make it as more \u0027atomic\u0027\nas possible.\n\nChange-Id: I895bee811496f920b075880a6c438c53f7ecb2ca\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/243c09d5054fd2e6ebecf36731c7758aac1c9a8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/243c09d5054fd2e6ebecf36731c7758aac1c9a8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
