)]}'
{"id":"openstack%2Fmistral~653710","triplet_id":"openstack%2Fmistral~master~If0d4403f5c61883ecfec4cfa14b98cc39aae5618","project":"openstack/mistral","branch":"master","topic":"remove_lookup_utils","hashtags":[],"change_id":"If0d4403f5c61883ecfec4cfa14b98cc39aae5618","subject":"Get rid of lookup utils","status":"MERGED","created":"2019-04-18 10:18:49.000000000","updated":"2019-05-23 13:03:53.000000000","submitted":"2019-05-23 13:03:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653710-1558616633558-35c0a956","meta_rev_id":"f3a8c231a6a8e0693e50fd6595dc9b9942f4116d","_number":653710,"virtual_id_number":653710,"owner":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-05-23 13:03:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-23 06:32:27.000000000","_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2019-05-23 11:15:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-05-23 10:42:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-05-23 10:42:57.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"date":"2019-05-23 10:43:06.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2019-05-23 10:43:11.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-05-23 11:23:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 10:42:57.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-05-23 10:43:06.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-05-23 10:43:11.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-05-23 11:15:57.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-05-23 11:23:30.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-05-23 13:03:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"679234dd1139540fd6da1b58070e9cdb8224a10b","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-04-18 10:18:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2ede74929769559e17fd4fdd7749283571bbd80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 12:33:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/653710/1/check/openstack-tox-pep8/7d2cb95/ : FAILURE in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/10/653710/1/check/openstack-tox-py27/ea34ce5/ : SUCCESS in 9m 38s\n- openstack-tox-py36 http://logs.openstack.org/10/653710/1/check/openstack-tox-py36/dba7167/ : SUCCESS in 8m 53s\n- openstack-tox-py37 http://logs.openstack.org/10/653710/1/check/openstack-tox-py37/9e741c7/ : SUCCESS in 9m 36s\n- openstack-tox-docs http://logs.openstack.org/10/653710/1/check/openstack-tox-docs/13661b3/html/ : SUCCESS in 5m 30s\n- openstack-tox-cover http://logs.openstack.org/10/653710/1/check/openstack-tox-cover/cb8e34a/cover/ : SUCCESS in 10m 00s (non-voting)\n- mistral-devstack http://logs.openstack.org/10/653710/1/check/mistral-devstack/6f74811/ : SUCCESS in 44m 56s\n- mistral-devstack-non-apache http://logs.openstack.org/10/653710/1/check/mistral-devstack-non-apache/5255de3/ : SUCCESS in 47m 14s\n- mistral-devstack-kombu http://logs.openstack.org/10/653710/1/check/mistral-devstack-kombu/68ebbf4/ : SUCCESS in 51m 27s\n- mistral-tox-unit-mysql http://logs.openstack.org/10/653710/1/check/mistral-tox-unit-mysql/e380e1d/ : FAILURE in 27m 23s\n- mistral-tox-unit-postgresql http://logs.openstack.org/10/653710/1/check/mistral-tox-unit-postgresql/4b99fa5/ : FAILURE in 29m 23s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/10/653710/1/check/tripleo-ci-centos-7-scenario003-standalone/bcafb09/ : SUCCESS in 1h 08m 59s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/653710/1/check/tripleo-ci-centos-7-containers-multinode/69ba6a7/ : SUCCESS in 1h 58m 24s\n- mistral-rally-task http://logs.openstack.org/10/653710/1/check/mistral-rally-task/841a7d5/ : SUCCESS in 1h 12m 34s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/10/653710/1/check/openstack-tox-lower-constraints/b9bae19/ : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":1},{"id":"757cc302f873c747b97d39de83f53a661f275c9d","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-05-20 15:56:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4abb643ab845cd6e6ac13e8fb1a1fac65f07bb12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 18:17:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/653710/2/check/openstack-tox-pep8/fdcdfb4/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/10/653710/2/check/openstack-tox-py27/847fe11/ : SUCCESS in 8m 33s\n- openstack-tox-py36 http://logs.openstack.org/10/653710/2/check/openstack-tox-py36/f0ea2f5/ : SUCCESS in 8m 35s\n- openstack-tox-py37 http://logs.openstack.org/10/653710/2/check/openstack-tox-py37/84807f8/ : FAILURE in 9m 14s\n- openstack-tox-docs http://logs.openstack.org/10/653710/2/check/openstack-tox-docs/c585e30/html/ : SUCCESS in 4m 32s\n- openstack-tox-cover http://logs.openstack.org/10/653710/2/check/openstack-tox-cover/30610ee/ : FAILURE in 9m 00s (non-voting)\n- mistral-devstack http://logs.openstack.org/10/653710/2/check/mistral-devstack/9d5e6fe/ : SUCCESS in 50m 51s\n- mistral-devstack-non-apache http://logs.openstack.org/10/653710/2/check/mistral-devstack-non-apache/825ce16/ : SUCCESS in 48m 13s\n- mistral-devstack-kombu http://logs.openstack.org/10/653710/2/check/mistral-devstack-kombu/f7cc0da/ : SUCCESS in 48m 16s\n- mistral-tox-unit-mysql http://logs.openstack.org/10/653710/2/check/mistral-tox-unit-mysql/9a12bc6/ : SUCCESS in 27m 09s\n- mistral-tox-unit-postgresql http://logs.openstack.org/10/653710/2/check/mistral-tox-unit-postgresql/0bea74c/ : SUCCESS in 26m 52s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/10/653710/2/check/tripleo-ci-centos-7-scenario003-standalone/eeb9efe/ : SUCCESS in 1h 13m 01s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/653710/2/check/tripleo-ci-centos-7-containers-multinode/b48a817/ : SUCCESS in 2h 08m 02s\n- mistral-rally-task http://logs.openstack.org/10/653710/2/check/mistral-rally-task/ee04501/ : SUCCESS in 1h 08m 24s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/10/653710/2/check/openstack-tox-lower-constraints/d8ddd9f/ : SUCCESS in 10m 43s","accounts_in_message":[],"_revision_number":2},{"id":"1e138efdc9be6837eb1327276a85b364587354c4","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-05-21 08:37:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dfc191794dffbf0127738d0c9a4233da98a26d85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:47:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/653710/2/check/openstack-tox-pep8/0044822/ : SUCCESS in 5m 16s\n- openstack-tox-py27 http://logs.openstack.org/10/653710/2/check/openstack-tox-py27/c9bb3fc/ : SUCCESS in 9m 23s\n- openstack-tox-py36 http://logs.openstack.org/10/653710/2/check/openstack-tox-py36/42fd156/ : SUCCESS in 8m 52s\n- openstack-tox-py37 http://logs.openstack.org/10/653710/2/check/openstack-tox-py37/b54f3b3/ : SUCCESS in 9m 51s\n- openstack-tox-docs http://logs.openstack.org/10/653710/2/check/openstack-tox-docs/bce56e7/html/ : SUCCESS in 5m 01s\n- openstack-tox-cover http://logs.openstack.org/10/653710/2/check/openstack-tox-cover/819bd57/cover/ : SUCCESS in 8m 54s (non-voting)\n- mistral-devstack http://logs.openstack.org/10/653710/2/check/mistral-devstack/9522f2a/ : SUCCESS in 45m 51s\n- mistral-devstack-non-apache http://logs.openstack.org/10/653710/2/check/mistral-devstack-non-apache/6cd3301/ : SUCCESS in 41m 48s\n- mistral-devstack-kombu http://logs.openstack.org/10/653710/2/check/mistral-devstack-kombu/5b04a28/ : SUCCESS in 46m 20s\n- mistral-tox-unit-mysql http://logs.openstack.org/10/653710/2/check/mistral-tox-unit-mysql/e6edbea/ : FAILURE in 27m 39s\n- mistral-tox-unit-postgresql http://logs.openstack.org/10/653710/2/check/mistral-tox-unit-postgresql/a36950a/ : SUCCESS in 25m 46s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/10/653710/2/check/tripleo-ci-centos-7-scenario003-standalone/17717a6/ : SUCCESS in 1h 13m 02s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/653710/2/check/tripleo-ci-centos-7-containers-multinode/a4220e6/ : SUCCESS in 2h 02m 05s\n- mistral-rally-task http://logs.openstack.org/10/653710/2/check/mistral-rally-task/85bd7ef/ : FAILURE in 1h 11m 12s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/10/653710/2/check/openstack-tox-lower-constraints/b886be7/ : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":2},{"id":"d0d644ac2e0e6d010e90803ccfac0edf520f03d5","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-05-22 14:50:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8da43abe761e9c6f61cd105746c2b2a12e4990cd","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-05-22 15:12:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"150e001ffe6e7f6dadf4c5aa7fb972e05319375a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 18:45:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/653710/4/check/openstack-tox-pep8/ab86fcd/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/10/653710/4/check/openstack-tox-py27/23bd3c8/ : SUCCESS in 8m 38s\n- openstack-tox-py36 http://logs.openstack.org/10/653710/4/check/openstack-tox-py36/55d5c7b/ : SUCCESS in 8m 31s\n- openstack-tox-py37 http://logs.openstack.org/10/653710/4/check/openstack-tox-py37/812921e/ : SUCCESS in 9m 02s\n- openstack-tox-docs http://logs.openstack.org/10/653710/4/check/openstack-tox-docs/351e92f/html/ : SUCCESS in 6m 09s\n- openstack-tox-cover http://logs.openstack.org/10/653710/4/check/openstack-tox-cover/8b37a2b/cover/ : SUCCESS in 8m 46s (non-voting)\n- mistral-devstack http://logs.openstack.org/10/653710/4/check/mistral-devstack/5046892/ : SUCCESS in 42m 46s\n- mistral-devstack-non-apache http://logs.openstack.org/10/653710/4/check/mistral-devstack-non-apache/5bd2011/ : SUCCESS in 43m 13s\n- mistral-devstack-kombu http://logs.openstack.org/10/653710/4/check/mistral-devstack-kombu/5752546/ : SUCCESS in 42m 58s\n- mistral-tox-unit-mysql http://logs.openstack.org/10/653710/4/check/mistral-tox-unit-mysql/3fe129e/ : SUCCESS in 29m 57s\n- mistral-tox-unit-postgresql http://logs.openstack.org/10/653710/4/check/mistral-tox-unit-postgresql/bd7d04b/ : SUCCESS in 27m 47s\n- tripleo-ci-centos-7-scenario003-standalone finger://ze12.openstack.org/a6fe98d311474c4c80c9037624f4c9b3 : RETRY_LIMIT in 29m 14s\n- tripleo-ci-centos-7-containers-multinode finger://ze04.openstack.org/042502184b5d46cfa35ee9a6985994d7 : RETRY_LIMIT in 36m 01s\n- mistral-rally-task http://logs.openstack.org/10/653710/4/check/mistral-rally-task/a4e1900/ : SUCCESS in 1h 11m 45s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/10/653710/4/check/openstack-tox-lower-constraints/3c7db37/ : SUCCESS in 11m 09s","accounts_in_message":[],"_revision_number":4},{"id":"eea56730e8d228fb07f65417a098663627221d04","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-05-22 23:12:55.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"11bebb92d64595c5d8c7fa2f360f32d21dc67a3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 01:45:52.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/653710/4/check/openstack-tox-pep8/4d279cd/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/10/653710/4/check/openstack-tox-py27/f5add39/ : SUCCESS in 8m 59s\n- openstack-tox-py36 http://logs.openstack.org/10/653710/4/check/openstack-tox-py36/a9e327d/ : SUCCESS in 9m 13s\n- openstack-tox-py37 http://logs.openstack.org/10/653710/4/check/openstack-tox-py37/9427950/ : SUCCESS in 12m 05s\n- openstack-tox-docs http://logs.openstack.org/10/653710/4/check/openstack-tox-docs/6c60810/html/ : SUCCESS in 5m 08s\n- openstack-tox-cover http://logs.openstack.org/10/653710/4/check/openstack-tox-cover/195578e/cover/ : SUCCESS in 9m 34s (non-voting)\n- mistral-devstack http://logs.openstack.org/10/653710/4/check/mistral-devstack/8fd83d5/ : SUCCESS in 36m 49s\n- mistral-devstack-non-apache http://logs.openstack.org/10/653710/4/check/mistral-devstack-non-apache/113830f/ : SUCCESS in 39m 54s\n- mistral-devstack-kombu http://logs.openstack.org/10/653710/4/check/mistral-devstack-kombu/7af7606/ : SUCCESS in 41m 58s\n- mistral-tox-unit-mysql http://logs.openstack.org/10/653710/4/check/mistral-tox-unit-mysql/6d8b542/ : FAILURE in 28m 10s\n- mistral-tox-unit-postgresql http://logs.openstack.org/10/653710/4/check/mistral-tox-unit-postgresql/2b5b47d/ : SUCCESS in 31m 49s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/10/653710/4/check/tripleo-ci-centos-7-scenario003-standalone/da79ecf/ : SUCCESS in 1h 06m 41s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/653710/4/check/tripleo-ci-centos-7-containers-multinode/9bbc6bf/ : SUCCESS in 2h 08m 38s\n- mistral-rally-task http://logs.openstack.org/10/653710/4/check/mistral-rally-task/c84ac48/ : SUCCESS in 1h 12m 18s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/10/653710/4/check/openstack-tox-lower-constraints/0d85f8a/ : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":4},{"id":"f128a6dfdb96fa7b36c123a39b4133a33173ea37","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-05-23 06:32:27.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"187f79f6e649b4ad5f4b5c7811ebfd43c6030a6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 09:07:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/653710/4/check/openstack-tox-pep8/5df59c5/ : SUCCESS in 9m 27s\n- openstack-tox-py27 http://logs.openstack.org/10/653710/4/check/openstack-tox-py27/4ef2a37/ : SUCCESS in 11m 24s\n- openstack-tox-py36 http://logs.openstack.org/10/653710/4/check/openstack-tox-py36/72494e2/ : SUCCESS in 12m 44s\n- openstack-tox-py37 http://logs.openstack.org/10/653710/4/check/openstack-tox-py37/2868b29/ : SUCCESS in 14m 04s\n- openstack-tox-docs http://logs.openstack.org/10/653710/4/check/openstack-tox-docs/f62369d/html/ : SUCCESS in 8m 33s\n- openstack-tox-cover http://logs.openstack.org/10/653710/4/check/openstack-tox-cover/7f336cb/cover/ : SUCCESS in 9m 02s (non-voting)\n- mistral-devstack http://logs.openstack.org/10/653710/4/check/mistral-devstack/5adddd1/ : SUCCESS in 46m 23s\n- mistral-devstack-non-apache http://logs.openstack.org/10/653710/4/check/mistral-devstack-non-apache/d23e215/ : SUCCESS in 45m 12s\n- mistral-devstack-kombu http://logs.openstack.org/10/653710/4/check/mistral-devstack-kombu/313e27a/ : SUCCESS in 45m 11s\n- mistral-tox-unit-mysql http://logs.openstack.org/10/653710/4/check/mistral-tox-unit-mysql/a30b43f/ : SUCCESS in 27m 30s\n- mistral-tox-unit-postgresql http://logs.openstack.org/10/653710/4/check/mistral-tox-unit-postgresql/568871b/ : SUCCESS in 26m 17s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/10/653710/4/check/tripleo-ci-centos-7-scenario003-standalone/2aca406/ : SUCCESS in 1h 17m 26s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/653710/4/check/tripleo-ci-centos-7-containers-multinode/04e7022/ : SUCCESS in 2h 26m 40s\n- mistral-rally-task http://logs.openstack.org/10/653710/4/check/mistral-rally-task/4ddcf0e/ : SUCCESS in 1h 02m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/10/653710/4/check/openstack-tox-lower-constraints/a01f8cc/ : SUCCESS in 11m 20s","accounts_in_message":[],"_revision_number":4},{"id":"5a103ff7dc8f73beba44fce9d81cd5b0e258d49e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-23 10:42:50.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM, thanks","accounts_in_message":[],"_revision_number":4},{"id":"966f9ce43b6d4f0b50f469584499b3defdcecbfc","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-05-23 11:15:57.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"62aceb575a3a3efdbc6205dfd84228275cd9b027","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-05-23 11:16:26.000000000","message":"Patch Set 4:\n\nLooks good but I\u0027d love to get another core have a look at it before +W it","accounts_in_message":[],"_revision_number":4},{"id":"eb3fcd363075639e190640d46b22e40e1f50c8ed","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-23 11:23:30.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"17986186d52bbe08279bea7cf736bc9a19c426ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 11:24:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6c51d5ae1d397cd70f6901c999d9519adb5d047b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 13:03:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"81c45813f235219667d5795710fc521ea8d9c8bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 13:03:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/653710/4/gate/openstack-tox-pep8/876b46d/ : SUCCESS in 5m 59s\n- openstack-tox-py27 http://logs.openstack.org/10/653710/4/gate/openstack-tox-py27/1caaade/ : SUCCESS in 8m 51s\n- openstack-tox-py36 http://logs.openstack.org/10/653710/4/gate/openstack-tox-py36/dbcc0de/ : SUCCESS in 8m 23s\n- openstack-tox-py37 http://logs.openstack.org/10/653710/4/gate/openstack-tox-py37/74696c7/ : SUCCESS in 8m 33s\n- openstack-tox-docs http://logs.openstack.org/10/653710/4/gate/openstack-tox-docs/5f2206c/html/ : SUCCESS in 4m 28s\n- mistral-devstack http://logs.openstack.org/10/653710/4/gate/mistral-devstack/2686927/ : SUCCESS in 41m 21s\n- mistral-devstack-non-apache http://logs.openstack.org/10/653710/4/gate/mistral-devstack-non-apache/37c4a00/ : SUCCESS in 42m 36s\n- mistral-tox-unit-mysql http://logs.openstack.org/10/653710/4/gate/mistral-tox-unit-mysql/9b5f3e7/ : SUCCESS in 27m 25s\n- mistral-tox-unit-postgresql http://logs.openstack.org/10/653710/4/gate/mistral-tox-unit-postgresql/0beeeff/ : SUCCESS in 26m 56s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/10/653710/4/gate/tripleo-ci-centos-7-scenario003-standalone/b5597b7/ : SUCCESS in 1h 04m 40s\n- mistral-devstack-kombu http://logs.openstack.org/10/653710/4/gate/mistral-devstack-kombu/43e7b3a/ : SUCCESS in 43m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/653710/4/gate/openstack-tox-lower-constraints/8df2131/ : SUCCESS in 10m 45s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/10/653710/4/gate/tripleo-ci-centos-7-undercloud-containers/5fd6ef0/ : SUCCESS in 1h 32m 01s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"84b8e92acc142d9cf13c44104cf66edb08b4a9de","revisions":{"4eff2ca7ffa5a734693bc770b4ff994a36e15c24":{"kind":"REWORK","_number":1,"created":"2019-04-18 10:18:49.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/10/653710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/10/653710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/10/653710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/10/653710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/10/653710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/10/653710/1"}}},"commit":{"parents":[{"commit":"9c5ce2921232181974b064d5af321881bc8dac53","subject":"WIP: POC: rework joining mechanism","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9c5ce2921232181974b064d5af321881bc8dac53"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-04-18 10:16:15.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-04-18 10:16:15.000000000","tz":120},"subject":"Disable task execution caching in lookup_utils","message":"Disable task execution caching in lookup_utils\n\nWith the new joining mechanism task execution cache becomes\nuseless and can be disabled to save some RAM.\n\nChange-Id: If0d4403f5c61883ecfec4cfa14b98cc39aae5618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4eff2ca7ffa5a734693bc770b4ff994a36e15c24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4eff2ca7ffa5a734693bc770b4ff994a36e15c24"}]},"branch":"refs/heads/master"},"83293ec30d1d5cffc78989d02206ffc0c172feea":{"kind":"REWORK","_number":2,"created":"2019-05-20 15:56:16.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/10/653710/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/10/653710/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/10/653710/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/10/653710/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/10/653710/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/10/653710/2"}}},"commit":{"parents":[{"commit":"26b9cc2bac9d4c959a2497342ecf63d23af34196","subject":"Merge \"Rework joining mechanism\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/26b9cc2bac9d4c959a2497342ecf63d23af34196"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-04-18 10:16:15.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-05-20 15:52:26.000000000","tz":120},"subject":"WIP: Get rid of lookup utils","message":"WIP: Get rid of lookup utils\n\nWith the new joining mechanism lookup utils for task executions\nbecome useless and can be removed from the codebase.\nNow all requests to the database will be perforemed directly from\na workflow controller.\n\nAction defention cache was moved to mistral/engine/actions.py,\nbecause it\u0027s used only there.\n\nChange-Id: If0d4403f5c61883ecfec4cfa14b98cc39aae5618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/83293ec30d1d5cffc78989d02206ffc0c172feea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/83293ec30d1d5cffc78989d02206ffc0c172feea"}]},"branch":"refs/heads/master"},"c40f71fa7d0517f3005377985a5c0ccec886d389":{"kind":"REWORK","_number":3,"created":"2019-05-22 14:50:27.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/10/653710/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/10/653710/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/10/653710/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/10/653710/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/10/653710/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/10/653710/3"}}},"commit":{"parents":[{"commit":"fefdcd77cd757b4cd3376122f04ef58f8d37ab41","subject":"Merge \"Optimize searching of upstream task executions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fefdcd77cd757b4cd3376122f04ef58f8d37ab41"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-04-18 10:16:15.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-05-22 14:47:17.000000000","tz":120},"subject":"Get rid of lookup utils","message":"Get rid of lookup utils\n\nWith the new joining mechanism lookup utils for task executions\nbecome useless and can be removed from the codebase.\nNow all requests to the database will be perforemed directly from\na workflow controller.\n\nAction defention cache was moved to mistral/engine/actions.py,\nbecause it\u0027s needed only there.\n\nChange-Id: If0d4403f5c61883ecfec4cfa14b98cc39aae5618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c40f71fa7d0517f3005377985a5c0ccec886d389"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c40f71fa7d0517f3005377985a5c0ccec886d389"}]},"branch":"refs/heads/master"},"84b8e92acc142d9cf13c44104cf66edb08b4a9de":{"kind":"REWORK","_number":4,"created":"2019-05-22 15:12:02.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/10/653710/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/10/653710/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/10/653710/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/10/653710/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/10/653710/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/10/653710/4"}}},"commit":{"parents":[{"commit":"fefdcd77cd757b4cd3376122f04ef58f8d37ab41","subject":"Merge \"Optimize searching of upstream task executions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fefdcd77cd757b4cd3376122f04ef58f8d37ab41"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-04-18 10:16:15.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-05-22 15:11:49.000000000","tz":120},"subject":"Get rid of lookup utils","message":"Get rid of lookup utils\n\nWith the new joining mechanism lookup utils for task executions\nbecome useless and can be removed from the codebase.\nNow all requests to the database will be perforemed directly from\na workflow controller.\n\nAction defention cache was moved to mistral/engine/actions.py,\nbecause it\u0027s needed only there.\n\nChange-Id: If0d4403f5c61883ecfec4cfa14b98cc39aae5618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/84b8e92acc142d9cf13c44104cf66edb08b4a9de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/84b8e92acc142d9cf13c44104cf66edb08b4a9de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
