)]}'
{"id":"openstack%2Fmistral~654855","triplet_id":"openstack%2Fmistral~master~I52d44a333d2eaa4a2e65c150e3640a4a84e291aa","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I52d44a333d2eaa4a2e65c150e3640a4a84e291aa","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 08:45:13.000000000","updated":"2019-04-24 15:11:35.000000000","submitted":"2019-04-24 15:11:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654855-1556118695986-eaa21e4c","meta_rev_id":"73eebb19ad8f90ef87ab37fc20a0615622c5556e","_number":654855,"virtual_id_number":654855,"owner":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-24 11:21:46.000000000","_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-04-24 15:11:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":2,"date":"2019-04-24 08:28:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":1,"date":"2019-04-24 08:28:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 08:28:50.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-04-24 15:11:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f86edd8cd3066a9be1b5cd517263017d3137169","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-23 08:45:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc81aa72a30465f31e215e63048b1e9fe6429abf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 11:22:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/55/654855/1/check/openstack-tox-pep8/7cd79a9/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/55/654855/1/check/openstack-tox-py27/fe00181/ : SUCCESS in 9m 42s\n- openstack-tox-py36 http://logs.openstack.org/55/654855/1/check/openstack-tox-py36/2b4c920/ : SUCCESS in 9m 28s\n- openstack-tox-py37 http://logs.openstack.org/55/654855/1/check/openstack-tox-py37/52b9b92/ : SUCCESS in 9m 48s\n- openstack-tox-docs http://logs.openstack.org/55/654855/1/check/openstack-tox-docs/fcfad65/html/ : SUCCESS in 4m 41s\n- build-openstack-releasenotes http://logs.openstack.org/55/654855/1/check/build-openstack-releasenotes/8a06d4c/html/ : SUCCESS in 3m 57s\n- openstack-tox-cover http://logs.openstack.org/55/654855/1/check/openstack-tox-cover/04f7408/cover/ : SUCCESS in 9m 20s (non-voting)\n- mistral-devstack http://logs.openstack.org/55/654855/1/check/mistral-devstack/03a57c6/ : SUCCESS in 45m 25s\n- mistral-devstack-non-apache http://logs.openstack.org/55/654855/1/check/mistral-devstack-non-apache/1c8c44a/ : SUCCESS in 40m 44s\n- mistral-devstack-kombu http://logs.openstack.org/55/654855/1/check/mistral-devstack-kombu/045a8d9/ : SUCCESS in 42m 58s\n- mistral-tox-unit-mysql http://logs.openstack.org/55/654855/1/check/mistral-tox-unit-mysql/47c6a92/ : FAILURE in 27m 24s\n- mistral-tox-unit-postgresql http://logs.openstack.org/55/654855/1/check/mistral-tox-unit-postgresql/78c2499/ : SUCCESS in 26m 36s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/55/654855/1/check/tripleo-ci-centos-7-scenario003-standalone/490e62d/ : SUCCESS in 1h 13m 22s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/55/654855/1/check/tripleo-ci-centos-7-containers-multinode/4b8910c/ : SUCCESS in 2h 17m 42s\n- mistral-rally-task http://logs.openstack.org/55/654855/1/check/mistral-rally-task/a353136/ : SUCCESS in 1h 11m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/55/654855/1/check/openstack-tox-lower-constraints/f347cdd/ : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":1},{"id":"116ba85f7fe1e28c49f4a2c0d250c1c4450147b4","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-24 07:44:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c48b7859522a724489f8ba1b22759feead0e5c64","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-24 07:44:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5b529bdf8d13b12e5de9cbb928436542ee217d6e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-04-24 08:28:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"74c386cd632f0991ff8ce1af811133770fb901d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 10:18:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/55/654855/3/check/openstack-tox-pep8/9471e31/ : SUCCESS in 5m 50s\n- openstack-tox-py27 http://logs.openstack.org/55/654855/3/check/openstack-tox-py27/be2ea59/ : SUCCESS in 9m 14s\n- openstack-tox-py36 http://logs.openstack.org/55/654855/3/check/openstack-tox-py36/9430b85/ : SUCCESS in 8m 36s\n- openstack-tox-py37 http://logs.openstack.org/55/654855/3/check/openstack-tox-py37/4ae606e/ : SUCCESS in 9m 36s\n- openstack-tox-docs http://logs.openstack.org/55/654855/3/check/openstack-tox-docs/070c2af/html/ : SUCCESS in 5m 36s\n- build-openstack-releasenotes http://logs.openstack.org/55/654855/3/check/build-openstack-releasenotes/c48bfde/html/ : SUCCESS in 4m 10s\n- openstack-tox-cover http://logs.openstack.org/55/654855/3/check/openstack-tox-cover/c3fb791/cover/ : SUCCESS in 10m 35s (non-voting)\n- mistral-devstack http://logs.openstack.org/55/654855/3/check/mistral-devstack/36eafda/ : SUCCESS in 45m 50s\n- mistral-devstack-non-apache http://logs.openstack.org/55/654855/3/check/mistral-devstack-non-apache/654ca6b/ : FAILURE in 35m 16s\n- mistral-devstack-kombu http://logs.openstack.org/55/654855/3/check/mistral-devstack-kombu/61a706e/ : SUCCESS in 41m 47s\n- mistral-tox-unit-mysql http://logs.openstack.org/55/654855/3/check/mistral-tox-unit-mysql/14955e5/ : SUCCESS in 27m 47s\n- mistral-tox-unit-postgresql http://logs.openstack.org/55/654855/3/check/mistral-tox-unit-postgresql/aab5e75/ : SUCCESS in 27m 09s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/55/654855/3/check/tripleo-ci-centos-7-scenario003-standalone/13fdaeb/ : SUCCESS in 1h 11m 28s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/55/654855/3/check/tripleo-ci-centos-7-containers-multinode/f919543/ : SUCCESS in 2h 16m 32s\n- mistral-rally-task http://logs.openstack.org/55/654855/3/check/mistral-rally-task/9bc82dc/ : SUCCESS in 1h 16m 52s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/55/654855/3/check/openstack-tox-lower-constraints/38b5975/ : SUCCESS in 11m 04s","accounts_in_message":[],"_revision_number":3},{"id":"34f3281427f1c909cc49c516d8c318a9888207c5","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-24 11:21:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"19010c6c8a208d008880793996c898ee92bdcd8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 13:35:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/55/654855/3/check/openstack-tox-pep8/4c4ea65/ : SUCCESS in 6m 12s\n- openstack-tox-py27 http://logs.openstack.org/55/654855/3/check/openstack-tox-py27/127c5b5/ : SUCCESS in 8m 54s\n- openstack-tox-py36 http://logs.openstack.org/55/654855/3/check/openstack-tox-py36/9ef50be/ : SUCCESS in 8m 43s\n- openstack-tox-py37 http://logs.openstack.org/55/654855/3/check/openstack-tox-py37/317d3a5/ : SUCCESS in 9m 25s\n- openstack-tox-docs http://logs.openstack.org/55/654855/3/check/openstack-tox-docs/012e5c7/html/ : SUCCESS in 5m 18s\n- build-openstack-releasenotes http://logs.openstack.org/55/654855/3/check/build-openstack-releasenotes/d0309f7/html/ : SUCCESS in 4m 09s\n- openstack-tox-cover http://logs.openstack.org/55/654855/3/check/openstack-tox-cover/5069e21/cover/ : SUCCESS in 9m 01s (non-voting)\n- mistral-devstack http://logs.openstack.org/55/654855/3/check/mistral-devstack/75ff7d0/ : SUCCESS in 38m 10s\n- mistral-devstack-non-apache http://logs.openstack.org/55/654855/3/check/mistral-devstack-non-apache/3360f84/ : SUCCESS in 36m 00s\n- mistral-devstack-kombu http://logs.openstack.org/55/654855/3/check/mistral-devstack-kombu/038858b/ : SUCCESS in 40m 02s\n- mistral-tox-unit-mysql http://logs.openstack.org/55/654855/3/check/mistral-tox-unit-mysql/1ba552b/ : SUCCESS in 27m 26s\n- mistral-tox-unit-postgresql http://logs.openstack.org/55/654855/3/check/mistral-tox-unit-postgresql/3d95e93/ : SUCCESS in 26m 22s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/55/654855/3/check/tripleo-ci-centos-7-scenario003-standalone/061ba03/ : SUCCESS in 1h 28m 51s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/55/654855/3/check/tripleo-ci-centos-7-containers-multinode/78bd1f4/ : SUCCESS in 2h 04m 31s\n- mistral-rally-task http://logs.openstack.org/55/654855/3/check/mistral-rally-task/080f1ff/ : SUCCESS in 1h 06m 22s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/55/654855/3/check/openstack-tox-lower-constraints/f9a4388/ : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":3},{"id":"5aaec267bd95c20b194b782327782b557db986ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 13:36:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"06c6ac82831e26464fc47a7d1f52bb558d7134ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 15:11:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/55/654855/3/gate/openstack-tox-pep8/10020b6/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/55/654855/3/gate/openstack-tox-py27/afc07a5/ : SUCCESS in 8m 54s\n- openstack-tox-py36 http://logs.openstack.org/55/654855/3/gate/openstack-tox-py36/4678315/ : SUCCESS in 10m 18s\n- openstack-tox-py37 http://logs.openstack.org/55/654855/3/gate/openstack-tox-py37/6ade335/ : SUCCESS in 10m 31s\n- openstack-tox-docs http://logs.openstack.org/55/654855/3/gate/openstack-tox-docs/2dd8f26/html/ : SUCCESS in 4m 30s\n- build-openstack-releasenotes http://logs.openstack.org/55/654855/3/gate/build-openstack-releasenotes/5c1d255/html/ : SUCCESS in 4m 02s\n- mistral-devstack http://logs.openstack.org/55/654855/3/gate/mistral-devstack/8d7fd39/ : SUCCESS in 49m 21s\n- mistral-devstack-non-apache http://logs.openstack.org/55/654855/3/gate/mistral-devstack-non-apache/83697ed/ : SUCCESS in 45m 02s\n- mistral-tox-unit-mysql http://logs.openstack.org/55/654855/3/gate/mistral-tox-unit-mysql/d9f4fde/ : SUCCESS in 27m 24s\n- mistral-tox-unit-postgresql http://logs.openstack.org/55/654855/3/gate/mistral-tox-unit-postgresql/73ddaee/ : SUCCESS in 27m 33s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/55/654855/3/gate/tripleo-ci-centos-7-scenario003-standalone/75b0f0e/ : SUCCESS in 1h 14m 56s\n- mistral-devstack-kombu http://logs.openstack.org/55/654855/3/gate/mistral-devstack-kombu/3133060/ : SUCCESS in 44m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/654855/3/gate/openstack-tox-lower-constraints/4a5ade0/ : SUCCESS in 9m 55s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/55/654855/3/gate/tripleo-ci-centos-7-undercloud-containers/5c72e8b/ : SUCCESS in 1h 33m 12s","accounts_in_message":[],"_revision_number":3},{"id":"73eebb19ad8f90ef87ab37fc20a0615622c5556e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 15:11:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"46fd1b8a759b37a19907d92b957719adbe2c22c5","revisions":{"f259490f2117e4a0111b4e87cd7f1e6591b5d9ee":{"kind":"REWORK","_number":1,"created":"2019-04-23 08:45:13.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/55/654855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/55/654855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/654855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/654855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/654855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/55/654855/1"}}},"commit":{"parents":[{"commit":"a461e079168f4cbbe00f6552f34aef2a4c75ce6e","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a461e079168f4cbbe00f6552f34aef2a4c75ce6e"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:45:05.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:45:05.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I52d44a333d2eaa4a2e65c150e3640a4a84e291aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f259490f2117e4a0111b4e87cd7f1e6591b5d9ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f259490f2117e4a0111b4e87cd7f1e6591b5d9ee"}]},"branch":"refs/heads/master"},"f09964681e0d5ed72e4a9c1077193a33a962ccce":{"kind":"REWORK","_number":2,"created":"2019-04-24 07:44:06.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/55/654855/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/55/654855/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/654855/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/654855/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/654855/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/55/654855/2"}}},"commit":{"parents":[{"commit":"a461e079168f4cbbe00f6552f34aef2a4c75ce6e","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a461e079168f4cbbe00f6552f34aef2a4c75ce6e"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:45:05.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-24 07:43:55.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I52d44a333d2eaa4a2e65c150e3640a4a84e291aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f09964681e0d5ed72e4a9c1077193a33a962ccce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f09964681e0d5ed72e4a9c1077193a33a962ccce"}]},"branch":"refs/heads/master"},"46fd1b8a759b37a19907d92b957719adbe2c22c5":{"kind":"REWORK","_number":3,"created":"2019-04-24 07:44:33.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/55/654855/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/55/654855/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/654855/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/654855/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/654855/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/55/654855/3"}}},"commit":{"parents":[{"commit":"a461e079168f4cbbe00f6552f34aef2a4c75ce6e","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a461e079168f4cbbe00f6552f34aef2a4c75ce6e"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:45:05.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-24 07:44:25.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I52d44a333d2eaa4a2e65c150e3640a4a84e291aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/46fd1b8a759b37a19907d92b957719adbe2c22c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/46fd1b8a759b37a19907d92b957719adbe2c22c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
