)]}'
{"id":"openstack%2Fmistral~656930","triplet_id":"openstack%2Fmistral~master~Ie6a073dc286f0d8704c67f8295dfd76bce7897bb","project":"openstack/mistral","branch":"master","topic":"fix_filter_import","hashtags":[],"change_id":"Ie6a073dc286f0d8704c67f8295dfd76bce7897bb","subject":"Fix the import for filter_utils","status":"MERGED","created":"2019-05-03 08:47:57.000000000","updated":"2019-05-13 04:05:08.000000000","submitted":"2019-05-13 04:05:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656930-1557720308735-1a85bf1c","meta_rev_id":"b49937708dafc0ed4b0b48937d063d4a5160d28e","_number":656930,"virtual_id_number":656930,"owner":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-06 12:49:47.000000000","_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-05-13 04:05:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-05-13 02:31:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-05-13 02:31:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 02:31:36.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-05-13 04:05:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eda130cf87e086b7cf3deb2f40f3c99832d1ce4b","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-05-03 08:47:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aeb33aaad8a5c961523ec3b89622886583778a23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 10:55:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/30/656930/1/check/openstack-tox-pep8/a6f15b7/ : SUCCESS in 7m 08s\n- openstack-tox-py27 http://logs.openstack.org/30/656930/1/check/openstack-tox-py27/a30bb77/ : SUCCESS in 10m 23s\n- openstack-tox-py36 http://logs.openstack.org/30/656930/1/check/openstack-tox-py36/55a0bad/ : SUCCESS in 10m 17s\n- openstack-tox-py37 http://logs.openstack.org/30/656930/1/check/openstack-tox-py37/b9b6aae/ : SUCCESS in 10m 32s\n- openstack-tox-docs http://logs.openstack.org/30/656930/1/check/openstack-tox-docs/e71e045/html/ : SUCCESS in 4m 28s\n- openstack-tox-cover http://logs.openstack.org/30/656930/1/check/openstack-tox-cover/4d1cecd/cover/ : SUCCESS in 9m 29s (non-voting)\n- mistral-devstack http://logs.openstack.org/30/656930/1/check/mistral-devstack/1bba6bf/ : SUCCESS in 43m 30s\n- mistral-devstack-non-apache http://logs.openstack.org/30/656930/1/check/mistral-devstack-non-apache/0295699/ : POST_FAILURE in 58m 20s\n- mistral-devstack-kombu http://logs.openstack.org/30/656930/1/check/mistral-devstack-kombu/5f9981f/ : SUCCESS in 44m 38s\n- mistral-tox-unit-mysql http://logs.openstack.org/30/656930/1/check/mistral-tox-unit-mysql/0af63fe/ : SUCCESS in 29m 44s\n- mistral-tox-unit-postgresql http://logs.openstack.org/30/656930/1/check/mistral-tox-unit-postgresql/3351c7e/ : SUCCESS in 25m 59s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/30/656930/1/check/tripleo-ci-centos-7-scenario003-standalone/2b55ab3/ : SUCCESS in 1h 11m 49s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/30/656930/1/check/tripleo-ci-centos-7-containers-multinode/ed8d737/ : SUCCESS in 2h 01m 50s\n- mistral-rally-task http://logs.openstack.org/30/656930/1/check/mistral-rally-task/4bb3a3c/ : SUCCESS in 58m 15s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/30/656930/1/check/openstack-tox-lower-constraints/4110a77/ : SUCCESS in 11m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4ff6127c3e8851fa0eeac59b107d57a46cc6d291","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-05-06 10:28:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"85e9134ae85c4ed5fadef492dbbfb7021708c32e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 12:30:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/30/656930/2/check/openstack-tox-pep8/611c695/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/30/656930/2/check/openstack-tox-py27/e60aa61/ : SUCCESS in 8m 38s\n- openstack-tox-py36 http://logs.openstack.org/30/656930/2/check/openstack-tox-py36/f3e34d3/ : SUCCESS in 8m 51s\n- openstack-tox-py37 http://logs.openstack.org/30/656930/2/check/openstack-tox-py37/47d15f7/ : SUCCESS in 8m 54s\n- openstack-tox-docs http://logs.openstack.org/30/656930/2/check/openstack-tox-docs/40c4e19/html/ : SUCCESS in 4m 45s\n- openstack-tox-cover http://logs.openstack.org/30/656930/2/check/openstack-tox-cover/d7cab81/cover/ : SUCCESS in 9m 08s (non-voting)\n- mistral-devstack http://logs.openstack.org/30/656930/2/check/mistral-devstack/41828ef/ : SUCCESS in 44m 38s\n- mistral-devstack-non-apache http://logs.openstack.org/30/656930/2/check/mistral-devstack-non-apache/9670767/ : SUCCESS in 42m 33s\n- mistral-devstack-kombu http://logs.openstack.org/30/656930/2/check/mistral-devstack-kombu/f767ee7/ : SUCCESS in 45m 39s\n- mistral-tox-unit-mysql http://logs.openstack.org/30/656930/2/check/mistral-tox-unit-mysql/30a46c7/ : FAILURE in 27m 09s\n- mistral-tox-unit-postgresql http://logs.openstack.org/30/656930/2/check/mistral-tox-unit-postgresql/26abbf5/ : SUCCESS in 29m 02s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/30/656930/2/check/tripleo-ci-centos-7-scenario003-standalone/0ed9685/ : SUCCESS in 1h 45m 57s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/30/656930/2/check/tripleo-ci-centos-7-containers-multinode/0e6f7bb/ : SUCCESS in 1h 56m 14s\n- mistral-rally-task http://logs.openstack.org/30/656930/2/check/mistral-rally-task/0c1b68d/ : SUCCESS in 1h 01m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/30/656930/2/check/openstack-tox-lower-constraints/a839d5e/ : SUCCESS in 10m 17s","accounts_in_message":[],"_revision_number":2},{"id":"6be4237cee54250e6955df7ff1707484565189c5","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-05-06 12:49:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"162b657a8c76f8fd469eeff58cc449835c5cf2b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 15:14:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/656930/2/check/openstack-tox-pep8/5fbd85f/ : SUCCESS in 7m 29s\n- openstack-tox-py27 http://logs.openstack.org/30/656930/2/check/openstack-tox-py27/104ea77/ : SUCCESS in 8m 43s\n- openstack-tox-py36 http://logs.openstack.org/30/656930/2/check/openstack-tox-py36/296cdf9/ : SUCCESS in 8m 39s\n- openstack-tox-py37 http://logs.openstack.org/30/656930/2/check/openstack-tox-py37/79b094d/ : SUCCESS in 9m 17s\n- openstack-tox-docs http://logs.openstack.org/30/656930/2/check/openstack-tox-docs/b10afa2/html/ : SUCCESS in 5m 03s\n- openstack-tox-cover http://logs.openstack.org/30/656930/2/check/openstack-tox-cover/2a34703/cover/ : SUCCESS in 10m 31s (non-voting)\n- mistral-devstack http://logs.openstack.org/30/656930/2/check/mistral-devstack/c9f5df8/ : SUCCESS in 44m 58s\n- mistral-devstack-non-apache http://logs.openstack.org/30/656930/2/check/mistral-devstack-non-apache/f8aee59/ : SUCCESS in 43m 00s\n- mistral-devstack-kombu http://logs.openstack.org/30/656930/2/check/mistral-devstack-kombu/0c5dcf7/ : SUCCESS in 48m 56s\n- mistral-tox-unit-mysql http://logs.openstack.org/30/656930/2/check/mistral-tox-unit-mysql/f22342b/ : SUCCESS in 28m 47s\n- mistral-tox-unit-postgresql http://logs.openstack.org/30/656930/2/check/mistral-tox-unit-postgresql/62694c2/ : SUCCESS in 28m 59s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/30/656930/2/check/tripleo-ci-centos-7-scenario003-standalone/730b5fd/ : SUCCESS in 1h 06m 19s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/30/656930/2/check/tripleo-ci-centos-7-containers-multinode/45641be/ : SUCCESS in 2h 02m 13s\n- mistral-rally-task http://logs.openstack.org/30/656930/2/check/mistral-rally-task/5974fe4/ : FAILURE in 1h 13m 39s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/30/656930/2/check/openstack-tox-lower-constraints/3db007a/ : SUCCESS in 12m 53s","accounts_in_message":[],"_revision_number":2},{"id":"80fc8ab59e298ea2a750925b849a0ab500164520","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-13 02:31:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a1e1679282911582ee375920aa3315bf21602c45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 02:31:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eb2436fa4ad63905a2437e498d88782a048378f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 04:05:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cf20cd62576a53776ec9429eb5ba6b6c2fd23094","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 04:05:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/656930/2/gate/openstack-tox-pep8/5bb106f/ : SUCCESS in 6m 04s\n- openstack-tox-py27 http://logs.openstack.org/30/656930/2/gate/openstack-tox-py27/1be70e0/ : SUCCESS in 9m 48s\n- openstack-tox-py36 http://logs.openstack.org/30/656930/2/gate/openstack-tox-py36/d0bf4d0/ : SUCCESS in 9m 09s\n- openstack-tox-py37 http://logs.openstack.org/30/656930/2/gate/openstack-tox-py37/701917c/ : SUCCESS in 10m 03s\n- openstack-tox-docs http://logs.openstack.org/30/656930/2/gate/openstack-tox-docs/334ce53/html/ : SUCCESS in 4m 30s\n- mistral-devstack http://logs.openstack.org/30/656930/2/gate/mistral-devstack/0b4a868/ : SUCCESS in 43m 17s\n- mistral-devstack-non-apache http://logs.openstack.org/30/656930/2/gate/mistral-devstack-non-apache/e24329a/ : SUCCESS in 41m 22s\n- mistral-tox-unit-mysql http://logs.openstack.org/30/656930/2/gate/mistral-tox-unit-mysql/b149c97/ : SUCCESS in 31m 48s\n- mistral-tox-unit-postgresql http://logs.openstack.org/30/656930/2/gate/mistral-tox-unit-postgresql/c1fea43/ : SUCCESS in 31m 26s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/30/656930/2/gate/tripleo-ci-centos-7-scenario003-standalone/0afc4ec/ : SUCCESS in 1h 13m 36s\n- mistral-devstack-kombu http://logs.openstack.org/30/656930/2/gate/mistral-devstack-kombu/d6be943/ : SUCCESS in 43m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/656930/2/gate/openstack-tox-lower-constraints/2d23a6c/ : SUCCESS in 10m 39s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/30/656930/2/gate/tripleo-ci-centos-7-undercloud-containers/25a7b3c/ : SUCCESS in 1h 21m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a83b4a6702b424c2b47b9d5568abe7db898680ce","revisions":{"86da45546e01985c5a03bfd40304f199bf3a444f":{"kind":"REWORK","_number":1,"created":"2019-05-03 08:47:57.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/30/656930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/30/656930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/656930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/656930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/656930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/30/656930/1"}}},"commit":{"parents":[{"commit":"13d8f532b9a1899b79ee4060b878f5cd33a0792d","subject":"Docs: Change description from on-success to on-error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/13d8f532b9a1899b79ee4060b878f5cd33a0792d"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-05-03 08:44:30.000000000","tz":120},"committer":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-05-03 08:44:30.000000000","tz":120},"subject":"Fix the import for filter_utils","message":"Fix the import for filter_utils\n\nChanged from importing utils and calling it via utils.filter_utils\nto a proper import utils.__init__.py didn\u0027t export it so the way it\nwas called probably worked for python 2 only.\n\nThis way is a more ubiquitous way of calling it.\n\nChange-Id: Ie6a073dc286f0d8704c67f8295dfd76bce7897bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/86da45546e01985c5a03bfd40304f199bf3a444f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/86da45546e01985c5a03bfd40304f199bf3a444f"}]},"branch":"refs/heads/master"},"a83b4a6702b424c2b47b9d5568abe7db898680ce":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-06 10:28:00.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/30/656930/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/30/656930/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/656930/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/656930/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/656930/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/30/656930/2"}}},"commit":{"parents":[{"commit":"2c7056b7095c56873772ad9bf79a7a1549a74e60","subject":"Merge \"Adds secrets_retrieve to the list of available actions in barbican\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2c7056b7095c56873772ad9bf79a7a1549a74e60"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-05-03 08:44:30.000000000","tz":120},"committer":{"name":"Adriano Petrich","email":"apetrich@redhat.com","date":"2019-05-06 10:28:00.000000000","tz":0},"subject":"Fix the import for filter_utils","message":"Fix the import for filter_utils\n\nChanged from importing utils and calling it via utils.filter_utils\nto a proper import utils.__init__.py didn\u0027t export it so the way it\nwas called probably worked for python 2 only.\n\nThis way is a more ubiquitous way of calling it.\n\nChange-Id: Ie6a073dc286f0d8704c67f8295dfd76bce7897bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a83b4a6702b424c2b47b9d5568abe7db898680ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a83b4a6702b424c2b47b9d5568abe7db898680ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
