)]}'
{"id":"openstack%2Fmistral~657376","triplet_id":"openstack%2Fmistral~master~Ia6c954e688589f69a7463f1b8e02244d029e8b7a","project":"openstack/mistral","branch":"master","topic":"return_root_exec","hashtags":[],"change_id":"Ia6c954e688589f69a7463f1b8e02244d029e8b7a","subject":"Adding root_execution_id to the jinja executor function","status":"MERGED","created":"2019-05-06 11:42:11.000000000","updated":"2019-05-07 09:29:18.000000000","submitted":"2019-05-07 09:29:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657376-1557221358198-1a85bf1c","meta_rev_id":"f1e590a8c313d69e4ec41c374209d4d69d1f7e34","_number":657376,"virtual_id_number":657376,"owner":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-05-07 09:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-05-07 07:48:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-05-07 07:48:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-07 07:48:40.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-05-07 09:29:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d9bc24a0d8d8542004de1109371689b26b82a1d","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-05-06 11:42:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68afd164ca6a697b21aab6b3902779fa1dff65ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 13:53:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/76/657376/1/check/openstack-tox-pep8/4e4f99e/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/76/657376/1/check/openstack-tox-py27/cd3b6b9/ : SUCCESS in 8m 47s\n- openstack-tox-py36 http://logs.openstack.org/76/657376/1/check/openstack-tox-py36/a8d93a6/ : SUCCESS in 8m 45s\n- openstack-tox-py37 http://logs.openstack.org/76/657376/1/check/openstack-tox-py37/fc00f69/ : SUCCESS in 9m 40s\n- openstack-tox-docs http://logs.openstack.org/76/657376/1/check/openstack-tox-docs/e1e0855/html/ : SUCCESS in 5m 02s\n- openstack-tox-cover http://logs.openstack.org/76/657376/1/check/openstack-tox-cover/3fa195e/cover/ : SUCCESS in 8m 49s (non-voting)\n- mistral-devstack http://logs.openstack.org/76/657376/1/check/mistral-devstack/57dd571/ : SUCCESS in 43m 17s\n- mistral-devstack-non-apache http://logs.openstack.org/76/657376/1/check/mistral-devstack-non-apache/0cc52dc/ : SUCCESS in 43m 16s\n- mistral-devstack-kombu http://logs.openstack.org/76/657376/1/check/mistral-devstack-kombu/7e4d2a4/ : SUCCESS in 42m 43s\n- mistral-tox-unit-mysql http://logs.openstack.org/76/657376/1/check/mistral-tox-unit-mysql/7fa1e2f/ : SUCCESS in 27m 57s\n- mistral-tox-unit-postgresql http://logs.openstack.org/76/657376/1/check/mistral-tox-unit-postgresql/1d562d0/ : SUCCESS in 27m 13s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/76/657376/1/check/tripleo-ci-centos-7-scenario003-standalone/0f98d06/ : SUCCESS in 1h 03m 44s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/76/657376/1/check/tripleo-ci-centos-7-containers-multinode/75ded52/ : SUCCESS in 2h 00m 24s\n- mistral-rally-task http://logs.openstack.org/76/657376/1/check/mistral-rally-task/8d4b5d4/ : SUCCESS in 55m 40s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/76/657376/1/check/openstack-tox-lower-constraints/d4852ba/ : SUCCESS in 10m 15s","accounts_in_message":[],"_revision_number":1},{"id":"78dec289ef0e477ef03b893487bee76a7f4c4008","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-07 07:48:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"23c148a6e4c727016ff1dad4d036192ea3047c43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 07:48:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f88136dd6a1587febe37024da678c62b730f5d24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 09:29:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/76/657376/1/gate/openstack-tox-pep8/12f9f5e/ : SUCCESS in 5m 05s\n- openstack-tox-py27 http://logs.openstack.org/76/657376/1/gate/openstack-tox-py27/d6145df/ : SUCCESS in 9m 23s\n- openstack-tox-py36 http://logs.openstack.org/76/657376/1/gate/openstack-tox-py36/7a93ba6/ : SUCCESS in 8m 39s\n- openstack-tox-py37 http://logs.openstack.org/76/657376/1/gate/openstack-tox-py37/f500570/ : SUCCESS in 8m 19s\n- openstack-tox-docs http://logs.openstack.org/76/657376/1/gate/openstack-tox-docs/7536672/html/ : SUCCESS in 5m 32s\n- mistral-devstack http://logs.openstack.org/76/657376/1/gate/mistral-devstack/6983fca/ : SUCCESS in 43m 43s\n- mistral-devstack-non-apache http://logs.openstack.org/76/657376/1/gate/mistral-devstack-non-apache/76282a5/ : SUCCESS in 43m 05s\n- mistral-tox-unit-mysql http://logs.openstack.org/76/657376/1/gate/mistral-tox-unit-mysql/1543b4e/ : SUCCESS in 28m 57s\n- mistral-tox-unit-postgresql http://logs.openstack.org/76/657376/1/gate/mistral-tox-unit-postgresql/3cd7fd6/ : SUCCESS in 26m 29s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/76/657376/1/gate/tripleo-ci-centos-7-scenario003-standalone/966e2ac/ : SUCCESS in 1h 20m 32s\n- mistral-devstack-kombu http://logs.openstack.org/76/657376/1/gate/mistral-devstack-kombu/cd9ed20/ : SUCCESS in 44m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/657376/1/gate/openstack-tox-lower-constraints/5792993/ : SUCCESS in 9m 41s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/76/657376/1/gate/tripleo-ci-centos-7-undercloud-containers/5756962/ : SUCCESS in 1h 29m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b66bb1d32ef186f7a83b1e2a07b851641c64b1d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 09:29:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f172e5da24b306b81ce617d4aee58470d33f4513","revisions":{"f172e5da24b306b81ce617d4aee58470d33f4513":{"kind":"REWORK","_number":1,"created":"2019-05-06 11:42:11.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/76/657376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/657376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/657376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/657376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/657376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/657376/1"}}},"commit":{"parents":[{"commit":"2c7056b7095c56873772ad9bf79a7a1549a74e60","subject":"Merge \"Adds secrets_retrieve to the list of available actions in barbican\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2c7056b7095c56873772ad9bf79a7a1549a74e60"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-05-06 11:39:48.000000000","tz":120},"committer":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-05-06 11:39:48.000000000","tz":120},"subject":"Adding root_execution_id to the jinja executor function","message":"Adding root_execution_id to the jinja executor function\n\nThis makes getting a root_execution_id possible without having to go\nthrough filtering and querying the executions search.\n\nChange-Id: Ia6c954e688589f69a7463f1b8e02244d029e8b7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f172e5da24b306b81ce617d4aee58470d33f4513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f172e5da24b306b81ce617d4aee58470d33f4513"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
