)]}'
{"id":"openstack%2Fmistral~660551","triplet_id":"openstack%2Fmistral~master~I26bbb2e5ef5f8a5f963b47385c81a64c0227f7b8","project":"openstack/mistral","branch":"master","topic":"Bug#cinderclient","hashtags":[],"change_id":"I26bbb2e5ef5f8a5f963b47385c81a64c0227f7b8","subject":"Blacklist python-cinderclient 4.0.0","status":"MERGED","created":"2019-05-22 03:28:22.000000000","updated":"2019-05-28 12:29:53.000000000","submitted":"2019-05-28 12:29:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660551-1559046593908-03a25ead","meta_rev_id":"ed311c49201a1570a1da6007ba675bb27567bb6d","_number":660551,"virtual_id_number":660551,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-05-28 12:29:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-28 06:31:15.000000000","_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"recommended":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-05-28 10:23:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":2,"date":"2019-05-28 03:33:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-05-28 05:29:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-05-28 03:25:23.000000000","_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"date":"2019-05-28 03:25:23.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2019-05-28 03:25:23.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"date":"2019-05-28 03:25:23.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-05-28 05:29:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-28 03:25:23.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-05-28 03:25:23.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-05-28 03:25:23.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-05-28 03:25:23.000000000","updated_by":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"reviewer":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"state":"REVIEWER"},{"updated":"2019-05-28 03:33:49.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-05-28 05:29:13.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-05-28 10:23:54.000000000","updated_by":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"reviewer":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"state":"REVIEWER"},{"updated":"2019-05-28 12:29:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1fcf9489bc0d8fde6e7dd2fc47d05fef50b0ad3b","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-05-22 03:28:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eadcfec2b0886a79be666f335a1d70bfac586501","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 04:43:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/51/660551/1/check/openstack-tox-pep8/3509093/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/51/660551/1/check/openstack-tox-py27/0035fb9/ : SUCCESS in 8m 50s\n- openstack-tox-py36 http://logs.openstack.org/51/660551/1/check/openstack-tox-py36/602915a/ : SUCCESS in 8m 33s\n- openstack-tox-py37 http://logs.openstack.org/51/660551/1/check/openstack-tox-py37/19f6194/ : SUCCESS in 8m 46s\n- openstack-tox-docs http://logs.openstack.org/51/660551/1/check/openstack-tox-docs/d700691/html/ : SUCCESS in 4m 46s\n- requirements-check http://logs.openstack.org/51/660551/1/check/requirements-check/cd23075/ : FAILURE in 3m 24s\n- openstack-tox-cover http://logs.openstack.org/51/660551/1/check/openstack-tox-cover/3034ea8/cover/ : SUCCESS in 9m 03s (non-voting)\n- mistral-devstack http://logs.openstack.org/51/660551/1/check/mistral-devstack/ffa3923/ : SUCCESS in 36m 31s\n- mistral-devstack-non-apache http://logs.openstack.org/51/660551/1/check/mistral-devstack-non-apache/f5ea71d/ : SUCCESS in 43m 25s\n- mistral-devstack-kombu http://logs.openstack.org/51/660551/1/check/mistral-devstack-kombu/5ecaa59/ : SUCCESS in 39m 29s\n- mistral-tox-unit-mysql http://logs.openstack.org/51/660551/1/check/mistral-tox-unit-mysql/2ec42be/ : SUCCESS in 27m 20s\n- mistral-tox-unit-postgresql http://logs.openstack.org/51/660551/1/check/mistral-tox-unit-postgresql/9aef728/ : SUCCESS in 26m 15s\n- mistral-rally-task http://logs.openstack.org/51/660551/1/check/mistral-rally-task/ec4ac8f/ : SUCCESS in 1h 09m 00s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/51/660551/1/check/openstack-tox-lower-constraints/523612e/ : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d05477f9eafa3da346becd66f0e2bbced182a107","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-05-27 04:40:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"953e570f44d636150f5ee185454728abe22421ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 05:51:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/51/660551/2/check/openstack-tox-pep8/f5ba875/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/51/660551/2/check/openstack-tox-py27/af94208/ : SUCCESS in 8m 09s\n- openstack-tox-py36 http://logs.openstack.org/51/660551/2/check/openstack-tox-py36/63163ac/ : SUCCESS in 8m 10s\n- openstack-tox-py37 http://logs.openstack.org/51/660551/2/check/openstack-tox-py37/0c2a6b1/ : SUCCESS in 8m 38s\n- openstack-tox-docs http://logs.openstack.org/51/660551/2/check/openstack-tox-docs/467cc1b/html/ : SUCCESS in 4m 16s\n- requirements-check http://logs.openstack.org/51/660551/2/check/requirements-check/bdc4724/ : FAILURE in 2m 58s\n- openstack-tox-cover http://logs.openstack.org/51/660551/2/check/openstack-tox-cover/601f842/cover/ : SUCCESS in 8m 38s (non-voting)\n- mistral-devstack http://logs.openstack.org/51/660551/2/check/mistral-devstack/1409d84/ : SUCCESS in 42m 44s\n- mistral-devstack-non-apache http://logs.openstack.org/51/660551/2/check/mistral-devstack-non-apache/1ec0517/ : SUCCESS in 41m 48s\n- mistral-devstack-kombu http://logs.openstack.org/51/660551/2/check/mistral-devstack-kombu/88cd602/ : SUCCESS in 41m 29s\n- mistral-tox-unit-mysql http://logs.openstack.org/51/660551/2/check/mistral-tox-unit-mysql/a6f9495/ : SUCCESS in 26m 56s\n- mistral-tox-unit-postgresql http://logs.openstack.org/51/660551/2/check/mistral-tox-unit-postgresql/45473cb/ : SUCCESS in 26m 10s\n- mistral-rally-task http://logs.openstack.org/51/660551/2/check/mistral-rally-task/8c73baa/ : SUCCESS in 1h 05m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/51/660551/2/check/openstack-tox-lower-constraints/f6e522b/ : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2588bbf10629fa4bfc3f95cb242831e04dae870e","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-05-27 08:01:00.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f426bde061512410da1929bd991a5094c4bffbfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 09:16:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/51/660551/3/check/openstack-tox-pep8/f53e5c7/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/51/660551/3/check/openstack-tox-py27/a264786/ : SUCCESS in 8m 41s\n- openstack-tox-py36 http://logs.openstack.org/51/660551/3/check/openstack-tox-py36/4bf83ec/ : SUCCESS in 8m 40s\n- openstack-tox-py37 http://logs.openstack.org/51/660551/3/check/openstack-tox-py37/9087def/ : SUCCESS in 9m 07s\n- openstack-tox-docs http://logs.openstack.org/51/660551/3/check/openstack-tox-docs/719fbcf/html/ : SUCCESS in 5m 34s\n- requirements-check http://logs.openstack.org/51/660551/3/check/requirements-check/3c9bba3/ : SUCCESS in 3m 20s\n- openstack-tox-cover http://logs.openstack.org/51/660551/3/check/openstack-tox-cover/edc14e5/cover/ : SUCCESS in 9m 14s (non-voting)\n- mistral-devstack http://logs.openstack.org/51/660551/3/check/mistral-devstack/d115315/ : SUCCESS in 45m 53s\n- mistral-devstack-non-apache http://logs.openstack.org/51/660551/3/check/mistral-devstack-non-apache/eea4dbc/ : SUCCESS in 41m 17s\n- mistral-devstack-kombu http://logs.openstack.org/51/660551/3/check/mistral-devstack-kombu/1321f56/ : SUCCESS in 42m 25s\n- mistral-tox-unit-mysql http://logs.openstack.org/51/660551/3/check/mistral-tox-unit-mysql/60b8bcf/ : SUCCESS in 28m 27s\n- mistral-tox-unit-postgresql http://logs.openstack.org/51/660551/3/check/mistral-tox-unit-postgresql/4b2bc31/ : SUCCESS in 25m 38s\n- mistral-rally-task http://logs.openstack.org/51/660551/3/check/mistral-rally-task/cca4a98/ : SUCCESS in 1h 09m 00s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/51/660551/3/check/openstack-tox-lower-constraints/fadccbc/ : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":3},{"id":"1076101229b9aa09d7fd0a5bdddea1460b588829","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-05-28 03:33:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"51912abda7a3ad4249144707a194dfb148e9deed","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-28 05:29:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a40d6692cead742b6e1fea0b35aec623abbd52c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 05:29:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"925b1cb814613607635bb4b0f7d71926b4e3217c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 06:30:40.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/51/660551/3/gate/openstack-tox-pep8/2099f32/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/51/660551/3/gate/openstack-tox-py27/ec7c889/ : SUCCESS in 9m 10s\n- openstack-tox-py36 http://logs.openstack.org/51/660551/3/gate/openstack-tox-py36/8633baa/ : SUCCESS in 8m 03s\n- openstack-tox-py37 http://logs.openstack.org/51/660551/3/gate/openstack-tox-py37/3254176/ : SUCCESS in 9m 42s\n- openstack-tox-docs http://logs.openstack.org/51/660551/3/gate/openstack-tox-docs/86679a9/html/ : SUCCESS in 4m 51s\n- requirements-check http://logs.openstack.org/51/660551/3/gate/requirements-check/d46dc7a/ : SUCCESS in 2m 58s\n- mistral-devstack http://logs.openstack.org/51/660551/3/gate/mistral-devstack/23aee97/ : SUCCESS in 42m 14s\n- mistral-devstack-non-apache http://logs.openstack.org/51/660551/3/gate/mistral-devstack-non-apache/3afbff3/ : SUCCESS in 42m 37s\n- mistral-tox-unit-mysql http://logs.openstack.org/51/660551/3/gate/mistral-tox-unit-mysql/683af96/ : SUCCESS in 29m 49s\n- mistral-tox-unit-postgresql http://logs.openstack.org/51/660551/3/gate/mistral-tox-unit-postgresql/df2e1b6/ : FAILURE in 27m 53s\n- mistral-devstack-kombu http://logs.openstack.org/51/660551/3/gate/mistral-devstack-kombu/c32ea12/ : SUCCESS in 42m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/660551/3/gate/openstack-tox-lower-constraints/de42f9d/ : SUCCESS in 11m 09s","accounts_in_message":[],"_revision_number":3},{"id":"d191a39552375332d10b0e5dd238abebf52e4ee2","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-05-28 06:31:15.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"04a3e86b6e4d019b9fa5ca9e5886069905101b8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 07:26:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/51/660551/3/check/openstack-tox-pep8/e611173/ : SUCCESS in 6m 50s\n- openstack-tox-py27 http://logs.openstack.org/51/660551/3/check/openstack-tox-py27/3bc6023/ : SUCCESS in 8m 57s\n- openstack-tox-py36 http://logs.openstack.org/51/660551/3/check/openstack-tox-py36/c31d79f/ : SUCCESS in 9m 14s\n- openstack-tox-py37 http://logs.openstack.org/51/660551/3/check/openstack-tox-py37/88a7b6b/ : SUCCESS in 10m 17s\n- openstack-tox-docs http://logs.openstack.org/51/660551/3/check/openstack-tox-docs/094db9c/html/ : SUCCESS in 5m 08s\n- requirements-check http://logs.openstack.org/51/660551/3/check/requirements-check/53b53cb/ : SUCCESS in 3m 47s\n- openstack-tox-cover http://logs.openstack.org/51/660551/3/check/openstack-tox-cover/2d01a03/cover/ : SUCCESS in 8m 32s (non-voting)\n- mistral-devstack http://logs.openstack.org/51/660551/3/check/mistral-devstack/8ce5433/ : SUCCESS in 35m 01s\n- mistral-devstack-non-apache http://logs.openstack.org/51/660551/3/check/mistral-devstack-non-apache/c31194d/ : SUCCESS in 39m 44s\n- mistral-devstack-kombu http://logs.openstack.org/51/660551/3/check/mistral-devstack-kombu/fb6475e/ : SUCCESS in 41m 17s\n- mistral-tox-unit-mysql http://logs.openstack.org/51/660551/3/check/mistral-tox-unit-mysql/854702a/ : SUCCESS in 27m 54s\n- mistral-tox-unit-postgresql http://logs.openstack.org/51/660551/3/check/mistral-tox-unit-postgresql/d62478c/ : SUCCESS in 25m 34s\n- mistral-rally-task http://logs.openstack.org/51/660551/3/check/mistral-rally-task/2411743/ : SUCCESS in 53m 42s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/51/660551/3/check/openstack-tox-lower-constraints/25a060d/ : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":3},{"id":"cd505bcb9ec26da5978e7a3b476f0f6027340c62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 07:26:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a48a2654daa730eaba9f9a5a246c0c68a2213aa6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 08:17:05.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/51/660551/3/gate/openstack-tox-pep8/7c3f936/ : SUCCESS in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/51/660551/3/gate/openstack-tox-py27/33313e2/ : SUCCESS in 9m 33s\n- openstack-tox-py36 http://logs.openstack.org/51/660551/3/gate/openstack-tox-py36/46c082b/ : SUCCESS in 10m 06s\n- openstack-tox-py37 http://logs.openstack.org/51/660551/3/gate/openstack-tox-py37/fbc8f7a/ : SUCCESS in 10m 32s\n- openstack-tox-docs http://logs.openstack.org/51/660551/3/gate/openstack-tox-docs/b1b70cd/html/ : SUCCESS in 5m 11s\n- requirements-check http://logs.openstack.org/51/660551/3/gate/requirements-check/2b96ed3/ : SUCCESS in 3m 38s\n- mistral-devstack http://logs.openstack.org/51/660551/3/gate/mistral-devstack/b331cb7/ : SUCCESS in 43m 15s\n- mistral-devstack-non-apache http://logs.openstack.org/51/660551/3/gate/mistral-devstack-non-apache/fe431fd/ : SUCCESS in 42m 55s\n- mistral-tox-unit-mysql http://logs.openstack.org/51/660551/3/gate/mistral-tox-unit-mysql/05b5ba4/ : FAILURE in 28m 50s\n- mistral-tox-unit-postgresql http://logs.openstack.org/51/660551/3/gate/mistral-tox-unit-postgresql/d64546f/ : SUCCESS in 27m 12s\n- mistral-devstack-kombu http://logs.openstack.org/51/660551/3/gate/mistral-devstack-kombu/e32df70/ : SUCCESS in 39m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/660551/3/gate/openstack-tox-lower-constraints/0687dc6/ : SUCCESS in 9m 30s","accounts_in_message":[],"_revision_number":3},{"id":"eb78c9a258a7f21b21486581b97aff04f808b805","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-28 08:57:10.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8cc2dda7676c73563bffc11fbb85dae4bd2f296a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 10:21:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/51/660551/3/check/openstack-tox-pep8/04ef420/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/51/660551/3/check/openstack-tox-py27/241d408/ : SUCCESS in 9m 02s\n- openstack-tox-py36 http://logs.openstack.org/51/660551/3/check/openstack-tox-py36/ce567a4/ : SUCCESS in 8m 42s\n- openstack-tox-py37 http://logs.openstack.org/51/660551/3/check/openstack-tox-py37/9a728fb/ : SUCCESS in 9m 45s\n- openstack-tox-docs http://logs.openstack.org/51/660551/3/check/openstack-tox-docs/6010aab/html/ : SUCCESS in 4m 48s\n- requirements-check http://logs.openstack.org/51/660551/3/check/requirements-check/7cb39c9/ : SUCCESS in 3m 26s\n- openstack-tox-cover http://logs.openstack.org/51/660551/3/check/openstack-tox-cover/4764bef/cover/ : SUCCESS in 9m 17s (non-voting)\n- mistral-devstack http://logs.openstack.org/51/660551/3/check/mistral-devstack/e927972/ : SUCCESS in 40m 53s\n- mistral-devstack-non-apache http://logs.openstack.org/51/660551/3/check/mistral-devstack-non-apache/a472534/ : SUCCESS in 40m 48s\n- mistral-devstack-kombu http://logs.openstack.org/51/660551/3/check/mistral-devstack-kombu/16f9b80/ : FAILURE in 54m 04s\n- mistral-tox-unit-mysql http://logs.openstack.org/51/660551/3/check/mistral-tox-unit-mysql/c27701e/ : FAILURE in 26m 49s\n- mistral-tox-unit-postgresql http://logs.openstack.org/51/660551/3/check/mistral-tox-unit-postgresql/99eeda2/ : SUCCESS in 26m 32s\n- mistral-rally-task http://logs.openstack.org/51/660551/3/check/mistral-rally-task/3a05699/ : FAILURE in 1h 08m 47s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/51/660551/3/check/openstack-tox-lower-constraints/b6f7887/ : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":3},{"id":"3adbe3e8cdd895fb4e00629d743dc001907fc04e","author":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"date":"2019-05-28 10:23:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"93d0699d61cbc74b34ec5353d26771edd5163e72","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-28 10:25:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"dc0918431b924141e8f2cf73d1ec53f65c3cf6dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 11:37:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/51/660551/3/check/openstack-tox-pep8/dcdbf62/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/51/660551/3/check/openstack-tox-py27/8d154f4/ : SUCCESS in 8m 55s\n- openstack-tox-py36 http://logs.openstack.org/51/660551/3/check/openstack-tox-py36/893fcdf/ : SUCCESS in 8m 46s\n- openstack-tox-py37 http://logs.openstack.org/51/660551/3/check/openstack-tox-py37/ab209f2/ : SUCCESS in 8m 22s\n- openstack-tox-docs http://logs.openstack.org/51/660551/3/check/openstack-tox-docs/531d50d/html/ : SUCCESS in 4m 46s\n- requirements-check http://logs.openstack.org/51/660551/3/check/requirements-check/4f70680/ : SUCCESS in 3m 19s\n- openstack-tox-cover http://logs.openstack.org/51/660551/3/check/openstack-tox-cover/76ca6db/cover/ : SUCCESS in 8m 57s (non-voting)\n- mistral-devstack http://logs.openstack.org/51/660551/3/check/mistral-devstack/febc7cd/ : SUCCESS in 39m 58s\n- mistral-devstack-non-apache http://logs.openstack.org/51/660551/3/check/mistral-devstack-non-apache/07a4746/ : SUCCESS in 48m 23s\n- mistral-devstack-kombu http://logs.openstack.org/51/660551/3/check/mistral-devstack-kombu/e3c11a7/ : SUCCESS in 43m 29s\n- mistral-tox-unit-mysql http://logs.openstack.org/51/660551/3/check/mistral-tox-unit-mysql/fc50081/ : SUCCESS in 26m 47s\n- mistral-tox-unit-postgresql http://logs.openstack.org/51/660551/3/check/mistral-tox-unit-postgresql/0d5532f/ : SUCCESS in 26m 19s\n- mistral-rally-task http://logs.openstack.org/51/660551/3/check/mistral-rally-task/659f565/ : SUCCESS in 1h 03m 36s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/51/660551/3/check/openstack-tox-lower-constraints/4c4bb5b/ : SUCCESS in 11m 50s","accounts_in_message":[],"_revision_number":3},{"id":"322803d6c139dde280ebed3d497111f96c654d6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 11:38:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cabb06ec39367c6bcf36e83e3b217732f2ddd5f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 12:29:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/51/660551/3/gate/openstack-tox-pep8/723477e/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/51/660551/3/gate/openstack-tox-py27/9dc840e/ : SUCCESS in 8m 11s\n- openstack-tox-py36 http://logs.openstack.org/51/660551/3/gate/openstack-tox-py36/d69302c/ : SUCCESS in 8m 39s\n- openstack-tox-py37 http://logs.openstack.org/51/660551/3/gate/openstack-tox-py37/be1970b/ : SUCCESS in 8m 47s\n- openstack-tox-docs http://logs.openstack.org/51/660551/3/gate/openstack-tox-docs/d9c2b24/html/ : SUCCESS in 4m 36s\n- requirements-check http://logs.openstack.org/51/660551/3/gate/requirements-check/88ff942/ : SUCCESS in 3m 25s\n- mistral-devstack http://logs.openstack.org/51/660551/3/gate/mistral-devstack/ee592e5/ : SUCCESS in 42m 26s\n- mistral-devstack-non-apache http://logs.openstack.org/51/660551/3/gate/mistral-devstack-non-apache/5eea51b/ : SUCCESS in 46m 03s\n- mistral-tox-unit-mysql http://logs.openstack.org/51/660551/3/gate/mistral-tox-unit-mysql/f820ad2/ : SUCCESS in 28m 33s\n- mistral-tox-unit-postgresql http://logs.openstack.org/51/660551/3/gate/mistral-tox-unit-postgresql/990b278/ : SUCCESS in 26m 12s\n- mistral-devstack-kombu http://logs.openstack.org/51/660551/3/gate/mistral-devstack-kombu/6f69d2e/ : SUCCESS in 43m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/660551/3/gate/openstack-tox-lower-constraints/efc3840/ : SUCCESS in 10m 54s","accounts_in_message":[],"_revision_number":3},{"id":"5c89650c7b8bbdc819b469499cae21d2916c17b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 12:29:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"424a0a55a0146f26b6c8d3885e31bddad5050a9d","revisions":{"d5e4652a6c186ec2e3e7ba130525b060dc1f78cd":{"kind":"REWORK","_number":1,"created":"2019-05-22 03:28:22.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/51/660551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/51/660551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/660551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/660551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/660551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/51/660551/1"}}},"commit":{"parents":[{"commit":"26b9cc2bac9d4c959a2497342ecf63d23af34196","subject":"Merge \"Rework joining mechanism\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/26b9cc2bac9d4c959a2497342ecf63d23af34196"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-05-22 03:08:20.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-05-22 03:10:18.000000000","tz":480},"subject":"Blacklist python-cinderclient 4.0.0","message":"Blacklist python-cinderclient 4.0.0\n\nThis release of the Cinder client broke support for the v3\nvolume-transfer APIs unless microversion 3.55 or higher was requested.\ndepend on https://review.opendev.org/#/c/587877/\n\nChange-Id: I26bbb2e5ef5f8a5f963b47385c81a64c0227f7b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d5e4652a6c186ec2e3e7ba130525b060dc1f78cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d5e4652a6c186ec2e3e7ba130525b060dc1f78cd"}]},"branch":"refs/heads/master"},"99cbd612aeeb6d9a8213ba3e7a997a31d247f44f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-27 04:40:11.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/51/660551/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/51/660551/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/660551/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/660551/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/660551/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/51/660551/2"}}},"commit":{"parents":[{"commit":"84b8e92acc142d9cf13c44104cf66edb08b4a9de","subject":"Get rid of lookup utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/84b8e92acc142d9cf13c44104cf66edb08b4a9de"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-05-22 03:08:20.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-05-27 04:40:11.000000000","tz":0},"subject":"Blacklist python-cinderclient 4.0.0","message":"Blacklist python-cinderclient 4.0.0\n\nThis release of the Cinder client broke support for the v3\nvolume-transfer APIs unless microversion 3.55 or higher was requested.\ndepend on https://review.opendev.org/#/c/587877/\n\nChange-Id: I26bbb2e5ef5f8a5f963b47385c81a64c0227f7b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99cbd612aeeb6d9a8213ba3e7a997a31d247f44f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99cbd612aeeb6d9a8213ba3e7a997a31d247f44f"}]},"branch":"refs/heads/master"},"424a0a55a0146f26b6c8d3885e31bddad5050a9d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-27 08:01:00.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/51/660551/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/51/660551/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/660551/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/660551/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/660551/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/51/660551/3"}}},"commit":{"parents":[{"commit":"49a276538343fed3dda2ad0d54b4b6b51a8972ee","subject":"Merge \"Constraint networkx to \u003c2.3 for Python 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/49a276538343fed3dda2ad0d54b4b6b51a8972ee"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-05-22 03:08:20.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-05-27 08:01:00.000000000","tz":0},"subject":"Blacklist python-cinderclient 4.0.0","message":"Blacklist python-cinderclient 4.0.0\n\nThis release of the Cinder client broke support for the v3\nvolume-transfer APIs unless microversion 3.55 or higher was requested.\ndepend on https://review.opendev.org/#/c/587877/\n\nChange-Id: I26bbb2e5ef5f8a5f963b47385c81a64c0227f7b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/424a0a55a0146f26b6c8d3885e31bddad5050a9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/424a0a55a0146f26b6c8d3885e31bddad5050a9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
