)]}'
{"id":"openstack%2Fmistral~661714","triplet_id":"openstack%2Fmistral~stable%2Fpike~I0ed4e690f67966ba2d145ad6430b517bd896ced6","project":"openstack/mistral","branch":"stable/pike","topic":"bug/1811775-stable/pike","hashtags":[],"change_id":"I0ed4e690f67966ba2d145ad6430b517bd896ced6","subject":"Process all task batches in wf output evaluation","status":"MERGED","created":"2019-05-28 10:04:53.000000000","updated":"2019-05-28 14:50:20.000000000","submitted":"2019-05-28 14:50:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661714-1559055020424-d087b668","meta_rev_id":"a4dac1e23c59aeb4692b071413b0404eea7fc86a","_number":661714,"virtual_id_number":661714,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-05-28 14:50:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":2,"date":"2019-05-28 10:08:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-05-28 10:04:53.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":1,"date":"2019-05-28 10:08:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-28 10:04:53.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-05-28 14:50:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84ccc99a10ae9e37473b3786c907125cd4ce0206","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-28 10:04:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d22fd65956f2d6161e26ccc241f23369dc881e2a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-28 10:08:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d69df7c07cb4593a2917293258360e53787c4bf6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 12:29:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/14/661714/1/check/openstack-tox-pep8/8c748b6/ : SUCCESS in 5m 15s\n- openstack-tox-py27 http://logs.openstack.org/14/661714/1/check/openstack-tox-py27/de7e422/ : SUCCESS in 7m 34s\n- openstack-tox-py35 http://logs.openstack.org/14/661714/1/check/openstack-tox-py35/83801bb/ : SUCCESS in 7m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/14/661714/1/check/build-openstack-sphinx-docs/324c86c/html/ : SUCCESS in 4m 45s\n- openstack-tox-cover http://logs.openstack.org/14/661714/1/check/openstack-tox-cover/3bc0759/cover/ : SUCCESS in 10m 49s (non-voting)\n- rally-dsvm-mistral-task http://logs.openstack.org/14/661714/1/check/rally-dsvm-mistral-task/5929d45/ : FAILURE in 27m 47s (non-voting)\n- mistral-devstack-dsvm http://logs.openstack.org/14/661714/1/check/mistral-devstack-dsvm/393066d/ : FAILURE in 39m 58s (non-voting)\n- mistral-devstack-dsvm-non-apache http://logs.openstack.org/14/661714/1/check/mistral-devstack-dsvm-non-apache/031fa4c/ : FAILURE in 39m 03s (non-voting)\n- mistral-tox-unit-mysql http://logs.openstack.org/14/661714/1/check/mistral-tox-unit-mysql/6cf45fa/ : FAILURE in 3m 42s (non-voting)\n- mistral-tox-unit-postgresql http://logs.openstack.org/14/661714/1/check/mistral-tox-unit-postgresql/573284e/ : FAILURE in 3m 30s (non-voting)\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/14/661714/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/a684887/ : SUCCESS in 2h 14m 48s (non-voting)\n- mistral-devstack-dsvm-kombu http://logs.openstack.org/14/661714/1/check/mistral-devstack-dsvm-kombu/f4e6aff/ : FAILURE in 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e8d896de861bfc994fc71ca5042d90b180c01ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 12:30:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"acb240bc1611156c1d0c19952f48f01a1db7e4d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 14:50:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/14/661714/1/gate/openstack-tox-pep8/99c7c4c/ : SUCCESS in 5m 12s\n- openstack-tox-py27 http://logs.openstack.org/14/661714/1/gate/openstack-tox-py27/f5cae5e/ : SUCCESS in 7m 27s\n- openstack-tox-py35 http://logs.openstack.org/14/661714/1/gate/openstack-tox-py35/780ecaf/ : SUCCESS in 8m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/14/661714/1/gate/build-openstack-sphinx-docs/9407f40/html/ : SUCCESS in 4m 27s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/14/661714/1/gate/tripleo-ci-centos-7-scenario003-multinode-oooq-container/df57ba4/ : SUCCESS in 2h 18m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9f570d0d839418ef432be6ecc5ad60905a6f5eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 14:50:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b078bc581b450149c46e063529baa4ac38a1b7f","revisions":{"4b078bc581b450149c46e063529baa4ac38a1b7f":{"kind":"REWORK","_number":1,"created":"2019-05-28 10:04:53.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/14/661714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/661714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/661714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/661714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/661714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/661714/1"}}},"commit":{"parents":[{"commit":"b239a10cda7c6e38b5a85e25087e3e57c49d5ec2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b239a10cda7c6e38b5a85e25087e3e57c49d5ec2"}]}],"author":{"name":"Andras Kovi","email":"akovi@nokia.com","date":"2019-01-15 08:33:06.000000000","tz":60},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-05-28 10:04:53.000000000","tz":0},"subject":"Process all task batches in wf output evaluation","message":"Process all task batches in wf output evaluation\n\nAll batches must be processed in workflow output evaluation. An\nempty batch means only that no tasks were end tasks in the queried\nslice.\n\nCloses-Bug: 1811775\nChange-Id: I0ed4e690f67966ba2d145ad6430b517bd896ced6\n(cherry picked from commit 81af1b48384d04400a3b988aca7d4b1fa9596af5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4b078bc581b450149c46e063529baa4ac38a1b7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4b078bc581b450149c46e063529baa4ac38a1b7f"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
