)]}'
{"id":"openstack%2Fmistral~662102","triplet_id":"openstack%2Fmistral~master~Ic73f2112406e681ae8a2aa67bcbccebe488fc03c","project":"openstack/mistral","branch":"master","topic":"rework_finding_joins","hashtags":[],"change_id":"Ic73f2112406e681ae8a2aa67bcbccebe488fc03c","subject":"Rework finding indirectly affected created joins","status":"MERGED","created":"2019-05-29 22:17:04.000000000","updated":"2019-05-30 09:21:50.000000000","submitted":"2019-05-30 09:21:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662102-1559208110668-01fcf80d","meta_rev_id":"feda400407b076bb65db300f91d8a2cf6110b792","_number":662102,"virtual_id_number":662102,"owner":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-05-30 09:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-05-30 07:31:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-05-30 05:10:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"date":"2019-05-30 05:09:43.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-05-30 07:31:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2019-05-30 05:09:40.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2019-05-30 05:09:36.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-30 05:09:36.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-05-30 05:09:40.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-05-30 05:09:43.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-05-30 05:10:20.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-05-30 07:31:52.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-05-30 09:21:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab814175e780a7b258306041eb2de5eb4743560a","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-05-29 22:17:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3be14bc298e8a87afbb53075234605e0df950e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 01:17:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/02/662102/1/check/openstack-tox-pep8/9c05f79/ : SUCCESS in 5m 43s\n- openstack-tox-py27 http://logs.openstack.org/02/662102/1/check/openstack-tox-py27/bc5b480/ : SUCCESS in 9m 09s\n- openstack-tox-py36 http://logs.openstack.org/02/662102/1/check/openstack-tox-py36/6ec756d/ : SUCCESS in 8m 48s\n- openstack-tox-py37 http://logs.openstack.org/02/662102/1/check/openstack-tox-py37/5eb94be/ : SUCCESS in 9m 46s\n- openstack-tox-docs http://logs.openstack.org/02/662102/1/check/openstack-tox-docs/5839bdd/html/ : SUCCESS in 5m 05s\n- openstack-tox-cover http://logs.openstack.org/02/662102/1/check/openstack-tox-cover/ce11f0a/cover/ : SUCCESS in 9m 49s (non-voting)\n- mistral-devstack http://logs.openstack.org/02/662102/1/check/mistral-devstack/c01c94e/ : SUCCESS in 43m 13s\n- mistral-devstack-non-apache http://logs.openstack.org/02/662102/1/check/mistral-devstack-non-apache/bbad583/ : SUCCESS in 42m 34s\n- mistral-devstack-kombu http://logs.openstack.org/02/662102/1/check/mistral-devstack-kombu/ea4bd5d/ : SUCCESS in 39m 37s\n- mistral-tox-unit-mysql http://logs.openstack.org/02/662102/1/check/mistral-tox-unit-mysql/8242b0c/ : SUCCESS in 30m 35s\n- mistral-tox-unit-postgresql http://logs.openstack.org/02/662102/1/check/mistral-tox-unit-postgresql/00a89f9/ : SUCCESS in 29m 05s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/02/662102/1/check/tripleo-ci-centos-7-scenario003-standalone/8b59ba0/ : SUCCESS in 1h 07m 33s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/02/662102/1/check/tripleo-ci-centos-7-containers-multinode/d94cf13/ : SUCCESS in 2h 09m 30s\n- mistral-rally-task http://logs.openstack.org/02/662102/1/check/mistral-rally-task/b3c62f7/ : SUCCESS in 1h 02m 19s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/02/662102/1/check/openstack-tox-lower-constraints/ce7ffc3/ : SUCCESS in 9m 47s","accounts_in_message":[],"_revision_number":1},{"id":"de1683a958725aa927721122d3f9e14a52151c90","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-30 05:10:20.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM! Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"766ded398e5b785c58925a25c9f9e929c004293c","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-05-30 07:31:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"03ae626104dcfc4c9390238b4f40d03a6832ea09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 07:32:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a8b07ac42a882ff8da966bb44fcab11a1f5da5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 09:21:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/02/662102/1/gate/openstack-tox-pep8/8a29447/ : SUCCESS in 6m 20s\n- openstack-tox-py27 http://logs.openstack.org/02/662102/1/gate/openstack-tox-py27/bc69018/ : SUCCESS in 8m 56s\n- openstack-tox-py36 http://logs.openstack.org/02/662102/1/gate/openstack-tox-py36/2cbfe38/ : SUCCESS in 8m 37s\n- openstack-tox-py37 http://logs.openstack.org/02/662102/1/gate/openstack-tox-py37/93b30f1/ : SUCCESS in 8m 42s\n- openstack-tox-docs http://logs.openstack.org/02/662102/1/gate/openstack-tox-docs/86dbe7f/html/ : SUCCESS in 5m 02s\n- mistral-devstack http://logs.openstack.org/02/662102/1/gate/mistral-devstack/86c9d3f/ : SUCCESS in 45m 03s\n- mistral-devstack-non-apache http://logs.openstack.org/02/662102/1/gate/mistral-devstack-non-apache/569a50a/ : SUCCESS in 40m 36s\n- mistral-tox-unit-mysql http://logs.openstack.org/02/662102/1/gate/mistral-tox-unit-mysql/ed7be58/ : SUCCESS in 27m 13s\n- mistral-tox-unit-postgresql http://logs.openstack.org/02/662102/1/gate/mistral-tox-unit-postgresql/557e05d/ : SUCCESS in 28m 11s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/02/662102/1/gate/tripleo-ci-centos-7-scenario003-standalone/a1fd5df/ : SUCCESS in 1h 12m 34s\n- mistral-devstack-kombu http://logs.openstack.org/02/662102/1/gate/mistral-devstack-kombu/6a5cae8/ : SUCCESS in 44m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/662102/1/gate/openstack-tox-lower-constraints/3c3c12c/ : SUCCESS in 9m 34s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/02/662102/1/gate/tripleo-ci-centos-7-undercloud-containers/5123457/ : SUCCESS in 1h 36m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0cde46f17a29a3331e9603b889ea57872f288764","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 09:21:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c52688523ce47f1b58e370768932bf3bc0f7798d","revisions":{"c52688523ce47f1b58e370768932bf3bc0f7798d":{"kind":"REWORK","_number":1,"created":"2019-05-29 22:17:04.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/02/662102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/02/662102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/662102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/662102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/662102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/02/662102/1"}}},"commit":{"parents":[{"commit":"49a276538343fed3dda2ad0d54b4b6b51a8972ee","subject":"Merge \"Constraint networkx to \u003c2.3 for Python 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/49a276538343fed3dda2ad0d54b4b6b51a8972ee"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-05-29 21:58:00.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-05-29 21:58:00.000000000","tz":120},"subject":"Rework finding indirectly affected created joins","message":"Rework finding indirectly affected created joins\n\nCurrent algorithm of finding the joins requires a recursive search\nthrough the execution graph, which leads to a large number of calls\nto the database.\n\nTo optimize it we introduce a new algorithm that requires only one db\nrequest. It downloads all potencial join task ids and names, and then\nanalyzes them without any additional db calls.\n\nChange-Id: Ic73f2112406e681ae8a2aa67bcbccebe488fc03c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c52688523ce47f1b58e370768932bf3bc0f7798d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c52688523ce47f1b58e370768932bf3bc0f7798d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
