)]}'
{"id":"openstack%2Fmistral~662124","triplet_id":"openstack%2Fmistral~master~Id87f5834492a69cc2af2278e7736149fcb65d73d","project":"openstack/mistral","branch":"master","topic":"workflow-input-dict-cut","hashtags":[],"change_id":"Id87f5834492a69cc2af2278e7736149fcb65d73d","subject":"feat: add workflow input dict support unicode in py2","status":"ABANDONED","created":"2019-05-30 02:32:15.000000000","updated":"2019-05-31 07:09:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"df3b87a2818a00f423a85c2c5e98dfeae91d623f","_number":662124,"virtual_id_number":662124,"owner":{"_account_id":30524,"name":"abner zhao","email":"opsabnerzhao@gmail.com","username":"abner"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-05-30 05:33:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-30 05:33:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3267a9fe203987c2a1b6c725e92fb08a858a4f7","author":{"_account_id":30524,"name":"abner zhao","email":"opsabnerzhao@gmail.com","username":"abner"},"date":"2019-05-30 02:32:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1af93d291567938c70e2dfdfcb79053f580d9ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 05:33:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/24/662124/1/check/openstack-tox-pep8/110e800/ : FAILURE in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/24/662124/1/check/openstack-tox-py27/88d1424/ : SUCCESS in 8m 42s\n- openstack-tox-py36 http://logs.openstack.org/24/662124/1/check/openstack-tox-py36/06ede17/ : FAILURE in 16m 57s\n- openstack-tox-py37 http://logs.openstack.org/24/662124/1/check/openstack-tox-py37/e4ab300/ : FAILURE in 17m 59s\n- openstack-tox-docs http://logs.openstack.org/24/662124/1/check/openstack-tox-docs/70d9ee5/html/ : SUCCESS in 6m 33s\n- openstack-tox-cover http://logs.openstack.org/24/662124/1/check/openstack-tox-cover/1b38be2/ : FAILURE in 18m 19s (non-voting)\n- mistral-devstack http://logs.openstack.org/24/662124/1/check/mistral-devstack/e9546d8/ : SUCCESS in 48m 53s\n- mistral-devstack-non-apache http://logs.openstack.org/24/662124/1/check/mistral-devstack-non-apache/7dbadf6/ : SUCCESS in 54m 07s\n- mistral-devstack-kombu http://logs.openstack.org/24/662124/1/check/mistral-devstack-kombu/6553084/ : SUCCESS in 54m 27s\n- mistral-tox-unit-mysql http://logs.openstack.org/24/662124/1/check/mistral-tox-unit-mysql/2d8b12c/ : FAILURE in 40m 00s\n- mistral-tox-unit-postgresql http://logs.openstack.org/24/662124/1/check/mistral-tox-unit-postgresql/9bc2dc9/ : FAILURE in 39m 23s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/24/662124/1/check/tripleo-ci-centos-7-scenario003-standalone/94613ce/ : SUCCESS in 1h 13m 00s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/662124/1/check/tripleo-ci-centos-7-containers-multinode/a529959/ : SUCCESS in 2h 00m 47s\n- mistral-rally-task http://logs.openstack.org/24/662124/1/check/mistral-rally-task/7e79e78/ : FAILURE in 1h 19m 09s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/662124/1/check/openstack-tox-lower-constraints/84b33d3/ : FAILURE in 19m 11s","accounts_in_message":[],"_revision_number":1},{"id":"138775494a2ba57f64bf1f8f89d697de984e59cc","author":{"_account_id":30524,"name":"abner zhao","email":"opsabnerzhao@gmail.com","username":"abner"},"date":"2019-05-31 07:09:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2db65e8f5d7177681313b0ed09bb94e8108da7ae","revisions":{"2db65e8f5d7177681313b0ed09bb94e8108da7ae":{"kind":"REWORK","_number":1,"created":"2019-05-30 02:32:15.000000000","uploader":{"_account_id":30524,"name":"abner zhao","email":"opsabnerzhao@gmail.com","username":"abner"},"ref":"refs/changes/24/662124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/24/662124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/662124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/662124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/24/662124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/24/662124/1"}}},"commit":{"parents":[{"commit":"26eb92a3ae3c7f22cc527555ed94817655faa007","subject":"feat: Adds workflow input dict Chinese parameter support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/26eb92a3ae3c7f22cc527555ed94817655faa007"}]}],"author":{"name":"abner","email":"opsabnerzhao@gmail.com","date":"2019-05-30 02:31:35.000000000","tz":480},"committer":{"name":"abner","email":"opsabnerzhao@gmail.com","date":"2019-05-30 02:31:35.000000000","tz":480},"subject":"feat: add workflow input dict support unicode in py2","message":"feat: add workflow input dict support unicode in py2\n\nChange-Id: Id87f5834492a69cc2af2278e7736149fcb65d73d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2db65e8f5d7177681313b0ed09bb94e8108da7ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2db65e8f5d7177681313b0ed09bb94e8108da7ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
