)]}'
{"id":"openstack%2Fmistral~662397","triplet_id":"openstack%2Fmistral~master~Ib06c6ee9e0227063b173f4b6ddc47558872fd5c7","project":"openstack/mistral","branch":"master","topic":"workflow-input-dict-support-unicode","hashtags":[],"change_id":"Ib06c6ee9e0227063b173f4b6ddc47558872fd5c7","subject":"feat:add workflow input dict support unicode in py2","status":"ABANDONED","created":"2019-05-31 07:03:46.000000000","updated":"2026-03-28 19:37:16.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"47ea25c3cb2979a1e9559b6a4b9ad15ced745125","_number":662397,"virtual_id_number":662397,"owner":{"_account_id":30524,"name":"abner zhao","email":"opsabnerzhao@gmail.com","username":"abner"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":1,"date":"2019-06-05 04:05:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"disliked":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-06-12 09:46:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":1,"date":"2019-06-11 07:07:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":-1,"date":"2019-08-28 05:47:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":1,"date":"2019-06-12 09:41:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-28 05:26:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 04:05:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-11 07:07:14.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-06-12 09:41:44.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-06-12 09:46:37.000000000","updated_by":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"reviewer":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"state":"REVIEWER"},{"updated":"2019-08-28 05:26:08.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-08-28 05:47:37.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"}],"messages":[{"id":"da4c14b820cf8fbc0bc08f69ee2a718a255a9233","author":{"_account_id":30524,"name":"abner zhao","email":"opsabnerzhao@gmail.com","username":"abner"},"date":"2019-05-31 07:03:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6064e151c923bc700cbcb71bea8f22f42fe15fb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 09:16:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/97/662397/1/check/openstack-tox-pep8/bf30257/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/97/662397/1/check/openstack-tox-py27/cc672a8/ : SUCCESS in 9m 15s\n- openstack-tox-py36 http://logs.openstack.org/97/662397/1/check/openstack-tox-py36/ea4f9f1/ : SUCCESS in 9m 29s\n- openstack-tox-py37 http://logs.openstack.org/97/662397/1/check/openstack-tox-py37/dcf978c/ : SUCCESS in 8m 28s\n- openstack-tox-docs http://logs.openstack.org/97/662397/1/check/openstack-tox-docs/187cd34/html/ : SUCCESS in 4m 26s\n- openstack-tox-cover http://logs.openstack.org/97/662397/1/check/openstack-tox-cover/01bf7bf/cover/ : SUCCESS in 10m 14s (non-voting)\n- mistral-devstack http://logs.openstack.org/97/662397/1/check/mistral-devstack/937236b/ : SUCCESS in 43m 54s\n- mistral-devstack-non-apache http://logs.openstack.org/97/662397/1/check/mistral-devstack-non-apache/e66d73c/ : SUCCESS in 42m 28s\n- mistral-devstack-kombu http://logs.openstack.org/97/662397/1/check/mistral-devstack-kombu/0f1cc38/ : SUCCESS in 43m 42s\n- mistral-tox-unit-mysql http://logs.openstack.org/97/662397/1/check/mistral-tox-unit-mysql/9bcbb01/ : SUCCESS in 31m 37s\n- mistral-tox-unit-postgresql http://logs.openstack.org/97/662397/1/check/mistral-tox-unit-postgresql/31236d6/ : SUCCESS in 25m 50s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/97/662397/1/check/tripleo-ci-centos-7-scenario003-standalone/87f77b0/ : SUCCESS in 1h 16m 08s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/662397/1/check/tripleo-ci-centos-7-containers-multinode/f8c0f08/ : SUCCESS in 2h 06m 45s\n- mistral-rally-task http://logs.openstack.org/97/662397/1/check/mistral-rally-task/daa129a/ : SUCCESS in 51m 57s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/97/662397/1/check/openstack-tox-lower-constraints/6c1e2b5/ : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f352a686cdb02dfa90fd0eda65c20d478b2dc73a","author":{"_account_id":30524,"name":"abner zhao","email":"opsabnerzhao@gmail.com","username":"abner"},"date":"2019-05-31 11:24:42.000000000","message":"Removed Verified+1 by Zuul (22348)\n","accounts_in_message":[],"_revision_number":1},{"id":"84d4e2594cbc856cef64eb872e99e1675483608a","author":{"_account_id":30524,"name":"abner zhao","email":"opsabnerzhao@gmail.com","username":"abner"},"date":"2019-06-05 01:46:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"077eede5ddfa2b433df509be499a0252fa261be6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 04:05:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/97/662397/2/check/openstack-tox-pep8/483f78d/ : SUCCESS in 7m 34s\n- openstack-tox-py27 http://logs.openstack.org/97/662397/2/check/openstack-tox-py27/388e327/ : SUCCESS in 8m 39s\n- openstack-tox-py36 http://logs.openstack.org/97/662397/2/check/openstack-tox-py36/38d632f/ : SUCCESS in 10m 35s\n- openstack-tox-py37 http://logs.openstack.org/97/662397/2/check/openstack-tox-py37/3e0d23a/ : SUCCESS in 8m 59s\n- openstack-tox-docs http://logs.openstack.org/97/662397/2/check/openstack-tox-docs/e70313f/html/ : SUCCESS in 7m 03s\n- openstack-tox-cover http://logs.openstack.org/97/662397/2/check/openstack-tox-cover/90f3a30/cover/ : SUCCESS in 8m 59s (non-voting)\n- mistral-devstack http://logs.openstack.org/97/662397/2/check/mistral-devstack/1911189/ : SUCCESS in 37m 57s\n- mistral-devstack-non-apache http://logs.openstack.org/97/662397/2/check/mistral-devstack-non-apache/54340b9/ : SUCCESS in 1h 02m 13s\n- mistral-devstack-kombu http://logs.openstack.org/97/662397/2/check/mistral-devstack-kombu/887eafa/ : SUCCESS in 38m 05s\n- mistral-tox-unit-mysql http://logs.openstack.org/97/662397/2/check/mistral-tox-unit-mysql/972daaf/ : SUCCESS in 36m 33s\n- mistral-tox-unit-postgresql http://logs.openstack.org/97/662397/2/check/mistral-tox-unit-postgresql/2344d6d/ : SUCCESS in 27m 12s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/97/662397/2/check/tripleo-ci-centos-7-scenario003-standalone/c055508/ : SUCCESS in 1h 16m 58s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/662397/2/check/tripleo-ci-centos-7-containers-multinode/df8206d/ : SUCCESS in 2h 03m 00s\n- mistral-rally-task http://logs.openstack.org/97/662397/2/check/mistral-rally-task/2792022/ : SUCCESS in 1h 08m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/97/662397/2/check/openstack-tox-lower-constraints/ce9fee8/ : SUCCESS in 10m 55s","accounts_in_message":[],"_revision_number":2},{"id":"4332bb14329f2d18278e8c4517789ea108d7e5c9","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-11 07:07:14.000000000","message":"Patch Set 2: Code-Review+1\n\nThis is not the only place, where this error could be occured, but I\u0027m OK with fixing just this case for now.","accounts_in_message":[],"_revision_number":2},{"id":"b0d14cbf9510dfccb8fe68e9981fbc4ba83a94a9","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-06-12 09:41:44.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6b0bad9f615c75699084ba99fe21643b64b84143","author":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"date":"2019-06-12 09:46:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4611e31f3444d1a3ea4a1ee00e6cd48b716f6ee5","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-28 05:26:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6523ea23dd655367dcef25d0926adcad56270343","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-28 05:47:37.000000000","message":"Patch Set 2: Code-Review-1\n\nPlease make this change in mistral-lib. The point is that the function was effectively moved over there. https://review.opendev.org/#/c/676373 removes it from mistral completely.","accounts_in_message":[],"_revision_number":2},{"id":"47ea25c3cb2979a1e9559b6a4b9ad15ced745125","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-03-28 19:37:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9f548ac0e4eeea3bde913085450bd525958f6bda","revisions":{"9df384dbd5a92965e2a396a8bc7c0d2e9be3cccb":{"kind":"REWORK","_number":1,"created":"2019-05-31 07:03:46.000000000","uploader":{"_account_id":30524,"name":"abner zhao","email":"opsabnerzhao@gmail.com","username":"abner"},"ref":"refs/changes/97/662397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/97/662397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/662397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/662397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/662397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/97/662397/1"}}},"commit":{"parents":[{"commit":"654075e553c2b87090a73f1e983dc6e40313832b","subject":"Merge \"Rework finding indirectly affected created joins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/654075e553c2b87090a73f1e983dc6e40313832b"}]}],"author":{"name":"Abnerzhao","email":"Abnerzhao@MacBook-Pro.local","date":"2019-05-31 07:02:29.000000000","tz":480},"committer":{"name":"abner","email":"opsabnerzhao@gmail.com","date":"2019-05-31 07:03:43.000000000","tz":480},"subject":"feat:add workflow input dict support unicode in py2","message":"feat:add workflow input dict support unicode in py2\n\nChange-Id: Ib06c6ee9e0227063b173f4b6ddc47558872fd5c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9df384dbd5a92965e2a396a8bc7c0d2e9be3cccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9df384dbd5a92965e2a396a8bc7c0d2e9be3cccb"}]},"branch":"refs/heads/master"},"9f548ac0e4eeea3bde913085450bd525958f6bda":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-05 01:46:41.000000000","uploader":{"_account_id":30524,"name":"abner zhao","email":"opsabnerzhao@gmail.com","username":"abner"},"ref":"refs/changes/97/662397/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/97/662397/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/662397/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/662397/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/662397/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/97/662397/2"}}},"commit":{"parents":[{"commit":"ea4681500a73eec8356678ec949f9bdb78271d23","subject":"Merge \"Move action caching to db layer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ea4681500a73eec8356678ec949f9bdb78271d23"}]}],"author":{"name":"Abnerzhao","email":"Abnerzhao@MacBook-Pro.local","date":"2019-05-31 07:02:29.000000000","tz":480},"committer":{"name":"abner zhao","email":"opsabnerzhao@gmail.com","date":"2019-06-05 01:46:41.000000000","tz":0},"subject":"feat:add workflow input dict support unicode in py2","message":"feat:add workflow input dict support unicode in py2\n\nChange-Id: Ib06c6ee9e0227063b173f4b6ddc47558872fd5c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f548ac0e4eeea3bde913085450bd525958f6bda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f548ac0e4eeea3bde913085450bd525958f6bda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
