)]}'
{"id":"openstack%2Fmistral~662648","triplet_id":"openstack%2Fmistral~master~I8c470cbb424e63c16e599758bb93192985666353","project":"openstack/mistral","branch":"master","topic":"move_action_caching","hashtags":[],"change_id":"I8c470cbb424e63c16e599758bb93192985666353","subject":"Move action caching to db layer","status":"MERGED","created":"2019-06-02 20:41:21.000000000","updated":"2019-06-04 15:05:52.000000000","submitted":"2019-06-04 15:05:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662648-1559660752938-2368eebd","meta_rev_id":"c0195ec34e6586ce7e443c3df24db2720eb12c09","_number":662648,"virtual_id_number":662648,"owner":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-06-04 15:05:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-03 18:44:31.000000000","_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-06-03 10:13:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":2,"date":"2019-06-03 18:55:20.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-06-03 04:32:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"date":"2019-06-03 06:22:07.000000000","_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"date":"2019-06-03 06:22:07.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-06-03 10:13:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":1,"date":"2019-06-03 18:55:20.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"date":"2019-06-03 06:22:07.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 04:32:53.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-06-03 06:22:07.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-06-03 06:22:07.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-06-03 06:22:07.000000000","updated_by":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"reviewer":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"state":"REVIEWER"},{"updated":"2019-06-03 10:13:33.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-06-03 18:55:20.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-06-04 15:05:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f3727f6f1af0cad788470abb7617de34be776b1","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-02 20:41:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c837a6aa3aea6ab85339b432dc3ee51049d0235","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 22:37:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/48/662648/1/check/openstack-tox-pep8/a953aff/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/48/662648/1/check/openstack-tox-py27/90ed6e7/ : SUCCESS in 9m 03s\n- openstack-tox-py36 http://logs.openstack.org/48/662648/1/check/openstack-tox-py36/8f1ba70/ : SUCCESS in 8m 26s\n- openstack-tox-py37 http://logs.openstack.org/48/662648/1/check/openstack-tox-py37/6a3068b/ : SUCCESS in 8m 38s\n- openstack-tox-docs http://logs.openstack.org/48/662648/1/check/openstack-tox-docs/eacbc9f/html/ : SUCCESS in 4m 38s\n- openstack-tox-cover http://logs.openstack.org/48/662648/1/check/openstack-tox-cover/c3bff3b/cover/ : SUCCESS in 9m 13s (non-voting)\n- mistral-devstack http://logs.openstack.org/48/662648/1/check/mistral-devstack/9feae2e/ : SUCCESS in 44m 37s\n- mistral-devstack-non-apache http://logs.openstack.org/48/662648/1/check/mistral-devstack-non-apache/52ba5f2/ : SUCCESS in 42m 38s\n- mistral-devstack-kombu http://logs.openstack.org/48/662648/1/check/mistral-devstack-kombu/61d9084/ : SUCCESS in 43m 20s\n- mistral-tox-unit-mysql http://logs.openstack.org/48/662648/1/check/mistral-tox-unit-mysql/8dcd7a6/ : SUCCESS in 27m 06s\n- mistral-tox-unit-postgresql http://logs.openstack.org/48/662648/1/check/mistral-tox-unit-postgresql/d3c2674/ : SUCCESS in 26m 30s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/48/662648/1/check/tripleo-ci-centos-7-scenario003-standalone/a053263/ : SUCCESS in 1h 09m 38s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/48/662648/1/check/tripleo-ci-centos-7-containers-multinode/065eb24/ : SUCCESS in 1h 50m 50s\n- mistral-rally-task http://logs.openstack.org/48/662648/1/check/mistral-rally-task/6c3cd45/ : FAILURE in 1h 14m 12s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/48/662648/1/check/openstack-tox-lower-constraints/c2b94cc/ : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":1},{"id":"491d3a3ea7691dfea33ad4373c7a1e4e6c2e0066","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-03 04:32:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"63b80b147350e10df2258758d16941185ecd9868","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-06-03 10:13:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d626c20fb845aa6aaac453432cc515b8282d7b19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 10:13:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af574061b08ffa8f5e0d1268ed65e87bef95318f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 12:04:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/48/662648/1/gate/openstack-tox-pep8/3599aa5/ : SUCCESS in 5m 57s\n- openstack-tox-py27 http://logs.openstack.org/48/662648/1/gate/openstack-tox-py27/7191ec1/ : FAILURE in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/48/662648/1/gate/openstack-tox-py36/1a410d5/ : FAILURE in 3m 28s\n- openstack-tox-py37 http://logs.openstack.org/48/662648/1/gate/openstack-tox-py37/97a2880/ : FAILURE in 3m 54s\n- openstack-tox-docs http://logs.openstack.org/48/662648/1/gate/openstack-tox-docs/6bd75de/ : POST_FAILURE in 3m 21s\n- mistral-devstack http://logs.openstack.org/48/662648/1/gate/mistral-devstack/1d5b160/ : SUCCESS in 42m 56s\n- mistral-devstack-non-apache http://logs.openstack.org/48/662648/1/gate/mistral-devstack-non-apache/e02c83a/ : SUCCESS in 36m 18s\n- mistral-tox-unit-mysql http://logs.openstack.org/48/662648/1/gate/mistral-tox-unit-mysql/c6f387c/ : SUCCESS in 27m 24s\n- mistral-tox-unit-postgresql http://logs.openstack.org/48/662648/1/gate/mistral-tox-unit-postgresql/30af28e/ : SUCCESS in 25m 49s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/48/662648/1/gate/tripleo-ci-centos-7-scenario003-standalone/f32b66e/ : SUCCESS in 1h 15m 15s\n- mistral-devstack-kombu http://logs.openstack.org/48/662648/1/gate/mistral-devstack-kombu/17583c0/ : POST_FAILURE in 49m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/662648/1/gate/openstack-tox-lower-constraints/01186cb/ : SUCCESS in 9m 33s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/48/662648/1/gate/tripleo-ci-centos-7-undercloud-containers/0ec8507/ : SUCCESS in 1h 35m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2a811bb852738835c8e8096a5aaabaafc0815ba5","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-03 18:44:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"267df6f27d4648393cf5f64bc22859aa76a84698","author":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"date":"2019-06-03 18:55:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7827a4e18201e4cc1a63554ef7786e3a376344f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 21:02:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/48/662648/1/check/openstack-tox-pep8/30e7b25/ : SUCCESS in 5m 21s\n- openstack-tox-py27 http://logs.openstack.org/48/662648/1/check/openstack-tox-py27/9ecbaec/ : SUCCESS in 8m 44s\n- openstack-tox-py36 http://logs.openstack.org/48/662648/1/check/openstack-tox-py36/38bd482/ : SUCCESS in 8m 54s\n- openstack-tox-py37 http://logs.openstack.org/48/662648/1/check/openstack-tox-py37/f5409fd/ : SUCCESS in 11m 56s\n- openstack-tox-docs http://logs.openstack.org/48/662648/1/check/openstack-tox-docs/8c4978b/html/ : SUCCESS in 6m 25s\n- openstack-tox-cover http://logs.openstack.org/48/662648/1/check/openstack-tox-cover/8597a1b/cover/ : SUCCESS in 10m 35s (non-voting)\n- mistral-devstack http://logs.openstack.org/48/662648/1/check/mistral-devstack/5fdc29f/ : SUCCESS in 46m 20s\n- mistral-devstack-non-apache http://logs.openstack.org/48/662648/1/check/mistral-devstack-non-apache/031d211/ : SUCCESS in 39m 30s\n- mistral-devstack-kombu http://logs.openstack.org/48/662648/1/check/mistral-devstack-kombu/f10dccd/ : SUCCESS in 43m 49s\n- mistral-tox-unit-mysql http://logs.openstack.org/48/662648/1/check/mistral-tox-unit-mysql/69f5378/ : SUCCESS in 28m 07s\n- mistral-tox-unit-postgresql http://logs.openstack.org/48/662648/1/check/mistral-tox-unit-postgresql/49bd974/ : SUCCESS in 33m 36s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/48/662648/1/check/tripleo-ci-centos-7-scenario003-standalone/c3dc275/ : SUCCESS in 1h 22m 59s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/48/662648/1/check/tripleo-ci-centos-7-containers-multinode/91e67f6/ : SUCCESS in 2h 08m 29s\n- mistral-rally-task http://logs.openstack.org/48/662648/1/check/mistral-rally-task/19c4f80/ : SUCCESS in 58m 45s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/48/662648/1/check/openstack-tox-lower-constraints/dc74894/ : SUCCESS in 12m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e6978f622792fc7c9a3db7a89e642b3560b0fbb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 21:02:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc1bd830b3e1a2eabee83eca68b7edaa5f405ceb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 22:43:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/48/662648/1/gate/openstack-tox-pep8/470c7ad/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/48/662648/1/gate/openstack-tox-py27/5ae2440/ : SUCCESS in 8m 25s\n- openstack-tox-py36 http://logs.openstack.org/48/662648/1/gate/openstack-tox-py36/967f856/ : SUCCESS in 8m 23s\n- openstack-tox-py37 http://logs.openstack.org/48/662648/1/gate/openstack-tox-py37/44a06c5/ : SUCCESS in 8m 37s\n- openstack-tox-docs http://logs.openstack.org/48/662648/1/gate/openstack-tox-docs/6568e0f/html/ : SUCCESS in 5m 11s\n- mistral-devstack http://logs.openstack.org/48/662648/1/gate/mistral-devstack/4f268f8/ : SUCCESS in 41m 35s\n- mistral-devstack-non-apache http://logs.openstack.org/48/662648/1/gate/mistral-devstack-non-apache/1d7ca71/ : SUCCESS in 34m 38s\n- mistral-tox-unit-mysql http://logs.openstack.org/48/662648/1/gate/mistral-tox-unit-mysql/c22f9c9/ : FAILURE in 36m 25s\n- mistral-tox-unit-postgresql http://logs.openstack.org/48/662648/1/gate/mistral-tox-unit-postgresql/59d49ca/ : SUCCESS in 27m 42s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/48/662648/1/gate/tripleo-ci-centos-7-scenario003-standalone/548382b/ : SUCCESS in 1h 14m 33s\n- mistral-devstack-kombu http://logs.openstack.org/48/662648/1/gate/mistral-devstack-kombu/432ae18/ : SUCCESS in 41m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/662648/1/gate/openstack-tox-lower-constraints/9e685bb/ : SUCCESS in 10m 33s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/48/662648/1/gate/tripleo-ci-centos-7-undercloud-containers/eb83f68/ : SUCCESS in 1h 25m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9c9b4dc3b7fc60a345d1b75a201466aa4f03d3e4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-04 05:00:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f325fc3a8f4d2c5e35203cee083ac5e7702bf3fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 07:12:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/48/662648/1/check/openstack-tox-pep8/7f220a6/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/48/662648/1/check/openstack-tox-py27/4620d5f/ : SUCCESS in 8m 58s\n- openstack-tox-py36 http://logs.openstack.org/48/662648/1/check/openstack-tox-py36/6ee770d/ : SUCCESS in 8m 36s\n- openstack-tox-py37 http://logs.openstack.org/48/662648/1/check/openstack-tox-py37/d157247/ : SUCCESS in 8m 54s\n- openstack-tox-docs http://logs.openstack.org/48/662648/1/check/openstack-tox-docs/2768cc6/html/ : SUCCESS in 4m 31s\n- openstack-tox-cover http://logs.openstack.org/48/662648/1/check/openstack-tox-cover/1ffcdd0/cover/ : SUCCESS in 9m 15s (non-voting)\n- mistral-devstack http://logs.openstack.org/48/662648/1/check/mistral-devstack/7aab57d/ : SUCCESS in 42m 58s\n- mistral-devstack-non-apache http://logs.openstack.org/48/662648/1/check/mistral-devstack-non-apache/aba65ad/ : SUCCESS in 43m 41s\n- mistral-devstack-kombu http://logs.openstack.org/48/662648/1/check/mistral-devstack-kombu/b95c669/ : SUCCESS in 42m 54s\n- mistral-tox-unit-mysql http://logs.openstack.org/48/662648/1/check/mistral-tox-unit-mysql/c9e8526/ : SUCCESS in 27m 52s\n- mistral-tox-unit-postgresql http://logs.openstack.org/48/662648/1/check/mistral-tox-unit-postgresql/c6a7856/ : SUCCESS in 28m 24s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/48/662648/1/check/tripleo-ci-centos-7-scenario003-standalone/a37c7d7/ : SUCCESS in 1h 13m 38s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/48/662648/1/check/tripleo-ci-centos-7-containers-multinode/538eb16/ : SUCCESS in 2h 06m 30s\n- mistral-rally-task http://logs.openstack.org/48/662648/1/check/mistral-rally-task/7f0f95a/ : SUCCESS in 1h 05m 33s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/48/662648/1/check/openstack-tox-lower-constraints/8bbd2d4/ : SUCCESS in 12m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a16c756948a04072635ea810f77a5c3a290dee99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 07:12:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"38784d67f0696ee3d29c147baa4bff6b7c91b726","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 08:44:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/48/662648/1/gate/openstack-tox-pep8/46c502d/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/48/662648/1/gate/openstack-tox-py27/08f0412/ : SUCCESS in 9m 20s\n- openstack-tox-py36 http://logs.openstack.org/48/662648/1/gate/openstack-tox-py36/41638be/ : SUCCESS in 9m 08s\n- openstack-tox-py37 http://logs.openstack.org/48/662648/1/gate/openstack-tox-py37/2add5ac/ : SUCCESS in 10m 08s\n- openstack-tox-docs http://logs.openstack.org/48/662648/1/gate/openstack-tox-docs/39a1c4f/html/ : SUCCESS in 5m 03s\n- mistral-devstack http://logs.openstack.org/48/662648/1/gate/mistral-devstack/a89aca9/ : SUCCESS in 48m 19s\n- mistral-devstack-non-apache http://logs.openstack.org/48/662648/1/gate/mistral-devstack-non-apache/733d0dc/ : SUCCESS in 44m 10s\n- mistral-tox-unit-mysql http://logs.openstack.org/48/662648/1/gate/mistral-tox-unit-mysql/5ff8a37/ : RETRY_LIMIT in 1m 38s\n- mistral-tox-unit-postgresql http://logs.openstack.org/48/662648/1/gate/mistral-tox-unit-postgresql/0560518/ : SUCCESS in 26m 31s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/48/662648/1/gate/tripleo-ci-centos-7-scenario003-standalone/1ca7684/ : SUCCESS in 1h 16m 44s\n- mistral-devstack-kombu http://logs.openstack.org/48/662648/1/gate/mistral-devstack-kombu/a0d19ab/ : SUCCESS in 47m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/662648/1/gate/openstack-tox-lower-constraints/d06a710/ : SUCCESS in 10m 00s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/48/662648/1/gate/tripleo-ci-centos-7-undercloud-containers/846b905/ : SUCCESS in 1h 25m 53s","accounts_in_message":[],"_revision_number":1},{"id":"8a6919d58f3226590bfdc11ab9f82732f3c3cbb2","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-04 08:46:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d56e6ec6b81da42f53bc09f510f270959385fcba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 10:58:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/48/662648/1/check/openstack-tox-pep8/876b86d/ : SUCCESS in 6m 03s\n- openstack-tox-py27 http://logs.openstack.org/48/662648/1/check/openstack-tox-py27/7c07cc9/ : SUCCESS in 8m 49s\n- openstack-tox-py36 http://logs.openstack.org/48/662648/1/check/openstack-tox-py36/8ba1ebb/ : SUCCESS in 8m 56s\n- openstack-tox-py37 http://logs.openstack.org/48/662648/1/check/openstack-tox-py37/aeb9d24/ : SUCCESS in 8m 59s\n- openstack-tox-docs http://logs.openstack.org/48/662648/1/check/openstack-tox-docs/e77e0af/html/ : SUCCESS in 4m 58s\n- openstack-tox-cover http://logs.openstack.org/48/662648/1/check/openstack-tox-cover/d2ca9c0/cover/ : SUCCESS in 8m 39s (non-voting)\n- mistral-devstack http://logs.openstack.org/48/662648/1/check/mistral-devstack/397bde1/ : SUCCESS in 42m 14s\n- mistral-devstack-non-apache http://logs.openstack.org/48/662648/1/check/mistral-devstack-non-apache/37edeb7/ : RETRY_LIMIT in 3m 29s\n- mistral-devstack-kombu http://logs.openstack.org/48/662648/1/check/mistral-devstack-kombu/a6dc9a1/ : SUCCESS in 39m 37s\n- mistral-tox-unit-mysql http://logs.openstack.org/48/662648/1/check/mistral-tox-unit-mysql/737ac21/ : SUCCESS in 26m 35s\n- mistral-tox-unit-postgresql http://logs.openstack.org/48/662648/1/check/mistral-tox-unit-postgresql/4f75cfc/ : SUCCESS in 27m 00s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/48/662648/1/check/tripleo-ci-centos-7-scenario003-standalone/a57a204/ : SUCCESS in 1h 34m 28s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/48/662648/1/check/tripleo-ci-centos-7-containers-multinode/b6f40cd/ : SUCCESS in 2h 05m 51s\n- mistral-rally-task http://logs.openstack.org/48/662648/1/check/mistral-rally-task/032121a/ : SUCCESS in 1h 03m 19s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/48/662648/1/check/openstack-tox-lower-constraints/9ab4c18/ : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":1},{"id":"527b5f5e3a08a8b0a6c0c60f09004e2379ad3890","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-04 11:09:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ce1d30ae7ee1b10d0afa468a495c1944443195e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 13:33:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/48/662648/1/check/openstack-tox-pep8/45b550f/ : SUCCESS in 6m 29s\n- openstack-tox-py27 http://logs.openstack.org/48/662648/1/check/openstack-tox-py27/b50dc0b/ : SUCCESS in 11m 26s\n- openstack-tox-py36 http://logs.openstack.org/48/662648/1/check/openstack-tox-py36/9560a9b/ : SUCCESS in 8m 41s\n- openstack-tox-py37 http://logs.openstack.org/48/662648/1/check/openstack-tox-py37/c4921b2/ : SUCCESS in 12m 29s\n- openstack-tox-docs http://logs.openstack.org/48/662648/1/check/openstack-tox-docs/bfc653b/html/ : SUCCESS in 8m 06s\n- openstack-tox-cover http://logs.openstack.org/48/662648/1/check/openstack-tox-cover/3f7f4e4/cover/ : SUCCESS in 8m 46s (non-voting)\n- mistral-devstack http://logs.openstack.org/48/662648/1/check/mistral-devstack/c31a29d/ : SUCCESS in 44m 29s\n- mistral-devstack-non-apache http://logs.openstack.org/48/662648/1/check/mistral-devstack-non-apache/9dfa24d/ : SUCCESS in 44m 56s\n- mistral-devstack-kombu http://logs.openstack.org/48/662648/1/check/mistral-devstack-kombu/f38c4c1/ : SUCCESS in 40m 32s\n- mistral-tox-unit-mysql http://logs.openstack.org/48/662648/1/check/mistral-tox-unit-mysql/627e3c0/ : SUCCESS in 27m 21s\n- mistral-tox-unit-postgresql http://logs.openstack.org/48/662648/1/check/mistral-tox-unit-postgresql/d6a6681/ : SUCCESS in 30m 54s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/48/662648/1/check/tripleo-ci-centos-7-scenario003-standalone/e43c138/ : SUCCESS in 1h 16m 51s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/48/662648/1/check/tripleo-ci-centos-7-containers-multinode/d4c94b8/ : SUCCESS in 2h 08m 29s\n- mistral-rally-task http://logs.openstack.org/48/662648/1/check/mistral-rally-task/7052d31/ : RETRY_LIMIT in 3m 22s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/48/662648/1/check/openstack-tox-lower-constraints/513d77d/ : SUCCESS in 10m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1559e3a6d971f15f2d29388f21071033f1e4c029","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 13:33:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b0262a7afeee37b58e480e896201058ae084210","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 15:05:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/48/662648/1/gate/openstack-tox-pep8/138f821/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/48/662648/1/gate/openstack-tox-py27/b19f574/ : SUCCESS in 10m 48s\n- openstack-tox-py36 http://logs.openstack.org/48/662648/1/gate/openstack-tox-py36/fa1612c/ : SUCCESS in 8m 42s\n- openstack-tox-py37 http://logs.openstack.org/48/662648/1/gate/openstack-tox-py37/b32be85/ : SUCCESS in 9m 10s\n- openstack-tox-docs http://logs.openstack.org/48/662648/1/gate/openstack-tox-docs/5a53928/html/ : SUCCESS in 5m 03s\n- mistral-devstack http://logs.openstack.org/48/662648/1/gate/mistral-devstack/b763081/ : SUCCESS in 44m 37s\n- mistral-devstack-non-apache http://logs.openstack.org/48/662648/1/gate/mistral-devstack-non-apache/525ce58/ : SUCCESS in 48m 39s\n- mistral-tox-unit-mysql http://logs.openstack.org/48/662648/1/gate/mistral-tox-unit-mysql/0f91080/ : SUCCESS in 30m 12s\n- mistral-tox-unit-postgresql http://logs.openstack.org/48/662648/1/gate/mistral-tox-unit-postgresql/f91220e/ : SUCCESS in 30m 11s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/48/662648/1/gate/tripleo-ci-centos-7-scenario003-standalone/4ee6e65/ : SUCCESS in 1h 17m 30s\n- mistral-devstack-kombu http://logs.openstack.org/48/662648/1/gate/mistral-devstack-kombu/9ce40af/ : SUCCESS in 41m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/662648/1/gate/openstack-tox-lower-constraints/96d2b4f/ : SUCCESS in 9m 45s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/48/662648/1/gate/tripleo-ci-centos-7-undercloud-containers/e551d6d/ : SUCCESS in 1h 24m 56s","accounts_in_message":[],"_revision_number":1},{"id":"87f6d18a43f52d9140a184d9e5bed858bb3bfa38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 15:05:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92528a82efea6a3501be7c10a7a397022955c8ae","revisions":{"92528a82efea6a3501be7c10a7a397022955c8ae":{"kind":"REWORK","_number":1,"created":"2019-06-02 20:41:21.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/48/662648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/48/662648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/48/662648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/48/662648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/48/662648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/48/662648/1"}}},"commit":{"parents":[{"commit":"e25da33607667767d0e34b576f1dd86d0638b854","subject":"Merge \"Fix adhoc action lookup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e25da33607667767d0e34b576f1dd86d0638b854"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-02 20:38:02.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-02 20:38:02.000000000","tz":120},"subject":"Move action caching to db layer","message":"Move action caching to db layer\n\nNow action caching is used only in engine.actions, but actually\nthere are a lot of place where this cache can be useful.\n\nThis commit moves the caching code to the db layer, that allows it to\nbe used in all other places, where load_action_definition is called.\n\nChange-Id: I8c470cbb424e63c16e599758bb93192985666353\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/92528a82efea6a3501be7c10a7a397022955c8ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/92528a82efea6a3501be7c10a7a397022955c8ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
