)]}'
{"id":"openstack%2Fmistral~662650","triplet_id":"openstack%2Fmistral~master~If1adb4ea408248e766bff449637de2b1b1356738","project":"openstack/mistral","branch":"master","topic":"remove_find_task_execution_by_name","hashtags":[],"change_id":"If1adb4ea408248e766bff449637de2b1b1356738","subject":"Prepare cache for _is_upstream_task_execution","status":"MERGED","created":"2019-06-02 21:21:27.000000000","updated":"2019-06-04 13:27:44.000000000","submitted":"2019-06-04 13:27:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662650-1559654864153-b9ca8759","meta_rev_id":"6525f52430d8e8e7126e2aa62dd0b351086e1f8d","_number":662650,"virtual_id_number":662650,"owner":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-06-04 13:27:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-03 18:43:55.000000000","_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-06-03 10:19:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-06-03 04:30:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"date":"2019-06-03 06:21:52.000000000","_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"date":"2019-06-03 06:21:52.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-06-03 10:19:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2019-06-03 06:21:52.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"date":"2019-06-03 06:21:52.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 04:30:05.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-06-03 06:21:52.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-06-03 06:21:52.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-06-03 06:21:52.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-06-03 06:21:52.000000000","updated_by":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"reviewer":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"state":"REVIEWER"},{"updated":"2019-06-03 10:19:02.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-06-04 13:27:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0a16386db71552e8bcb9d0e2d25167841fec9806","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-02 21:21:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3487b3f298b1bebc72d981bc6a6faeef0a150d6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 23:22:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/50/662650/1/check/openstack-tox-pep8/e009db3/ : SUCCESS in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/50/662650/1/check/openstack-tox-py27/05a6be5/ : SUCCESS in 8m 49s\n- openstack-tox-py36 http://logs.openstack.org/50/662650/1/check/openstack-tox-py36/7c4259d/ : SUCCESS in 8m 13s\n- openstack-tox-py37 http://logs.openstack.org/50/662650/1/check/openstack-tox-py37/82d73f2/ : SUCCESS in 8m 16s\n- openstack-tox-docs http://logs.openstack.org/50/662650/1/check/openstack-tox-docs/019694e/html/ : SUCCESS in 4m 39s\n- openstack-tox-cover http://logs.openstack.org/50/662650/1/check/openstack-tox-cover/cc1b5b5/cover/ : SUCCESS in 8m 43s (non-voting)\n- mistral-devstack http://logs.openstack.org/50/662650/1/check/mistral-devstack/0afce61/ : SUCCESS in 43m 23s\n- mistral-devstack-non-apache http://logs.openstack.org/50/662650/1/check/mistral-devstack-non-apache/0f5f7f7/ : SUCCESS in 43m 35s\n- mistral-devstack-kombu http://logs.openstack.org/50/662650/1/check/mistral-devstack-kombu/dc52041/ : SUCCESS in 41m 48s\n- mistral-tox-unit-mysql http://logs.openstack.org/50/662650/1/check/mistral-tox-unit-mysql/a2152e3/ : SUCCESS in 26m 52s\n- mistral-tox-unit-postgresql http://logs.openstack.org/50/662650/1/check/mistral-tox-unit-postgresql/5871c4b/ : SUCCESS in 25m 59s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/50/662650/1/check/tripleo-ci-centos-7-scenario003-standalone/a209052/ : SUCCESS in 1h 10m 15s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/662650/1/check/tripleo-ci-centos-7-containers-multinode/1089e84/ : SUCCESS in 1h 55m 20s\n- mistral-rally-task http://logs.openstack.org/50/662650/1/check/mistral-rally-task/02e8a05/ : FAILURE in 1h 18m 44s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/50/662650/1/check/openstack-tox-lower-constraints/f820f30/ : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":1},{"id":"dbbe61e505f6a5d6205137f3a217a3df75b0b1fd","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-03 04:30:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"57d5f7592f2fb1c9edd8cce041bbcea781d22d7d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-03 06:20:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"552ca3876682fea3a49f484d04fc968376290750","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 08:54:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/50/662650/1/check/openstack-tox-pep8/2b597c5/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/50/662650/1/check/openstack-tox-py27/fc7c1e6/ : SUCCESS in 8m 10s\n- openstack-tox-py36 http://logs.openstack.org/50/662650/1/check/openstack-tox-py36/cf44b73/ : SUCCESS in 8m 53s\n- openstack-tox-py37 http://logs.openstack.org/50/662650/1/check/openstack-tox-py37/75c886c/ : SUCCESS in 8m 17s\n- openstack-tox-docs http://logs.openstack.org/50/662650/1/check/openstack-tox-docs/2dd7b0b/html/ : SUCCESS in 4m 56s\n- openstack-tox-cover http://logs.openstack.org/50/662650/1/check/openstack-tox-cover/e9e5843/cover/ : SUCCESS in 13m 54s (non-voting)\n- mistral-devstack http://logs.openstack.org/50/662650/1/check/mistral-devstack/01c83c8/ : SUCCESS in 43m 32s\n- mistral-devstack-non-apache http://logs.openstack.org/50/662650/1/check/mistral-devstack-non-apache/1dece6e/ : SUCCESS in 36m 57s\n- mistral-devstack-kombu http://logs.openstack.org/50/662650/1/check/mistral-devstack-kombu/3821d2c/ : POST_FAILURE in 54m 15s\n- mistral-tox-unit-mysql http://logs.openstack.org/50/662650/1/check/mistral-tox-unit-mysql/910c299/ : SUCCESS in 27m 08s\n- mistral-tox-unit-postgresql http://logs.openstack.org/50/662650/1/check/mistral-tox-unit-postgresql/9c44c34/ : SUCCESS in 25m 56s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/50/662650/1/check/tripleo-ci-centos-7-scenario003-standalone/cd6c487/ : SUCCESS in 1h 14m 03s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/662650/1/check/tripleo-ci-centos-7-containers-multinode/79e821c/ : SUCCESS in 1h 59m 57s\n- mistral-rally-task http://logs.openstack.org/50/662650/1/check/mistral-rally-task/d2767dd/ : SUCCESS in 1h 09m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/50/662650/1/check/openstack-tox-lower-constraints/9b499c3/ : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a7fc7cda077e442ddde01073e5fd8d73fc7fa7f4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-03 08:54:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0b5562c80221460770146100fc74da6ccba967d2","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-06-03 10:19:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8192e6843ef6b83176172d2331ef72b35ded9d63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 10:56:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/50/662650/1/check/openstack-tox-pep8/1f007d1/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/50/662650/1/check/openstack-tox-py27/18c3430/ : SUCCESS in 9m 15s\n- openstack-tox-py36 http://logs.openstack.org/50/662650/1/check/openstack-tox-py36/4a0e819/ : SUCCESS in 8m 51s\n- openstack-tox-py37 http://logs.openstack.org/50/662650/1/check/openstack-tox-py37/e3ed165/ : SUCCESS in 9m 34s\n- openstack-tox-docs http://logs.openstack.org/50/662650/1/check/openstack-tox-docs/85bdc95/html/ : SUCCESS in 5m 04s\n- openstack-tox-cover http://logs.openstack.org/50/662650/1/check/openstack-tox-cover/8eaeb8c/cover/ : SUCCESS in 9m 35s (non-voting)\n- mistral-devstack http://logs.openstack.org/50/662650/1/check/mistral-devstack/9b2adcb/ : SUCCESS in 43m 52s\n- mistral-devstack-non-apache http://logs.openstack.org/50/662650/1/check/mistral-devstack-non-apache/0d94db7/ : SUCCESS in 43m 01s\n- mistral-devstack-kombu http://logs.openstack.org/50/662650/1/check/mistral-devstack-kombu/dbe9b94/ : SUCCESS in 43m 20s\n- mistral-tox-unit-mysql http://logs.openstack.org/50/662650/1/check/mistral-tox-unit-mysql/d7f7f1c/ : SUCCESS in 27m 55s\n- mistral-tox-unit-postgresql http://logs.openstack.org/50/662650/1/check/mistral-tox-unit-postgresql/1a899fd/ : SUCCESS in 26m 17s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/50/662650/1/check/tripleo-ci-centos-7-scenario003-standalone/5af156b/ : SUCCESS in 1h 10m 37s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/662650/1/check/tripleo-ci-centos-7-containers-multinode/3e5502d/ : SUCCESS in 1h 54m 11s\n- mistral-rally-task http://logs.openstack.org/50/662650/1/check/mistral-rally-task/d5ba948/ : SUCCESS in 1h 10m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/50/662650/1/check/openstack-tox-lower-constraints/98fe0ca/ : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c72f8b81c58d1b8db4f0819d4ab98a1dc40914c4","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-03 18:43:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"570134bd5510bfc3ffc33a677265984c4c1fae88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 21:01:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/50/662650/1/check/openstack-tox-pep8/0249f31/ : SUCCESS in 11m 27s\n- openstack-tox-py27 http://logs.openstack.org/50/662650/1/check/openstack-tox-py27/435a878/ : SUCCESS in 9m 17s\n- openstack-tox-py36 http://logs.openstack.org/50/662650/1/check/openstack-tox-py36/d0d877a/ : SUCCESS in 8m 26s\n- openstack-tox-py37 http://logs.openstack.org/50/662650/1/check/openstack-tox-py37/ecc0d27/ : SUCCESS in 8m 43s\n- openstack-tox-docs http://logs.openstack.org/50/662650/1/check/openstack-tox-docs/a3cb64c/html/ : SUCCESS in 6m 15s\n- openstack-tox-cover http://logs.openstack.org/50/662650/1/check/openstack-tox-cover/7395116/cover/ : SUCCESS in 8m 39s (non-voting)\n- mistral-devstack http://logs.openstack.org/50/662650/1/check/mistral-devstack/afb3d39/ : SUCCESS in 39m 54s\n- mistral-devstack-non-apache http://logs.openstack.org/50/662650/1/check/mistral-devstack-non-apache/b9ec211/ : SUCCESS in 42m 45s\n- mistral-devstack-kombu http://logs.openstack.org/50/662650/1/check/mistral-devstack-kombu/e2f21ab/ : SUCCESS in 44m 35s\n- mistral-tox-unit-mysql http://logs.openstack.org/50/662650/1/check/mistral-tox-unit-mysql/f4df458/ : SUCCESS in 31m 16s\n- mistral-tox-unit-postgresql http://logs.openstack.org/50/662650/1/check/mistral-tox-unit-postgresql/d28be89/ : SUCCESS in 30m 42s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/50/662650/1/check/tripleo-ci-centos-7-scenario003-standalone/c55c836/ : SUCCESS in 1h 22m 00s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/662650/1/check/tripleo-ci-centos-7-containers-multinode/2686a73/ : SUCCESS in 2h 07m 21s\n- mistral-rally-task http://logs.openstack.org/50/662650/1/check/mistral-rally-task/da7643e/ : SUCCESS in 58m 17s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/50/662650/1/check/openstack-tox-lower-constraints/5d5193c/ : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":1},{"id":"92d7b6be9b245c56cf65fa31c8d77c7d4f9b3e8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 11:10:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e1caf1b1bf93e774d5eb2cfdd2426a4033e97df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 13:27:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/50/662650/1/gate/openstack-tox-pep8/207423e/ : SUCCESS in 5m 25s\n- openstack-tox-py27 http://logs.openstack.org/50/662650/1/gate/openstack-tox-py27/58ffd91/ : SUCCESS in 9m 19s\n- openstack-tox-py36 http://logs.openstack.org/50/662650/1/gate/openstack-tox-py36/b959618/ : SUCCESS in 8m 15s\n- openstack-tox-py37 http://logs.openstack.org/50/662650/1/gate/openstack-tox-py37/8f8aae5/ : SUCCESS in 10m 08s\n- openstack-tox-docs http://logs.openstack.org/50/662650/1/gate/openstack-tox-docs/4cfcd00/html/ : SUCCESS in 4m 58s\n- mistral-devstack http://logs.openstack.org/50/662650/1/gate/mistral-devstack/e55613a/ : SUCCESS in 44m 08s\n- mistral-devstack-non-apache http://logs.openstack.org/50/662650/1/gate/mistral-devstack-non-apache/d1f5d42/ : SUCCESS in 42m 44s\n- mistral-tox-unit-mysql http://logs.openstack.org/50/662650/1/gate/mistral-tox-unit-mysql/6ffead3/ : SUCCESS in 27m 58s\n- mistral-tox-unit-postgresql http://logs.openstack.org/50/662650/1/gate/mistral-tox-unit-postgresql/049a059/ : SUCCESS in 26m 01s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/50/662650/1/gate/tripleo-ci-centos-7-scenario003-standalone/fe623d3/ : SUCCESS in 1h 16m 09s\n- mistral-devstack-kombu http://logs.openstack.org/50/662650/1/gate/mistral-devstack-kombu/fa61b0e/ : SUCCESS in 43m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/662650/1/gate/openstack-tox-lower-constraints/156d5cd/ : SUCCESS in 10m 15s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/50/662650/1/gate/tripleo-ci-centos-7-undercloud-containers/82b057a/ : SUCCESS in 1h 35m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e836f61557971eda2812fe6c55f39c96d258aa5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 13:27:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58b714eb16ce389fb4faa7598cc0c74ad351fc30","revisions":{"58b714eb16ce389fb4faa7598cc0c74ad351fc30":{"kind":"REWORK","_number":1,"created":"2019-06-02 21:21:27.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/50/662650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/50/662650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/50/662650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/50/662650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/50/662650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/50/662650/1"}}},"commit":{"parents":[{"commit":"cd19e4869781bf4ff8df7c47c39791e824e835f6","subject":"Remove _find_task_execution_by_name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cd19e4869781bf4ff8df7c47c39791e824e835f6"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-02 21:18:07.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-02 21:18:07.000000000","tz":120},"subject":"Prepare cache for _is_upstream_task_execution","message":"Prepare cache for _is_upstream_task_execution\n\nThis trick is similar to what we do in _get_join_logical_state.\nInstead of bypassing the entire workflow graph and downloading\ntasks executions one by one, we prepare a lightweight cache with\na single database query and then analyze it on the spot.\n\nChange-Id: If1adb4ea408248e766bff449637de2b1b1356738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/58b714eb16ce389fb4faa7598cc0c74ad351fc30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/58b714eb16ce389fb4faa7598cc0c74ad351fc30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
