)]}'
{"id":"openstack%2Fmistral~663229","triplet_id":"openstack%2Fmistral~stable%2Fstein~Ic2e0ad75b666dd4c258764018dd68d0ef0b032a8","project":"openstack/mistral","branch":"stable/stein","topic":"fix_action_execution_controller_test-stable/stein","hashtags":[],"change_id":"Ic2e0ad75b666dd4c258764018dd68d0ef0b032a8","subject":"Fix an action execution controller test","status":"MERGED","created":"2019-06-05 04:31:33.000000000","updated":"2019-06-05 07:31:06.000000000","submitted":"2019-06-05 07:31:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663229-1559719866204-717f2137","meta_rev_id":"776524e3e0f63fdfde6170dc9f9059c490ea1422","_number":663229,"virtual_id_number":663229,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-06-05 07:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-06-05 04:38:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-06-05 04:38:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 07:31:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a2a375e18e0a3b662be49993d40af8bd83a0085","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-05 04:31:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4fa84b07a95cb35f273fd331d059f7b859963d0b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-05 04:31:56.000000000","message":"Patch Set 1:\n\nThe same problem exists in stein so we need to backport the patch.","accounts_in_message":[],"_revision_number":1},{"id":"db603b46f4e291a30b316e1334b722b59afa2f97","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-05 04:38:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"30ec4f5aab71b953bedaa862949c19cc80f0fa22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 06:38:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/29/663229/1/check/openstack-tox-pep8/c0be22f/ : SUCCESS in 5m 11s\n- openstack-tox-py27 http://logs.openstack.org/29/663229/1/check/openstack-tox-py27/0f99c00/ : SUCCESS in 8m 25s\n- openstack-tox-py35 http://logs.openstack.org/29/663229/1/check/openstack-tox-py35/c5e6215/ : SUCCESS in 8m 46s\n- openstack-tox-py36 http://logs.openstack.org/29/663229/1/check/openstack-tox-py36/6490379/ : SUCCESS in 8m 55s\n- openstack-tox-py37 http://logs.openstack.org/29/663229/1/check/openstack-tox-py37/a63d0c8/ : SUCCESS in 9m 37s\n- openstack-tox-docs http://logs.openstack.org/29/663229/1/check/openstack-tox-docs/8196915/html/ : SUCCESS in 4m 44s\n- openstack-tox-cover http://logs.openstack.org/29/663229/1/check/openstack-tox-cover/17efff0/cover/ : SUCCESS in 8m 25s (non-voting)\n- mistral-devstack http://logs.openstack.org/29/663229/1/check/mistral-devstack/eff63ed/ : SUCCESS in 43m 11s\n- mistral-devstack-non-apache http://logs.openstack.org/29/663229/1/check/mistral-devstack-non-apache/881be72/ : SUCCESS in 40m 12s\n- mistral-devstack-kombu http://logs.openstack.org/29/663229/1/check/mistral-devstack-kombu/275c82b/ : SUCCESS in 40m 22s\n- mistral-tox-unit-mysql http://logs.openstack.org/29/663229/1/check/mistral-tox-unit-mysql/2bdcde1/ : SUCCESS in 28m 28s\n- mistral-tox-unit-postgresql http://logs.openstack.org/29/663229/1/check/mistral-tox-unit-postgresql/dc9bb99/ : SUCCESS in 26m 03s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/29/663229/1/check/tripleo-ci-centos-7-scenario003-standalone/15f9aa8/ : SUCCESS in 1h 14m 52s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/29/663229/1/check/tripleo-ci-centos-7-containers-multinode/32b8103/ : SUCCESS in 2h 01m 38s (non-voting)\n- mistral-rally-task http://logs.openstack.org/29/663229/1/check/mistral-rally-task/6a1127d/ : SUCCESS in 1h 02m 39s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/29/663229/1/check/openstack-tox-lower-constraints/89bad12/ : SUCCESS in 9m 55s","accounts_in_message":[],"_revision_number":1},{"id":"bb6d4460c4ec71ec1d8addab711a93e3b9e9ff2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 06:38:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ca07a147473813c6eb24a0ca6e6e900fe5cd827","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 07:31:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/29/663229/1/gate/openstack-tox-pep8/289f986/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/29/663229/1/gate/openstack-tox-py27/8a64c5c/ : SUCCESS in 8m 38s\n- openstack-tox-py35 http://logs.openstack.org/29/663229/1/gate/openstack-tox-py35/17e58f9/ : SUCCESS in 9m 50s\n- openstack-tox-py36 http://logs.openstack.org/29/663229/1/gate/openstack-tox-py36/a784a0a/ : SUCCESS in 9m 06s\n- openstack-tox-py37 http://logs.openstack.org/29/663229/1/gate/openstack-tox-py37/c15d690/ : SUCCESS in 8m 17s\n- openstack-tox-docs http://logs.openstack.org/29/663229/1/gate/openstack-tox-docs/542a2f7/html/ : SUCCESS in 8m 17s\n- mistral-devstack http://logs.openstack.org/29/663229/1/gate/mistral-devstack/65725ee/ : SUCCESS in 40m 46s\n- mistral-devstack-non-apache http://logs.openstack.org/29/663229/1/gate/mistral-devstack-non-apache/f18af8b/ : SUCCESS in 45m 39s\n- mistral-tox-unit-mysql http://logs.openstack.org/29/663229/1/gate/mistral-tox-unit-mysql/cbcb316/ : SUCCESS in 27m 45s\n- mistral-tox-unit-postgresql http://logs.openstack.org/29/663229/1/gate/mistral-tox-unit-postgresql/4a98c55/ : SUCCESS in 26m 14s\n- mistral-devstack-kombu http://logs.openstack.org/29/663229/1/gate/mistral-devstack-kombu/be83229/ : SUCCESS in 45m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/663229/1/gate/openstack-tox-lower-constraints/f085318/ : SUCCESS in 10m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a0d47a08b77d9afd9a7418af08e66d07f559189c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 07:31:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa504e885532f8a704e88d252003bb51c09798cd","revisions":{"fa504e885532f8a704e88d252003bb51c09798cd":{"kind":"REWORK","_number":1,"created":"2019-06-05 04:31:33.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/29/663229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/29/663229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/29/663229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/29/663229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/29/663229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/29/663229/1"}}},"commit":{"parents":[{"commit":"448f30e10128f03546c78f27748dc4c1601e9877","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/448f30e10128f03546c78f27748dc4c1601e9877"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-04-17 11:16:29.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-06-05 04:31:33.000000000","tz":0},"subject":"Fix an action execution controller test","message":"Fix an action execution controller test\n\n* test_get_all_with_and_without_output failed sometimes failed on\n  Python 3.7 with a weird error coming from simplejson library:\n  \"AttributeError: \u0027MagicMock\u0027 object has no attribute \u0027items\u0027\"\n  It seems like that the mock object used in the test got broken\n  magically sometimes after the first use. So splitting the test\n  into two solved the issue.\n\nChange-Id: Ic2e0ad75b666dd4c258764018dd68d0ef0b032a8\n(cherry picked from commit 9a60cd62d2d3ebd7706cdc4dec9f3d6f5ab5c9f8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fa504e885532f8a704e88d252003bb51c09798cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fa504e885532f8a704e88d252003bb51c09798cd"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
