)]}'
{"id":"openstack%2Fmistral~663350","triplet_id":"openstack%2Fmistral~stable%2Frocky~I976fe5723a8bad81f1c6e30618475aa34e3e2050","project":"openstack/mistral","branch":"stable/rocky","topic":"add_release_note-stable/rocky","hashtags":[],"change_id":"I976fe5723a8bad81f1c6e30618475aa34e3e2050","subject":"Add release notes to changes to wflanguage execution object","status":"MERGED","created":"2019-06-05 15:00:00.000000000","updated":"2019-06-06 10:09:46.000000000","submitted":"2019-06-06 10:09:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663350-1559815786914-20b3d2df","meta_rev_id":"e65afc7dcf5ead04b3279a8f5d0651794203a377","_number":663350,"virtual_id_number":663350,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-06-06 10:09:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-06-06 04:53:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-06-05 15:00:00.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-06-06 04:53:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 15:00:00.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-06-06 04:53:28.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-06-06 10:09:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"83de0313c1cea46d5c5cb271b05df58fad78df6b","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-06-05 15:00:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ecd59db35a97edb7dc6e8516eb0fc90296077bb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 15:37:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/50/663350/1/check/openstack-tox-pep8/f09ca5f/ : SUCCESS in 5m 12s\n- build-openstack-sphinx-docs http://logs.openstack.org/50/663350/1/check/build-openstack-sphinx-docs/41853b7/html/ : SUCCESS in 6m 36s\n- build-openstack-releasenotes http://logs.openstack.org/50/663350/1/check/build-openstack-releasenotes/4dbdf39/html/ : SUCCESS in 4m 12s\n- openstack-tox-cover http://logs.openstack.org/50/663350/1/check/openstack-tox-cover/8d3e41f/cover/ : SUCCESS in 8m 25s (non-voting)\n- mistral-tox-unit-mysql http://logs.openstack.org/50/663350/1/check/mistral-tox-unit-mysql/1a0336e/ : SUCCESS in 26m 24s\n- mistral-tox-unit-postgresql http://logs.openstack.org/50/663350/1/check/mistral-tox-unit-postgresql/7e16115/ : SUCCESS in 27m 45s","accounts_in_message":[],"_revision_number":1},{"id":"17fd9fb988f1f3c7e118c22076908feddda71335","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-06 04:53:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2121b2dd1a38e25029c11dc3239a5676edec66db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 04:53:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f163e9911a91a6ab27d3c9040c54c56edecc0b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 05:41:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/50/663350/1/gate/openstack-tox-pep8/f794e2b/ : SUCCESS in 5m 04s\n- build-openstack-sphinx-docs http://logs.openstack.org/50/663350/1/gate/build-openstack-sphinx-docs/ed4fc70/html/ : SUCCESS in 5m 13s\n- build-openstack-releasenotes http://logs.openstack.org/50/663350/1/gate/build-openstack-releasenotes/eea2835/html/ : SUCCESS in 3m 56s\n- mistral-tox-unit-mysql http://logs.openstack.org/50/663350/1/gate/mistral-tox-unit-mysql/a990166/ : FAILURE in 26m 41s\n- mistral-tox-unit-postgresql http://logs.openstack.org/50/663350/1/gate/mistral-tox-unit-postgresql/21d8463/ : SUCCESS in 25m 21s","accounts_in_message":[],"_revision_number":1},{"id":"214ecd6be2c30714a609628d13415ed39b6704a4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-06 08:38:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a12dc339019f4e9b3476dc1c0e45bbf19d55ca6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 09:17:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/50/663350/1/check/openstack-tox-pep8/f88752b/ : SUCCESS in 5m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/50/663350/1/check/build-openstack-sphinx-docs/c1d0ebc/html/ : SUCCESS in 5m 06s\n- build-openstack-releasenotes http://logs.openstack.org/50/663350/1/check/build-openstack-releasenotes/70a829d/html/ : SUCCESS in 4m 07s\n- openstack-tox-cover http://logs.openstack.org/50/663350/1/check/openstack-tox-cover/780824e/cover/ : SUCCESS in 9m 09s (non-voting)\n- mistral-tox-unit-mysql http://logs.openstack.org/50/663350/1/check/mistral-tox-unit-mysql/15a271d/ : SUCCESS in 26m 23s\n- mistral-tox-unit-postgresql http://logs.openstack.org/50/663350/1/check/mistral-tox-unit-postgresql/31fa9e6/ : SUCCESS in 24m 56s","accounts_in_message":[],"_revision_number":1},{"id":"dd4c3d2a84362afcb30e265f1fd2ee1e8e870182","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 09:17:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"842d35f8d32250c9b5dea2e82e1f39af181e028d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 10:09:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/50/663350/1/gate/openstack-tox-pep8/c2adae1/ : SUCCESS in 6m 08s\n- build-openstack-sphinx-docs http://logs.openstack.org/50/663350/1/gate/build-openstack-sphinx-docs/3e6f1be/html/ : SUCCESS in 4m 32s\n- build-openstack-releasenotes http://logs.openstack.org/50/663350/1/gate/build-openstack-releasenotes/f9cd038/html/ : SUCCESS in 3m 54s\n- mistral-tox-unit-mysql http://logs.openstack.org/50/663350/1/gate/mistral-tox-unit-mysql/0305181/ : SUCCESS in 29m 26s\n- mistral-tox-unit-postgresql http://logs.openstack.org/50/663350/1/gate/mistral-tox-unit-postgresql/2c8d592/ : SUCCESS in 25m 11s","accounts_in_message":[],"_revision_number":1},{"id":"190935a6610dae3d801da3be221b9d62436d930b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 10:09:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9e61b80ad9f762d4acb07466b7fa548addcbd5c","revisions":{"c9e61b80ad9f762d4acb07466b7fa548addcbd5c":{"kind":"REWORK","_number":1,"created":"2019-06-05 15:00:00.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/50/663350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/50/663350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/50/663350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/50/663350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/50/663350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/50/663350/1"}}},"commit":{"parents":[{"commit":"84b9ac24e46ecc9129c64c6013e8522ee6dee461","subject":"Merge \"Adding root_execution_id to the jinja executor function\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/84b9ac24e46ecc9129c64c6013e8522ee6dee461"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-03 11:58:43.000000000","tz":120},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-06-05 15:00:00.000000000","tz":0},"subject":"Add release notes to changes to wflanguage execution object","message":"Add release notes to changes to wflanguage execution object\n\nChange-Id: I976fe5723a8bad81f1c6e30618475aa34e3e2050\ndepends-on: Ia6c954e688589f69a7463f1b8e02244d029e8b7a\n(cherry picked from commit 9289f7d28ec861a6dd9fd7b73bf3d84066c74726)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c9e61b80ad9f762d4acb07466b7fa548addcbd5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c9e61b80ad9f762d4acb07466b7fa548addcbd5c"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
