)]}'
{"id":"openstack%2Fmistral~663351","triplet_id":"openstack%2Fmistral~stable%2Fqueens~I976fe5723a8bad81f1c6e30618475aa34e3e2050","project":"openstack/mistral","branch":"stable/queens","topic":"add_release_note-stable/stein-stable/queens","hashtags":[],"change_id":"I976fe5723a8bad81f1c6e30618475aa34e3e2050","subject":"Add release notes to changes to wflanguage execution object","status":"MERGED","created":"2019-06-05 15:00:05.000000000","updated":"2019-06-06 05:41:30.000000000","submitted":"2019-06-06 05:41:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663351-1559799690768-d9d96933","meta_rev_id":"af231ba2d7e5d2a72a1bf7f7dc8cd6cce54ddf30","_number":663351,"virtual_id_number":663351,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-06-06 05:41:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-06-06 04:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-06-05 15:00:05.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-06-06 04:52:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 15:00:05.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-06-06 04:52:53.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-06-06 05:41:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"338131aa3fd7d7791db0da736a299f6daedcd7ed","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-06-05 15:00:05.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"314f090592f2998df925011e1af3c6fa7bbe2a35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 16:44:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/51/663351/1/check/openstack-tox-pep8/c4a4bb3/ : SUCCESS in 5m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/51/663351/1/check/build-openstack-sphinx-docs/9676a8b/html/ : SUCCESS in 5m 15s\n- build-openstack-releasenotes http://logs.openstack.org/51/663351/1/check/build-openstack-releasenotes/5b64bd9/html/ : SUCCESS in 6m 04s\n- mistral-tox-unit-mysql http://logs.openstack.org/51/663351/1/check/mistral-tox-unit-mysql/df75da0/ : SUCCESS in 23m 42s\n- mistral-tox-unit-postgresql http://logs.openstack.org/51/663351/1/check/mistral-tox-unit-postgresql/7a495d7/ : SUCCESS in 28m 04s","accounts_in_message":[],"_revision_number":1},{"id":"2b60efbebbfccf75de1dd9f3c07b170a45070b87","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-06 04:52:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"74fc86f1b91001925531fe189e5cbd2605a59ae3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 04:53:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f8a5ac9b96d3a4d0cf379b5095287f8e896c4720","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 05:41:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/51/663351/1/gate/openstack-tox-pep8/8f0601b/ : SUCCESS in 5m 46s\n- build-openstack-sphinx-docs http://logs.openstack.org/51/663351/1/gate/build-openstack-sphinx-docs/918750a/html/ : SUCCESS in 5m 19s\n- build-openstack-releasenotes http://logs.openstack.org/51/663351/1/gate/build-openstack-releasenotes/3848239/html/ : SUCCESS in 3m 54s\n- mistral-tox-unit-mysql http://logs.openstack.org/51/663351/1/gate/mistral-tox-unit-mysql/8729f4c/ : SUCCESS in 25m 27s\n- mistral-tox-unit-postgresql http://logs.openstack.org/51/663351/1/gate/mistral-tox-unit-postgresql/efd50d1/ : SUCCESS in 26m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5eb71966662864c59e3d4070f7d999d0d3e52e9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 05:41:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da0e2639a8d89e95ac3011d0c21696d37b03301e","revisions":{"da0e2639a8d89e95ac3011d0c21696d37b03301e":{"kind":"REWORK","_number":1,"created":"2019-06-05 15:00:05.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/51/663351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/51/663351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/663351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/663351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/663351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/51/663351/1"}}},"commit":{"parents":[{"commit":"7fa29766cdc3ae8606ada2d9f2bc7be51217adb4","subject":"Merge \"Adding root_execution_id to the jinja executor function\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7fa29766cdc3ae8606ada2d9f2bc7be51217adb4"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-03 11:58:43.000000000","tz":120},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-06-05 15:00:05.000000000","tz":0},"subject":"Add release notes to changes to wflanguage execution object","message":"Add release notes to changes to wflanguage execution object\n\nChange-Id: I976fe5723a8bad81f1c6e30618475aa34e3e2050\ndepends-on: Ia6c954e688589f69a7463f1b8e02244d029e8b7a\n(cherry picked from commit 9289f7d28ec861a6dd9fd7b73bf3d84066c74726)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da0e2639a8d89e95ac3011d0c21696d37b03301e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da0e2639a8d89e95ac3011d0c21696d37b03301e"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
