)]}'
{"id":"openstack%2Fmistral~663455","triplet_id":"openstack%2Fmistral~master~I4cb2d85c193531648540a37caca9ee868ede3a08","project":"openstack/mistral","branch":"master","topic":"update_last_heartbeat","hashtags":[],"change_id":"I4cb2d85c193531648540a37caca9ee868ede3a08","subject":"Rework updating action executions heartbeats","status":"MERGED","created":"2019-06-05 22:56:30.000000000","updated":"2019-06-06 23:03:47.000000000","submitted":"2019-06-06 23:03:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663455-1559862227135-3605573f","meta_rev_id":"e94c2d62772f1f1d08fc9c28121408c442d334b9","_number":663455,"virtual_id_number":663455,"owner":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-06-06 23:03:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-06 18:51:09.000000000","_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-06-06 07:26:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-06-06 07:26:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 07:26:52.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-06-06 23:03:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c9bcb354d53b0f0009556469ee1cee588e8f357","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-05 22:56:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc3294c191f0d6fff31e0ac1104579c74c1835d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 01:16:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/55/663455/1/check/openstack-tox-pep8/d02f318/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/55/663455/1/check/openstack-tox-py27/fdf91e3/ : SUCCESS in 10m 04s\n- openstack-tox-py36 http://logs.openstack.org/55/663455/1/check/openstack-tox-py36/e4f2ac5/ : SUCCESS in 8m 32s\n- openstack-tox-py37 http://logs.openstack.org/55/663455/1/check/openstack-tox-py37/954406d/ : SUCCESS in 9m 38s\n- openstack-tox-docs http://logs.openstack.org/55/663455/1/check/openstack-tox-docs/9e8dcf9/html/ : SUCCESS in 5m 08s\n- openstack-tox-cover http://logs.openstack.org/55/663455/1/check/openstack-tox-cover/3840ecd/cover/ : SUCCESS in 9m 58s (non-voting)\n- mistral-devstack http://logs.openstack.org/55/663455/1/check/mistral-devstack/b72ac6b/ : SUCCESS in 40m 34s\n- mistral-devstack-non-apache http://logs.openstack.org/55/663455/1/check/mistral-devstack-non-apache/b7f6d9a/ : SUCCESS in 46m 28s\n- mistral-devstack-kombu http://logs.openstack.org/55/663455/1/check/mistral-devstack-kombu/b5a0dda/ : SUCCESS in 44m 02s\n- mistral-tox-unit-mysql http://logs.openstack.org/55/663455/1/check/mistral-tox-unit-mysql/bed0bc1/ : SUCCESS in 27m 58s\n- mistral-tox-unit-postgresql http://logs.openstack.org/55/663455/1/check/mistral-tox-unit-postgresql/f613489/ : SUCCESS in 25m 59s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/55/663455/1/check/tripleo-ci-centos-7-scenario003-standalone/c5c31c3/ : SUCCESS in 1h 16m 10s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/55/663455/1/check/tripleo-ci-centos-7-containers-multinode/bb045d6/ : SUCCESS in 1h 52m 07s\n- mistral-rally-task http://logs.openstack.org/55/663455/1/check/mistral-rally-task/6c7b7e3/ : SUCCESS in 57m 54s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/55/663455/1/check/openstack-tox-lower-constraints/655d659/ : SUCCESS in 12m 47s","accounts_in_message":[],"_revision_number":1},{"id":"46a07c649df10715599e21fd46e6f44209fcff6c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-06 05:01:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nOverall good, minor comments.","accounts_in_message":[],"_revision_number":1},{"id":"31f831ff843b16d99f4d1ff6019d12d57a91230f","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-06 07:04:33.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a6f98de09c95da6e0565e526516342c44d8286ad","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-06 07:26:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"385030e5af9516a729ceb4fed9396b0a0677aab1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 09:18:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/55/663455/2/check/openstack-tox-pep8/e4ff5e1/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/55/663455/2/check/openstack-tox-py27/8a80ae5/ : SUCCESS in 11m 15s\n- openstack-tox-py36 http://logs.openstack.org/55/663455/2/check/openstack-tox-py36/8d6fa43/ : SUCCESS in 8m 48s\n- openstack-tox-py37 http://logs.openstack.org/55/663455/2/check/openstack-tox-py37/0a11f7a/ : SUCCESS in 13m 17s\n- openstack-tox-docs http://logs.openstack.org/55/663455/2/check/openstack-tox-docs/5e42a06/html/ : SUCCESS in 5m 20s\n- openstack-tox-cover http://logs.openstack.org/55/663455/2/check/openstack-tox-cover/068589a/cover/ : SUCCESS in 8m 48s (non-voting)\n- mistral-devstack http://logs.openstack.org/55/663455/2/check/mistral-devstack/a3065e8/ : SUCCESS in 45m 07s\n- mistral-devstack-non-apache http://logs.openstack.org/55/663455/2/check/mistral-devstack-non-apache/c518122/ : SUCCESS in 44m 55s\n- mistral-devstack-kombu http://logs.openstack.org/55/663455/2/check/mistral-devstack-kombu/253ed7b/ : SUCCESS in 44m 10s\n- mistral-tox-unit-mysql http://logs.openstack.org/55/663455/2/check/mistral-tox-unit-mysql/98b21da/ : SUCCESS in 29m 38s\n- mistral-tox-unit-postgresql http://logs.openstack.org/55/663455/2/check/mistral-tox-unit-postgresql/13353ca/ : SUCCESS in 26m 13s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/55/663455/2/check/tripleo-ci-centos-7-scenario003-standalone/d47f2de/ : SUCCESS in 1h 15m 07s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/55/663455/2/check/tripleo-ci-centos-7-containers-multinode/7bfa34a/ : SUCCESS in 1h 56m 23s\n- mistral-rally-task http://logs.openstack.org/55/663455/2/check/mistral-rally-task/1f86f46/ : FAILURE in 1h 01m 42s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/55/663455/2/check/openstack-tox-lower-constraints/0c96a1a/ : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":2},{"id":"db0af14514aaa97ee0dc6112746424075a8e8cbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 09:18:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eee5f370a8f65acc6cc1aeebef605ee553c67367","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 10:46:09.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/55/663455/2/gate/openstack-tox-pep8/551e196/ : SUCCESS in 6m 29s\n- openstack-tox-py27 http://logs.openstack.org/55/663455/2/gate/openstack-tox-py27/65122e1/ : SUCCESS in 8m 53s\n- openstack-tox-py36 http://logs.openstack.org/55/663455/2/gate/openstack-tox-py36/abcfd9b/ : SUCCESS in 8m 44s\n- openstack-tox-py37 http://logs.openstack.org/55/663455/2/gate/openstack-tox-py37/bac1a8a/ : SUCCESS in 8m 53s\n- openstack-tox-docs http://logs.openstack.org/55/663455/2/gate/openstack-tox-docs/f556fd2/html/ : SUCCESS in 4m 57s\n- mistral-devstack http://logs.openstack.org/55/663455/2/gate/mistral-devstack/3dc1c04/ : SUCCESS in 46m 08s\n- mistral-devstack-non-apache http://logs.openstack.org/55/663455/2/gate/mistral-devstack-non-apache/5219e19/ : SUCCESS in 35m 58s\n- mistral-tox-unit-mysql http://logs.openstack.org/55/663455/2/gate/mistral-tox-unit-mysql/29e982b/ : FAILURE in 27m 17s\n- mistral-tox-unit-postgresql http://logs.openstack.org/55/663455/2/gate/mistral-tox-unit-postgresql/ee8c5c5/ : SUCCESS in 26m 55s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/55/663455/2/gate/tripleo-ci-centos-7-scenario003-standalone/59b929b/ : SUCCESS in 1h 18m 11s\n- mistral-devstack-kombu http://logs.openstack.org/55/663455/2/gate/mistral-devstack-kombu/68af5d3/ : SUCCESS in 42m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/663455/2/gate/openstack-tox-lower-constraints/40106ca/ : SUCCESS in 10m 40s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/55/663455/2/gate/tripleo-ci-centos-7-undercloud-containers/568097b/ : SUCCESS in 1h 26m 11s","accounts_in_message":[],"_revision_number":2},{"id":"c310c8228e8388f38d452e0c6ee9cd032a8abd8b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-06 10:55:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d8c19a94f3d0632b53c5d1ba685b4add82856aba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 13:16:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/55/663455/2/check/openstack-tox-pep8/fc04a68/ : SUCCESS in 4m 50s\n- openstack-tox-py27 http://logs.openstack.org/55/663455/2/check/openstack-tox-py27/3535ab1/ : SUCCESS in 8m 50s\n- openstack-tox-py36 http://logs.openstack.org/55/663455/2/check/openstack-tox-py36/76fd176/ : SUCCESS in 9m 21s\n- openstack-tox-py37 http://logs.openstack.org/55/663455/2/check/openstack-tox-py37/bac058e/ : SUCCESS in 10m 37s\n- openstack-tox-docs http://logs.openstack.org/55/663455/2/check/openstack-tox-docs/7d8c4fa/html/ : SUCCESS in 4m 47s\n- openstack-tox-cover http://logs.openstack.org/55/663455/2/check/openstack-tox-cover/5597a00/cover/ : SUCCESS in 10m 41s (non-voting)\n- mistral-devstack http://logs.openstack.org/55/663455/2/check/mistral-devstack/e2d55bf/ : FAILURE in 28m 50s\n- mistral-devstack-non-apache http://logs.openstack.org/55/663455/2/check/mistral-devstack-non-apache/6ad4db6/ : SUCCESS in 48m 46s\n- mistral-devstack-kombu http://logs.openstack.org/55/663455/2/check/mistral-devstack-kombu/d556438/ : SUCCESS in 46m 06s\n- mistral-tox-unit-mysql http://logs.openstack.org/55/663455/2/check/mistral-tox-unit-mysql/c502c3e/ : SUCCESS in 27m 37s\n- mistral-tox-unit-postgresql http://logs.openstack.org/55/663455/2/check/mistral-tox-unit-postgresql/2df3cf2/ : SUCCESS in 26m 43s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/55/663455/2/check/tripleo-ci-centos-7-scenario003-standalone/ef4b2f6/ : SUCCESS in 1h 17m 15s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/55/663455/2/check/tripleo-ci-centos-7-containers-multinode/ec60117/ : SUCCESS in 2h 02m 34s\n- mistral-rally-task http://logs.openstack.org/55/663455/2/check/mistral-rally-task/41e8fcf/ : SUCCESS in 1h 06m 09s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/55/663455/2/check/openstack-tox-lower-constraints/62bc8f0/ : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":2},{"id":"15aec55cb908bda5aa6fb4e41632cba7e699f5b2","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-06 13:34:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"93721b80d3987f76a045589b41b7aa2341b97cb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 16:20:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/55/663455/2/check/openstack-tox-pep8/3f07a00/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/55/663455/2/check/openstack-tox-py27/bb50ee9/ : SUCCESS in 8m 40s\n- openstack-tox-py36 http://logs.openstack.org/55/663455/2/check/openstack-tox-py36/c029909/ : SUCCESS in 9m 25s\n- openstack-tox-py37 http://logs.openstack.org/55/663455/2/check/openstack-tox-py37/6c32d7d/ : SUCCESS in 8m 40s\n- openstack-tox-docs http://logs.openstack.org/55/663455/2/check/openstack-tox-docs/a32768c/html/ : SUCCESS in 5m 18s\n- openstack-tox-cover http://logs.openstack.org/55/663455/2/check/openstack-tox-cover/123f804/cover/ : SUCCESS in 8m 38s (non-voting)\n- mistral-devstack http://logs.openstack.org/55/663455/2/check/mistral-devstack/f2f0273/ : SUCCESS in 46m 48s\n- mistral-devstack-non-apache http://logs.openstack.org/55/663455/2/check/mistral-devstack-non-apache/35d9a47/ : SUCCESS in 47m 58s\n- mistral-devstack-kombu http://logs.openstack.org/55/663455/2/check/mistral-devstack-kombu/f80d289/ : SUCCESS in 46m 59s\n- mistral-tox-unit-mysql http://logs.openstack.org/55/663455/2/check/mistral-tox-unit-mysql/1eb2444/ : SUCCESS in 28m 07s\n- mistral-tox-unit-postgresql http://logs.openstack.org/55/663455/2/check/mistral-tox-unit-postgresql/eae2ca8/ : FAILURE in 27m 42s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/55/663455/2/check/tripleo-ci-centos-7-scenario003-standalone/2946263/ : SUCCESS in 1h 12m 45s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/55/663455/2/check/tripleo-ci-centos-7-containers-multinode/3c45dc6/ : SUCCESS in 2h 11m 30s\n- mistral-rally-task http://logs.openstack.org/55/663455/2/check/mistral-rally-task/b0b010b/ : SUCCESS in 1h 06m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/55/663455/2/check/openstack-tox-lower-constraints/81982af/ : SUCCESS in 10m 27s","accounts_in_message":[],"_revision_number":2},{"id":"01059bcd2d36bc2867230d23badbcfb989aeb923","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-06 18:51:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"01ebaa814a41ec56233909682a2a8bc8e698def3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 21:26:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/55/663455/2/check/openstack-tox-pep8/f7947df/ : SUCCESS in 7m 56s\n- openstack-tox-py27 http://logs.openstack.org/55/663455/2/check/openstack-tox-py27/92976e0/ : SUCCESS in 10m 01s\n- openstack-tox-py36 http://logs.openstack.org/55/663455/2/check/openstack-tox-py36/98ac9a5/ : SUCCESS in 8m 45s\n- openstack-tox-py37 http://logs.openstack.org/55/663455/2/check/openstack-tox-py37/2814952/ : SUCCESS in 9m 17s\n- openstack-tox-docs http://logs.openstack.org/55/663455/2/check/openstack-tox-docs/402985a/html/ : SUCCESS in 4m 49s\n- openstack-tox-cover http://logs.openstack.org/55/663455/2/check/openstack-tox-cover/3396a0b/cover/ : SUCCESS in 8m 38s (non-voting)\n- mistral-devstack http://logs.openstack.org/55/663455/2/check/mistral-devstack/6b7322d/ : SUCCESS in 46m 46s\n- mistral-devstack-non-apache http://logs.openstack.org/55/663455/2/check/mistral-devstack-non-apache/e4cfd4d/ : SUCCESS in 51m 18s\n- mistral-devstack-kombu http://logs.openstack.org/55/663455/2/check/mistral-devstack-kombu/0e2b99b/ : SUCCESS in 47m 45s\n- mistral-tox-unit-mysql http://logs.openstack.org/55/663455/2/check/mistral-tox-unit-mysql/672116b/ : SUCCESS in 27m 53s\n- mistral-tox-unit-postgresql http://logs.openstack.org/55/663455/2/check/mistral-tox-unit-postgresql/f975809/ : SUCCESS in 27m 19s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/55/663455/2/check/tripleo-ci-centos-7-scenario003-standalone/e0b21ee/ : SUCCESS in 1h 14m 30s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/55/663455/2/check/tripleo-ci-centos-7-containers-multinode/093911b/ : SUCCESS in 1h 55m 16s\n- mistral-rally-task http://logs.openstack.org/55/663455/2/check/mistral-rally-task/e245d4a/ : SUCCESS in 1h 19m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/55/663455/2/check/openstack-tox-lower-constraints/6e5b5b5/ : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":2},{"id":"e7e512b6add776e02900a650e309ce3ed18b7091","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 21:26:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"59fbb9b539e38fef4d9a0efcb563b2c48f8c2e29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 23:03:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/55/663455/2/gate/openstack-tox-pep8/7d22d7e/ : SUCCESS in 5m 07s\n- openstack-tox-py27 http://logs.openstack.org/55/663455/2/gate/openstack-tox-py27/8cb7dbd/ : SUCCESS in 8m 21s\n- openstack-tox-py36 http://logs.openstack.org/55/663455/2/gate/openstack-tox-py36/8805d08/ : SUCCESS in 8m 42s\n- openstack-tox-py37 http://logs.openstack.org/55/663455/2/gate/openstack-tox-py37/cba1411/ : SUCCESS in 8m 34s\n- openstack-tox-docs http://logs.openstack.org/55/663455/2/gate/openstack-tox-docs/051edfb/html/ : SUCCESS in 4m 58s\n- mistral-devstack http://logs.openstack.org/55/663455/2/gate/mistral-devstack/36690e2/ : SUCCESS in 34m 43s\n- mistral-devstack-non-apache http://logs.openstack.org/55/663455/2/gate/mistral-devstack-non-apache/f7a2845/ : SUCCESS in 42m 57s\n- mistral-tox-unit-mysql http://logs.openstack.org/55/663455/2/gate/mistral-tox-unit-mysql/86928a5/ : SUCCESS in 31m 12s\n- mistral-tox-unit-postgresql http://logs.openstack.org/55/663455/2/gate/mistral-tox-unit-postgresql/03d7134/ : SUCCESS in 26m 12s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/55/663455/2/gate/tripleo-ci-centos-7-scenario003-standalone/b957ffd/ : SUCCESS in 1h 14m 42s\n- mistral-devstack-kombu http://logs.openstack.org/55/663455/2/gate/mistral-devstack-kombu/ff3537d/ : SUCCESS in 45m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/663455/2/gate/openstack-tox-lower-constraints/8364f6a/ : SUCCESS in 9m 38s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/55/663455/2/gate/tripleo-ci-centos-7-undercloud-containers/34523ba/ : SUCCESS in 1h 29m 02s","accounts_in_message":[],"_revision_number":2},{"id":"47c3069833448c4257eee41eecacd1dbaf1d64a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 23:03:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e5031e26f12295fb4379cfe99ca0d7b4f53cd6e6","revisions":{"001728f2d0d568e91476d9020c79dfa63da3cd7f":{"kind":"REWORK","_number":1,"created":"2019-06-05 22:56:30.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/55/663455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/55/663455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/663455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/663455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/663455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/55/663455/1"}}},"commit":{"parents":[{"commit":"55c79906848ac37c4ad4f56eca694640adbf2dd7","subject":"Merge \"Remove _get_next_clauses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55c79906848ac37c4ad4f56eca694640adbf2dd7"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-05 22:44:23.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-05 22:56:13.000000000","tz":120},"subject":"Rework updating action executions heartbeats","message":"Rework updating action executions heartbeats\n\nNew function update_action_execution_heartbeat was introduced in db api\nto optimize updating of heartbeats.\n\nNow to perform the update mistral downloads an action execution object,\nthen modifies it, and finally  uploads back to the db using ORM.\nThis function removes initial object fetching and changes ORM update to\nsqlalchemy core.\n\nChange-Id: I4cb2d85c193531648540a37caca9ee868ede3a08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/001728f2d0d568e91476d9020c79dfa63da3cd7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/001728f2d0d568e91476d9020c79dfa63da3cd7f"}]},"branch":"refs/heads/master"},"e5031e26f12295fb4379cfe99ca0d7b4f53cd6e6":{"kind":"REWORK","_number":2,"created":"2019-06-06 07:04:33.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/55/663455/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/55/663455/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/663455/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/663455/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/55/663455/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/55/663455/2"}}},"commit":{"parents":[{"commit":"55c79906848ac37c4ad4f56eca694640adbf2dd7","subject":"Merge \"Remove _get_next_clauses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55c79906848ac37c4ad4f56eca694640adbf2dd7"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-05 22:44:23.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-06 07:04:27.000000000","tz":0},"subject":"Rework updating action executions heartbeats","message":"Rework updating action executions heartbeats\n\nNew function update_action_execution_heartbeat was introduced in db api\nto optimize updating of heartbeats.\n\nNow to perform the update mistral downloads an action execution object,\nthen modifies it, and finally  uploads back to the db using ORM.\nThis function removes initial object fetching and changes ORM update to\nsqlalchemy core.\n\nChange-Id: I4cb2d85c193531648540a37caca9ee868ede3a08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e5031e26f12295fb4379cfe99ca0d7b4f53cd6e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e5031e26f12295fb4379cfe99ca0d7b4f53cd6e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
