)]}'
{"id":"openstack%2Fmistral~663753","triplet_id":"openstack%2Fmistral~master~I0762a465742fcb1d1dc8d4d00d81cf8288b842ca","project":"openstack/mistral","branch":"master","topic":"notification_reduce_object","hashtags":[],"change_id":"I0762a465742fcb1d1dc8d4d00d81cf8288b842ca","subject":"Reduce number of fields in notification data","status":"MERGED","created":"2019-06-06 19:36:34.000000000","updated":"2019-06-10 12:33:32.000000000","submitted":"2019-06-10 12:33:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663753-1560170012239-d9d96933","meta_rev_id":"6246399eb7a55b02018b04de0b82ed495bf1e51d","_number":663753,"virtual_id_number":663753,"owner":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-06-10 12:33:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2019-06-10 08:36:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2019-06-10 08:36:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 08:36:39.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-06-10 12:33:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d96fd600250afc16b12c7e4bded6cbda61c83c8","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-06 19:36:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe2d96e5d6f131a536a4ceadccffcb08a30c40f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 21:46:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/53/663753/1/check/openstack-tox-pep8/dd2b706/ : SUCCESS in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/53/663753/1/check/openstack-tox-py27/ba5aba3/ : SUCCESS in 8m 59s\n- openstack-tox-py36 http://logs.openstack.org/53/663753/1/check/openstack-tox-py36/256962e/ : SUCCESS in 9m 28s\n- openstack-tox-py37 http://logs.openstack.org/53/663753/1/check/openstack-tox-py37/3099b53/ : SUCCESS in 8m 53s\n- openstack-tox-docs http://logs.openstack.org/53/663753/1/check/openstack-tox-docs/37cf086/html/ : SUCCESS in 4m 54s\n- openstack-tox-cover http://logs.openstack.org/53/663753/1/check/openstack-tox-cover/218dc06/cover/ : SUCCESS in 11m 22s (non-voting)\n- mistral-devstack http://logs.openstack.org/53/663753/1/check/mistral-devstack/75201f1/ : SUCCESS in 46m 40s\n- mistral-devstack-non-apache http://logs.openstack.org/53/663753/1/check/mistral-devstack-non-apache/a59625b/ : SUCCESS in 36m 48s\n- mistral-devstack-kombu http://logs.openstack.org/53/663753/1/check/mistral-devstack-kombu/daf8c86/ : SUCCESS in 37m 00s\n- mistral-tox-unit-mysql http://logs.openstack.org/53/663753/1/check/mistral-tox-unit-mysql/0ab4b7b/ : FAILURE in 28m 17s\n- mistral-tox-unit-postgresql http://logs.openstack.org/53/663753/1/check/mistral-tox-unit-postgresql/37a1d4c/ : SUCCESS in 27m 52s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/53/663753/1/check/tripleo-ci-centos-7-scenario003-standalone/6c84484/ : SUCCESS in 1h 16m 39s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/53/663753/1/check/tripleo-ci-centos-7-containers-multinode/303ef52/ : SUCCESS in 2h 02m 21s\n- mistral-rally-task http://logs.openstack.org/53/663753/1/check/mistral-rally-task/4d960fd/ : FAILURE in 1h 20m 58s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/53/663753/1/check/openstack-tox-lower-constraints/2af0c5c/ : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c276b558af03d558d4bcdc83d9a8be44adacef99","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-06 21:58:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"00887ccb1a6d3ed9f7f975e42e39be83f26fdb1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 00:47:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/53/663753/2/check/openstack-tox-pep8/e069cd3/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/53/663753/2/check/openstack-tox-py27/6b908c1/ : SUCCESS in 9m 09s\n- openstack-tox-py36 http://logs.openstack.org/53/663753/2/check/openstack-tox-py36/72511ba/ : SUCCESS in 8m 39s\n- openstack-tox-py37 http://logs.openstack.org/53/663753/2/check/openstack-tox-py37/2580845/ : SUCCESS in 9m 15s\n- openstack-tox-docs http://logs.openstack.org/53/663753/2/check/openstack-tox-docs/76aa5e0/html/ : SUCCESS in 4m 57s\n- openstack-tox-cover http://logs.openstack.org/53/663753/2/check/openstack-tox-cover/15cd129/cover/ : SUCCESS in 11m 28s (non-voting)\n- mistral-devstack http://logs.openstack.org/53/663753/2/check/mistral-devstack/9c88335/ : FAILURE in 41m 28s\n- mistral-devstack-non-apache http://logs.openstack.org/53/663753/2/check/mistral-devstack-non-apache/f559a85/ : SUCCESS in 42m 09s\n- mistral-devstack-kombu http://logs.openstack.org/53/663753/2/check/mistral-devstack-kombu/de6b2d3/ : SUCCESS in 44m 43s\n- mistral-tox-unit-mysql http://logs.openstack.org/53/663753/2/check/mistral-tox-unit-mysql/f0c2baa/ : SUCCESS in 28m 37s\n- mistral-tox-unit-postgresql http://logs.openstack.org/53/663753/2/check/mistral-tox-unit-postgresql/5447daa/ : SUCCESS in 26m 28s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/53/663753/2/check/tripleo-ci-centos-7-scenario003-standalone/8456df1/ : SUCCESS in 1h 21m 37s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/53/663753/2/check/tripleo-ci-centos-7-containers-multinode/9985a75/ : SUCCESS in 2h 07m 49s\n- mistral-rally-task http://logs.openstack.org/53/663753/2/check/mistral-rally-task/9a2e4b3/ : FAILURE in 1h 16m 13s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/53/663753/2/check/openstack-tox-lower-constraints/522401c/ : SUCCESS in 11m 14s","accounts_in_message":[],"_revision_number":2},{"id":"0a1b4ab89689a46bee6c8386457ea01c821f4f68","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-07 09:26:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fd0c6767fe7c4065da7c7bd19754e72779e9c770","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 11:29:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/53/663753/2/check/openstack-tox-pep8/36ddf6e/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/53/663753/2/check/openstack-tox-py27/2806857/ : SUCCESS in 8m 49s\n- openstack-tox-py36 http://logs.openstack.org/53/663753/2/check/openstack-tox-py36/9d81e82/ : SUCCESS in 9m 30s\n- openstack-tox-py37 http://logs.openstack.org/53/663753/2/check/openstack-tox-py37/aa9363c/ : SUCCESS in 9m 40s\n- openstack-tox-docs http://logs.openstack.org/53/663753/2/check/openstack-tox-docs/e20af2e/html/ : SUCCESS in 5m 05s\n- openstack-tox-cover http://logs.openstack.org/53/663753/2/check/openstack-tox-cover/300a5b1/cover/ : SUCCESS in 9m 29s (non-voting)\n- mistral-devstack http://logs.openstack.org/53/663753/2/check/mistral-devstack/7a29549/ : SUCCESS in 43m 02s\n- mistral-devstack-non-apache http://logs.openstack.org/53/663753/2/check/mistral-devstack-non-apache/fe5af9d/ : SUCCESS in 46m 35s\n- mistral-devstack-kombu http://logs.openstack.org/53/663753/2/check/mistral-devstack-kombu/6ab8a45/ : SUCCESS in 45m 42s\n- mistral-tox-unit-mysql http://logs.openstack.org/53/663753/2/check/mistral-tox-unit-mysql/e7b72aa/ : SUCCESS in 30m 19s\n- mistral-tox-unit-postgresql http://logs.openstack.org/53/663753/2/check/mistral-tox-unit-postgresql/b187c51/ : SUCCESS in 26m 22s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/53/663753/2/check/tripleo-ci-centos-7-scenario003-standalone/ab1fa49/ : SUCCESS in 1h 11m 14s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/53/663753/2/check/tripleo-ci-centos-7-containers-multinode/05fada6/ : SUCCESS in 1h 56m 03s\n- mistral-rally-task http://logs.openstack.org/53/663753/2/check/mistral-rally-task/46919c7/ : SUCCESS in 1h 05m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/53/663753/2/check/openstack-tox-lower-constraints/4ac4016/ : SUCCESS in 11m 12s","accounts_in_message":[],"_revision_number":2},{"id":"2e76baf8ae98a84d261bbcd39afd5bb59a8aa4e5","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-10 04:03:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6960d485d1f4c74339e0d71a23ec5c97cc91856c","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-10 08:17:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"33a8fc787bfd13258a12b1e9ec555c8be50998eb","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-10 08:36:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f1c6ae476b7597c56e32912175bd5bf4c9c20296","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 10:27:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/53/663753/3/check/openstack-tox-pep8/5daad93/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/53/663753/3/check/openstack-tox-py27/d8b6f73/ : SUCCESS in 8m 39s\n- openstack-tox-py36 http://logs.openstack.org/53/663753/3/check/openstack-tox-py36/2df88a0/ : SUCCESS in 8m 18s\n- openstack-tox-py37 http://logs.openstack.org/53/663753/3/check/openstack-tox-py37/c139df5/ : SUCCESS in 9m 36s\n- openstack-tox-docs http://logs.openstack.org/53/663753/3/check/openstack-tox-docs/af84d6b/html/ : SUCCESS in 4m 44s\n- openstack-tox-cover http://logs.openstack.org/53/663753/3/check/openstack-tox-cover/aa16dc8/cover/ : SUCCESS in 9m 20s (non-voting)\n- mistral-devstack http://logs.openstack.org/53/663753/3/check/mistral-devstack/13a5bad/ : SUCCESS in 42m 38s\n- mistral-devstack-non-apache http://logs.openstack.org/53/663753/3/check/mistral-devstack-non-apache/cabeb9f/ : SUCCESS in 44m 12s\n- mistral-devstack-kombu http://logs.openstack.org/53/663753/3/check/mistral-devstack-kombu/fe8758a/ : SUCCESS in 42m 53s\n- mistral-tox-unit-mysql http://logs.openstack.org/53/663753/3/check/mistral-tox-unit-mysql/8e15e5e/ : SUCCESS in 26m 59s\n- mistral-tox-unit-postgresql http://logs.openstack.org/53/663753/3/check/mistral-tox-unit-postgresql/db3326a/ : SUCCESS in 26m 23s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/53/663753/3/check/tripleo-ci-centos-7-scenario003-standalone/5b61f49/ : SUCCESS in 1h 13m 50s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/53/663753/3/check/tripleo-ci-centos-7-containers-multinode/5a97c60/ : SUCCESS in 2h 02m 45s\n- mistral-rally-task http://logs.openstack.org/53/663753/3/check/mistral-rally-task/f8f03b4/ : SUCCESS in 1h 08m 33s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/53/663753/3/check/openstack-tox-lower-constraints/b4cd551/ : SUCCESS in 10m 28s","accounts_in_message":[],"_revision_number":3},{"id":"ce4ac9901977965ffab72d7165f338a4294ef16b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 10:27:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a702a100f52cf52c04f5ed9bf6ccdf6689385042","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 12:33:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/53/663753/3/gate/openstack-tox-pep8/966cf20/ : SUCCESS in 6m 26s\n- openstack-tox-py27 http://logs.openstack.org/53/663753/3/gate/openstack-tox-py27/83f0e26/ : SUCCESS in 8m 30s\n- openstack-tox-py36 http://logs.openstack.org/53/663753/3/gate/openstack-tox-py36/2a6e3d1/ : SUCCESS in 8m 43s\n- openstack-tox-py37 http://logs.openstack.org/53/663753/3/gate/openstack-tox-py37/b0b673f/ : SUCCESS in 8m 54s\n- openstack-tox-docs http://logs.openstack.org/53/663753/3/gate/openstack-tox-docs/099a533/html/ : SUCCESS in 7m 01s\n- mistral-devstack http://logs.openstack.org/53/663753/3/gate/mistral-devstack/bae265a/ : SUCCESS in 50m 21s\n- mistral-devstack-non-apache http://logs.openstack.org/53/663753/3/gate/mistral-devstack-non-apache/6e7b13e/ : SUCCESS in 44m 11s\n- mistral-tox-unit-mysql http://logs.openstack.org/53/663753/3/gate/mistral-tox-unit-mysql/d78b26d/ : SUCCESS in 35m 27s\n- mistral-tox-unit-postgresql http://logs.openstack.org/53/663753/3/gate/mistral-tox-unit-postgresql/22ad88b/ : SUCCESS in 27m 05s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/53/663753/3/gate/tripleo-ci-centos-7-scenario003-standalone/85b442f/ : SUCCESS in 1h 20m 39s\n- mistral-devstack-kombu http://logs.openstack.org/53/663753/3/gate/mistral-devstack-kombu/0e94de8/ : SUCCESS in 53m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/663753/3/gate/openstack-tox-lower-constraints/95f7cbb/ : SUCCESS in 11m 42s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/53/663753/3/gate/tripleo-ci-centos-7-undercloud-containers/ee4e778/ : SUCCESS in 1h 27m 45s","accounts_in_message":[],"_revision_number":3},{"id":"2e64629b03aa77042a7f5bb38cb03f65036b8774","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 12:33:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bea98f6f8525308e926cf430b13fba6762f86c96","revisions":{"3fb7aed266ab36a4c0ab37ca9dbe4e23fd070bdd":{"kind":"REWORK","_number":1,"created":"2019-06-06 19:36:34.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/53/663753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/663753/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/663753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/663753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/663753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/663753/1"}}},"commit":{"parents":[{"commit":"f85caa02c9aaa64db4300b43854327d132a71ed4","subject":"Merge \"Skip context evaluation for non-conditional transitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f85caa02c9aaa64db4300b43854327d132a71ed4"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 19:14:53.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 19:36:07.000000000","tz":180},"subject":"Reduce number of fields in notification data","message":"Reduce number of fields in notification data\n\nIt is very important to prevent sending huge messages via rpc.\nNotification data has a lot of potentially large fields, such as output,\nspec and etc. Most of them are useless as a part of notification message\nso we should omit them.\n\nChange-Id: I0762a465742fcb1d1dc8d4d00d81cf8288b842ca\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3fb7aed266ab36a4c0ab37ca9dbe4e23fd070bdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3fb7aed266ab36a4c0ab37ca9dbe4e23fd070bdd"}]},"branch":"refs/heads/master"},"cacd891911f804522c7c5bb81c962d1381a2bc37":{"kind":"REWORK","_number":2,"created":"2019-06-06 21:58:22.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/53/663753/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/663753/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/663753/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/663753/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/663753/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/663753/2"}}},"commit":{"parents":[{"commit":"f85caa02c9aaa64db4300b43854327d132a71ed4","subject":"Merge \"Skip context evaluation for non-conditional transitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f85caa02c9aaa64db4300b43854327d132a71ed4"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 19:14:53.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 21:58:08.000000000","tz":180},"subject":"Reduce number of fields in notification data","message":"Reduce number of fields in notification data\n\nIt is very important to prevent sending huge messages via rpc.\nNotification data has a lot of potentially large fields, such as output,\nspec and etc. Most of them are useless as a part of notification message\nso we should omit them.\n\nChange-Id: I0762a465742fcb1d1dc8d4d00d81cf8288b842ca\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cacd891911f804522c7c5bb81c962d1381a2bc37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cacd891911f804522c7c5bb81c962d1381a2bc37"}]},"branch":"refs/heads/master"},"bea98f6f8525308e926cf430b13fba6762f86c96":{"kind":"REWORK","_number":3,"created":"2019-06-10 08:17:58.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/53/663753/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/663753/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/663753/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/663753/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/663753/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/663753/3"}}},"commit":{"parents":[{"commit":"f85caa02c9aaa64db4300b43854327d132a71ed4","subject":"Merge \"Skip context evaluation for non-conditional transitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f85caa02c9aaa64db4300b43854327d132a71ed4"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 19:14:53.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-10 08:19:05.000000000","tz":180},"subject":"Reduce number of fields in notification data","message":"Reduce number of fields in notification data\n\nIt is very important to prevent sending huge messages via rpc.\nNotification data has a lot of potentially large fields, such as output,\nspec and etc. Most of them are useless as a part of notification message\nso we should omit them.\n\nChange-Id: I0762a465742fcb1d1dc8d4d00d81cf8288b842ca\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bea98f6f8525308e926cf430b13fba6762f86c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bea98f6f8525308e926cf430b13fba6762f86c96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
