)]}'
{"id":"openstack%2Fmistral~663768","triplet_id":"openstack%2Fmistral~master~Ifdf032a9e1377ed73b80cc26fd659aabb417cfc5","project":"openstack/mistral","branch":"master","topic":"task_context_in_policies","hashtags":[],"change_id":"Ifdf032a9e1377ed73b80cc26fd659aabb417cfc5","subject":"Allow to use \u0027task()\u0027 yaql function in task policies","status":"MERGED","created":"2019-06-06 20:43:34.000000000","updated":"2019-06-07 11:03:12.000000000","submitted":"2019-06-07 11:03:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663768-1559905392329-3605573f","meta_rev_id":"4d11de6227a42a27d3071e26840d19fd648d2a92","_number":663768,"virtual_id_number":663768,"owner":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-06-07 11:03:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-06-07 09:29:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-06-07 09:29:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 09:29:16.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-06-07 11:03:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"045a41f2303556e1f30829ca810771a2f934b6cb","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-06 20:43:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9a1e428ce2f8e2545f0b20c9ffdee2874dc20dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 23:24:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/663768/1/check/openstack-tox-pep8/0988a9f/ : SUCCESS in 7m 13s\n- openstack-tox-py27 http://logs.openstack.org/68/663768/1/check/openstack-tox-py27/e7a403b/ : SUCCESS in 8m 35s\n- openstack-tox-py36 http://logs.openstack.org/68/663768/1/check/openstack-tox-py36/15a5947/ : SUCCESS in 9m 50s\n- openstack-tox-py37 http://logs.openstack.org/68/663768/1/check/openstack-tox-py37/aac7c1b/ : SUCCESS in 9m 25s\n- openstack-tox-docs http://logs.openstack.org/68/663768/1/check/openstack-tox-docs/f802f3d/html/ : SUCCESS in 5m 39s\n- openstack-tox-cover http://logs.openstack.org/68/663768/1/check/openstack-tox-cover/3e015b1/cover/ : SUCCESS in 9m 34s (non-voting)\n- mistral-devstack http://logs.openstack.org/68/663768/1/check/mistral-devstack/a6ee842/ : SUCCESS in 1h 01m 06s\n- mistral-devstack-non-apache http://logs.openstack.org/68/663768/1/check/mistral-devstack-non-apache/b5448a5/ : SUCCESS in 42m 10s\n- mistral-devstack-kombu http://logs.openstack.org/68/663768/1/check/mistral-devstack-kombu/9c12d22/ : SUCCESS in 42m 53s\n- mistral-tox-unit-mysql http://logs.openstack.org/68/663768/1/check/mistral-tox-unit-mysql/8b393a7/ : SUCCESS in 27m 34s\n- mistral-tox-unit-postgresql http://logs.openstack.org/68/663768/1/check/mistral-tox-unit-postgresql/421b772/ : SUCCESS in 31m 38s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/68/663768/1/check/tripleo-ci-centos-7-scenario003-standalone/a8aa67f/ : SUCCESS in 1h 12m 35s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/68/663768/1/check/tripleo-ci-centos-7-containers-multinode/f3841ab/ : SUCCESS in 2h 06m 03s\n- mistral-rally-task http://logs.openstack.org/68/663768/1/check/mistral-rally-task/116a3c6/ : SUCCESS in 56m 15s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/663768/1/check/openstack-tox-lower-constraints/2b9cc13/ : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3ec6dc47b3dcd29e762b41486e746683f986ae52","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-07 09:29:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c04ad4717256370b2d944976e084768dd199cfb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 09:29:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"396f503b4ff639b451388e05cd8b013d5d4d4d77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 11:03:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2766ddf987f7c6483ff163514b84eee70533f4bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 11:03:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/663768/1/gate/openstack-tox-pep8/32d4c90/ : SUCCESS in 8m 39s\n- openstack-tox-py27 http://logs.openstack.org/68/663768/1/gate/openstack-tox-py27/fd600df/ : SUCCESS in 8m 42s\n- openstack-tox-py36 http://logs.openstack.org/68/663768/1/gate/openstack-tox-py36/58ecfc6/ : SUCCESS in 8m 38s\n- openstack-tox-py37 http://logs.openstack.org/68/663768/1/gate/openstack-tox-py37/7d0ac6b/ : SUCCESS in 9m 17s\n- openstack-tox-docs http://logs.openstack.org/68/663768/1/gate/openstack-tox-docs/9cf1d75/html/ : SUCCESS in 7m 00s\n- mistral-devstack http://logs.openstack.org/68/663768/1/gate/mistral-devstack/f6f7073/ : SUCCESS in 51m 49s\n- mistral-devstack-non-apache http://logs.openstack.org/68/663768/1/gate/mistral-devstack-non-apache/b209c5f/ : SUCCESS in 43m 41s\n- mistral-tox-unit-mysql http://logs.openstack.org/68/663768/1/gate/mistral-tox-unit-mysql/9a73053/ : SUCCESS in 27m 39s\n- mistral-tox-unit-postgresql http://logs.openstack.org/68/663768/1/gate/mistral-tox-unit-postgresql/cb657d8/ : SUCCESS in 31m 55s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/68/663768/1/gate/tripleo-ci-centos-7-scenario003-standalone/6ce5f04/ : SUCCESS in 1h 23m 35s\n- mistral-devstack-kombu http://logs.openstack.org/68/663768/1/gate/mistral-devstack-kombu/865597c/ : SUCCESS in 45m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/663768/1/gate/openstack-tox-lower-constraints/87e5aa7/ : SUCCESS in 9m 25s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/68/663768/1/gate/tripleo-ci-centos-7-undercloud-containers/6a18051/ : SUCCESS in 1h 24m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f86cf1f5c77bdaeb89ca81b6198b1a47ef4c3c2","revisions":{"5f86cf1f5c77bdaeb89ca81b6198b1a47ef4c3c2":{"kind":"REWORK","_number":1,"created":"2019-06-06 20:43:34.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/68/663768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/68/663768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/663768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/663768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/663768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/68/663768/1"}}},"commit":{"parents":[{"commit":"f85caa02c9aaa64db4300b43854327d132a71ed4","subject":"Merge \"Skip context evaluation for non-conditional transitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f85caa02c9aaa64db4300b43854327d132a71ed4"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 20:43:16.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 20:43:16.000000000","tz":180},"subject":"Allow to use \u0027task()\u0027 yaql function in task policies","message":"Allow to use \u0027task()\u0027 yaql function in task policies\n\nChange-Id: Ifdf032a9e1377ed73b80cc26fd659aabb417cfc5\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5f86cf1f5c77bdaeb89ca81b6198b1a47ef4c3c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5f86cf1f5c77bdaeb89ca81b6198b1a47ef4c3c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
