)]}'
{"id":"openstack%2Fmistral~663776","triplet_id":"openstack%2Fmistral~master~I57b4f3d1533982d3b9b7063925f8d70f044aefea","project":"openstack/mistral","branch":"master","topic":"bp/fail-on-policy","hashtags":[],"change_id":"I57b4f3d1533982d3b9b7063925f8d70f044aefea","subject":"Fail-on policy","status":"MERGED","created":"2019-06-06 21:30:38.000000000","updated":"2019-08-12 12:58:38.000000000","submitted":"2019-08-12 12:58:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663776-1565614718895-674d1078","meta_rev_id":"18f1cdf9b1cb795d5339909b6b1d001e13123379","_number":663776,"virtual_id_number":663776,"owner":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-08-12 12:58:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-11 09:44:15.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-08-12 03:28:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-08-12 03:28:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 08:17:54.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-06-13 08:17:54.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-06-13 08:17:54.000000000","updated_by":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"reviewer":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"state":"REVIEWER"},{"updated":"2019-08-08 07:27:27.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-08-08 16:12:11.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-08-11 09:44:15.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-08-12 03:28:29.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-08-12 12:58:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dbd7f92c5a005a7ccfc3c1e5e5d737860a5620a7","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-06 21:30:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b5e237a6a2fb8ffcaea0c6bcd1209f979de8ab6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 23:31:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/76/663776/1/check/openstack-tox-pep8/954b397/ : FAILURE in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/76/663776/1/check/openstack-tox-py27/0605c9d/ : SUCCESS in 7m 53s\n- openstack-tox-py36 http://logs.openstack.org/76/663776/1/check/openstack-tox-py36/1e3459f/ : SUCCESS in 8m 15s\n- openstack-tox-py37 http://logs.openstack.org/76/663776/1/check/openstack-tox-py37/b0ca6c2/ : SUCCESS in 11m 15s\n- openstack-tox-docs http://logs.openstack.org/76/663776/1/check/openstack-tox-docs/4dd5c68/html/ : SUCCESS in 4m 43s\n- openstack-tox-cover http://logs.openstack.org/76/663776/1/check/openstack-tox-cover/3b49fc7/cover/ : SUCCESS in 8m 43s (non-voting)\n- mistral-devstack http://logs.openstack.org/76/663776/1/check/mistral-devstack/9f64484/ : SUCCESS in 47m 22s\n- mistral-devstack-non-apache http://logs.openstack.org/76/663776/1/check/mistral-devstack-non-apache/35049d5/ : SUCCESS in 48m 53s\n- mistral-devstack-kombu http://logs.openstack.org/76/663776/1/check/mistral-devstack-kombu/edcf01e/ : SUCCESS in 41m 44s\n- mistral-tox-unit-mysql http://logs.openstack.org/76/663776/1/check/mistral-tox-unit-mysql/0ca247d/ : FAILURE in 28m 35s\n- mistral-tox-unit-postgresql http://logs.openstack.org/76/663776/1/check/mistral-tox-unit-postgresql/a803d0b/ : FAILURE in 26m 28s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/76/663776/1/check/tripleo-ci-centos-7-scenario003-standalone/b4209fe/ : SUCCESS in 1h 22m 18s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/76/663776/1/check/tripleo-ci-centos-7-containers-multinode/99c9e64/ : SUCCESS in 1h 54m 13s\n- mistral-rally-task http://logs.openstack.org/76/663776/1/check/mistral-rally-task/a0b91bd/ : SUCCESS in 1h 12m 42s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/76/663776/1/check/openstack-tox-lower-constraints/ccba1ac/ : SUCCESS in 10m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8ad836d1475c52cd2a6a18e0bc500c63bccee78c","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-07 07:55:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc4c5d860562177fc4c9b0e2ab0f5482a1ebd3bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 10:14:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/76/663776/2/check/openstack-tox-pep8/23bb0f1/ : SUCCESS in 5m 51s\n- openstack-tox-py27 http://logs.openstack.org/76/663776/2/check/openstack-tox-py27/88651d7/ : SUCCESS in 12m 15s\n- openstack-tox-py36 http://logs.openstack.org/76/663776/2/check/openstack-tox-py36/0a05916/ : SUCCESS in 9m 59s\n- openstack-tox-py37 http://logs.openstack.org/76/663776/2/check/openstack-tox-py37/5e5cbcb/ : SUCCESS in 9m 56s\n- openstack-tox-docs http://logs.openstack.org/76/663776/2/check/openstack-tox-docs/eaf9f18/html/ : SUCCESS in 5m 26s\n- openstack-tox-cover http://logs.openstack.org/76/663776/2/check/openstack-tox-cover/5644736/cover/ : SUCCESS in 8m 59s (non-voting)\n- mistral-devstack http://logs.openstack.org/76/663776/2/check/mistral-devstack/65f891e/ : SUCCESS in 42m 13s\n- mistral-devstack-non-apache http://logs.openstack.org/76/663776/2/check/mistral-devstack-non-apache/d10eaf9/ : SUCCESS in 50m 07s\n- mistral-devstack-kombu http://logs.openstack.org/76/663776/2/check/mistral-devstack-kombu/1739fbf/ : SUCCESS in 48m 36s\n- mistral-tox-unit-mysql http://logs.openstack.org/76/663776/2/check/mistral-tox-unit-mysql/6f1e118/ : FAILURE in 32m 02s\n- mistral-tox-unit-postgresql http://logs.openstack.org/76/663776/2/check/mistral-tox-unit-postgresql/9294445/ : SUCCESS in 26m 42s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/76/663776/2/check/tripleo-ci-centos-7-scenario003-standalone/d2c6d90/ : SUCCESS in 1h 18m 27s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/76/663776/2/check/tripleo-ci-centos-7-containers-multinode/43b7c13/ : SUCCESS in 2h 08m 22s\n- mistral-rally-task http://logs.openstack.org/76/663776/2/check/mistral-rally-task/576ba97/ : SUCCESS in 1h 05m 37s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/76/663776/2/check/openstack-tox-lower-constraints/11f5c79/ : SUCCESS in 9m 52s","accounts_in_message":[],"_revision_number":2},{"id":"c69e3b2df30da756e673c16e783ba0c6d3313b6c","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-07 10:35:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"640084ea26cd9d03c89de365c46e51c6a945be75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:03:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/76/663776/2/check/openstack-tox-pep8/60640f6/ : SUCCESS in 5m 11s\n- openstack-tox-py27 http://logs.openstack.org/76/663776/2/check/openstack-tox-py27/f81c94e/ : SUCCESS in 8m 29s\n- openstack-tox-py36 http://logs.openstack.org/76/663776/2/check/openstack-tox-py36/36123ea/ : SUCCESS in 8m 54s\n- openstack-tox-py37 http://logs.openstack.org/76/663776/2/check/openstack-tox-py37/72860e3/ : SUCCESS in 9m 41s\n- openstack-tox-docs http://logs.openstack.org/76/663776/2/check/openstack-tox-docs/0a8dc29/html/ : SUCCESS in 5m 34s\n- openstack-tox-cover http://logs.openstack.org/76/663776/2/check/openstack-tox-cover/5c5fc59/cover/ : SUCCESS in 9m 13s (non-voting)\n- mistral-devstack http://logs.openstack.org/76/663776/2/check/mistral-devstack/5d149e4/ : SUCCESS in 48m 11s\n- mistral-devstack-non-apache http://logs.openstack.org/76/663776/2/check/mistral-devstack-non-apache/e712de8/ : SUCCESS in 39m 02s\n- mistral-devstack-kombu http://logs.openstack.org/76/663776/2/check/mistral-devstack-kombu/0fb26cf/ : SUCCESS in 43m 52s\n- mistral-tox-unit-mysql http://logs.openstack.org/76/663776/2/check/mistral-tox-unit-mysql/5045bd6/ : SUCCESS in 31m 58s\n- mistral-tox-unit-postgresql http://logs.openstack.org/76/663776/2/check/mistral-tox-unit-postgresql/6700a22/ : SUCCESS in 26m 09s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/76/663776/2/check/tripleo-ci-centos-7-scenario003-standalone/63d24ca/ : SUCCESS in 1h 14m 50s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/76/663776/2/check/tripleo-ci-centos-7-containers-multinode/15aefa7/ : SUCCESS in 1h 57m 46s\n- mistral-rally-task http://logs.openstack.org/76/663776/2/check/mistral-rally-task/278f532/ : SUCCESS in 52m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/76/663776/2/check/openstack-tox-lower-constraints/23eb6a4/ : SUCCESS in 10m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a643246986a0d26b55e5cf974a0afdb05538e7f5","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-07-11 23:29:22.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b7f091064ba009d1e5c0004186fda32b6c24737d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 02:08:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/76/663776/3/check/openstack-tox-pep8/37bea8b/ : SUCCESS in 5m 51s\n- openstack-tox-py27 http://logs.openstack.org/76/663776/3/check/openstack-tox-py27/1e49683/ : SUCCESS in 8m 03s\n- openstack-tox-py36 http://logs.openstack.org/76/663776/3/check/openstack-tox-py36/8e11981/ : SUCCESS in 8m 43s\n- openstack-tox-py37 http://logs.openstack.org/76/663776/3/check/openstack-tox-py37/6040ac7/ : SUCCESS in 8m 42s\n- openstack-tox-docs http://logs.openstack.org/76/663776/3/check/openstack-tox-docs/0429b8b/html/ : SUCCESS in 4m 35s\n- openstack-tox-cover http://logs.openstack.org/76/663776/3/check/openstack-tox-cover/6f0fb0a/ : FAILURE in 8m 31s (non-voting)\n- mistral-devstack http://logs.openstack.org/76/663776/3/check/mistral-devstack/ad617d2/ : SUCCESS in 47m 52s\n- mistral-devstack-non-apache http://logs.openstack.org/76/663776/3/check/mistral-devstack-non-apache/f3ee211/ : SUCCESS in 47m 44s\n- mistral-devstack-kombu http://logs.openstack.org/76/663776/3/check/mistral-devstack-kombu/1f563e2/ : SUCCESS in 42m 28s\n- mistral-tox-unit-mysql http://logs.openstack.org/76/663776/3/check/mistral-tox-unit-mysql/e910b42/ : SUCCESS in 31m 25s\n- mistral-tox-unit-postgresql http://logs.openstack.org/76/663776/3/check/mistral-tox-unit-postgresql/39d0e10/ : SUCCESS in 28m 21s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/76/663776/3/check/tripleo-ci-centos-7-scenario003-standalone/18af4df/ : SUCCESS in 1h 24m 37s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/76/663776/3/check/tripleo-ci-centos-7-containers-multinode/5f3588d/ : SUCCESS in 2h 24m 52s\n- mistral-rally-task http://logs.openstack.org/76/663776/3/check/mistral-rally-task/50c04d1/ : SUCCESS in 58m 56s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/76/663776/3/check/openstack-tox-lower-constraints/b9c66a1/ : SUCCESS in 10m 24s","accounts_in_message":[],"_revision_number":3},{"id":"4475d823aa17ba8e6d6722b6f85dec73fcdd868d","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-07-17 10:50:04.000000000","message":"Patch Set 3: Code-Review-1\n\nGreat feature. In order to merge it, please add this to the docs. It can be a new change depending on this one or you can include it here. Thank you!","accounts_in_message":[],"_revision_number":3},{"id":"ad5130b778233a7f8ec4a6a849f1b15d64a1d664","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-08-03 22:05:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"45c63e12f3e6a8d717c2fdaa92514c17128918ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 00:21:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/76/663776/4/check/openstack-tox-pep8/3824cd6/ : SUCCESS in 4m 13s\n- openstack-tox-py27 https://logs.opendev.org/76/663776/4/check/openstack-tox-py27/8bf2300/ : SUCCESS in 7m 18s\n- openstack-tox-py36 https://logs.opendev.org/76/663776/4/check/openstack-tox-py36/8f6abd6/ : SUCCESS in 8m 17s\n- openstack-tox-py37 https://logs.opendev.org/76/663776/4/check/openstack-tox-py37/959e4f8/ : SUCCESS in 7m 51s\n- openstack-tox-docs https://logs.opendev.org/76/663776/4/check/openstack-tox-docs/805a595/html/ : SUCCESS in 3m 33s\n- openstack-tox-cover https://logs.opendev.org/76/663776/4/check/openstack-tox-cover/f4f4506/cover/ : SUCCESS in 7m 45s (non-voting)\n- mistral-devstack https://logs.opendev.org/76/663776/4/check/mistral-devstack/2aa540f/ : SUCCESS in 39m 18s\n- mistral-devstack-non-apache https://logs.opendev.org/76/663776/4/check/mistral-devstack-non-apache/c60da4d/ : SUCCESS in 37m 33s\n- mistral-devstack-kombu https://logs.opendev.org/76/663776/4/check/mistral-devstack-kombu/6119138/ : SUCCESS in 35m 17s\n- mistral-tox-unit-mysql https://logs.opendev.org/76/663776/4/check/mistral-tox-unit-mysql/81ccc8f/ : SUCCESS in 28m 29s\n- mistral-tox-unit-postgresql https://logs.opendev.org/76/663776/4/check/mistral-tox-unit-postgresql/d849c8b/ : SUCCESS in 26m 17s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/76/663776/4/check/tripleo-ci-centos-7-scenario003-standalone/3ea3561/ : SUCCESS in 1h 08m 25s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/76/663776/4/check/tripleo-ci-centos-7-containers-multinode/c003d6d/ : SUCCESS in 2h 14m 25s\n- mistral-rally-task https://logs.opendev.org/76/663776/4/check/mistral-rally-task/1969972/ : SUCCESS in 58m 53s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/76/663776/4/check/openstack-tox-lower-constraints/2289b1a/ : SUCCESS in 9m 22s","accounts_in_message":[],"_revision_number":4},{"id":"3f0dad4287c3e093008599dc93a47797fc9b9893","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-08-08 07:27:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e16ff62bf9d14d5c0a5f5c32104ecb2a2222160a","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-08-08 07:56:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7f743ce0afb94d27ec7820d9edf021419dd36872","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 07:56:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2d887f0f693377e89bfc24b50c59fedd4f6cf1d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 09:31:03.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/76/663776/4/gate/openstack-tox-pep8/de86003/ : SUCCESS in 5m 40s\n- openstack-tox-py27 https://logs.opendev.org/76/663776/4/gate/openstack-tox-py27/b8dc852/ : SUCCESS in 8m 42s\n- openstack-tox-py36 https://logs.opendev.org/76/663776/4/gate/openstack-tox-py36/8c748ad/ : SUCCESS in 8m 49s\n- openstack-tox-py37 https://logs.opendev.org/76/663776/4/gate/openstack-tox-py37/0e9e6d6/ : SUCCESS in 8m 51s\n- openstack-tox-docs https://logs.opendev.org/76/663776/4/gate/openstack-tox-docs/c0d69b3/html/ : SUCCESS in 5m 25s\n- mistral-devstack https://logs.opendev.org/76/663776/4/gate/mistral-devstack/2cd8df0/ : SUCCESS in 40m 47s\n- mistral-devstack-non-apache https://logs.opendev.org/76/663776/4/gate/mistral-devstack-non-apache/86987f9/ : SUCCESS in 43m 14s\n- mistral-tox-unit-mysql https://logs.opendev.org/76/663776/4/gate/mistral-tox-unit-mysql/a0a6bc2/ : SUCCESS in 28m 28s\n- mistral-tox-unit-postgresql https://logs.opendev.org/76/663776/4/gate/mistral-tox-unit-postgresql/870b16b/ : SUCCESS in 29m 13s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/76/663776/4/gate/tripleo-ci-centos-7-scenario003-standalone/491d833/ : FAILURE in 25m 24s\n- mistral-devstack-kombu https://logs.opendev.org/76/663776/4/gate/mistral-devstack-kombu/05e475d/ : SUCCESS in 51m 39s\n- openstack-tox-lower-constraints https://logs.opendev.org/76/663776/4/gate/openstack-tox-lower-constraints/153155a/ : SUCCESS in 12m 27s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/76/663776/4/gate/tripleo-ci-centos-7-undercloud-containers/ba8467c/ : FAILURE in 28m 07s","accounts_in_message":[],"_revision_number":4},{"id":"f7a7d072111163c16358834be8ea3325695d62e4","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-08-08 10:05:31.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"eb7fce2559fb8df549af39a52111296f09ade399","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 11:28:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/76/663776/4/check/openstack-tox-pep8/3877d6a/ : SUCCESS in 7m 44s\n- openstack-tox-py27 https://logs.opendev.org/76/663776/4/check/openstack-tox-py27/7e82857/ : SUCCESS in 10m 05s\n- openstack-tox-py36 https://logs.opendev.org/76/663776/4/check/openstack-tox-py36/2853df7/ : SUCCESS in 8m 40s\n- openstack-tox-py37 https://logs.opendev.org/76/663776/4/check/openstack-tox-py37/4ba17b8/ : SUCCESS in 11m 27s\n- openstack-tox-docs https://logs.opendev.org/76/663776/4/check/openstack-tox-docs/ed80edd/html/ : SUCCESS in 4m 55s\n- openstack-tox-cover https://logs.opendev.org/76/663776/4/check/openstack-tox-cover/03bb22e/cover/ : SUCCESS in 8m 50s (non-voting)\n- mistral-devstack https://logs.opendev.org/76/663776/4/check/mistral-devstack/510e4be/ : SUCCESS in 44m 16s\n- mistral-devstack-non-apache https://logs.opendev.org/76/663776/4/check/mistral-devstack-non-apache/b1b7336/ : SUCCESS in 51m 41s\n- mistral-devstack-kombu https://logs.opendev.org/76/663776/4/check/mistral-devstack-kombu/aefcf83/ : SUCCESS in 54m 47s\n- mistral-tox-unit-mysql https://logs.opendev.org/76/663776/4/check/mistral-tox-unit-mysql/6a38b2f/ : SUCCESS in 32m 10s\n- mistral-tox-unit-postgresql https://logs.opendev.org/76/663776/4/check/mistral-tox-unit-postgresql/dbb4018/ : SUCCESS in 26m 40s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/76/663776/4/check/tripleo-ci-centos-7-scenario003-standalone/33e369c/ : FAILURE in 25m 11s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/76/663776/4/check/tripleo-ci-centos-7-containers-multinode/12ba3b7/ : FAILURE in 29m 11s\n- mistral-rally-task https://logs.opendev.org/76/663776/4/check/mistral-rally-task/d4d6e81/ : SUCCESS in 1h 13m 57s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/76/663776/4/check/openstack-tox-lower-constraints/138d2ac/ : SUCCESS in 11m 52s","accounts_in_message":[],"_revision_number":4},{"id":"a97566f50b8e3d4cf77eeb454367ae81a2ec91d3","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-08 11:50:43.000000000","message":"Patch Set 4:\n\nthis is the error \n\nWarning, treated as error:\n/builddir/build/BUILD/mistral-8.1.0.dev170/doc/source/user/wf_namespaces.rst:134:Error in \"code-block\" directive:\n1 argument(s) required, 0 supplied.\n.. code-block::\n wf1 -\u003e wf2 -\u003e wf3\nRPM build errors:\nerror: Bad exit status from /var/tmp/rpm-tmp.NiA3ec (%install)\n    Bad exit status from /var/tmp/rpm-tmp.NiA3ec (%install)\nChild return code was: 1\nEXCEPTION: [Error()]\nTraceback (most recent call last):\n  File \"/usr/lib/python2.7/site-packages/mockbuild/trace_decorator.py\", line 88, in trace\n    result \u003d func(*args, **kw)\n  File \"/usr/lib/python2.7/site-packages/mockbuild/util.py\", line 547, in do\n    raise exception.Error(\"Command failed. See logs for output.\\n # %s\" % (command,), child.returncode)\nError: Command failed. See logs for output.\n # bash --login -c /usr/bin/rpmbuild -bb --target x86_64 --nodeps /builddir/build/SPECS/openstack-mistral.spec","accounts_in_message":[],"_revision_number":4},{"id":"4dc2426b52e10b8ca4d34bca19b302d0f3fbeb11","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-08 11:51:39.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"d9dc7c123d5fcc40c1aa4220149a470b10de37b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 13:07:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/76/663776/5/check/openstack-tox-pep8/4f0b826/ : SUCCESS in 7m 49s\n- openstack-tox-py27 https://logs.opendev.org/76/663776/5/check/openstack-tox-py27/2d4e2c2/ : SUCCESS in 9m 08s\n- openstack-tox-py36 https://logs.opendev.org/76/663776/5/check/openstack-tox-py36/dbcf3be/ : SUCCESS in 9m 24s\n- openstack-tox-py37 https://logs.opendev.org/76/663776/5/check/openstack-tox-py37/96387d6/ : SUCCESS in 11m 52s\n- openstack-tox-docs https://logs.opendev.org/76/663776/5/check/openstack-tox-docs/850aaab/html/ : SUCCESS in 5m 10s\n- openstack-tox-cover https://logs.opendev.org/76/663776/5/check/openstack-tox-cover/3e203ff/cover/ : SUCCESS in 8m 33s (non-voting)\n- mistral-devstack https://logs.opendev.org/76/663776/5/check/mistral-devstack/f7fce4e/ : SUCCESS in 44m 57s\n- mistral-devstack-non-apache https://logs.opendev.org/76/663776/5/check/mistral-devstack-non-apache/d09b14f/ : SUCCESS in 1h 07m 10s\n- mistral-devstack-kombu https://logs.opendev.org/76/663776/5/check/mistral-devstack-kombu/9276713/ : FAILURE in 38m 42s\n- mistral-tox-unit-mysql https://logs.opendev.org/76/663776/5/check/mistral-tox-unit-mysql/aabd340/ : SUCCESS in 30m 22s\n- mistral-tox-unit-postgresql https://logs.opendev.org/76/663776/5/check/mistral-tox-unit-postgresql/634b3fb/ : SUCCESS in 27m 38s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/76/663776/5/check/tripleo-ci-centos-7-scenario003-standalone/947f7d2/ : FAILURE in 21m 03s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/76/663776/5/check/tripleo-ci-centos-7-containers-multinode/ea6768a/ : FAILURE in 29m 13s\n- mistral-rally-task https://logs.opendev.org/76/663776/5/check/mistral-rally-task/c923e36/ : SUCCESS in 1h 02m 37s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/76/663776/5/check/openstack-tox-lower-constraints/7e8f930/ : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":5},{"id":"a9347d6a0e2e7bdbf0d6f123cba258139825e66f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-08 13:10:44.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"405fba124ac9f3cf259a4c610601facf040a7a7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:58:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/76/663776/6/check/openstack-tox-pep8/2cc7de7/ : SUCCESS in 4m 40s\n- openstack-tox-py27 https://logs.opendev.org/76/663776/6/check/openstack-tox-py27/197a9f6/ : SUCCESS in 8m 55s\n- openstack-tox-py36 https://logs.opendev.org/76/663776/6/check/openstack-tox-py36/94ed4f9/ : SUCCESS in 8m 29s\n- openstack-tox-py37 https://logs.opendev.org/76/663776/6/check/openstack-tox-py37/9da8ec4/ : SUCCESS in 10m 15s\n- openstack-tox-docs https://logs.opendev.org/76/663776/6/check/openstack-tox-docs/4a55414/html/ : SUCCESS in 5m 45s\n- openstack-tox-cover https://logs.opendev.org/76/663776/6/check/openstack-tox-cover/6ceae56/cover/ : SUCCESS in 10m 08s (non-voting)\n- mistral-devstack https://logs.opendev.org/76/663776/6/check/mistral-devstack/b942924/ : SUCCESS in 45m 31s\n- mistral-devstack-non-apache https://logs.opendev.org/76/663776/6/check/mistral-devstack-non-apache/5cbee57/ : SUCCESS in 59m 08s\n- mistral-devstack-kombu https://logs.opendev.org/76/663776/6/check/mistral-devstack-kombu/8806b8e/ : SUCCESS in 46m 01s\n- mistral-tox-unit-mysql https://logs.opendev.org/76/663776/6/check/mistral-tox-unit-mysql/f904a4c/ : SUCCESS in 30m 09s\n- mistral-tox-unit-postgresql https://logs.opendev.org/76/663776/6/check/mistral-tox-unit-postgresql/0a64ee1/ : SUCCESS in 29m 22s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/76/663776/6/check/tripleo-ci-centos-7-scenario003-standalone/d037906/ : SUCCESS in 1h 20m 57s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/76/663776/6/check/tripleo-ci-centos-7-containers-multinode/e7f7d79/ : SUCCESS in 2h 39m 26s\n- mistral-rally-task https://logs.opendev.org/76/663776/6/check/mistral-rally-task/da83f52/ : SUCCESS in 1h 02m 21s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/76/663776/6/check/openstack-tox-lower-constraints/f6a8af9/ : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":6},{"id":"12cfb807fa6e8822e911de8645c9caa416095169","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-08-08 16:12:11.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"35744fe9e5b63c735c180318a5a51c702170d814","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-09 05:04:08.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"2cbedb937c037e5e8196c93cac03c08f7e193fc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 05:04:19.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"856929ca5b04cec5992e3e6ccc5408105bc7118e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 06:51:04.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/76/663776/6/gate/openstack-tox-pep8/c6e22c4/ : SUCCESS in 5m 44s\n- openstack-tox-py27 https://logs.opendev.org/76/663776/6/gate/openstack-tox-py27/15a9a83/ : SUCCESS in 9m 21s\n- openstack-tox-py36 https://logs.opendev.org/76/663776/6/gate/openstack-tox-py36/adb3654/ : SUCCESS in 9m 34s\n- openstack-tox-py37 https://logs.opendev.org/76/663776/6/gate/openstack-tox-py37/a0ef2e7/ : SUCCESS in 10m 46s\n- openstack-tox-docs https://logs.opendev.org/76/663776/6/gate/openstack-tox-docs/938fd8a/ : FAILURE in 5m 28s\n- mistral-devstack https://logs.opendev.org/76/663776/6/gate/mistral-devstack/907b87c/ : SUCCESS in 43m 07s\n- mistral-devstack-non-apache https://logs.opendev.org/76/663776/6/gate/mistral-devstack-non-apache/f9ef5e4/ : SUCCESS in 40m 52s\n- mistral-tox-unit-mysql https://logs.opendev.org/76/663776/6/gate/mistral-tox-unit-mysql/d37d76f/ : SUCCESS in 30m 37s\n- mistral-tox-unit-postgresql https://logs.opendev.org/76/663776/6/gate/mistral-tox-unit-postgresql/eaa9391/ : SUCCESS in 27m 36s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/76/663776/6/gate/tripleo-ci-centos-7-scenario003-standalone/da40208/ : SUCCESS in 1h 23m 33s\n- mistral-devstack-kombu https://logs.opendev.org/76/663776/6/gate/mistral-devstack-kombu/bffd6d6/ : SUCCESS in 42m 55s\n- openstack-tox-lower-constraints https://logs.opendev.org/76/663776/6/gate/openstack-tox-lower-constraints/4e91473/ : SUCCESS in 9m 57s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/76/663776/6/gate/tripleo-ci-centos-7-undercloud-containers/1741ec1/ : SUCCESS in 1h 22m 22s","accounts_in_message":[],"_revision_number":6},{"id":"e9bf32fd8bb180d42923832baf55ab1b5dc19538","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-09 07:07:38.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"52b2b3f6b63678bc1088d2e25482829e56577861","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 09:33:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/76/663776/6/check/openstack-tox-pep8/33b8255/ : SUCCESS in 4m 57s\n- openstack-tox-py27 https://logs.opendev.org/76/663776/6/check/openstack-tox-py27/960b288/ : SUCCESS in 10m 08s\n- openstack-tox-py36 https://logs.opendev.org/76/663776/6/check/openstack-tox-py36/1a3829f/ : SUCCESS in 9m 31s\n- openstack-tox-py37 https://logs.opendev.org/76/663776/6/check/openstack-tox-py37/fa613dd/ : SUCCESS in 10m 30s\n- openstack-tox-docs https://logs.opendev.org/76/663776/6/check/openstack-tox-docs/aeb5023/ : FAILURE in 4m 52s\n- openstack-tox-cover https://logs.opendev.org/76/663776/6/check/openstack-tox-cover/91e691b/cover/ : SUCCESS in 9m 00s (non-voting)\n- mistral-devstack https://logs.opendev.org/76/663776/6/check/mistral-devstack/e979391/ : SUCCESS in 50m 21s\n- mistral-devstack-non-apache https://logs.opendev.org/76/663776/6/check/mistral-devstack-non-apache/80a1ba0/ : SUCCESS in 52m 05s\n- mistral-devstack-kombu https://logs.opendev.org/76/663776/6/check/mistral-devstack-kombu/527a7cf/ : SUCCESS in 45m 40s\n- mistral-tox-unit-mysql https://logs.opendev.org/76/663776/6/check/mistral-tox-unit-mysql/010f043/ : SUCCESS in 28m 14s\n- mistral-tox-unit-postgresql https://logs.opendev.org/76/663776/6/check/mistral-tox-unit-postgresql/d8dd9cd/ : SUCCESS in 31m 00s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/76/663776/6/check/tripleo-ci-centos-7-scenario003-standalone/734d30e/ : SUCCESS in 1h 23m 19s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/76/663776/6/check/tripleo-ci-centos-7-containers-multinode/ccbf065/ : FAILURE in 2h 22m 42s\n- mistral-rally-task https://logs.opendev.org/76/663776/6/check/mistral-rally-task/b1eead7/ : SUCCESS in 1h 02m 58s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/76/663776/6/check/openstack-tox-lower-constraints/5999aff/ : SUCCESS in 11m 35s","accounts_in_message":[],"_revision_number":6},{"id":"2b5a9acc3bbb04c9908fa3cb7449ae1fac751880","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-08-09 12:01:30.000000000","message":"Patch Set 6:\n\nThere\u0027s an weird error on the tox docs\n\nhttps://logs.opendev.org/76/663776/6/check/openstack-tox-docs/aeb5023/job-output.txt.gz#_2019-08-09_07_18_31_256551","accounts_in_message":[],"_revision_number":6},{"id":"1785b681a18ecb6f3f4a21efc837f178339aecd4","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-11 06:57:15.000000000","message":"Patch Set 6:\n\n\u003e There\u0027s an weird error on the tox docs\n \u003e \n \u003e https://logs.opendev.org/76/663776/6/check/openstack-tox-docs/aeb5023/job-output.txt.gz#_2019-08-09_07_18_31_256551\n\nthere is an issue with zaqarclient and jsonschema see this:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-August/008395.html","accounts_in_message":[],"_revision_number":6},{"id":"4d07591813951fcf1ddb2051a684dcafacac7456","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-11 07:22:01.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"712e0e03674e1cdd9f305fe1c46e86235a131b3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-11 09:42:49.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/76/663776/7/check/openstack-tox-pep8/7ff2a44/ : SUCCESS in 4m 04s\n- openstack-tox-py27 https://logs.opendev.org/76/663776/7/check/openstack-tox-py27/090aa18/ : SUCCESS in 9m 55s\n- openstack-tox-py36 https://logs.opendev.org/76/663776/7/check/openstack-tox-py36/a5725a3/ : SUCCESS in 9m 17s\n- openstack-tox-py37 https://logs.opendev.org/76/663776/7/check/openstack-tox-py37/fcdabb8/ : SUCCESS in 9m 50s\n- openstack-tox-docs https://logs.opendev.org/76/663776/7/check/openstack-tox-docs/396421a/html/ : SUCCESS in 4m 07s\n- requirements-check https://logs.opendev.org/76/663776/7/check/requirements-check/5d98a85/ : SUCCESS in 3m 38s\n- openstack-tox-cover https://logs.opendev.org/76/663776/7/check/openstack-tox-cover/32ec698/cover/ : SUCCESS in 9m 56s (non-voting)\n- mistral-devstack https://logs.opendev.org/76/663776/7/check/mistral-devstack/862af9a/ : SUCCESS in 42m 37s\n- mistral-devstack-non-apache https://logs.opendev.org/76/663776/7/check/mistral-devstack-non-apache/5d89086/ : SUCCESS in 43m 04s\n- mistral-devstack-kombu https://logs.opendev.org/76/663776/7/check/mistral-devstack-kombu/5e9cb9a/ : SUCCESS in 43m 21s\n- mistral-tox-unit-mysql https://logs.opendev.org/76/663776/7/check/mistral-tox-unit-mysql/f0f6a40/ : SUCCESS in 31m 13s\n- mistral-tox-unit-postgresql https://logs.opendev.org/76/663776/7/check/mistral-tox-unit-postgresql/1fdcedd/ : SUCCESS in 29m 42s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/76/663776/7/check/tripleo-ci-centos-7-scenario003-standalone/6f04363/ : FAILURE in 12m 45s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/76/663776/7/check/tripleo-ci-centos-7-containers-multinode/cb39ea6/ : SUCCESS in 2h 11m 47s\n- mistral-rally-task https://logs.opendev.org/76/663776/7/check/mistral-rally-task/df30c13/ : SUCCESS in 1h 00m 00s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/76/663776/7/check/openstack-tox-lower-constraints/aac76e0/ : SUCCESS in 9m 18s","accounts_in_message":[],"_revision_number":7},{"id":"63e5d3418fde1719017099d2f51e119d8b9bcb96","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-11 09:44:15.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"f68c28f38e5f26e71452f2576265e534766ad98e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-11 11:57:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/76/663776/7/check/openstack-tox-pep8/d7f3cb8/ : SUCCESS in 4m 17s\n- openstack-tox-py27 https://logs.opendev.org/76/663776/7/check/openstack-tox-py27/f7f6435/ : SUCCESS in 8m 02s\n- openstack-tox-py36 https://logs.opendev.org/76/663776/7/check/openstack-tox-py36/ccbba9e/ : SUCCESS in 7m 44s\n- openstack-tox-py37 https://logs.opendev.org/76/663776/7/check/openstack-tox-py37/e9418f5/ : SUCCESS in 8m 16s\n- openstack-tox-docs https://logs.opendev.org/76/663776/7/check/openstack-tox-docs/2896f85/html/ : SUCCESS in 4m 03s\n- requirements-check https://logs.opendev.org/76/663776/7/check/requirements-check/34ec111/ : SUCCESS in 2m 28s\n- openstack-tox-cover https://logs.opendev.org/76/663776/7/check/openstack-tox-cover/8ccaf97/cover/ : SUCCESS in 8m 02s (non-voting)\n- mistral-devstack https://logs.opendev.org/76/663776/7/check/mistral-devstack/8581b70/ : SUCCESS in 47m 53s\n- mistral-devstack-non-apache https://logs.opendev.org/76/663776/7/check/mistral-devstack-non-apache/e482eb7/ : SUCCESS in 45m 49s\n- mistral-devstack-kombu https://logs.opendev.org/76/663776/7/check/mistral-devstack-kombu/972e9a1/ : SUCCESS in 41m 29s\n- mistral-tox-unit-mysql https://logs.opendev.org/76/663776/7/check/mistral-tox-unit-mysql/06d9b2f/ : SUCCESS in 27m 12s\n- mistral-tox-unit-postgresql https://logs.opendev.org/76/663776/7/check/mistral-tox-unit-postgresql/1efef8f/ : SUCCESS in 26m 13s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/76/663776/7/check/tripleo-ci-centos-7-scenario003-standalone/fdc33ff/ : SUCCESS in 1h 04m 05s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/76/663776/7/check/tripleo-ci-centos-7-containers-multinode/085fc0f/ : SUCCESS in 2h 08m 32s\n- mistral-rally-task https://logs.opendev.org/76/663776/7/check/mistral-rally-task/b0d9b29/ : SUCCESS in 1h 02m 45s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/76/663776/7/check/openstack-tox-lower-constraints/298cd38/ : SUCCESS in 10m 31s","accounts_in_message":[],"_revision_number":7},{"id":"847aaea8b0b3191b018c3d0202096ec85291a50e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-12 03:28:29.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"5b65699f544e8e3ba9056b85efc7fe13da078164","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 03:28:40.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f0cea37ccb82b9d2618b1705d87393006d6edbe7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 05:39:47.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/76/663776/7/gate/openstack-tox-pep8/0b35b1c/ : SUCCESS in 4m 09s\n- openstack-tox-py27 https://logs.opendev.org/76/663776/7/gate/openstack-tox-py27/d5beb4a/ : SUCCESS in 7m 02s\n- openstack-tox-py36 https://logs.opendev.org/76/663776/7/gate/openstack-tox-py36/704d90c/ : SUCCESS in 7m 15s\n- openstack-tox-py37 https://logs.opendev.org/76/663776/7/gate/openstack-tox-py37/88170e3/ : SUCCESS in 8m 06s\n- openstack-tox-docs https://logs.opendev.org/76/663776/7/gate/openstack-tox-docs/f7807b1/html/ : SUCCESS in 3m 40s\n- requirements-check https://logs.opendev.org/76/663776/7/gate/requirements-check/edc585d/ : SUCCESS in 2m 39s\n- mistral-devstack https://logs.opendev.org/76/663776/7/gate/mistral-devstack/fa614d5/ : SUCCESS in 38m 24s\n- mistral-devstack-non-apache https://logs.opendev.org/76/663776/7/gate/mistral-devstack-non-apache/2bb9134/ : SUCCESS in 39m 58s\n- mistral-tox-unit-mysql https://logs.opendev.org/76/663776/7/gate/mistral-tox-unit-mysql/2889f67/ : FAILURE in 25m 57s\n- mistral-tox-unit-postgresql https://logs.opendev.org/76/663776/7/gate/mistral-tox-unit-postgresql/a2db97a/ : SUCCESS in 25m 26s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/76/663776/7/gate/tripleo-ci-centos-7-scenario003-standalone/03d83b6/ : SUCCESS in 1h 12m 51s\n- mistral-devstack-kombu https://logs.opendev.org/76/663776/7/gate/mistral-devstack-kombu/e87670e/ : SUCCESS in 43m 14s\n- openstack-tox-lower-constraints https://logs.opendev.org/76/663776/7/gate/openstack-tox-lower-constraints/b65b309/ : SUCCESS in 9m 00s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/76/663776/7/gate/tripleo-ci-centos-7-undercloud-containers/27a5ddd/ : FAILURE in 2h 10m 10s","accounts_in_message":[],"_revision_number":7},{"id":"ae32028913b4afd0d533aacd923176ec49127ea2","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-12 05:57:46.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"07a741ad81c9cf6796b8a8911f99da00671d4e75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 08:16:46.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/76/663776/7/check/openstack-tox-pep8/f383631/ : SUCCESS in 4m 33s\n- openstack-tox-py27 https://logs.opendev.org/76/663776/7/check/openstack-tox-py27/632a8f6/ : SUCCESS in 8m 16s\n- openstack-tox-py36 https://logs.opendev.org/76/663776/7/check/openstack-tox-py36/3a066a1/ : SUCCESS in 10m 03s\n- openstack-tox-py37 https://logs.opendev.org/76/663776/7/check/openstack-tox-py37/fd6d066/ : SUCCESS in 9m 59s\n- openstack-tox-docs https://logs.opendev.org/76/663776/7/check/openstack-tox-docs/d88e3d1/html/ : SUCCESS in 4m 19s\n- requirements-check https://logs.opendev.org/76/663776/7/check/requirements-check/2b4845e/ : SUCCESS in 2m 53s\n- openstack-tox-cover https://logs.opendev.org/76/663776/7/check/openstack-tox-cover/02ee8c5/cover/ : SUCCESS in 9m 40s (non-voting)\n- mistral-devstack https://logs.opendev.org/76/663776/7/check/mistral-devstack/c512f9f/ : SUCCESS in 44m 32s\n- mistral-devstack-non-apache https://logs.opendev.org/76/663776/7/check/mistral-devstack-non-apache/afac639/ : FAILURE in 32m 03s\n- mistral-devstack-kombu https://logs.opendev.org/76/663776/7/check/mistral-devstack-kombu/f61a824/ : SUCCESS in 41m 59s\n- mistral-tox-unit-mysql https://logs.opendev.org/76/663776/7/check/mistral-tox-unit-mysql/cc06822/ : SUCCESS in 28m 41s\n- mistral-tox-unit-postgresql https://logs.opendev.org/76/663776/7/check/mistral-tox-unit-postgresql/8bf1fd7/ : SUCCESS in 26m 38s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/76/663776/7/check/tripleo-ci-centos-7-scenario003-standalone/1b1488f/ : FAILURE in 11m 12s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/76/663776/7/check/tripleo-ci-centos-7-containers-multinode/bff3553/ : SUCCESS in 2h 16m 48s\n- mistral-rally-task https://logs.opendev.org/76/663776/7/check/mistral-rally-task/cbf69b3/ : SUCCESS in 1h 02m 05s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/76/663776/7/check/openstack-tox-lower-constraints/c9da13d/ : SUCCESS in 9m 56s","accounts_in_message":[],"_revision_number":7},{"id":"374d102389b56be62806a441e052d5c548b09f1b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-12 08:44:01.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"34bf5be681af12bd86654a089b955c32a56395df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 11:07:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/76/663776/7/check/openstack-tox-pep8/00e7f9f/ : SUCCESS in 5m 21s\n- openstack-tox-py27 https://logs.opendev.org/76/663776/7/check/openstack-tox-py27/ac7f31f/ : SUCCESS in 9m 00s\n- openstack-tox-py36 https://logs.opendev.org/76/663776/7/check/openstack-tox-py36/c26739a/ : SUCCESS in 7m 41s\n- openstack-tox-py37 https://logs.opendev.org/76/663776/7/check/openstack-tox-py37/63d04c5/ : SUCCESS in 8m 48s\n- openstack-tox-docs https://logs.opendev.org/76/663776/7/check/openstack-tox-docs/aebb057/html/ : SUCCESS in 4m 44s\n- requirements-check https://logs.opendev.org/76/663776/7/check/requirements-check/1892d0a/ : SUCCESS in 2m 28s\n- openstack-tox-cover https://logs.opendev.org/76/663776/7/check/openstack-tox-cover/8b5b1a5/cover/ : SUCCESS in 9m 10s (non-voting)\n- mistral-devstack https://logs.opendev.org/76/663776/7/check/mistral-devstack/c7639b8/ : SUCCESS in 43m 32s\n- mistral-devstack-non-apache https://logs.opendev.org/76/663776/7/check/mistral-devstack-non-apache/4cb38db/ : SUCCESS in 47m 16s\n- mistral-devstack-kombu https://logs.opendev.org/76/663776/7/check/mistral-devstack-kombu/722206f/ : SUCCESS in 45m 14s\n- mistral-tox-unit-mysql https://logs.opendev.org/76/663776/7/check/mistral-tox-unit-mysql/d07958a/ : SUCCESS in 28m 35s\n- mistral-tox-unit-postgresql https://logs.opendev.org/76/663776/7/check/mistral-tox-unit-postgresql/a94dc7b/ : SUCCESS in 26m 50s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/76/663776/7/check/tripleo-ci-centos-7-scenario003-standalone/4667fba/ : SUCCESS in 1h 11m 59s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/76/663776/7/check/tripleo-ci-centos-7-containers-multinode/7aeb00b/ : SUCCESS in 2h 17m 41s\n- mistral-rally-task https://logs.opendev.org/76/663776/7/check/mistral-rally-task/1ebc508/ : SUCCESS in 1h 16m 14s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/76/663776/7/check/openstack-tox-lower-constraints/78d3ce2/ : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":7},{"id":"e1f41f9ed1530bea49da253aefc4df1f7a70e5f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 11:07:19.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"777ec85fa6fc78c6448da92a3a24953c4cd0abe8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 12:58:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"c23d57aa1f31b0ad6d7b0e663ba5c82e532aca0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 12:58:38.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/76/663776/7/gate/openstack-tox-pep8/aaf348b/ : SUCCESS in 8m 38s\n- openstack-tox-py27 https://logs.opendev.org/76/663776/7/gate/openstack-tox-py27/8b74832/ : SUCCESS in 12m 30s\n- openstack-tox-py36 https://logs.opendev.org/76/663776/7/gate/openstack-tox-py36/a063187/ : SUCCESS in 9m 05s\n- openstack-tox-py37 https://logs.opendev.org/76/663776/7/gate/openstack-tox-py37/059890f/ : SUCCESS in 14m 13s\n- openstack-tox-docs https://logs.opendev.org/76/663776/7/gate/openstack-tox-docs/76522c5/html/ : SUCCESS in 4m 30s\n- requirements-check https://logs.opendev.org/76/663776/7/gate/requirements-check/8bf6c9d/ : SUCCESS in 2m 43s\n- mistral-devstack https://logs.opendev.org/76/663776/7/gate/mistral-devstack/d60f110/ : SUCCESS in 50m 25s\n- mistral-devstack-non-apache https://logs.opendev.org/76/663776/7/gate/mistral-devstack-non-apache/3ef32c9/ : SUCCESS in 51m 36s\n- mistral-tox-unit-mysql https://logs.opendev.org/76/663776/7/gate/mistral-tox-unit-mysql/61713ca/ : SUCCESS in 28m 09s\n- mistral-tox-unit-postgresql https://logs.opendev.org/76/663776/7/gate/mistral-tox-unit-postgresql/ff91962/ : SUCCESS in 27m 44s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/76/663776/7/gate/tripleo-ci-centos-7-scenario003-standalone/62a669a/ : SUCCESS in 1h 20m 25s\n- mistral-devstack-kombu https://logs.opendev.org/76/663776/7/gate/mistral-devstack-kombu/8db9e65/ : SUCCESS in 53m 40s\n- openstack-tox-lower-constraints https://logs.opendev.org/76/663776/7/gate/openstack-tox-lower-constraints/8a1ff76/ : SUCCESS in 14m 05s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/76/663776/7/gate/tripleo-ci-centos-7-undercloud-containers/f65c960/ : SUCCESS in 1h 46m 00s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"4e926a1f136a54ddf83e9eae43c33d9636d48aaf","revisions":{"3c18cf20d10189c164a5dea769fea2532f589ebb":{"kind":"REWORK","_number":1,"created":"2019-06-06 21:30:38.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/76/663776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/663776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/663776/1"}}},"commit":{"parents":[{"commit":"f85caa02c9aaa64db4300b43854327d132a71ed4","subject":"Merge \"Skip context evaluation for non-conditional transitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f85caa02c9aaa64db4300b43854327d132a71ed4"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 21:30:20.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 21:30:20.000000000","tz":180},"subject":"Fail-on policy","message":"Fail-on policy\n\nFail-on policy allows to fail success tasks by condition. It is useful\nin cases we have to fail task if its result is unacceptable and it makes\nworkflow definition more readable.\n\nChange-Id: I57b4f3d1533982d3b9b7063925f8d70f044aefea\nImplements: blueprint fail-on-policy\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3c18cf20d10189c164a5dea769fea2532f589ebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3c18cf20d10189c164a5dea769fea2532f589ebb"}]},"branch":"refs/heads/master"},"d0fbea703b6b009c8818ed26f22615dca923fedc":{"kind":"REWORK","_number":2,"created":"2019-06-07 07:55:00.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/76/663776/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/663776/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/663776/2"}}},"commit":{"parents":[{"commit":"f85caa02c9aaa64db4300b43854327d132a71ed4","subject":"Merge \"Skip context evaluation for non-conditional transitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f85caa02c9aaa64db4300b43854327d132a71ed4"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 21:30:20.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-07 07:54:35.000000000","tz":180},"subject":"Fail-on policy","message":"Fail-on policy\n\nFail-on policy allows to fail success tasks by condition. It is useful\nin cases we have to fail task if its result is unacceptable and it makes\nworkflow definition more readable.\n\nChange-Id: I57b4f3d1533982d3b9b7063925f8d70f044aefea\nImplements: blueprint fail-on-policy\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d0fbea703b6b009c8818ed26f22615dca923fedc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d0fbea703b6b009c8818ed26f22615dca923fedc"}]},"branch":"refs/heads/master"},"517b8e73dcb4c4b2b1a13376bc2c1a9118bc6661":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-11 23:29:22.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/76/663776/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/663776/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/663776/3"}}},"commit":{"parents":[{"commit":"6a9aa4c295b51e9b11ab4a30e5c002bcc2924609","subject":"Merge \"Initialize the lazily loaded execution \"input\" field in API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a9aa4c295b51e9b11ab4a30e5c002bcc2924609"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 21:30:20.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-07-11 23:29:22.000000000","tz":0},"subject":"Fail-on policy","message":"Fail-on policy\n\nFail-on policy allows to fail success tasks by condition. It is useful\nin cases we have to fail task if its result is unacceptable and it makes\nworkflow definition more readable.\n\nChange-Id: I57b4f3d1533982d3b9b7063925f8d70f044aefea\nImplements: blueprint fail-on-policy\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/517b8e73dcb4c4b2b1a13376bc2c1a9118bc6661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/517b8e73dcb4c4b2b1a13376bc2c1a9118bc6661"}]},"branch":"refs/heads/master"},"293f1ae451bde75a680485f394b2be0a112c2c7a":{"kind":"REWORK","_number":4,"created":"2019-08-03 22:05:59.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/76/663776/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/663776/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/663776/4"}}},"commit":{"parents":[{"commit":"1a2ff63733e53ee39b59f877c56065afb443343a","subject":"A1 B should be tasks rather than workflow levels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1a2ff63733e53ee39b59f877c56065afb443343a"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 21:30:20.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-08-03 22:05:46.000000000","tz":180},"subject":"Fail-on policy","message":"Fail-on policy\n\nFail-on policy allows to fail success tasks by condition. It is useful\nin cases we have to fail task if its result is unacceptable and it makes\nworkflow definition more readable.\n\nChange-Id: I57b4f3d1533982d3b9b7063925f8d70f044aefea\nImplements: blueprint fail-on-policy\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/293f1ae451bde75a680485f394b2be0a112c2c7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/293f1ae451bde75a680485f394b2be0a112c2c7a"}]},"branch":"refs/heads/master"},"e46cc8306383d14408c80750868c113f00afedd3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-08 11:51:39.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/76/663776/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/663776/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/663776/5"}}},"commit":{"parents":[{"commit":"ee94087c8972579ce658cf2d0bb59540d8f974e0","subject":"Merge \"Create docs for the workflow namespaces feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee94087c8972579ce658cf2d0bb59540d8f974e0"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 21:30:20.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-08 11:51:39.000000000","tz":0},"subject":"Fail-on policy","message":"Fail-on policy\n\nFail-on policy allows to fail success tasks by condition. It is useful\nin cases we have to fail task if its result is unacceptable and it makes\nworkflow definition more readable.\n\nChange-Id: I57b4f3d1533982d3b9b7063925f8d70f044aefea\nImplements: blueprint fail-on-policy\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e46cc8306383d14408c80750868c113f00afedd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e46cc8306383d14408c80750868c113f00afedd3"}]},"branch":"refs/heads/master"},"d3a721f3e22dc434b629aff432aa0c45a31c11a2":{"kind":"REWORK","_number":6,"created":"2019-08-08 13:10:44.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/76/663776/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/663776/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/663776/6"}}},"commit":{"parents":[{"commit":"ee94087c8972579ce658cf2d0bb59540d8f974e0","subject":"Merge \"Create docs for the workflow namespaces feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee94087c8972579ce658cf2d0bb59540d8f974e0"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 21:30:20.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-08 13:10:36.000000000","tz":0},"subject":"Fail-on policy","message":"Fail-on policy\n\nFail-on policy allows to fail success tasks by condition. It is useful\nin cases we have to fail task if its result is unacceptable and it makes\nworkflow definition more readable.\n\nChange-Id: I57b4f3d1533982d3b9b7063925f8d70f044aefea\nImplements: blueprint fail-on-policy\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d3a721f3e22dc434b629aff432aa0c45a31c11a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d3a721f3e22dc434b629aff432aa0c45a31c11a2"}]},"branch":"refs/heads/master"},"4e926a1f136a54ddf83e9eae43c33d9636d48aaf":{"kind":"REWORK","_number":7,"created":"2019-08-11 07:22:01.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/76/663776/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/663776/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/663776/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/663776/7"}}},"commit":{"parents":[{"commit":"ee94087c8972579ce658cf2d0bb59540d8f974e0","subject":"Merge \"Create docs for the workflow namespaces feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee94087c8972579ce658cf2d0bb59540d8f974e0"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 21:30:20.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-11 07:21:57.000000000","tz":0},"subject":"Fail-on policy","message":"Fail-on policy\n\nFail-on policy allows to fail success tasks by condition. It is useful\nin cases we have to fail task if its result is unacceptable and it makes\nworkflow definition more readable.\n\nChange-Id: I57b4f3d1533982d3b9b7063925f8d70f044aefea\nImplements: blueprint fail-on-policy\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e926a1f136a54ddf83e9eae43c33d9636d48aaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e926a1f136a54ddf83e9eae43c33d9636d48aaf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
