)]}'
{"id":"openstack%2Fmistral~663841","triplet_id":"openstack%2Fmistral~master~Idd4187f395aca9e359252f98cb657037bd14a998","project":"openstack/mistral","branch":"master","topic":"task_executions_cache","hashtags":[],"change_id":"Idd4187f395aca9e359252f98cb657037bd14a998","subject":"WIP: add lightweight task execution cache","status":"ABANDONED","created":"2019-06-07 06:26:31.000000000","updated":"2026-03-28 19:40:07.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"70b731e7809e1ac9843a8a1a54fc67332316fe34","_number":663841,"virtual_id_number":663841,"owner":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-07 08:50:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 08:50:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-09 08:10:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-05-09 08:10:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"CC"}],"messages":[{"id":"fd876fd174e0f4711a5de1350f3f4f43fc18e8dc","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-07 06:26:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22471d913adf2003180bc6565f3820a4d47ac861","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 08:50:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/41/663841/1/check/openstack-tox-pep8/b7bf626/ : SUCCESS in 5m 59s\n- openstack-tox-py27 http://logs.openstack.org/41/663841/1/check/openstack-tox-py27/211697d/ : SUCCESS in 9m 03s\n- openstack-tox-py36 http://logs.openstack.org/41/663841/1/check/openstack-tox-py36/4837ff9/ : SUCCESS in 8m 50s\n- openstack-tox-py37 http://logs.openstack.org/41/663841/1/check/openstack-tox-py37/b92d69a/ : SUCCESS in 9m 45s\n- openstack-tox-docs http://logs.openstack.org/41/663841/1/check/openstack-tox-docs/4052bee/html/ : SUCCESS in 5m 38s\n- openstack-tox-cover http://logs.openstack.org/41/663841/1/check/openstack-tox-cover/53405ff/cover/ : SUCCESS in 9m 52s (non-voting)\n- mistral-devstack http://logs.openstack.org/41/663841/1/check/mistral-devstack/0426888/ : SUCCESS in 46m 17s\n- mistral-devstack-non-apache http://logs.openstack.org/41/663841/1/check/mistral-devstack-non-apache/f1553fb/ : SUCCESS in 41m 41s\n- mistral-devstack-kombu http://logs.openstack.org/41/663841/1/check/mistral-devstack-kombu/d9d672d/ : SUCCESS in 42m 33s\n- mistral-tox-unit-mysql http://logs.openstack.org/41/663841/1/check/mistral-tox-unit-mysql/cc0304f/ : SUCCESS in 28m 59s\n- mistral-tox-unit-postgresql http://logs.openstack.org/41/663841/1/check/mistral-tox-unit-postgresql/336e471/ : SUCCESS in 27m 41s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/41/663841/1/check/tripleo-ci-centos-7-scenario003-standalone/1076958/ : SUCCESS in 1h 24m 51s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/41/663841/1/check/tripleo-ci-centos-7-containers-multinode/6cd6c71/ : SUCCESS in 2h 00m 23s\n- mistral-rally-task http://logs.openstack.org/41/663841/1/check/mistral-rally-task/f3ed462/ : SUCCESS in 51m 13s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/41/663841/1/check/openstack-tox-lower-constraints/9d65b34/ : SUCCESS in 13m 14s","accounts_in_message":[],"_revision_number":1},{"id":"dec2c6880872f137f4351d67d61b23b746c3bc1c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-09 08:10:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70b731e7809e1ac9843a8a1a54fc67332316fe34","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-03-28 19:40:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5359c7c76dc2004769b48021729cef6485d62cb1","revisions":{"5359c7c76dc2004769b48021729cef6485d62cb1":{"kind":"REWORK","_number":1,"created":"2019-06-07 06:26:31.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/41/663841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/41/663841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/41/663841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/41/663841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/41/663841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/41/663841/1"}}},"commit":{"parents":[{"commit":"f85caa02c9aaa64db4300b43854327d132a71ed4","subject":"Merge \"Skip context evaluation for non-conditional transitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f85caa02c9aaa64db4300b43854327d132a71ed4"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-07 06:22:03.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-07 06:22:03.000000000","tz":120},"subject":"WIP: add lightweight task execution cache","message":"WIP: add lightweight task execution cache\n\nIt\u0027s possible to store ids, names and states of completed task\nexecutions in local cache to prevent them from being fetched from\nthe database multiple times.\n\nChange-Id: Idd4187f395aca9e359252f98cb657037bd14a998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5359c7c76dc2004769b48021729cef6485d62cb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5359c7c76dc2004769b48021729cef6485d62cb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
