)]}'
{"id":"openstack%2Fmistral~663877","triplet_id":"openstack%2Fmistral~master~Ia7fb914843e56650c8f6ff5e8c16d1b57001c0fc","project":"openstack/mistral","branch":"master","topic":"wf_create_optimization","hashtags":[],"change_id":"Ia7fb914843e56650c8f6ff5e8c16d1b57001c0fc","subject":"Simple optimization of creating/updating workflows","status":"MERGED","created":"2019-06-07 09:54:26.000000000","updated":"2019-06-10 15:27:51.000000000","submitted":"2019-06-10 15:27:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663877-1560180471078-d9eb55b5","meta_rev_id":"bde4d99a0a3992d606b6f9c83a715eec4402891e","_number":663877,"virtual_id_number":663877,"owner":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-06-10 15:27:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-06-10 09:04:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-06-10 08:57:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-10 11:18:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"date":"2019-06-10 09:00:05.000000000","_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"date":"2019-06-10 09:00:05.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-06-10 09:04:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2019-06-10 09:00:05.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2019-06-10 09:00:05.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 08:57:22.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-06-10 09:00:05.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-06-10 09:00:05.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-06-10 09:00:05.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-06-10 09:00:05.000000000","updated_by":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"reviewer":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"state":"REVIEWER"},{"updated":"2019-06-10 09:04:06.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-06-10 11:18:04.000000000","updated_by":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"reviewer":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"state":"REVIEWER"},{"updated":"2019-06-10 15:27:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc2415aa81df1c11a45e3e7f0e38447bb9b3d70b","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-07 09:54:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdf6242828688ff5dde64c71a0ec7c6f0956b4d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 12:34:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/77/663877/1/check/openstack-tox-pep8/90adeae/ : SUCCESS in 6m 17s\n- openstack-tox-py27 http://logs.openstack.org/77/663877/1/check/openstack-tox-py27/64e14f8/ : SUCCESS in 9m 04s\n- openstack-tox-py36 http://logs.openstack.org/77/663877/1/check/openstack-tox-py36/e18e7f1/ : SUCCESS in 8m 29s\n- openstack-tox-py37 http://logs.openstack.org/77/663877/1/check/openstack-tox-py37/833ac05/ : SUCCESS in 9m 05s\n- openstack-tox-docs http://logs.openstack.org/77/663877/1/check/openstack-tox-docs/d90b3bc/html/ : SUCCESS in 6m 11s\n- openstack-tox-cover http://logs.openstack.org/77/663877/1/check/openstack-tox-cover/5b24002/cover/ : SUCCESS in 9m 32s (non-voting)\n- mistral-devstack http://logs.openstack.org/77/663877/1/check/mistral-devstack/d857bff/ : SUCCESS in 42m 41s\n- mistral-devstack-non-apache http://logs.openstack.org/77/663877/1/check/mistral-devstack-non-apache/7fa41d5/ : SUCCESS in 45m 11s\n- mistral-devstack-kombu http://logs.openstack.org/77/663877/1/check/mistral-devstack-kombu/d15a2b4/ : SUCCESS in 43m 04s\n- mistral-tox-unit-mysql http://logs.openstack.org/77/663877/1/check/mistral-tox-unit-mysql/e577f0e/ : SUCCESS in 35m 00s\n- mistral-tox-unit-postgresql http://logs.openstack.org/77/663877/1/check/mistral-tox-unit-postgresql/e38d63b/ : SUCCESS in 26m 24s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/77/663877/1/check/tripleo-ci-centos-7-scenario003-standalone/8c09387/ : SUCCESS in 1h 15m 50s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/77/663877/1/check/tripleo-ci-centos-7-containers-multinode/33b7880/ : SUCCESS in 2h 07m 08s\n- mistral-rally-task http://logs.openstack.org/77/663877/1/check/mistral-rally-task/fc28cae/ : SUCCESS in 1h 07m 04s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/77/663877/1/check/openstack-tox-lower-constraints/05a65c6/ : SUCCESS in 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"638669f4abd2dbcff105a2c3bcf91b9c1d815ec6","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-10 03:57:50.000000000","message":"Patch Set 1: Code-Review-1\n\nWhat do we get from this change? It doesn\u0027t seem like a huge optimisation because such operations are relatively rare. However, we blow up the code and make it less generic. Can you elaborate on that?","accounts_in_message":[],"_revision_number":1},{"id":"bf7cc6024527481dfad1c7375ac41fcc80807e08","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-10 06:45:39.000000000","message":"Patch Set 1:\n\n\u003e What do we get from this change? It doesn\u0027t seem like a huge\n \u003e optimisation because such operations are relatively rare. However,\n \u003e we blow up the code and make it less generic. Can you elaborate on\n \u003e that?\n\nI disagree with statement that this option is rare. There are a lot of cases, when user generates workflow right before its execution, so it is very important to not spend any extra time doing redundant job.\n\nWe should understand, that flows could be huge, it is not OK to spend additional 5 seconds to convertation every time we load flow.","accounts_in_message":[],"_revision_number":1},{"id":"40c234d043230747167f504d80dc7004319203ed","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-10 07:39:22.000000000","message":"Patch Set 1:\n\nOk. I\u0027m ok with the idea of the patch. I\u0027d rather try to find a bit more elegant solution. Please think.","accounts_in_message":[],"_revision_number":1},{"id":"aeacf069ec1e51352e9472273978011e3f0bc5aa","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-10 08:51:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c3f54967bf3a9e063b178a42c6ec87408fe09224","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-10 08:57:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2c9fe21437149e920eab965614d4acb671601b12","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-06-10 09:04:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c814e401fedaf52b62826c5c9ff871a160ab13f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 11:02:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/77/663877/2/check/openstack-tox-pep8/15db32c/ : SUCCESS in 7m 18s\n- openstack-tox-py27 http://logs.openstack.org/77/663877/2/check/openstack-tox-py27/2437075/ : SUCCESS in 9m 41s\n- openstack-tox-py36 http://logs.openstack.org/77/663877/2/check/openstack-tox-py36/93c5f01/ : SUCCESS in 9m 50s\n- openstack-tox-py37 http://logs.openstack.org/77/663877/2/check/openstack-tox-py37/9302e32/ : SUCCESS in 9m 30s\n- openstack-tox-docs http://logs.openstack.org/77/663877/2/check/openstack-tox-docs/d0b793b/html/ : SUCCESS in 4m 42s\n- openstack-tox-cover http://logs.openstack.org/77/663877/2/check/openstack-tox-cover/c6a1a46/cover/ : SUCCESS in 9m 14s (non-voting)\n- mistral-devstack http://logs.openstack.org/77/663877/2/check/mistral-devstack/708f330/ : SUCCESS in 48m 43s\n- mistral-devstack-non-apache http://logs.openstack.org/77/663877/2/check/mistral-devstack-non-apache/b73350b/ : SUCCESS in 43m 27s\n- mistral-devstack-kombu http://logs.openstack.org/77/663877/2/check/mistral-devstack-kombu/7d01f21/ : SUCCESS in 44m 56s\n- mistral-tox-unit-mysql http://logs.openstack.org/77/663877/2/check/mistral-tox-unit-mysql/9526924/ : FAILURE in 31m 43s\n- mistral-tox-unit-postgresql http://logs.openstack.org/77/663877/2/check/mistral-tox-unit-postgresql/cb19a5b/ : SUCCESS in 27m 34s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/77/663877/2/check/tripleo-ci-centos-7-scenario003-standalone/01151ef/ : SUCCESS in 1h 22m 59s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/77/663877/2/check/tripleo-ci-centos-7-containers-multinode/4c84453/ : SUCCESS in 2h 06m 13s\n- mistral-rally-task http://logs.openstack.org/77/663877/2/check/mistral-rally-task/d1485cc/ : SUCCESS in 1h 01m 09s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/77/663877/2/check/openstack-tox-lower-constraints/333cc58/ : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":2},{"id":"89a484a20fc8be07d02b1140ef2abe73f93371a8","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-10 11:18:04.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cf8d8383adba5854618552a8a8308a5c59e44ab3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:46:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/77/663877/2/check/openstack-tox-pep8/3c25282/ : SUCCESS in 6m 03s\n- openstack-tox-py27 http://logs.openstack.org/77/663877/2/check/openstack-tox-py27/94ae79e/ : SUCCESS in 9m 44s\n- openstack-tox-py36 http://logs.openstack.org/77/663877/2/check/openstack-tox-py36/ddc4300/ : SUCCESS in 9m 43s\n- openstack-tox-py37 http://logs.openstack.org/77/663877/2/check/openstack-tox-py37/0df0ab5/ : SUCCESS in 9m 42s\n- openstack-tox-docs http://logs.openstack.org/77/663877/2/check/openstack-tox-docs/b22f9e9/html/ : SUCCESS in 6m 41s\n- openstack-tox-cover http://logs.openstack.org/77/663877/2/check/openstack-tox-cover/01d710c/cover/ : SUCCESS in 11m 03s (non-voting)\n- mistral-devstack http://logs.openstack.org/77/663877/2/check/mistral-devstack/b64e9ef/ : SUCCESS in 43m 12s\n- mistral-devstack-non-apache http://logs.openstack.org/77/663877/2/check/mistral-devstack-non-apache/0795353/ : SUCCESS in 41m 17s\n- mistral-devstack-kombu http://logs.openstack.org/77/663877/2/check/mistral-devstack-kombu/9c3e544/ : SUCCESS in 38m 42s\n- mistral-tox-unit-mysql http://logs.openstack.org/77/663877/2/check/mistral-tox-unit-mysql/a36004c/ : SUCCESS in 29m 18s\n- mistral-tox-unit-postgresql http://logs.openstack.org/77/663877/2/check/mistral-tox-unit-postgresql/cc66d18/ : SUCCESS in 26m 31s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/77/663877/2/check/tripleo-ci-centos-7-scenario003-standalone/5ef85d8/ : SUCCESS in 1h 16m 07s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/77/663877/2/check/tripleo-ci-centos-7-containers-multinode/2076ee3/ : SUCCESS in 2h 16m 24s\n- mistral-rally-task http://logs.openstack.org/77/663877/2/check/mistral-rally-task/2e8d477/ : SUCCESS in 1h 08m 40s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/77/663877/2/check/openstack-tox-lower-constraints/552c9ab/ : SUCCESS in 11m 41s","accounts_in_message":[],"_revision_number":2},{"id":"db9b2779af29dc8953ab6195c795b7b33a0741d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:46:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0b12ce54e6d41972dcb498290d9516b8463ec83f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 15:27:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/77/663877/2/gate/openstack-tox-pep8/b857bd8/ : SUCCESS in 5m 41s\n- openstack-tox-py27 http://logs.openstack.org/77/663877/2/gate/openstack-tox-py27/7b0dc6a/ : SUCCESS in 8m 32s\n- openstack-tox-py36 http://logs.openstack.org/77/663877/2/gate/openstack-tox-py36/6a36faf/ : SUCCESS in 9m 48s\n- openstack-tox-py37 http://logs.openstack.org/77/663877/2/gate/openstack-tox-py37/d72927f/ : SUCCESS in 10m 05s\n- openstack-tox-docs http://logs.openstack.org/77/663877/2/gate/openstack-tox-docs/db80543/html/ : SUCCESS in 4m 44s\n- mistral-devstack http://logs.openstack.org/77/663877/2/gate/mistral-devstack/947ebaa/ : SUCCESS in 49m 42s\n- mistral-devstack-non-apache http://logs.openstack.org/77/663877/2/gate/mistral-devstack-non-apache/da1acf9/ : SUCCESS in 54m 40s\n- mistral-tox-unit-mysql http://logs.openstack.org/77/663877/2/gate/mistral-tox-unit-mysql/bf151e6/ : SUCCESS in 27m 13s\n- mistral-tox-unit-postgresql http://logs.openstack.org/77/663877/2/gate/mistral-tox-unit-postgresql/ceddd6f/ : SUCCESS in 27m 00s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/77/663877/2/gate/tripleo-ci-centos-7-scenario003-standalone/b979c27/ : SUCCESS in 1h 20m 17s\n- mistral-devstack-kombu http://logs.openstack.org/77/663877/2/gate/mistral-devstack-kombu/91985ff/ : SUCCESS in 43m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/663877/2/gate/openstack-tox-lower-constraints/af65b0b/ : SUCCESS in 10m 22s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/77/663877/2/gate/tripleo-ci-centos-7-undercloud-containers/d4e5ea2/ : SUCCESS in 1h 35m 25s","accounts_in_message":[],"_revision_number":2},{"id":"efa178753545036cd4f16e30b5ac9e18c10d7fb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 15:27:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9a2f3b5715e99fb8905cb34c97ac54a8af192b7e","revisions":{"c2ef44db040ce8182cb5c60668d7e07b65c02826":{"kind":"REWORK","_number":1,"created":"2019-06-07 09:54:26.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/77/663877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/663877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/663877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/663877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/663877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/663877/1"}}},"commit":{"parents":[{"commit":"f85caa02c9aaa64db4300b43854327d132a71ed4","subject":"Merge \"Skip context evaluation for non-conditional transitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f85caa02c9aaa64db4300b43854327d132a71ed4"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-07 09:54:13.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-07 09:54:13.000000000","tz":180},"subject":"Simple optimization of creating/updating workflows","message":"Simple optimization of creating/updating workflows\n\nIn case we load definition of one workflow, there is no need to convert\nit to yaml and back.\n\nChange-Id: Ia7fb914843e56650c8f6ff5e8c16d1b57001c0fc\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c2ef44db040ce8182cb5c60668d7e07b65c02826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c2ef44db040ce8182cb5c60668d7e07b65c02826"}]},"branch":"refs/heads/master"},"9a2f3b5715e99fb8905cb34c97ac54a8af192b7e":{"kind":"REWORK","_number":2,"created":"2019-06-10 08:51:40.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/77/663877/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/663877/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/663877/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/663877/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/663877/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/663877/2"}}},"commit":{"parents":[{"commit":"f85caa02c9aaa64db4300b43854327d132a71ed4","subject":"Merge \"Skip context evaluation for non-conditional transitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f85caa02c9aaa64db4300b43854327d132a71ed4"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-07 09:54:13.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-10 08:50:43.000000000","tz":180},"subject":"Simple optimization of creating/updating workflows","message":"Simple optimization of creating/updating workflows\n\nIn case we load definition of one workflow, there is no need to convert\nit to yaml and back.\n\nChange-Id: Ia7fb914843e56650c8f6ff5e8c16d1b57001c0fc\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9a2f3b5715e99fb8905cb34c97ac54a8af192b7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9a2f3b5715e99fb8905cb34c97ac54a8af192b7e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
