)]}'
{"id":"openstack%2Fmistral~664256","triplet_id":"openstack%2Fmistral~master~Ic9bf73736392e91ce02faa0b3a73b5b3ce40fd14","project":"openstack/mistral","branch":"master","topic":"store_only_conditional_transitions","hashtags":[],"change_id":"Ic9bf73736392e91ce02faa0b3a73b5b3ce40fd14","subject":"Direct workflow code cleanup and refactoring","status":"MERGED","created":"2019-06-10 13:17:06.000000000","updated":"2019-06-14 16:35:26.000000000","submitted":"2019-06-14 16:35:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664256-1560530126707-018dc456","meta_rev_id":"7363e173c7eae0a8f05b17069d69f4b408c1c7da","_number":664256,"virtual_id_number":664256,"owner":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-06-14 16:35:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-12 17:11:31.000000000","_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"all":[{"value":1,"date":"2019-06-13 08:17:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2019-06-13 13:07:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-06-13 12:20:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":1,"date":"2019-06-14 14:25:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-06-14 13:19:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 08:17:23.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-06-13 13:07:42.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-06-14 13:19:42.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-06-14 14:25:39.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-06-14 16:35:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16e4041c77adc37a8c56dcf518b1443dd6ece98b","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-10 13:17:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"454e5c3a9f59adf7f05cba76e366a3179508f81a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-10 14:48:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"689393b5aba95657aae3f010c837d16a8915f07a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 15:38:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/56/664256/1/check/openstack-tox-pep8/8b60528/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/56/664256/1/check/openstack-tox-py27/18ef4ac/ : SUCCESS in 8m 44s\n- openstack-tox-py36 http://logs.openstack.org/56/664256/1/check/openstack-tox-py36/f18b716/ : SUCCESS in 8m 34s\n- openstack-tox-py37 http://logs.openstack.org/56/664256/1/check/openstack-tox-py37/b2ade81/ : SUCCESS in 9m 21s\n- openstack-tox-docs http://logs.openstack.org/56/664256/1/check/openstack-tox-docs/230f52d/html/ : SUCCESS in 4m 49s\n- openstack-tox-cover http://logs.openstack.org/56/664256/1/check/openstack-tox-cover/210e46c/cover/ : SUCCESS in 9m 19s (non-voting)\n- mistral-devstack http://logs.openstack.org/56/664256/1/check/mistral-devstack/35c4188/ : SUCCESS in 46m 44s\n- mistral-devstack-non-apache http://logs.openstack.org/56/664256/1/check/mistral-devstack-non-apache/71aea2d/ : SUCCESS in 37m 27s\n- mistral-devstack-kombu http://logs.openstack.org/56/664256/1/check/mistral-devstack-kombu/8d307b1/ : SUCCESS in 39m 56s\n- mistral-tox-unit-mysql http://logs.openstack.org/56/664256/1/check/mistral-tox-unit-mysql/2195118/ : SUCCESS in 28m 12s\n- mistral-tox-unit-postgresql http://logs.openstack.org/56/664256/1/check/mistral-tox-unit-postgresql/99e5342/ : SUCCESS in 26m 35s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/56/664256/1/check/tripleo-ci-centos-7-scenario003-standalone/4256f89/ : SUCCESS in 1h 20m 58s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/56/664256/1/check/tripleo-ci-centos-7-containers-multinode/77ee29b/ : SUCCESS in 2h 08m 03s\n- mistral-rally-task http://logs.openstack.org/56/664256/1/check/mistral-rally-task/8789b5d/ : SUCCESS in 1h 10m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/664256/1/check/openstack-tox-lower-constraints/e5cbdbe/ : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3d65dce65b66a05c6e53e6f2ea6a68edd8bf9695","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-11 07:08:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0054aad544b20bdf122eb057ee625a3cbbbfc942","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-11 08:02:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b2892257fc65c75dd1fd41130e480ddcd4c62ba","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-11 09:40:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"678bb4a16d527e0a8f6f81fb1c99d69004ac49e4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-11 09:40:49.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2bfcc2de6f7fe514ad82b0b11811b925525e66f4","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-11 10:16:45.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"3e5bb422f6a8f9bf4019e7ea5b5f78901c7b18c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 12:41:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/56/664256/3/check/openstack-tox-pep8/4a5f661/ : SUCCESS in 6m 20s\n- openstack-tox-py27 http://logs.openstack.org/56/664256/3/check/openstack-tox-py27/4f002dc/ : SUCCESS in 8m 28s\n- openstack-tox-py36 http://logs.openstack.org/56/664256/3/check/openstack-tox-py36/e9f9995/ : SUCCESS in 8m 18s\n- openstack-tox-py37 http://logs.openstack.org/56/664256/3/check/openstack-tox-py37/a85720f/ : SUCCESS in 9m 21s\n- openstack-tox-docs http://logs.openstack.org/56/664256/3/check/openstack-tox-docs/5c4b1a9/html/ : SUCCESS in 5m 24s\n- openstack-tox-cover http://logs.openstack.org/56/664256/3/check/openstack-tox-cover/2f44bb0/cover/ : SUCCESS in 9m 45s (non-voting)\n- mistral-devstack http://logs.openstack.org/56/664256/3/check/mistral-devstack/29e0712/ : SUCCESS in 44m 10s\n- mistral-devstack-non-apache http://logs.openstack.org/56/664256/3/check/mistral-devstack-non-apache/fe69cec/ : SUCCESS in 47m 26s\n- mistral-devstack-kombu http://logs.openstack.org/56/664256/3/check/mistral-devstack-kombu/9680925/ : SUCCESS in 47m 31s\n- mistral-tox-unit-mysql http://logs.openstack.org/56/664256/3/check/mistral-tox-unit-mysql/4217f94/ : SUCCESS in 27m 19s\n- mistral-tox-unit-postgresql http://logs.openstack.org/56/664256/3/check/mistral-tox-unit-postgresql/d756330/ : SUCCESS in 26m 17s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/56/664256/3/check/tripleo-ci-centos-7-scenario003-standalone/b307d1f/ : SUCCESS in 1h 18m 50s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/56/664256/3/check/tripleo-ci-centos-7-containers-multinode/d97263e/ : SUCCESS in 2h 13m 10s\n- mistral-rally-task http://logs.openstack.org/56/664256/3/check/mistral-rally-task/cd75c31/ : SUCCESS in 1h 18m 08s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/664256/3/check/openstack-tox-lower-constraints/0550e3a/ : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":3},{"id":"d3d6299b34a7361c09204917b64e7e78add521b7","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-12 08:52:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"51c48aab2b170e42ad84b295df845c328a19b892","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 11:08:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/56/664256/4/check/openstack-tox-pep8/381d163/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/56/664256/4/check/openstack-tox-py27/d6d711d/ : SUCCESS in 8m 51s\n- openstack-tox-py36 http://logs.openstack.org/56/664256/4/check/openstack-tox-py36/5af003c/ : SUCCESS in 9m 25s\n- openstack-tox-py37 http://logs.openstack.org/56/664256/4/check/openstack-tox-py37/8ce8900/ : SUCCESS in 9m 07s\n- openstack-tox-docs http://logs.openstack.org/56/664256/4/check/openstack-tox-docs/4b6ea25/html/ : SUCCESS in 5m 09s\n- openstack-tox-cover http://logs.openstack.org/56/664256/4/check/openstack-tox-cover/f373743/cover/ : SUCCESS in 9m 25s (non-voting)\n- mistral-devstack http://logs.openstack.org/56/664256/4/check/mistral-devstack/0354601/ : SUCCESS in 45m 56s\n- mistral-devstack-non-apache http://logs.openstack.org/56/664256/4/check/mistral-devstack-non-apache/e73347a/ : SUCCESS in 42m 31s\n- mistral-devstack-kombu http://logs.openstack.org/56/664256/4/check/mistral-devstack-kombu/e97513c/ : SUCCESS in 44m 58s\n- mistral-tox-unit-mysql http://logs.openstack.org/56/664256/4/check/mistral-tox-unit-mysql/1ff576f/ : FAILURE in 43m 30s\n- mistral-tox-unit-postgresql http://logs.openstack.org/56/664256/4/check/mistral-tox-unit-postgresql/acdad25/ : SUCCESS in 26m 54s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/56/664256/4/check/tripleo-ci-centos-7-scenario003-standalone/de9c7e4/ : SUCCESS in 1h 20m 05s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/56/664256/4/check/tripleo-ci-centos-7-containers-multinode/7c628f6/ : SUCCESS in 2h 09m 28s\n- mistral-rally-task http://logs.openstack.org/56/664256/4/check/mistral-rally-task/b213e07/ : SUCCESS in 1h 22m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/664256/4/check/openstack-tox-lower-constraints/63f9de6/ : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":4},{"id":"fe6ea2da0bc0e4a780a34baf50b1187be94cba11","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-12 14:09:37.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1d9ec398bc89808b38d8c03ff31a512e1244a997","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 17:00:24.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/56/664256/4/check/openstack-tox-pep8/10c3608/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/56/664256/4/check/openstack-tox-py27/624184c/ : SUCCESS in 9m 38s\n- openstack-tox-py36 http://logs.openstack.org/56/664256/4/check/openstack-tox-py36/59b0419/ : SUCCESS in 10m 39s\n- openstack-tox-py37 http://logs.openstack.org/56/664256/4/check/openstack-tox-py37/5780f6c/ : SUCCESS in 8m 55s\n- openstack-tox-docs http://logs.openstack.org/56/664256/4/check/openstack-tox-docs/576f25c/html/ : SUCCESS in 6m 31s\n- openstack-tox-cover http://logs.openstack.org/56/664256/4/check/openstack-tox-cover/899a3b3/cover/ : SUCCESS in 9m 49s (non-voting)\n- mistral-devstack http://logs.openstack.org/56/664256/4/check/mistral-devstack/3c412ed/ : SUCCESS in 49m 36s\n- mistral-devstack-non-apache http://logs.openstack.org/56/664256/4/check/mistral-devstack-non-apache/0431c3e/ : SUCCESS in 45m 48s\n- mistral-devstack-kombu http://logs.openstack.org/56/664256/4/check/mistral-devstack-kombu/a3e86a7/ : SUCCESS in 47m 49s\n- mistral-tox-unit-mysql http://logs.openstack.org/56/664256/4/check/mistral-tox-unit-mysql/f0ca178/ : FAILURE in 31m 53s\n- mistral-tox-unit-postgresql http://logs.openstack.org/56/664256/4/check/mistral-tox-unit-postgresql/a233c70/ : SUCCESS in 27m 27s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/56/664256/4/check/tripleo-ci-centos-7-scenario003-standalone/6ccd2e8/ : SUCCESS in 1h 16m 49s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/56/664256/4/check/tripleo-ci-centos-7-containers-multinode/ab773c6/ : SUCCESS in 2h 12m 03s\n- mistral-rally-task http://logs.openstack.org/56/664256/4/check/mistral-rally-task/aa02304/ : SUCCESS in 1h 09m 12s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/664256/4/check/openstack-tox-lower-constraints/f5ead49/ : SUCCESS in 10m 30s","accounts_in_message":[],"_revision_number":4},{"id":"afd04a3d7353acad6793234b4d5529b8cee2b284","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-12 17:11:31.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c0b9e55b58122c7749d996320920c83f2e3e0b6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 19:45:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/56/664256/4/check/openstack-tox-pep8/3ea9256/ : SUCCESS in 5m 43s\n- openstack-tox-py27 http://logs.openstack.org/56/664256/4/check/openstack-tox-py27/3bf52d7/ : SUCCESS in 8m 30s\n- openstack-tox-py36 http://logs.openstack.org/56/664256/4/check/openstack-tox-py36/7d116fe/ : SUCCESS in 11m 16s\n- openstack-tox-py37 http://logs.openstack.org/56/664256/4/check/openstack-tox-py37/a256cf8/ : SUCCESS in 10m 12s\n- openstack-tox-docs http://logs.openstack.org/56/664256/4/check/openstack-tox-docs/65524fc/html/ : SUCCESS in 4m 54s\n- openstack-tox-cover http://logs.openstack.org/56/664256/4/check/openstack-tox-cover/2595d60/cover/ : SUCCESS in 9m 03s (non-voting)\n- mistral-devstack http://logs.openstack.org/56/664256/4/check/mistral-devstack/ccd4cc9/ : SUCCESS in 44m 12s\n- mistral-devstack-non-apache http://logs.openstack.org/56/664256/4/check/mistral-devstack-non-apache/6101bde/ : SUCCESS in 46m 09s\n- mistral-devstack-kombu http://logs.openstack.org/56/664256/4/check/mistral-devstack-kombu/0b3ed89/ : SUCCESS in 41m 05s\n- mistral-tox-unit-mysql http://logs.openstack.org/56/664256/4/check/mistral-tox-unit-mysql/8b00a3e/ : SUCCESS in 31m 26s\n- mistral-tox-unit-postgresql http://logs.openstack.org/56/664256/4/check/mistral-tox-unit-postgresql/3a54224/ : SUCCESS in 26m 35s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/56/664256/4/check/tripleo-ci-centos-7-scenario003-standalone/f4af30c/ : SUCCESS in 1h 21m 59s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/56/664256/4/check/tripleo-ci-centos-7-containers-multinode/617cd41/ : SUCCESS in 2h 02m 45s\n- mistral-rally-task http://logs.openstack.org/56/664256/4/check/mistral-rally-task/68a311b/ : SUCCESS in 1h 10m 03s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/664256/4/check/openstack-tox-lower-constraints/2f20b45/ : SUCCESS in 10m 14s","accounts_in_message":[],"_revision_number":4},{"id":"7c2601e0e2c2a402033fd7bdd2ab67137814ef53","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-13 08:17:23.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"bc4b16137218f755fc9a5be9f475c75e11ed8aa3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-13 12:20:02.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a13327179fe17d7fd60891647a8541e3e0693459","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-06-13 13:07:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a6fe011a7d5f3947417c92502943ff53c309462e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-14 13:19:42.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"dbae22b635dea6a49b8bcaa8f2258a58b96ce877","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 13:19:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b3590be7bb6b3a315db20c6af54f053e760c9ae4","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-06-14 14:25:39.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4989b24f5827f37a0937d38e3f5004b5ad846b41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 14:51:15.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5b6d0966d2ab5a80d1cbbdf5316671362264563d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 16:35:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"802952dd55142699e2eb1461663a65f024c4b2a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 16:35:26.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/56/664256/4/gate/openstack-tox-pep8/1d75c05/ : SUCCESS in 8m 05s\n- openstack-tox-py27 http://logs.openstack.org/56/664256/4/gate/openstack-tox-py27/aee39f5/ : SUCCESS in 8m 55s\n- openstack-tox-py36 http://logs.openstack.org/56/664256/4/gate/openstack-tox-py36/6631c80/ : SUCCESS in 9m 41s\n- openstack-tox-py37 http://logs.openstack.org/56/664256/4/gate/openstack-tox-py37/45bb8e3/ : SUCCESS in 13m 39s\n- openstack-tox-docs http://logs.openstack.org/56/664256/4/gate/openstack-tox-docs/6febac4/html/ : SUCCESS in 5m 30s\n- mistral-devstack http://logs.openstack.org/56/664256/4/gate/mistral-devstack/d470d54/ : SUCCESS in 42m 43s\n- mistral-devstack-non-apache http://logs.openstack.org/56/664256/4/gate/mistral-devstack-non-apache/4814a94/ : SUCCESS in 52m 35s\n- mistral-tox-unit-mysql http://logs.openstack.org/56/664256/4/gate/mistral-tox-unit-mysql/c1b0dd5/ : SUCCESS in 33m 45s\n- mistral-tox-unit-postgresql http://logs.openstack.org/56/664256/4/gate/mistral-tox-unit-postgresql/52b9b0d/ : SUCCESS in 31m 03s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/56/664256/4/gate/tripleo-ci-centos-7-scenario003-standalone/9fa0a21/ : SUCCESS in 1h 16m 22s\n- mistral-devstack-kombu http://logs.openstack.org/56/664256/4/gate/mistral-devstack-kombu/cd7d641/ : SUCCESS in 53m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/664256/4/gate/openstack-tox-lower-constraints/fcb18cf/ : SUCCESS in 12m 07s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/56/664256/4/gate/tripleo-ci-centos-7-undercloud-containers/9f9ec47/ : SUCCESS in 1h 37m 24s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c215c0520f078d47df9443b043e1d801073c1fc2","revisions":{"cdf8e69c2f501105e3297caf045e0c4a7ca1cb32":{"kind":"REWORK","_number":1,"created":"2019-06-10 13:17:06.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/56/664256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/664256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/664256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/664256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/664256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/664256/1"}}},"commit":{"parents":[{"commit":"1d6064eaec5a34e9c59898be8877f38333e4b908","subject":"Store next task names in DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1d6064eaec5a34e9c59898be8877f38333e4b908"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-10 13:13:59.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-10 13:13:59.000000000","tz":120},"subject":"Code cleanup","message":"Code cleanup\n\nWhen Ib3a684f63f05a7cfe846782774d2c68be78bebab is merged a lot of\ncode becomes useless and can be removed.\n\nChange-Id: Ic9bf73736392e91ce02faa0b3a73b5b3ce40fd14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cdf8e69c2f501105e3297caf045e0c4a7ca1cb32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cdf8e69c2f501105e3297caf045e0c4a7ca1cb32"}]},"branch":"refs/heads/master"},"18932941d816aa889d42b28755d1dea01613f99e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-11 09:40:45.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/56/664256/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/664256/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/664256/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/664256/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/664256/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/664256/2"}}},"commit":{"parents":[{"commit":"eb97410610a909fe88bda6b82258688921cebb05","subject":"Merge \"Store next task names in DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eb97410610a909fe88bda6b82258688921cebb05"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-10 13:13:59.000000000","tz":120},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-06-11 09:40:45.000000000","tz":0},"subject":"Code cleanup","message":"Code cleanup\n\nWhen Ib3a684f63f05a7cfe846782774d2c68be78bebab is merged a lot of\ncode becomes useless and can be removed.\n\nChange-Id: Ic9bf73736392e91ce02faa0b3a73b5b3ce40fd14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/18932941d816aa889d42b28755d1dea01613f99e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/18932941d816aa889d42b28755d1dea01613f99e"}]},"branch":"refs/heads/master"},"99e2690b48e5263c9a950def078ee5bcdbc16dae":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-11 10:16:45.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/56/664256/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/664256/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/664256/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/664256/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/664256/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/664256/3"}}},"commit":{"parents":[{"commit":"97d87b843f280d6a511c035d1ba7e1d586998b7b","subject":"Merge \"Change mistral opendev.org to releases.openstack.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/97d87b843f280d6a511c035d1ba7e1d586998b7b"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-10 13:13:59.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-11 10:16:15.000000000","tz":120},"subject":"Code cleanup","message":"Code cleanup\n\nWhen Ib3a684f63f05a7cfe846782774d2c68be78bebab is merged a lot of\ncode becomes useless and can be removed.\n\nChange-Id: Ic9bf73736392e91ce02faa0b3a73b5b3ce40fd14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e2690b48e5263c9a950def078ee5bcdbc16dae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e2690b48e5263c9a950def078ee5bcdbc16dae"}]},"branch":"refs/heads/master"},"c215c0520f078d47df9443b043e1d801073c1fc2":{"kind":"REWORK","_number":4,"created":"2019-06-12 08:52:46.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/56/664256/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/664256/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/664256/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/664256/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/664256/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/664256/4"}}},"commit":{"parents":[{"commit":"97d87b843f280d6a511c035d1ba7e1d586998b7b","subject":"Merge \"Change mistral opendev.org to releases.openstack.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/97d87b843f280d6a511c035d1ba7e1d586998b7b"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-10 13:13:59.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-12 08:51:54.000000000","tz":120},"subject":"Direct workflow code cleanup and refactoring","message":"Direct workflow code cleanup and refactoring\n\nWhen Ib3a684f63f05a7cfe846782774d2c68be78bebab is merged a lot of\ncode becomes useless and can be removed.\n\nChange-Id: Ic9bf73736392e91ce02faa0b3a73b5b3ce40fd14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c215c0520f078d47df9443b043e1d801073c1fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c215c0520f078d47df9443b043e1d801073c1fc2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
