)]}'
{"id":"openstack%2Fmistral~664464","triplet_id":"openstack%2Fmistral~master~I04f4e42878b81e82832593e89c6dbe9257bb53b9","project":"openstack/mistral","branch":"master","topic":"store_only_conditional_transitions","hashtags":[],"change_id":"I04f4e42878b81e82832593e89c6dbe9257bb53b9","subject":"WIP: Store in database only tasks with conditional transitions","status":"ABANDONED","created":"2019-06-10 23:19:23.000000000","updated":"2026-03-28 19:39:58.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7a48f372956393283ce79ac3604cfa8428536f9f","_number":664464,"virtual_id_number":664464,"owner":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":1,"date":"2019-06-12 12:38:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 03:44:34.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-06-12 12:38:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-09 08:10:09.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-05-09 08:10:09.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"CC"}],"messages":[{"id":"5eb739c2748b09fa466b2950ce925f792c0cdf5a","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-10 23:19:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"241b6100d0c0451c7839dc6f64a2ff7ad4eb009e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 01:04:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/64/664464/1/check/openstack-tox-pep8/5a85351/ : SUCCESS in 5m 48s\n- openstack-tox-py27 http://logs.openstack.org/64/664464/1/check/openstack-tox-py27/8ba414c/ : SUCCESS in 10m 35s\n- openstack-tox-py36 http://logs.openstack.org/64/664464/1/check/openstack-tox-py36/717ead6/ : SUCCESS in 8m 43s\n- openstack-tox-py37 http://logs.openstack.org/64/664464/1/check/openstack-tox-py37/e6f94cb/ : SUCCESS in 15m 43s\n- openstack-tox-docs http://logs.openstack.org/64/664464/1/check/openstack-tox-docs/13cdb4c/html/ : SUCCESS in 6m 36s\n- openstack-tox-cover http://logs.openstack.org/64/664464/1/check/openstack-tox-cover/fded4b5/cover/ : SUCCESS in 10m 15s (non-voting)\n- mistral-devstack http://logs.openstack.org/64/664464/1/check/mistral-devstack/9ddc6a5/ : SUCCESS in 1h 07m 52s\n- mistral-devstack-non-apache http://logs.openstack.org/64/664464/1/check/mistral-devstack-non-apache/9eb7e88/ : SUCCESS in 44m 50s\n- mistral-devstack-kombu http://logs.openstack.org/64/664464/1/check/mistral-devstack-kombu/d9c5252/ : SUCCESS in 1h 06m 10s\n- mistral-tox-unit-mysql http://logs.openstack.org/64/664464/1/check/mistral-tox-unit-mysql/e2a2250/ : SUCCESS in 28m 43s\n- mistral-tox-unit-postgresql http://logs.openstack.org/64/664464/1/check/mistral-tox-unit-postgresql/ed7ca97/ : SUCCESS in 32m 22s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/64/664464/1/check/tripleo-ci-centos-7-scenario003-standalone/73e8645/ : FAILURE in 57m 07s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/64/664464/1/check/tripleo-ci-centos-7-containers-multinode/2b81eac/ : FAILURE in 51m 29s\n- mistral-rally-task http://logs.openstack.org/64/664464/1/check/mistral-rally-task/f57c5a5/ : SUCCESS in 1h 15m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/64/664464/1/check/openstack-tox-lower-constraints/3955b87/ : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":1},{"id":"8bbfd483005901abef7398cbce32dc979911dc26","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-11 03:44:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3d49aab75786c4698c6ac991625b8e07d5572c63","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-11 10:16:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"46756f8f90d8da2e64e54651ba0124a315229c77","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-11 10:25:21.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"84cd03b307fa938ed6a4a0bf69439a45a8b04f54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 12:48:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/64/664464/3/check/openstack-tox-pep8/5e9c708/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/64/664464/3/check/openstack-tox-py27/45bc39a/ : SUCCESS in 8m 53s\n- openstack-tox-py36 http://logs.openstack.org/64/664464/3/check/openstack-tox-py36/c2cab1c/ : SUCCESS in 8m 41s\n- openstack-tox-py37 http://logs.openstack.org/64/664464/3/check/openstack-tox-py37/cf60b3e/ : SUCCESS in 9m 51s\n- openstack-tox-docs http://logs.openstack.org/64/664464/3/check/openstack-tox-docs/517293c/html/ : SUCCESS in 4m 48s\n- openstack-tox-cover http://logs.openstack.org/64/664464/3/check/openstack-tox-cover/baa7ce0/cover/ : SUCCESS in 9m 27s (non-voting)\n- mistral-devstack http://logs.openstack.org/64/664464/3/check/mistral-devstack/f5daa4c/ : SUCCESS in 42m 34s\n- mistral-devstack-non-apache http://logs.openstack.org/64/664464/3/check/mistral-devstack-non-apache/4b1e1a0/ : SUCCESS in 47m 50s\n- mistral-devstack-kombu http://logs.openstack.org/64/664464/3/check/mistral-devstack-kombu/fb56254/ : SUCCESS in 42m 49s\n- mistral-tox-unit-mysql http://logs.openstack.org/64/664464/3/check/mistral-tox-unit-mysql/1878010/ : SUCCESS in 31m 07s\n- mistral-tox-unit-postgresql http://logs.openstack.org/64/664464/3/check/mistral-tox-unit-postgresql/47eafab/ : SUCCESS in 27m 08s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/64/664464/3/check/tripleo-ci-centos-7-scenario003-standalone/b4754da/ : SUCCESS in 1h 33m 13s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/64/664464/3/check/tripleo-ci-centos-7-containers-multinode/dc068eb/ : SUCCESS in 2h 09m 18s\n- mistral-rally-task http://logs.openstack.org/64/664464/3/check/mistral-rally-task/a2a7547/ : SUCCESS in 1h 07m 03s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/64/664464/3/check/openstack-tox-lower-constraints/797482f/ : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":3},{"id":"8d45f78167c07e043a5297b39ee1f658ff2881bf","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-12 10:16:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"76ff0cad947ea3ec1758605b1bbcbb0baaf59b30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 12:38:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/64/664464/4/check/openstack-tox-pep8/c8b85bb/ : SUCCESS in 5m 21s\n- openstack-tox-py27 http://logs.openstack.org/64/664464/4/check/openstack-tox-py27/b07faf3/ : SUCCESS in 8m 27s\n- openstack-tox-py36 http://logs.openstack.org/64/664464/4/check/openstack-tox-py36/3053c8a/ : SUCCESS in 9m 06s\n- openstack-tox-py37 http://logs.openstack.org/64/664464/4/check/openstack-tox-py37/ce11e2d/ : SUCCESS in 10m 11s\n- openstack-tox-docs http://logs.openstack.org/64/664464/4/check/openstack-tox-docs/054cd63/html/ : SUCCESS in 4m 52s\n- openstack-tox-cover http://logs.openstack.org/64/664464/4/check/openstack-tox-cover/c4f3015/cover/ : SUCCESS in 10m 12s (non-voting)\n- mistral-devstack http://logs.openstack.org/64/664464/4/check/mistral-devstack/776089b/ : SUCCESS in 44m 15s\n- mistral-devstack-non-apache http://logs.openstack.org/64/664464/4/check/mistral-devstack-non-apache/47de748/ : SUCCESS in 53m 09s\n- mistral-devstack-kombu http://logs.openstack.org/64/664464/4/check/mistral-devstack-kombu/ba32d5a/ : SUCCESS in 47m 30s\n- mistral-tox-unit-mysql http://logs.openstack.org/64/664464/4/check/mistral-tox-unit-mysql/7f963c6/ : SUCCESS in 28m 27s\n- mistral-tox-unit-postgresql http://logs.openstack.org/64/664464/4/check/mistral-tox-unit-postgresql/22ec137/ : SUCCESS in 27m 02s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/64/664464/4/check/tripleo-ci-centos-7-scenario003-standalone/10d1d01/ : SUCCESS in 1h 21m 24s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/64/664464/4/check/tripleo-ci-centos-7-containers-multinode/f3a91f6/ : SUCCESS in 2h 14m 15s\n- mistral-rally-task http://logs.openstack.org/64/664464/4/check/mistral-rally-task/a61e104/ : SUCCESS in 1h 09m 55s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/64/664464/4/check/openstack-tox-lower-constraints/1b89e68/ : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":4},{"id":"01c8c36d6517db67597c05584fb158ce5eb0f9f9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-09 08:10:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7a48f372956393283ce79ac3604cfa8428536f9f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-03-28 19:39:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"061ef3dd746e4ade5467477b1a19f576e2321e8e","revisions":{"aca3e42248b102d59a96edb09486c63c24405100":{"kind":"REWORK","_number":1,"created":"2019-06-10 23:19:23.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/64/664464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/64/664464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/664464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/664464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/664464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/64/664464/1"}}},"commit":{"parents":[{"commit":"cdf8e69c2f501105e3297caf045e0c4a7ca1cb32","subject":"Code cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cdf8e69c2f501105e3297caf045e0c4a7ca1cb32"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-10 21:58:31.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-10 23:01:58.000000000","tz":120},"subject":"WIP: store in database only tasks with conditional transitions","message":"WIP: store in database only tasks with conditional transitions\n\nTo prevent storing duplicate data in the database, we will save only\nconditional transitions there, and get other transitions from the\nworkflow spec directly.\n\nThis commit also adds cache for \u0027on-success\u0027, \u0027on-error\u0027 and\n\u0027on-complete\u0027 clauses.\n\nChange-Id: I04f4e42878b81e82832593e89c6dbe9257bb53b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aca3e42248b102d59a96edb09486c63c24405100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aca3e42248b102d59a96edb09486c63c24405100"}]},"branch":"refs/heads/master"},"9dd634c127411cdade91e7992293909a88972772":{"kind":"REWORK","_number":2,"created":"2019-06-11 10:16:45.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/64/664464/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/64/664464/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/664464/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/664464/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/664464/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/64/664464/2"}}},"commit":{"parents":[{"commit":"99e2690b48e5263c9a950def078ee5bcdbc16dae","subject":"Code cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e2690b48e5263c9a950def078ee5bcdbc16dae"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-10 21:58:31.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-11 10:16:20.000000000","tz":120},"subject":"Store in database only tasks with conditional transitions","message":"Store in database only tasks with conditional transitions\n\nTo prevent storing duplicate data in the database, for direct workflows\nwe will save only conditional transitions, and get other transitions\nfrom the workflow spec directly.\n\nThis commit also adds cache for \u0027on-success\u0027, \u0027on-error\u0027 and\n\u0027on-complete\u0027 clauses.\n\nChange-Id: I04f4e42878b81e82832593e89c6dbe9257bb53b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9dd634c127411cdade91e7992293909a88972772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9dd634c127411cdade91e7992293909a88972772"}]},"branch":"refs/heads/master"},"c8b3165afa21be27a4ae8622ea52b8a126d7cd98":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-06-11 10:25:21.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/64/664464/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/64/664464/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/664464/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/664464/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/664464/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/64/664464/3"}}},"commit":{"parents":[{"commit":"99e2690b48e5263c9a950def078ee5bcdbc16dae","subject":"Code cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e2690b48e5263c9a950def078ee5bcdbc16dae"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-10 21:58:31.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-11 10:25:16.000000000","tz":0},"subject":"WIP: Store in database only tasks with conditional transitions","message":"WIP: Store in database only tasks with conditional transitions\n\nTo prevent storing duplicate data in the database, for direct workflows\nwe will save only conditional transitions, and get other transitions\nfrom the workflow spec directly.\n\nThis commit also adds cache for \u0027on-success\u0027, \u0027on-error\u0027 and\n\u0027on-complete\u0027 clauses.\n\nChange-Id: I04f4e42878b81e82832593e89c6dbe9257bb53b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c8b3165afa21be27a4ae8622ea52b8a126d7cd98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c8b3165afa21be27a4ae8622ea52b8a126d7cd98"}]},"branch":"refs/heads/master"},"061ef3dd746e4ade5467477b1a19f576e2321e8e":{"kind":"REWORK","_number":4,"created":"2019-06-12 10:16:42.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/64/664464/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/64/664464/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/664464/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/664464/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/664464/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/64/664464/4"}}},"commit":{"parents":[{"commit":"c215c0520f078d47df9443b043e1d801073c1fc2","subject":"Direct workflow code cleanup and refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c215c0520f078d47df9443b043e1d801073c1fc2"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-10 21:58:31.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-12 10:16:17.000000000","tz":120},"subject":"WIP: Store in database only tasks with conditional transitions","message":"WIP: Store in database only tasks with conditional transitions\n\nTo prevent storing duplicate data in the database, for direct workflows\nwe will save only conditional transitions, and get other transitions\nfrom the workflow spec directly.\n\nThis commit also adds cache for \u0027on-success\u0027, \u0027on-error\u0027 and\n\u0027on-complete\u0027 clauses.\n\nChange-Id: I04f4e42878b81e82832593e89c6dbe9257bb53b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/061ef3dd746e4ade5467477b1a19f576e2321e8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/061ef3dd746e4ade5467477b1a19f576e2321e8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
