)]}'
{"id":"openstack%2Fmistral~664490","triplet_id":"openstack%2Fmistral~master~Ic5faf47b13f38b54d5c9ae2d8476ba86c664b074","project":"openstack/mistral","branch":"master","topic":"mistral_constraints","hashtags":[],"change_id":"Ic5faf47b13f38b54d5c9ae2d8476ba86c664b074","subject":"Change mistral opendev.org to releases.openstack.org","status":"MERGED","created":"2019-06-11 03:09:56.000000000","updated":"2019-06-11 09:54:18.000000000","submitted":"2019-06-11 09:54:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664490-1560246858452-d90c8b36","meta_rev_id":"5f7441839a468555b6f69e6680e93a47260a1bde","_number":664490,"virtual_id_number":664490,"owner":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-06-11 09:54:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-06-11 08:52:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-06-11 08:52:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 08:52:38.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-06-11 09:54:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"11839fbee9d280260704f57b46703b65bfa7e009","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-06-11 03:09:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c6b3e510aaa760be03f4f92e9e2703194b5dcd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 04:10:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/90/664490/1/check/openstack-tox-pep8/7c91dcf/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/90/664490/1/check/openstack-tox-py27/9bee8f1/ : SUCCESS in 8m 52s\n- openstack-tox-py36 http://logs.openstack.org/90/664490/1/check/openstack-tox-py36/820a9b4/ : RETRY_LIMIT in 2m 36s\n- openstack-tox-py37 http://logs.openstack.org/90/664490/1/check/openstack-tox-py37/5442018/ : SUCCESS in 10m 29s\n- openstack-tox-docs http://logs.openstack.org/90/664490/1/check/openstack-tox-docs/3f39af3/ : RETRY_LIMIT in 1m 49s\n- build-openstack-releasenotes http://logs.openstack.org/90/664490/1/check/build-openstack-releasenotes/5dfd740/ : RETRY_LIMIT in 2m 06s\n- openstack-tox-cover http://logs.openstack.org/90/664490/1/check/openstack-tox-cover/e7593d2/cover/ : SUCCESS in 8m 36s (non-voting)\n- mistral-devstack http://logs.openstack.org/90/664490/1/check/mistral-devstack/e925300/ : SUCCESS in 44m 52s\n- mistral-devstack-non-apache http://logs.openstack.org/90/664490/1/check/mistral-devstack-non-apache/de32372/ : SUCCESS in 47m 27s\n- mistral-devstack-kombu http://logs.openstack.org/90/664490/1/check/mistral-devstack-kombu/99e648e/ : SUCCESS in 51m 48s\n- mistral-tox-unit-mysql http://logs.openstack.org/90/664490/1/check/mistral-tox-unit-mysql/e768f2c/ : SUCCESS in 27m 38s\n- mistral-tox-unit-postgresql http://logs.openstack.org/90/664490/1/check/mistral-tox-unit-postgresql/1f6fc86/ : SUCCESS in 25m 43s\n- mistral-rally-task http://logs.openstack.org/90/664490/1/check/mistral-rally-task/8a6c924/ : FAILURE in 8m 59s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/90/664490/1/check/openstack-tox-lower-constraints/27c269e/ : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"43382e141b73dd86e4f465b9da843fadd87f1110","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-11 06:39:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"07ca4d9df9a444c70a372fed2fb124316c39e015","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 08:36:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/90/664490/1/check/openstack-tox-pep8/b767707/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/90/664490/1/check/openstack-tox-py27/e70ac6d/ : SUCCESS in 8m 11s\n- openstack-tox-py36 http://logs.openstack.org/90/664490/1/check/openstack-tox-py36/c57ee0a/ : SUCCESS in 10m 41s\n- openstack-tox-py37 http://logs.openstack.org/90/664490/1/check/openstack-tox-py37/b2cd1ec/ : SUCCESS in 10m 25s\n- openstack-tox-docs http://logs.openstack.org/90/664490/1/check/openstack-tox-docs/88c0186/html/ : SUCCESS in 5m 06s\n- build-openstack-releasenotes http://logs.openstack.org/90/664490/1/check/build-openstack-releasenotes/8c16f84/html/ : SUCCESS in 4m 37s\n- openstack-tox-cover http://logs.openstack.org/90/664490/1/check/openstack-tox-cover/a512882/cover/ : SUCCESS in 11m 08s (non-voting)\n- mistral-devstack http://logs.openstack.org/90/664490/1/check/mistral-devstack/136a927/ : SUCCESS in 53m 56s\n- mistral-devstack-non-apache http://logs.openstack.org/90/664490/1/check/mistral-devstack-non-apache/4cac145/ : SUCCESS in 43m 45s\n- mistral-devstack-kombu http://logs.openstack.org/90/664490/1/check/mistral-devstack-kombu/f77a19d/ : SUCCESS in 42m 26s\n- mistral-tox-unit-mysql http://logs.openstack.org/90/664490/1/check/mistral-tox-unit-mysql/9968c3e/ : SUCCESS in 27m 13s\n- mistral-tox-unit-postgresql http://logs.openstack.org/90/664490/1/check/mistral-tox-unit-postgresql/70f572d/ : SUCCESS in 30m 00s\n- mistral-rally-task http://logs.openstack.org/90/664490/1/check/mistral-rally-task/a7e6587/ : SUCCESS in 1h 09m 21s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/90/664490/1/check/openstack-tox-lower-constraints/82840a0/ : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e89dc4f1b763060152e5a8fd81ae96e4f1e9aba2","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-11 08:52:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa0438bd07d18c6298d37d9d915aae310fa2b6b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 08:52:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c6b8534be60c7c6f7202e5973f22d6efd43d507","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 09:54:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/90/664490/1/gate/openstack-tox-pep8/24f735c/ : SUCCESS in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/90/664490/1/gate/openstack-tox-py27/94cee63/ : SUCCESS in 8m 53s\n- openstack-tox-py36 http://logs.openstack.org/90/664490/1/gate/openstack-tox-py36/cf8de74/ : SUCCESS in 8m 35s\n- openstack-tox-py37 http://logs.openstack.org/90/664490/1/gate/openstack-tox-py37/2c2c402/ : SUCCESS in 9m 19s\n- openstack-tox-docs http://logs.openstack.org/90/664490/1/gate/openstack-tox-docs/0535f9b/html/ : SUCCESS in 4m 54s\n- build-openstack-releasenotes http://logs.openstack.org/90/664490/1/gate/build-openstack-releasenotes/0dc67a9/html/ : SUCCESS in 6m 03s\n- mistral-devstack http://logs.openstack.org/90/664490/1/gate/mistral-devstack/14dc494/ : SUCCESS in 56m 56s\n- mistral-devstack-non-apache http://logs.openstack.org/90/664490/1/gate/mistral-devstack-non-apache/44d8375/ : SUCCESS in 47m 13s\n- mistral-tox-unit-mysql http://logs.openstack.org/90/664490/1/gate/mistral-tox-unit-mysql/3c42e5f/ : SUCCESS in 28m 05s\n- mistral-tox-unit-postgresql http://logs.openstack.org/90/664490/1/gate/mistral-tox-unit-postgresql/4ab71c9/ : SUCCESS in 27m 11s\n- mistral-devstack-kombu http://logs.openstack.org/90/664490/1/gate/mistral-devstack-kombu/146df54/ : SUCCESS in 49m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/664490/1/gate/openstack-tox-lower-constraints/7856fe1/ : SUCCESS in 10m 12s","accounts_in_message":[],"_revision_number":1},{"id":"566b6a568e0e0f53900373c96db44534802f828f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 09:54:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0cada98e14aab68ca93a1bb25923a0ba50025d3","revisions":{"a0cada98e14aab68ca93a1bb25923a0ba50025d3":{"kind":"REWORK","_number":1,"created":"2019-06-11 03:09:56.000000000","uploader":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"ref":"refs/changes/90/664490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/90/664490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/90/664490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/90/664490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/90/664490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/90/664490/1"}}},"commit":{"parents":[{"commit":"d1d4087a3fa08ad119b4edbd7a307b0bb991f062","subject":"Merge \"Simple optimization of creating/updating workflows\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d1d4087a3fa08ad119b4edbd7a307b0bb991f062"}]}],"author":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-06-11 03:09:26.000000000","tz":480},"committer":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-06-11 03:09:53.000000000","tz":480},"subject":"Change mistral opendev.org to releases.openstack.org","message":"Change mistral opendev.org to releases.openstack.org\n\nChange-Id: Ic5faf47b13f38b54d5c9ae2d8476ba86c664b074\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a0cada98e14aab68ca93a1bb25923a0ba50025d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a0cada98e14aab68ca93a1bb25923a0ba50025d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
