)]}'
{"id":"openstack%2Fmistral~664506","triplet_id":"openstack%2Fmistral~master~Iadf333fcc6f241d707daf2767bab9cfbf2715917","project":"openstack/mistral","branch":"master","topic":"fix_test_with_items_action_context","hashtags":[],"change_id":"Iadf333fcc6f241d707daf2767bab9cfbf2715917","subject":"Fix sporadically failing test test_with_items_action_context","status":"MERGED","created":"2019-06-11 07:07:30.000000000","updated":"2019-06-11 12:31:05.000000000","submitted":"2019-06-11 12:31:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664506-1560256265957-43f1f006","meta_rev_id":"a08b0a12f91a14f6ccee978bf88d4da1658cf789","_number":664506,"virtual_id_number":664506,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-06-11 12:31:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-06-11 08:01:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":1,"date":"2019-06-11 08:01:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-11 08:01:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-06-11 08:01:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 07:20:00.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-06-11 07:20:20.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-06-11 07:20:20.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-06-11 07:20:20.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-06-11 07:20:20.000000000","updated_by":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"reviewer":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"state":"REVIEWER"},{"updated":"2019-06-11 07:22:11.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-06-11 07:28:18.000000000","updated_by":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"reviewer":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"state":"REVIEWER"},{"updated":"2019-06-11 12:31:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14b226e84108984878d70ab3bac9b973d371c5a8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-11 07:07:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"807ad83e66fa81d535f0721027afa55d46aac22e","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-11 07:20:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6bb1f916a0eded764b9c8c43f781528faefc0275","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-06-11 07:22:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"72212b63637e9f5734181b855f49bd6128e4c047","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-11 07:28:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"86fdf604b52c4e99b4f160914d98324a8df4a291","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-11 07:35:23.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"092f639934a6a4331ca722df085d7d215f236db9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-11 08:01:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"96e3394ba64e12ce5f9a4831cad49929a1b4decc","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-11 08:01:03.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d20fa53ea1cb42c66e819054d198ad349b947c3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 10:33:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/06/664506/2/check/openstack-tox-pep8/8a4c58b/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/06/664506/2/check/openstack-tox-py27/2938969/ : SUCCESS in 9m 22s\n- openstack-tox-py36 http://logs.openstack.org/06/664506/2/check/openstack-tox-py36/69c6d95/ : SUCCESS in 8m 56s\n- openstack-tox-py37 http://logs.openstack.org/06/664506/2/check/openstack-tox-py37/fec6b24/ : SUCCESS in 10m 47s\n- openstack-tox-docs http://logs.openstack.org/06/664506/2/check/openstack-tox-docs/8f0940b/html/ : SUCCESS in 7m 04s\n- openstack-tox-cover http://logs.openstack.org/06/664506/2/check/openstack-tox-cover/0b4679d/cover/ : SUCCESS in 8m 58s (non-voting)\n- mistral-devstack http://logs.openstack.org/06/664506/2/check/mistral-devstack/7b4f731/ : SUCCESS in 44m 22s\n- mistral-devstack-non-apache http://logs.openstack.org/06/664506/2/check/mistral-devstack-non-apache/989297d/ : SUCCESS in 45m 05s\n- mistral-devstack-kombu http://logs.openstack.org/06/664506/2/check/mistral-devstack-kombu/c11171e/ : SUCCESS in 43m 41s\n- mistral-tox-unit-mysql http://logs.openstack.org/06/664506/2/check/mistral-tox-unit-mysql/a62dced/ : SUCCESS in 27m 42s\n- mistral-tox-unit-postgresql http://logs.openstack.org/06/664506/2/check/mistral-tox-unit-postgresql/06d8b9e/ : SUCCESS in 27m 06s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/06/664506/2/check/tripleo-ci-centos-7-scenario003-standalone/a7996bf/ : SUCCESS in 1h 35m 08s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/06/664506/2/check/tripleo-ci-centos-7-containers-multinode/6c25015/ : SUCCESS in 2h 11m 58s\n- mistral-rally-task http://logs.openstack.org/06/664506/2/check/mistral-rally-task/b6e6eb3/ : SUCCESS in 1h 05m 39s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/06/664506/2/check/openstack-tox-lower-constraints/5ed5b85/ : SUCCESS in 10m 17s","accounts_in_message":[],"_revision_number":2},{"id":"31ffd420c8aa9e479c590f42d289fbfa1ec0c2cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 10:33:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d7324c522280c33b9ed1d8015d8027b997df97cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 12:31:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"71c76ff62b2a345dc74f1fbc462416a6b4bb360a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 12:31:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/06/664506/2/gate/openstack-tox-pep8/99fdd08/ : SUCCESS in 6m 13s\n- openstack-tox-py27 http://logs.openstack.org/06/664506/2/gate/openstack-tox-py27/2ac9efc/ : SUCCESS in 8m 44s\n- openstack-tox-py36 http://logs.openstack.org/06/664506/2/gate/openstack-tox-py36/999c978/ : SUCCESS in 8m 59s\n- openstack-tox-py37 http://logs.openstack.org/06/664506/2/gate/openstack-tox-py37/21c81d6/ : SUCCESS in 9m 45s\n- openstack-tox-docs http://logs.openstack.org/06/664506/2/gate/openstack-tox-docs/fb459df/html/ : SUCCESS in 4m 48s\n- mistral-devstack http://logs.openstack.org/06/664506/2/gate/mistral-devstack/a5aa257/ : SUCCESS in 49m 10s\n- mistral-devstack-non-apache http://logs.openstack.org/06/664506/2/gate/mistral-devstack-non-apache/75af42b/ : SUCCESS in 46m 48s\n- mistral-tox-unit-mysql http://logs.openstack.org/06/664506/2/gate/mistral-tox-unit-mysql/d97b9be/ : SUCCESS in 28m 04s\n- mistral-tox-unit-postgresql http://logs.openstack.org/06/664506/2/gate/mistral-tox-unit-postgresql/219907a/ : SUCCESS in 25m 58s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/06/664506/2/gate/tripleo-ci-centos-7-scenario003-standalone/04d1929/ : SUCCESS in 1h 21m 28s\n- mistral-devstack-kombu http://logs.openstack.org/06/664506/2/gate/mistral-devstack-kombu/800d7a6/ : SUCCESS in 47m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/664506/2/gate/openstack-tox-lower-constraints/897daa9/ : SUCCESS in 10m 19s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/06/664506/2/gate/tripleo-ci-centos-7-undercloud-containers/fa71db4/ : SUCCESS in 1h 54m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aed4b18c3e90ade30d3af8ff99495682be75129c","revisions":{"2a2a5563104cfa9d0ae4f7fb53ce61e31758c394":{"kind":"REWORK","_number":1,"created":"2019-06-11 07:07:30.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/06/664506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/06/664506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/664506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/664506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/664506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/06/664506/1"}}},"commit":{"parents":[{"commit":"d1d4087a3fa08ad119b4edbd7a307b0bb991f062","subject":"Merge \"Simple optimization of creating/updating workflows\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d1d4087a3fa08ad119b4edbd7a307b0bb991f062"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-06-11 07:05:38.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-06-11 07:05:38.000000000","tz":420},"subject":"Fix sporadically failing test test_with_items_action_context","message":"Fix sporadically failing test test_with_items_action_context\n\n* The test fails sometimes if std.http returns something like 30x\n  status code. This patch fixes it by replacing std.http with\n  std.async_noop. The entire module though requires revisiting\n  and refactoring since lots of things don\u0027t look relevant\n  anymore.\n\nChange-Id: Iadf333fcc6f241d707daf2767bab9cfbf2715917\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2a2a5563104cfa9d0ae4f7fb53ce61e31758c394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2a2a5563104cfa9d0ae4f7fb53ce61e31758c394"}]},"branch":"refs/heads/master"},"aed4b18c3e90ade30d3af8ff99495682be75129c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-11 08:01:00.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/06/664506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/06/664506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/664506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/664506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/664506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/06/664506/2"}}},"commit":{"parents":[{"commit":"2a1063e7d4452c87c973d075947004c8bf668fca","subject":"Add python 3.7 classifier to setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2a1063e7d4452c87c973d075947004c8bf668fca"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-06-11 07:05:38.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-06-11 08:01:00.000000000","tz":0},"subject":"Fix sporadically failing test test_with_items_action_context","message":"Fix sporadically failing test test_with_items_action_context\n\n* The test fails sometimes if std.http returns something like 30x\n  status code. This patch fixes it by replacing std.http with\n  std.async_noop. The entire module though requires revisiting\n  and refactoring since lots of things don\u0027t look relevant\n  anymore.\n\nChange-Id: Iadf333fcc6f241d707daf2767bab9cfbf2715917\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aed4b18c3e90ade30d3af8ff99495682be75129c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aed4b18c3e90ade30d3af8ff99495682be75129c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
