)]}'
{"id":"openstack%2Fmistral~664763","triplet_id":"openstack%2Fmistral~master~I3bedeb6cc1141f01e8a4148cefb3b6d98d7dfb2f","project":"openstack/mistral","branch":"master","topic":"Bug#4","hashtags":[],"change_id":"I3bedeb6cc1141f01e8a4148cefb3b6d98d7dfb2f","subject":"Use openstack-python3-train-jobs for python3 test runtime","status":"MERGED","created":"2019-06-12 02:29:21.000000000","updated":"2019-06-12 08:34:00.000000000","submitted":"2019-06-12 08:33:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664763-1560328440020-48a02ede","meta_rev_id":"2dd27457a42e52fc64ed41992b9adf8a722407fa","_number":664763,"virtual_id_number":664763,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-06-12 08:33:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-06-12 06:34:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"date":"2019-06-12 06:28:28.000000000","_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"date":"2019-06-12 06:28:28.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-06-12 06:34:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2019-06-12 06:28:28.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"date":"2019-06-12 06:28:28.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"date":"2019-06-12 06:28:28.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 06:28:28.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-06-12 06:28:28.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-06-12 06:28:28.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-06-12 06:28:28.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-06-12 06:28:28.000000000","updated_by":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"reviewer":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"state":"REVIEWER"},{"updated":"2019-06-12 06:34:03.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-06-12 08:33:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d3e0e9a741e375215729bd363acb57658b94ecb3","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-12 02:29:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f62099465ca99025370dc17ab7a6087a3b0aef0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 05:40:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/63/664763/1/check/openstack-tox-pep8/e4d6ecf/ : SUCCESS in 5m 24s\n- openstack-tox-py27 http://logs.openstack.org/63/664763/1/check/openstack-tox-py27/750765d/ : SUCCESS in 10m 53s\n- openstack-tox-py36 http://logs.openstack.org/63/664763/1/check/openstack-tox-py36/2122267/ : SUCCESS in 8m 53s\n- openstack-tox-py37 http://logs.openstack.org/63/664763/1/check/openstack-tox-py37/f32970c/ : SUCCESS in 10m 59s\n- openstack-tox-docs http://logs.openstack.org/63/664763/1/check/openstack-tox-docs/833d363/html/ : SUCCESS in 4m 54s\n- mistral-devstack http://logs.openstack.org/63/664763/1/check/mistral-devstack/14cb797/ : SUCCESS in 37m 17s\n- mistral-devstack-non-apache http://logs.openstack.org/63/664763/1/check/mistral-devstack-non-apache/a627552/ : SUCCESS in 1h 03m 38s\n- mistral-devstack-kombu http://logs.openstack.org/63/664763/1/check/mistral-devstack-kombu/27f2f31/ : SUCCESS in 42m 39s\n- mistral-tox-unit-mysql http://logs.openstack.org/63/664763/1/check/mistral-tox-unit-mysql/dbc5ac7/ : SUCCESS in 27m 59s\n- mistral-tox-unit-postgresql http://logs.openstack.org/63/664763/1/check/mistral-tox-unit-postgresql/51ca45e/ : SUCCESS in 26m 45s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/63/664763/1/check/tripleo-ci-centos-7-scenario003-standalone/0f907a6/ : SUCCESS in 1h 16m 08s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/63/664763/1/check/tripleo-ci-centos-7-containers-multinode/1102b16/ : SUCCESS in 2h 08m 59s\n- mistral-rally-task http://logs.openstack.org/63/664763/1/check/mistral-rally-task/3465ad7/ : SUCCESS in 1h 09m 06s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/63/664763/1/check/openstack-tox-lower-constraints/7634500/ : SUCCESS in 11m 02s","accounts_in_message":[],"_revision_number":1},{"id":"48e50198be513ae9f78caa44f832674e127cf36c","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-06-12 06:34:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4a69eaf0263b6289e67450c99a9ab701aea81fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 06:34:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"60541034a3470fa43300902680d73a3fed0b8446","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 08:33:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/63/664763/1/gate/openstack-tox-pep8/bd919a1/ : SUCCESS in 5m 38s\n- openstack-tox-py27 http://logs.openstack.org/63/664763/1/gate/openstack-tox-py27/2a764be/ : SUCCESS in 8m 44s\n- openstack-tox-py36 http://logs.openstack.org/63/664763/1/gate/openstack-tox-py36/61d1e74/ : SUCCESS in 9m 00s\n- openstack-tox-py37 http://logs.openstack.org/63/664763/1/gate/openstack-tox-py37/14f57a1/ : SUCCESS in 9m 18s\n- openstack-tox-docs http://logs.openstack.org/63/664763/1/gate/openstack-tox-docs/125da54/html/ : SUCCESS in 5m 17s\n- mistral-devstack http://logs.openstack.org/63/664763/1/gate/mistral-devstack/63ee40e/ : SUCCESS in 42m 09s\n- mistral-devstack-non-apache http://logs.openstack.org/63/664763/1/gate/mistral-devstack-non-apache/39b6b3a/ : SUCCESS in 36m 19s\n- mistral-tox-unit-mysql http://logs.openstack.org/63/664763/1/gate/mistral-tox-unit-mysql/23f0ed7/ : SUCCESS in 27m 20s\n- mistral-tox-unit-postgresql http://logs.openstack.org/63/664763/1/gate/mistral-tox-unit-postgresql/20b5835/ : SUCCESS in 26m 54s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/63/664763/1/gate/tripleo-ci-centos-7-scenario003-standalone/a9a6ed4/ : SUCCESS in 1h 14m 44s\n- mistral-devstack-kombu http://logs.openstack.org/63/664763/1/gate/mistral-devstack-kombu/1de76ab/ : SUCCESS in 45m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/664763/1/gate/openstack-tox-lower-constraints/6397537/ : SUCCESS in 10m 19s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/63/664763/1/gate/tripleo-ci-centos-7-undercloud-containers/8ae3dff/ : SUCCESS in 1h 51m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0155169f5d5ac01e08bfb51e3dd36beff68bd833","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 08:34:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c5e948d3729d1ba1d41dfa0aee96333ad0f5db3","revisions":{"5c5e948d3729d1ba1d41dfa0aee96333ad0f5db3":{"kind":"REWORK","_number":1,"created":"2019-06-12 02:29:21.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/63/664763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/63/664763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/664763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/664763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/63/664763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/63/664763/1"}}},"commit":{"parents":[{"commit":"fbbb9790c7efe254ae903ae16d865f276c71007c","subject":"Merge \"Handle action inputs properly to prevent tasks stuck in RUNNING state\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fbbb9790c7efe254ae903ae16d865f276c71007c"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-12 02:10:16.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-12 02:10:22.000000000","tz":480},"subject":"Use openstack-python3-train-jobs for python3 test runtime","message":"Use openstack-python3-train-jobs for python3 test runtime\n\nDepends-On:https://review.opendev.org/#/c/641878/\n\nChange-Id: I3bedeb6cc1141f01e8a4148cefb3b6d98d7dfb2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5c5e948d3729d1ba1d41dfa0aee96333ad0f5db3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5c5e948d3729d1ba1d41dfa0aee96333ad0f5db3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
