)]}'
{"id":"openstack%2Fmistral~665553","triplet_id":"openstack%2Fmistral~master~I5cb144903cd2abb6eecfe32a13da4a2ebf7db3dc","project":"openstack/mistral","branch":"master","topic":"rework_getting_upstream_tasks","hashtags":[],"change_id":"I5cb144903cd2abb6eecfe32a13da4a2ebf7db3dc","subject":"Optimize finding upstream task executions","status":"MERGED","created":"2019-06-16 18:17:08.000000000","updated":"2019-06-20 12:42:46.000000000","submitted":"2019-06-20 12:42:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665553-1561034566471-b3e15709","meta_rev_id":"b557a62c3daec364c26bfbe3ee45e633c5d72ed1","_number":665553,"virtual_id_number":665553,"owner":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-06-20 12:42:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-20 09:16:47.000000000","_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"recommended":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":1,"date":"2019-06-17 09:34:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2019-06-20 07:27:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-06-17 08:24:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2019-06-20 07:27:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 08:24:47.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-06-17 09:34:36.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-06-20 07:27:02.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-06-20 12:42:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7f5f8f14d5fbf39c6769086e34f8b8db157558b","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-16 18:17:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80e65f4bfb1d08ec81c5c5e47f0b303e3dcb5e5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-16 20:14:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/53/665553/1/check/openstack-tox-pep8/9009062/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/53/665553/1/check/openstack-tox-py27/55efe9d/ : SUCCESS in 8m 16s\n- openstack-tox-py36 http://logs.openstack.org/53/665553/1/check/openstack-tox-py36/f946a76/ : SUCCESS in 8m 29s\n- openstack-tox-py37 http://logs.openstack.org/53/665553/1/check/openstack-tox-py37/3490c46/ : SUCCESS in 10m 16s\n- openstack-tox-docs http://logs.openstack.org/53/665553/1/check/openstack-tox-docs/fe3b2e5/html/ : SUCCESS in 4m 43s\n- openstack-tox-cover http://logs.openstack.org/53/665553/1/check/openstack-tox-cover/9868e61/cover/ : SUCCESS in 9m 00s (non-voting)\n- mistral-devstack http://logs.openstack.org/53/665553/1/check/mistral-devstack/9f04511/ : SUCCESS in 41m 33s\n- mistral-devstack-non-apache http://logs.openstack.org/53/665553/1/check/mistral-devstack-non-apache/527ae5f/ : SUCCESS in 45m 27s\n- mistral-devstack-kombu http://logs.openstack.org/53/665553/1/check/mistral-devstack-kombu/61a45ad/ : SUCCESS in 42m 20s\n- mistral-tox-unit-mysql http://logs.openstack.org/53/665553/1/check/mistral-tox-unit-mysql/c72db39/ : SUCCESS in 27m 38s\n- mistral-tox-unit-postgresql http://logs.openstack.org/53/665553/1/check/mistral-tox-unit-postgresql/0ab55cf/ : SUCCESS in 26m 34s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/53/665553/1/check/tripleo-ci-centos-7-scenario003-standalone/51af4d9/ : SUCCESS in 1h 10m 29s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/53/665553/1/check/tripleo-ci-centos-7-containers-multinode/4be566e/ : SUCCESS in 1h 51m 40s\n- mistral-rally-task http://logs.openstack.org/53/665553/1/check/mistral-rally-task/d1642b2/ : SUCCESS in 57m 15s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/53/665553/1/check/openstack-tox-lower-constraints/b5d87da/ : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1c841936e6859edfe198006ec6fdc7eed33f88c6","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-17 04:00:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nPlease clarify a couple of things.","accounts_in_message":[],"_revision_number":1},{"id":"0dc17aa47058f639a01de30c24402f3f9d287b74","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-17 08:24:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6934db80b4881c4ff5d0139cb867b4b13b7db25","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-17 09:34:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9dbb711ab04bce33a1de76fdef35361274faee1f","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-06-18 07:24:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThe code is fine and I will +2 it, but could we have some comments there?","accounts_in_message":[],"_revision_number":1},{"id":"047fd9b2866717d44202b8871d3547cd5598cd64","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-18 08:18:37.000000000","message":"Patch Set 1:\n\nMike, for me it\u0027s relatively OK to understand but can we really come up with some small comments where we deliver the main idea of the code?","accounts_in_message":[],"_revision_number":1},{"id":"de6c73a7057ada92bd1e9018300a8ced3507f3a3","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-18 11:44:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c1f56a63bdb882683c273651c184565cbf6c59b","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-06-20 07:27:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c382630da5e55eac4680d17cd72f1ba5e1fa3c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 07:27:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"251c83781beb17997f8ec4cc24d9983e23aa0faa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 09:00:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/53/665553/1/gate/openstack-tox-pep8/8389c6c/ : SUCCESS in 5m 47s\n- openstack-tox-py27 http://logs.openstack.org/53/665553/1/gate/openstack-tox-py27/c8bea19/ : SUCCESS in 15m 02s\n- openstack-tox-py36 http://logs.openstack.org/53/665553/1/gate/openstack-tox-py36/6f0baa8/ : SUCCESS in 17m 42s\n- openstack-tox-py37 http://logs.openstack.org/53/665553/1/gate/openstack-tox-py37/1fa8886/ : SUCCESS in 9m 49s\n- openstack-tox-docs http://logs.openstack.org/53/665553/1/gate/openstack-tox-docs/09604ca/html/ : SUCCESS in 6m 31s\n- mistral-devstack http://logs.openstack.org/53/665553/1/gate/mistral-devstack/a6dc414/ : SUCCESS in 1h 10m 29s\n- mistral-devstack-non-apache http://logs.openstack.org/53/665553/1/gate/mistral-devstack-non-apache/ab3a541/ : SUCCESS in 39m 42s\n- mistral-tox-unit-mysql http://logs.openstack.org/53/665553/1/gate/mistral-tox-unit-mysql/892b94a/ : SUCCESS in 28m 47s\n- mistral-tox-unit-postgresql http://logs.openstack.org/53/665553/1/gate/mistral-tox-unit-postgresql/fb9401c/ : SUCCESS in 27m 14s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/53/665553/1/gate/tripleo-ci-centos-7-scenario003-standalone/3c91eb1/ : SUCCESS in 1h 21m 54s\n- mistral-devstack-kombu http://logs.openstack.org/53/665553/1/gate/mistral-devstack-kombu/443be80/ : SUCCESS in 46m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/665553/1/gate/openstack-tox-lower-constraints/2263f68/ : SUCCESS in 10m 05s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/53/665553/1/gate/tripleo-ci-centos-7-undercloud-containers/2ca30f7/ : FAILURE in 1h 27m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0558484f94b5a9c41eb489c6bd650003aaf79a02","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-20 09:03:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"084d5fc78b41ded3ebf55fca8537941eb83a9b4a","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-20 09:16:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de994d0515ae9a53435d08b1b810841062da69d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 11:09:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/53/665553/1/check/openstack-tox-pep8/74c42f4/ : SUCCESS in 13m 28s\n- openstack-tox-py27 http://logs.openstack.org/53/665553/1/check/openstack-tox-py27/cb800b7/ : SUCCESS in 8m 08s\n- openstack-tox-py36 http://logs.openstack.org/53/665553/1/check/openstack-tox-py36/f3fb646/ : SUCCESS in 8m 42s\n- openstack-tox-py37 http://logs.openstack.org/53/665553/1/check/openstack-tox-py37/bcdc6d4/ : SUCCESS in 8m 51s\n- openstack-tox-docs http://logs.openstack.org/53/665553/1/check/openstack-tox-docs/c88ca30/html/ : SUCCESS in 4m 53s\n- openstack-tox-cover http://logs.openstack.org/53/665553/1/check/openstack-tox-cover/92e28e8/cover/ : SUCCESS in 11m 23s (non-voting)\n- mistral-devstack http://logs.openstack.org/53/665553/1/check/mistral-devstack/9f414c4/ : SUCCESS in 41m 00s\n- mistral-devstack-non-apache http://logs.openstack.org/53/665553/1/check/mistral-devstack-non-apache/752f74a/ : SUCCESS in 1h 35m 27s\n- mistral-devstack-kombu http://logs.openstack.org/53/665553/1/check/mistral-devstack-kombu/5261cdf/ : SUCCESS in 1h 35m 00s\n- mistral-tox-unit-mysql http://logs.openstack.org/53/665553/1/check/mistral-tox-unit-mysql/8a7607c/ : SUCCESS in 28m 15s\n- mistral-tox-unit-postgresql http://logs.openstack.org/53/665553/1/check/mistral-tox-unit-postgresql/199d7ca/ : SUCCESS in 28m 21s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/53/665553/1/check/tripleo-ci-centos-7-scenario003-standalone/c2ae50a/ : SUCCESS in 1h 14m 46s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/53/665553/1/check/tripleo-ci-centos-7-containers-multinode/1095c82/ : SUCCESS in 2h 01m 46s\n- mistral-rally-task http://logs.openstack.org/53/665553/1/check/mistral-rally-task/9d33efc/ : SUCCESS in 1h 05m 22s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/53/665553/1/check/openstack-tox-lower-constraints/aafacd3/ : SUCCESS in 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"2e61002d4ee9942edb03ee8877f6144ef0924cfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 11:09:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fad55628808243334d7907c6eed2b1be7b4d3878","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 12:42:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/53/665553/1/gate/openstack-tox-pep8/e76a8e2/ : SUCCESS in 14m 01s\n- openstack-tox-py27 http://logs.openstack.org/53/665553/1/gate/openstack-tox-py27/1b0865f/ : SUCCESS in 8m 18s\n- openstack-tox-py36 http://logs.openstack.org/53/665553/1/gate/openstack-tox-py36/6ce171e/ : SUCCESS in 11m 10s\n- openstack-tox-py37 http://logs.openstack.org/53/665553/1/gate/openstack-tox-py37/fd0153f/ : SUCCESS in 9m 29s\n- openstack-tox-docs http://logs.openstack.org/53/665553/1/gate/openstack-tox-docs/2dc5c84/html/ : SUCCESS in 5m 20s\n- mistral-devstack http://logs.openstack.org/53/665553/1/gate/mistral-devstack/2cece44/ : SUCCESS in 42m 11s\n- mistral-devstack-non-apache http://logs.openstack.org/53/665553/1/gate/mistral-devstack-non-apache/6541c77/ : SUCCESS in 43m 17s\n- mistral-tox-unit-mysql http://logs.openstack.org/53/665553/1/gate/mistral-tox-unit-mysql/7e53112/ : SUCCESS in 29m 52s\n- mistral-tox-unit-postgresql http://logs.openstack.org/53/665553/1/gate/mistral-tox-unit-postgresql/302a9b7/ : SUCCESS in 29m 56s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/53/665553/1/gate/tripleo-ci-centos-7-scenario003-standalone/5e51f45/ : SUCCESS in 1h 19m 51s\n- mistral-devstack-kombu http://logs.openstack.org/53/665553/1/gate/mistral-devstack-kombu/2d1ddd1/ : SUCCESS in 43m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/665553/1/gate/openstack-tox-lower-constraints/ff5e8d3/ : SUCCESS in 13m 25s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/53/665553/1/gate/tripleo-ci-centos-7-undercloud-containers/d6b67b7/ : SUCCESS in 1h 23m 46s","accounts_in_message":[],"_revision_number":1},{"id":"dfb28b2c4c6cab655f5346f52912e9e322cccc23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 12:42:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0fb101c47d0984cdcd34cb6c48ece066c82599e","revisions":{"b0fb101c47d0984cdcd34cb6c48ece066c82599e":{"kind":"REWORK","_number":1,"created":"2019-06-16 18:17:08.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/53/665553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/665553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/665553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/665553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/665553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/665553/1"}}},"commit":{"parents":[{"commit":"f4416678ba0ed1aa464cd62fe898f11871e152b9","subject":"Merge \"Direct workflow code cleanup and refactoring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f4416678ba0ed1aa464cd62fe898f11871e152b9"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-16 18:11:33.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-16 18:11:33.000000000","tz":120},"subject":"Optimize finding upstream task executions","message":"Optimize finding upstream task executions\n\nThis commit introduces next optimizations:\n1. No need to send a request if the list of inbound tasks is empty.\n2. Instead of checking already downloaded tasks, it is more reasonable\nto add filters to the request.\n3. Generation of task execution cache is meaningless here, because\nwe only check the nearest tasks without going up the graph.\n\nChange-Id: I5cb144903cd2abb6eecfe32a13da4a2ebf7db3dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b0fb101c47d0984cdcd34cb6c48ece066c82599e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b0fb101c47d0984cdcd34cb6c48ece066c82599e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
