)]}'
{"id":"openstack%2Fmistral~667039","triplet_id":"openstack%2Fmistral~master~I683c79fa0a3ab23a16c493ce2314a506dfee9749","project":"openstack/mistral","branch":"master","topic":"init_lazy_loaded_execution_input_field_in_api","hashtags":[],"change_id":"I683c79fa0a3ab23a16c493ce2314a506dfee9749","subject":"Initialize the lazily loaded execution \"input\" field in API","status":"MERGED","created":"2019-06-24 08:13:57.000000000","updated":"2019-06-28 15:52:49.000000000","submitted":"2019-06-28 15:52:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667039-1561737169121-a45a0009","meta_rev_id":"53b11fc5b0b668b4a63f7bb1f2c8ee0d9878764c","_number":667039,"virtual_id_number":667039,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-06-28 15:52:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-06-25 06:04:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":1,"date":"2019-06-25 13:06:09.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-25 05:54:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-06-24 12:28:12.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2019-06-24 12:28:12.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2019-06-24 12:28:12.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-06-25 06:42:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 12:28:12.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-06-24 12:28:12.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-06-24 12:28:12.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-06-25 05:54:40.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-06-25 06:04:53.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-06-25 13:06:09.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-06-28 15:52:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"873987bcc5f5436d49d811d7a9d2996578b55b11","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-24 08:13:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6bb82a7277be053a826c951a425d9d2090ae3f8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-24 10:06:35.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"640782ba318dfebd597f17c128d0fc161053c1df","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-24 10:19:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6a5f53ef112a90d30fca437a19d2a8926a9035ce","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-24 10:20:14.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"fc7d60a17c7bacd2551df1b5cb581879aa754012","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-24 12:26:51.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"ace9ecbb3a074509fd071254d51c26db2a78ced9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-24 12:27:18.000000000","message":"Patch Set 2:\n\nWe also need to cover this with tempest tests.","accounts_in_message":[],"_revision_number":2},{"id":"9f72140c116ff96df50468e188003974d8644f0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 12:50:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/39/667039/2/check/openstack-tox-pep8/f20b116/ : SUCCESS in 6m 36s\n- openstack-tox-py27 http://logs.openstack.org/39/667039/2/check/openstack-tox-py27/20b91ab/ : SUCCESS in 9m 48s\n- openstack-tox-py36 http://logs.openstack.org/39/667039/2/check/openstack-tox-py36/f123254/ : SUCCESS in 8m 39s\n- openstack-tox-py37 http://logs.openstack.org/39/667039/2/check/openstack-tox-py37/19422e9/ : SUCCESS in 8m 44s\n- openstack-tox-docs http://logs.openstack.org/39/667039/2/check/openstack-tox-docs/22e4c67/html/ : SUCCESS in 4m 50s\n- openstack-tox-cover http://logs.openstack.org/39/667039/2/check/openstack-tox-cover/765757e/cover/ : SUCCESS in 11m 59s (non-voting)\n- mistral-devstack http://logs.openstack.org/39/667039/2/check/mistral-devstack/63373d0/ : SUCCESS in 48m 02s\n- mistral-devstack-non-apache http://logs.openstack.org/39/667039/2/check/mistral-devstack-non-apache/005296a/ : SUCCESS in 45m 02s\n- mistral-devstack-kombu http://logs.openstack.org/39/667039/2/check/mistral-devstack-kombu/82ae37e/ : SUCCESS in 44m 03s\n- mistral-tox-unit-mysql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-mysql/a692b09/ : FAILURE in 30m 10s\n- mistral-tox-unit-postgresql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-postgresql/1a2e3a4/ : SUCCESS in 30m 26s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/667039/2/check/tripleo-ci-centos-7-scenario003-standalone/c4a7a61/ : SUCCESS in 1h 36m 12s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/667039/2/check/tripleo-ci-centos-7-containers-multinode/ce9a606/ : SUCCESS in 2h 23m 55s\n- mistral-rally-task http://logs.openstack.org/39/667039/2/check/mistral-rally-task/c5c581b/ : SUCCESS in 1h 06m 01s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/39/667039/2/check/openstack-tox-lower-constraints/9eea473/ : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":2},{"id":"c88d8579ab71ef422755ffcbc8ed9a379c7ed23e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-24 13:01:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fe3899a198b4fb946ab185ffa7f01772df45088c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 15:43:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/39/667039/2/check/openstack-tox-pep8/57c897c/ : SUCCESS in 5m 21s\n- openstack-tox-py27 http://logs.openstack.org/39/667039/2/check/openstack-tox-py27/6287cd6/ : SUCCESS in 9m 21s\n- openstack-tox-py36 http://logs.openstack.org/39/667039/2/check/openstack-tox-py36/370e4dd/ : SUCCESS in 8m 53s\n- openstack-tox-py37 http://logs.openstack.org/39/667039/2/check/openstack-tox-py37/1e66680/ : SUCCESS in 10m 16s\n- openstack-tox-docs http://logs.openstack.org/39/667039/2/check/openstack-tox-docs/1aad778/html/ : SUCCESS in 5m 13s\n- openstack-tox-cover http://logs.openstack.org/39/667039/2/check/openstack-tox-cover/9cfc3d1/cover/ : SUCCESS in 9m 49s (non-voting)\n- mistral-devstack http://logs.openstack.org/39/667039/2/check/mistral-devstack/7dcdd68/ : SUCCESS in 52m 12s\n- mistral-devstack-non-apache http://logs.openstack.org/39/667039/2/check/mistral-devstack-non-apache/1822e35/ : SUCCESS in 44m 05s\n- mistral-devstack-kombu http://logs.openstack.org/39/667039/2/check/mistral-devstack-kombu/593e2f0/ : SUCCESS in 43m 32s\n- mistral-tox-unit-mysql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-mysql/a155ace/ : SUCCESS in 27m 43s\n- mistral-tox-unit-postgresql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-postgresql/ab9ce34/ : SUCCESS in 27m 21s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/667039/2/check/tripleo-ci-centos-7-scenario003-standalone/bdf8189/ : SUCCESS in 1h 18m 09s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/667039/2/check/tripleo-ci-centos-7-containers-multinode/5d2a835/ : SUCCESS in 2h 15m 50s\n- mistral-rally-task http://logs.openstack.org/39/667039/2/check/mistral-rally-task/dc0e743/ : SUCCESS in 1h 01m 27s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/39/667039/2/check/openstack-tox-lower-constraints/96d6895/ : SUCCESS in 11m 49s","accounts_in_message":[],"_revision_number":2},{"id":"8c49e2728db75f6eb7ebfd24542acf1437951071","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-06-25 05:54:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ff381810fa55b5430401bab2a2078c58bcde2cd7","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-06-25 06:04:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2c4a76b063b234aaf8c5949c552e5baeaa45c9e0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-25 06:42:57.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"95a3d6eac2d900c3885e2f83adfa25143898090b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 06:43:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"945c2f68df60d1c35d9385ce11ec9aa9618d6c37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 08:53:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/39/667039/2/gate/openstack-tox-pep8/7c1991f/ : RETRY_LIMIT in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/39/667039/2/gate/openstack-tox-py27/387a288/ : RETRY_LIMIT in 2m 49s\n- openstack-tox-py36 http://logs.openstack.org/39/667039/2/gate/openstack-tox-py36/21a1dfe/ : RETRY_LIMIT in 4m 11s\n- openstack-tox-py37 http://logs.openstack.org/39/667039/2/gate/openstack-tox-py37/a57d3da/ : RETRY_LIMIT in 2m 33s\n- openstack-tox-docs http://logs.openstack.org/39/667039/2/gate/openstack-tox-docs/9978d32/html/ : SUCCESS in 3m 25s\n- mistral-devstack http://logs.openstack.org/39/667039/2/gate/mistral-devstack/a501690/ : SUCCESS in 49m 55s\n- mistral-devstack-non-apache http://logs.openstack.org/39/667039/2/gate/mistral-devstack-non-apache/ec98cb0/ : SUCCESS in 43m 14s\n- mistral-tox-unit-mysql http://logs.openstack.org/39/667039/2/gate/mistral-tox-unit-mysql/59b866c/ : RETRY_LIMIT in 3m 37s\n- mistral-tox-unit-postgresql http://logs.openstack.org/39/667039/2/gate/mistral-tox-unit-postgresql/7315722/ : RETRY_LIMIT in 3m 32s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/667039/2/gate/tripleo-ci-centos-7-scenario003-standalone/f6f32cf/ : SUCCESS in 1h 19m 55s\n- mistral-devstack-kombu http://logs.openstack.org/39/667039/2/gate/mistral-devstack-kombu/0de8b2d/ : SUCCESS in 49m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/39/667039/2/gate/openstack-tox-lower-constraints/3b81fb3/ : RETRY_LIMIT in 2m 03s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/39/667039/2/gate/tripleo-ci-centos-7-undercloud-containers/c724cd2/ : SUCCESS in 1h 52m 00s","accounts_in_message":[],"_revision_number":2},{"id":"d097e84ed9b0f7d8107e2dfe4338c64068aae9aa","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-06-25 08:54:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"68ef56d03abc0d5b946abf2644cadb49e65807bf","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-25 11:37:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"387698568b9bcea9daa95dee3390e8be4d338500","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 12:14:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/39/667039/2/check/openstack-tox-pep8/2b4b7d1/ : RETRY_LIMIT in 2m 01s\n- openstack-tox-py27 http://logs.openstack.org/39/667039/2/check/openstack-tox-py27/367daa6/ : RETRY_LIMIT in 2m 15s\n- openstack-tox-py36 http://logs.openstack.org/39/667039/2/check/openstack-tox-py36/1e5965b/ : RETRY_LIMIT in 2m 24s\n- openstack-tox-py37 http://logs.openstack.org/39/667039/2/check/openstack-tox-py37/27399db/ : RETRY_LIMIT in 3m 02s\n- openstack-tox-docs http://logs.openstack.org/39/667039/2/check/openstack-tox-docs/0b590f7/html/ : SUCCESS in 4m 06s\n- openstack-tox-cover http://logs.openstack.org/39/667039/2/check/openstack-tox-cover/2d94435/ : RETRY_LIMIT in 2m 26s (non-voting)\n- mistral-devstack http://logs.openstack.org/39/667039/2/check/mistral-devstack/3e95bdc/ : SUCCESS in 43m 49s\n- mistral-devstack-non-apache http://logs.openstack.org/39/667039/2/check/mistral-devstack-non-apache/45e6fc1/ : SUCCESS in 43m 49s\n- mistral-devstack-kombu http://logs.openstack.org/39/667039/2/check/mistral-devstack-kombu/856f27b/ : SUCCESS in 45m 11s\n- mistral-tox-unit-mysql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-mysql/23b00bf/ : RETRY_LIMIT in 2m 52s\n- mistral-tox-unit-postgresql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-postgresql/adf4904/ : RETRY_LIMIT in 2m 59s\n- tripleo-ci-centos-7-scenario003-standalone finger://ze02.openstack.org/39cabaa641804d7194698c72e713444a : RETRY_LIMIT in 38m 00s\n- tripleo-ci-centos-7-containers-multinode finger://ze12.openstack.org/cc190f9634fc4f8189f2b31a7a46cf4a : RETRY_LIMIT in 57m 30s\n- mistral-rally-task finger://ze01.openstack.org/d215afb2a1fc41baaae48682a37950f9 : RETRY_LIMIT in 1h 01m 34s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/39/667039/2/check/openstack-tox-lower-constraints/1b1358d/ : RETRY_LIMIT in 2m 56s","accounts_in_message":[],"_revision_number":2},{"id":"f0d0a62c4ddd8806e8c21961473ab161b24a3695","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-06-25 12:14:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"efc30768c47e4dd14079af2b2bd3da53e3b040fb","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-25 13:06:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"48c9f7be6b9052b4ea8640d5e7d54a03e82194c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 18:13:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/39/667039/2/check/openstack-tox-pep8/83fcf58/ : RETRY_LIMIT in 2m 14s\n- openstack-tox-py27 http://logs.openstack.org/39/667039/2/check/openstack-tox-py27/c84a5bc/ : RETRY_LIMIT in 2m 35s\n- openstack-tox-py36 http://logs.openstack.org/39/667039/2/check/openstack-tox-py36/7fe9f42/ : RETRY_LIMIT in 2m 10s\n- openstack-tox-py37 http://logs.openstack.org/39/667039/2/check/openstack-tox-py37/96fe378/ : RETRY_LIMIT in 2m 24s\n- openstack-tox-docs http://logs.openstack.org/39/667039/2/check/openstack-tox-docs/e52c221/html/ : SUCCESS in 7m 13s\n- openstack-tox-cover http://logs.openstack.org/39/667039/2/check/openstack-tox-cover/1839a25/ : RETRY_LIMIT in 2m 34s (non-voting)\n- mistral-devstack http://logs.openstack.org/39/667039/2/check/mistral-devstack/437fbff/ : SUCCESS in 46m 19s\n- mistral-devstack-non-apache http://logs.openstack.org/39/667039/2/check/mistral-devstack-non-apache/9c3befa/ : SUCCESS in 40m 09s\n- mistral-devstack-kombu http://logs.openstack.org/39/667039/2/check/mistral-devstack-kombu/ddea35a/ : SUCCESS in 46m 44s\n- mistral-tox-unit-mysql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-mysql/d145927/ : RETRY_LIMIT in 1m 58s\n- mistral-tox-unit-postgresql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-postgresql/091537b/ : RETRY_LIMIT in 2m 17s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/667039/2/check/tripleo-ci-centos-7-scenario003-standalone/e98be4a/ : SUCCESS in 1h 20m 46s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/667039/2/check/tripleo-ci-centos-7-containers-multinode/049e78a/ : SUCCESS in 2h 08m 55s\n- mistral-rally-task http://logs.openstack.org/39/667039/2/check/mistral-rally-task/1524ec4/ : SUCCESS in 1h 01m 10s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/39/667039/2/check/openstack-tox-lower-constraints/66a4cc1/ : RETRY_LIMIT in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"13f112531cf7b4090ed37eb591ef2da430911f25","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-06-26 06:22:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"67429af1198894c268bba48db53fcb2427835315","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 10:20:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/39/667039/2/check/openstack-tox-pep8/b8423b2/ : RETRY_LIMIT in 2m 15s\n- openstack-tox-py27 http://logs.openstack.org/39/667039/2/check/openstack-tox-py27/55ce1b4/ : RETRY_LIMIT in 2m 21s\n- openstack-tox-py36 http://logs.openstack.org/39/667039/2/check/openstack-tox-py36/3dd91df/ : RETRY_LIMIT in 2m 16s\n- openstack-tox-py37 http://logs.openstack.org/39/667039/2/check/openstack-tox-py37/60569fd/ : RETRY_LIMIT in 2m 03s\n- openstack-tox-docs http://logs.openstack.org/39/667039/2/check/openstack-tox-docs/b3c7bba/html/ : SUCCESS in 3m 15s\n- openstack-tox-cover http://logs.openstack.org/39/667039/2/check/openstack-tox-cover/3635d6c/ : RETRY_LIMIT in 2m 02s (non-voting)\n- mistral-devstack http://logs.openstack.org/39/667039/2/check/mistral-devstack/e4101c6/ : FAILURE in 52m 58s\n- mistral-devstack-non-apache http://logs.openstack.org/39/667039/2/check/mistral-devstack-non-apache/c5a810f/ : FAILURE in 37m 16s\n- mistral-devstack-kombu http://logs.openstack.org/39/667039/2/check/mistral-devstack-kombu/c1684ea/ : FAILURE in 36m 38s\n- mistral-tox-unit-mysql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-mysql/3246750/ : RETRY_LIMIT in 2m 12s\n- mistral-tox-unit-postgresql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-postgresql/3811923/ : RETRY_LIMIT in 2m 19s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/667039/2/check/tripleo-ci-centos-7-scenario003-standalone/21bec35/ : SUCCESS in 1h 15m 25s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/667039/2/check/tripleo-ci-centos-7-containers-multinode/5d6df6b/ : TIMED_OUT in 3h 25m 54s\n- mistral-rally-task http://logs.openstack.org/39/667039/2/check/mistral-rally-task/3d884d9/ : SUCCESS in 59m 16s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/39/667039/2/check/openstack-tox-lower-constraints/334e3c8/ : RETRY_LIMIT in 2m 01s","accounts_in_message":[],"_revision_number":2},{"id":"3b2be969416578fc30e5358e8c9a763c1138de60","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-06-27 05:59:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"11f3faa77ba836042e7a11d95f4be1d0f5fd8a3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 08:59:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/39/667039/2/check/openstack-tox-pep8/396ac3d/ : RETRY_LIMIT in 2m 07s\n- openstack-tox-py27 http://logs.openstack.org/39/667039/2/check/openstack-tox-py27/30b620d/ : RETRY_LIMIT in 2m 10s\n- openstack-tox-py36 http://logs.openstack.org/39/667039/2/check/openstack-tox-py36/a35d4e2/ : RETRY_LIMIT in 2m 04s\n- openstack-tox-py37 http://logs.openstack.org/39/667039/2/check/openstack-tox-py37/385bcc5/ : RETRY_LIMIT in 2m 16s\n- openstack-tox-docs http://logs.openstack.org/39/667039/2/check/openstack-tox-docs/6ffe8e2/html/ : SUCCESS in 3m 58s\n- openstack-tox-cover http://logs.openstack.org/39/667039/2/check/openstack-tox-cover/2d50c01/ : RETRY_LIMIT in 2m 22s (non-voting)\n- mistral-devstack http://logs.openstack.org/39/667039/2/check/mistral-devstack/5a4eb54/ : FAILURE in 38m 51s\n- mistral-devstack-non-apache http://logs.openstack.org/39/667039/2/check/mistral-devstack-non-apache/7fd88cf/ : FAILURE in 33m 45s\n- mistral-devstack-kombu http://logs.openstack.org/39/667039/2/check/mistral-devstack-kombu/0149e92/ : FAILURE in 42m 20s\n- mistral-tox-unit-mysql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-mysql/8bcb8b9/ : RETRY_LIMIT in 2m 15s\n- mistral-tox-unit-postgresql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-postgresql/cf2d652/ : RETRY_LIMIT in 1m 53s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/667039/2/check/tripleo-ci-centos-7-scenario003-standalone/08b260f/ : SUCCESS in 1h 18m 52s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/667039/2/check/tripleo-ci-centos-7-containers-multinode/8a18d39/ : SUCCESS in 2h 56m 01s\n- mistral-rally-task http://logs.openstack.org/39/667039/2/check/mistral-rally-task/25ecb33/ : SUCCESS in 1h 04m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/39/667039/2/check/openstack-tox-lower-constraints/c2d1091/ : RETRY_LIMIT in 2m 04s","accounts_in_message":[],"_revision_number":2},{"id":"bf413617d58f72737946063a166471f57a1a29ab","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-06-28 09:15:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bccce2eb5340182f9c7cc3f2cb4c6c5cccf19745","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 11:44:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/39/667039/2/check/openstack-tox-pep8/6265c8c/ : SUCCESS in 5m 45s\n- openstack-tox-py27 http://logs.openstack.org/39/667039/2/check/openstack-tox-py27/fba7b2f/ : SUCCESS in 7m 46s\n- openstack-tox-py36 http://logs.openstack.org/39/667039/2/check/openstack-tox-py36/9cd04ba/ : SUCCESS in 7m 23s\n- openstack-tox-py37 http://logs.openstack.org/39/667039/2/check/openstack-tox-py37/9193720/ : SUCCESS in 8m 13s\n- openstack-tox-docs http://logs.openstack.org/39/667039/2/check/openstack-tox-docs/70b21d7/html/ : SUCCESS in 3m 37s\n- openstack-tox-cover http://logs.openstack.org/39/667039/2/check/openstack-tox-cover/4303b7e/cover/ : SUCCESS in 8m 35s (non-voting)\n- mistral-devstack http://logs.openstack.org/39/667039/2/check/mistral-devstack/5969ee2/ : SUCCESS in 44m 41s\n- mistral-devstack-non-apache http://logs.openstack.org/39/667039/2/check/mistral-devstack-non-apache/289dcac/ : SUCCESS in 47m 30s\n- mistral-devstack-kombu http://logs.openstack.org/39/667039/2/check/mistral-devstack-kombu/8d11179/ : SUCCESS in 47m 05s\n- mistral-tox-unit-mysql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-mysql/cb0143c/ : SUCCESS in 26m 26s\n- mistral-tox-unit-postgresql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-postgresql/45241f2/ : SUCCESS in 26m 48s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/667039/2/check/tripleo-ci-centos-7-scenario003-standalone/ee24656/ : FAILURE in 17m 14s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/667039/2/check/tripleo-ci-centos-7-containers-multinode/9c0907b/ : SUCCESS in 2h 12m 48s\n- mistral-rally-task http://logs.openstack.org/39/667039/2/check/mistral-rally-task/428ed21/ : SUCCESS in 1h 09m 09s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/39/667039/2/check/openstack-tox-lower-constraints/fa5e36b/ : SUCCESS in 9m 49s","accounts_in_message":[],"_revision_number":2},{"id":"752069490e6683ad450eacbe2064d415e9c81e29","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-06-28 11:45:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"22f23425f5cec78be8d43caaaf3b14baada30442","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 14:10:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/39/667039/2/check/openstack-tox-pep8/b577121/ : SUCCESS in 5m 43s\n- openstack-tox-py27 http://logs.openstack.org/39/667039/2/check/openstack-tox-py27/460805c/ : SUCCESS in 8m 53s\n- openstack-tox-py36 http://logs.openstack.org/39/667039/2/check/openstack-tox-py36/3568e66/ : SUCCESS in 7m 34s\n- openstack-tox-py37 http://logs.openstack.org/39/667039/2/check/openstack-tox-py37/4202f76/ : SUCCESS in 8m 36s\n- openstack-tox-docs http://logs.openstack.org/39/667039/2/check/openstack-tox-docs/dc90aa3/html/ : SUCCESS in 3m 55s\n- openstack-tox-cover http://logs.openstack.org/39/667039/2/check/openstack-tox-cover/4b96aac/ : FAILURE in 7m 51s (non-voting)\n- mistral-devstack http://logs.openstack.org/39/667039/2/check/mistral-devstack/f88b38b/ : SUCCESS in 39m 13s\n- mistral-devstack-non-apache http://logs.openstack.org/39/667039/2/check/mistral-devstack-non-apache/a741c2a/ : SUCCESS in 43m 46s\n- mistral-devstack-kombu http://logs.openstack.org/39/667039/2/check/mistral-devstack-kombu/404b37e/ : SUCCESS in 45m 24s\n- mistral-tox-unit-mysql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-mysql/92e227b/ : SUCCESS in 27m 27s\n- mistral-tox-unit-postgresql http://logs.openstack.org/39/667039/2/check/mistral-tox-unit-postgresql/c438d2c/ : SUCCESS in 25m 53s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/667039/2/check/tripleo-ci-centos-7-scenario003-standalone/30b32e9/ : SUCCESS in 1h 18m 32s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/39/667039/2/check/tripleo-ci-centos-7-containers-multinode/7a52ed3/ : SUCCESS in 2h 09m 09s\n- mistral-rally-task http://logs.openstack.org/39/667039/2/check/mistral-rally-task/a24e563/ : SUCCESS in 55m 35s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/39/667039/2/check/openstack-tox-lower-constraints/659c823/ : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":2},{"id":"9e033698562c4eeb9b0230e49e4c08356b32fc40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 14:10:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e72d5ca59c90092629d8ec8dd311ebf391419eaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 15:52:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/39/667039/2/gate/openstack-tox-pep8/61ffab4/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/39/667039/2/gate/openstack-tox-py27/b22acdb/ : SUCCESS in 8m 10s\n- openstack-tox-py36 http://logs.openstack.org/39/667039/2/gate/openstack-tox-py36/21c218f/ : SUCCESS in 8m 10s\n- openstack-tox-py37 http://logs.openstack.org/39/667039/2/gate/openstack-tox-py37/b26d118/ : SUCCESS in 8m 34s\n- openstack-tox-docs http://logs.openstack.org/39/667039/2/gate/openstack-tox-docs/ac66711/html/ : SUCCESS in 4m 20s\n- mistral-devstack http://logs.openstack.org/39/667039/2/gate/mistral-devstack/67afe48/ : SUCCESS in 43m 32s\n- mistral-devstack-non-apache http://logs.openstack.org/39/667039/2/gate/mistral-devstack-non-apache/aa340ff/ : SUCCESS in 45m 02s\n- mistral-tox-unit-mysql http://logs.openstack.org/39/667039/2/gate/mistral-tox-unit-mysql/8d2fa76/ : SUCCESS in 27m 14s\n- mistral-tox-unit-postgresql http://logs.openstack.org/39/667039/2/gate/mistral-tox-unit-postgresql/0987f7e/ : SUCCESS in 26m 10s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/39/667039/2/gate/tripleo-ci-centos-7-scenario003-standalone/52c3461/ : SUCCESS in 1h 26m 07s\n- mistral-devstack-kombu http://logs.openstack.org/39/667039/2/gate/mistral-devstack-kombu/d953a2b/ : SUCCESS in 45m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/39/667039/2/gate/openstack-tox-lower-constraints/25e15c4/ : SUCCESS in 9m 54s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/39/667039/2/gate/tripleo-ci-centos-7-undercloud-containers/d8dfbea/ : SUCCESS in 1h 37m 20s","accounts_in_message":[],"_revision_number":2},{"id":"c2ac2301dbdb262a22c1c31ef7795f4d114a9bdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 15:52:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dfc3277ffd3b2217cbc45f8635e4008e0f9d1100","revisions":{"74b7e259186ee1abdc635ba0d9f8a717787258e6":{"kind":"REWORK","_number":1,"created":"2019-06-24 08:13:57.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/39/667039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/39/667039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/39/667039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/39/667039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/39/667039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/39/667039/1"}}},"commit":{"parents":[{"commit":"f75e7198c61da0547ab6ab7e96c797ac5c6657da","subject":"Merge \"Fix workflow execution cascade delete error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f75e7198c61da0547ab6ab7e96c797ac5c6657da"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-06-24 08:06:59.000000000","tz":180},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-06-24 08:06:59.000000000","tz":180},"subject":"Initialize the lazily loaded execution \"input\" field in API","message":"Initialize the lazily loaded execution \"input\" field in API\n\n* After the recent change that made all potentially heavy fields\n  of execution objects lazily loaded, some clients using Mistral\n  APIs started crashing because they expect to get \"input\" field\n  in JSON of execution objects that Mistral API returns. It wasn\u0027t\n  present though because we did not initialize it explicitly in\n  the API controller.\n* Unfortunately, there\u0027s no easy way now to cover this change in\n  the API tests just for how they are organized: they mock all\n  DB calls and return all fields already initialized. We may want\n  to refactor these tests moving forward.\n\nChange-Id: I683c79fa0a3ab23a16c493ce2314a506dfee9749\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/74b7e259186ee1abdc635ba0d9f8a717787258e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/74b7e259186ee1abdc635ba0d9f8a717787258e6"}]},"branch":"refs/heads/master"},"dfc3277ffd3b2217cbc45f8635e4008e0f9d1100":{"kind":"REWORK","_number":2,"created":"2019-06-24 10:19:09.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/39/667039/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/39/667039/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/39/667039/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/39/667039/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/39/667039/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/39/667039/2"}}},"commit":{"parents":[{"commit":"f75e7198c61da0547ab6ab7e96c797ac5c6657da","subject":"Merge \"Fix workflow execution cascade delete error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f75e7198c61da0547ab6ab7e96c797ac5c6657da"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-06-24 08:06:59.000000000","tz":180},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-06-24 10:18:50.000000000","tz":180},"subject":"Initialize the lazily loaded execution \"input\" field in API","message":"Initialize the lazily loaded execution \"input\" field in API\n\n* After the recent change that made all potentially heavy fields\n  of execution objects lazily loaded, some clients using Mistral\n  APIs started crashing because they expect to get \"input\" field\n  in JSON of execution objects that Mistral API returns. It wasn\u0027t\n  present though because we did not initialize it explicitly in\n  the API controller.\n* Unfortunately, there\u0027s no easy way now to cover this change in\n  the API tests just for how they are organized: they mock all\n  DB calls and return all fields already initialized. We may want\n  to refactor these tests moving forward.\n\nChange-Id: I683c79fa0a3ab23a16c493ce2314a506dfee9749\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dfc3277ffd3b2217cbc45f8635e4008e0f9d1100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dfc3277ffd3b2217cbc45f8635e4008e0f9d1100"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
