)]}'
{"id":"openstack%2Fmistral~669177","triplet_id":"openstack%2Fmistral~master~I7cba0d41204d33aaafd4cab00875fe76f357197a","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I7cba0d41204d33aaafd4cab00875fe76f357197a","subject":"Fix the pep8 failed","status":"ABANDONED","created":"2019-07-04 13:30:06.000000000","updated":"2019-08-13 15:53:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"631e9365156aaed2b33a87629f2a80ab7a363579","_number":669177,"virtual_id_number":669177,"owner":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":-1,"date":"2019-07-04 16:14:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-13 14:40:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 13:38:07.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-07-04 16:14:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-13 14:40:07.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"32a341b574f2ddb7e66cbcb6e819a22e2b00b531","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-07-04 13:30:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b0fca4aff570b351de96732adafd693c5db0333","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-07-04 13:38:07.000000000","message":"Patch Set 1:\n\nI think it would be great to add codes of fixed errors to the commit  message so we can learn something new and prevent similar errors in the future.","accounts_in_message":[],"_revision_number":1},{"id":"efc8c28e824abc3b485a0b8de8580257f1bc734c","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-07-04 13:39:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"67996e8ac1c656856bd7c0aa9c70b8864ea593fe","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-07-04 13:45:23.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1ae4a93f9a12711bf8a608edf6035abb39187ce4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 16:14:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/77/669177/3/check/openstack-tox-pep8/1976ad0/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/77/669177/3/check/openstack-tox-py27/8f1739c/ : SUCCESS in 8m 56s\n- openstack-tox-py36 http://logs.openstack.org/77/669177/3/check/openstack-tox-py36/163f15f/ : SUCCESS in 9m 35s\n- openstack-tox-py37 http://logs.openstack.org/77/669177/3/check/openstack-tox-py37/1db9847/ : SUCCESS in 9m 47s\n- openstack-tox-docs http://logs.openstack.org/77/669177/3/check/openstack-tox-docs/a03e4f5/html/ : SUCCESS in 5m 35s\n- build-openstack-releasenotes http://logs.openstack.org/77/669177/3/check/build-openstack-releasenotes/236c023/html/ : SUCCESS in 4m 15s\n- openstack-tox-cover http://logs.openstack.org/77/669177/3/check/openstack-tox-cover/ca57171/cover/ : SUCCESS in 8m 30s (non-voting)\n- mistral-devstack http://logs.openstack.org/77/669177/3/check/mistral-devstack/fcd51dc/ : SUCCESS in 39m 20s\n- mistral-devstack-non-apache http://logs.openstack.org/77/669177/3/check/mistral-devstack-non-apache/ebe0c43/ : SUCCESS in 49m 08s\n- mistral-devstack-kombu http://logs.openstack.org/77/669177/3/check/mistral-devstack-kombu/bfe66e0/ : SUCCESS in 45m 23s\n- mistral-tox-unit-mysql http://logs.openstack.org/77/669177/3/check/mistral-tox-unit-mysql/c0dade1/ : FAILURE in 29m 32s\n- mistral-tox-unit-postgresql http://logs.openstack.org/77/669177/3/check/mistral-tox-unit-postgresql/7aed527/ : SUCCESS in 27m 54s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/77/669177/3/check/tripleo-ci-centos-7-scenario003-standalone/cfc57d9/ : SUCCESS in 1h 25m 56s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/77/669177/3/check/tripleo-ci-centos-7-containers-multinode/166f789/ : SUCCESS in 2h 21m 22s\n- mistral-rally-task http://logs.openstack.org/77/669177/3/check/mistral-rally-task/2b2aef1/ : SUCCESS in 1h 02m 28s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/77/669177/3/check/openstack-tox-lower-constraints/5ac3b62/ : SUCCESS in 11m 03s","accounts_in_message":[],"_revision_number":3},{"id":"0a7f4475d1a47af4ac47552f358f75a19f759d8e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2019-08-13 14:40:07.000000000","message":"Patch Set 3: Code-Review-1\n\nfix  Merge conflict","accounts_in_message":[],"_revision_number":3},{"id":"631e9365156aaed2b33a87629f2a80ab7a363579","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-08-13 15:53:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"587b4749c9a031a2de15e42089dbf679aefef7f6","revisions":{"66091e438d13798ac7881939a66d60b19ca2bb65":{"kind":"REWORK","_number":1,"created":"2019-07-04 13:30:06.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/77/669177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/669177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/669177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/669177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/669177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/669177/1"}}},"commit":{"parents":[{"commit":"6a9aa4c295b51e9b11ab4a30e5c002bcc2924609","subject":"Merge \"Initialize the lazily loaded execution \"input\" field in API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a9aa4c295b51e9b11ab4a30e5c002bcc2924609"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-07-04 13:29:49.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-07-04 13:29:49.000000000","tz":480},"subject":"Fix the pep8 failed","message":"Fix the pep8 failed\n\nChange-Id: I7cba0d41204d33aaafd4cab00875fe76f357197a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66091e438d13798ac7881939a66d60b19ca2bb65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66091e438d13798ac7881939a66d60b19ca2bb65"}]},"branch":"refs/heads/master"},"ad9edeb8f42ec0c85350bbbb9843dc36cd14fac7":{"kind":"REWORK","_number":2,"created":"2019-07-04 13:39:52.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/77/669177/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/669177/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/669177/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/669177/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/669177/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/669177/2"}}},"commit":{"parents":[{"commit":"6a9aa4c295b51e9b11ab4a30e5c002bcc2924609","subject":"Merge \"Initialize the lazily loaded execution \"input\" field in API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a9aa4c295b51e9b11ab4a30e5c002bcc2924609"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-07-04 13:29:49.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-07-04 13:39:40.000000000","tz":480},"subject":"Fix the pep8 failed","message":"Fix the pep8 failed\n\nChange-Id: I7cba0d41204d33aaafd4cab00875fe76f357197a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ad9edeb8f42ec0c85350bbbb9843dc36cd14fac7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ad9edeb8f42ec0c85350bbbb9843dc36cd14fac7"}]},"branch":"refs/heads/master"},"587b4749c9a031a2de15e42089dbf679aefef7f6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-07-04 13:45:23.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/77/669177/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/669177/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/669177/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/669177/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/669177/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/669177/3"}}},"commit":{"parents":[{"commit":"6a9aa4c295b51e9b11ab4a30e5c002bcc2924609","subject":"Merge \"Initialize the lazily loaded execution \"input\" field in API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a9aa4c295b51e9b11ab4a30e5c002bcc2924609"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-07-04 13:29:49.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-07-04 13:43:19.000000000","tz":480},"subject":"Fix the pep8 failed","message":"Fix the pep8 failed\n\nwhen run `tox -e pep8` it will failed, the error is like[1]\n\n[1]: http://paste.openstack.org/show/753870/\n\nChange-Id: I7cba0d41204d33aaafd4cab00875fe76f357197a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/587b4749c9a031a2de15e42089dbf679aefef7f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/587b4749c9a031a2de15e42089dbf679aefef7f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
