)]}'
{"id":"openstack%2Fmistral~669388","triplet_id":"openstack%2Fmistral~master~Ieb3834672827521d209c5963f2bf1a28c26155d4","project":"openstack/mistral","branch":"master","topic":"python3-train","hashtags":[],"change_id":"Ieb3834672827521d209c5963f2bf1a28c26155d4","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 16:02:35.000000000","updated":"2019-08-16 05:51:15.000000000","submitted":"2019-08-16 05:51:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669388-1565934675186-8c476a95","meta_rev_id":"9c2de7fd7413de53945cb44f6c96a73b37047613","_number":669388,"virtual_id_number":669388,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":16635,"name":"deray","email":"debayan.ray@gmail.com","username":"deray"},{"value":2,"date":"2019-08-16 05:51:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-14 15:56:00.000000000","_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-08-15 04:29:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"date":"2019-08-15 19:07:27.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16635,"name":"deray","email":"debayan.ray@gmail.com","username":"deray"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-08-16 04:46:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":16635,"name":"deray","email":"debayan.ray@gmail.com","username":"deray"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":16635,"name":"deray","email":"debayan.ray@gmail.com","username":"deray"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 19:07:27.000000000","updated_by":{"_account_id":16635,"name":"deray","email":"debayan.ray@gmail.com","username":"deray"},"reviewer":{"_account_id":16635,"name":"deray","email":"debayan.ray@gmail.com","username":"deray"},"state":"REVIEWER"},{"updated":"2019-08-16 04:46:12.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-08-16 05:51:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fca539c35d9335f7a7316dfca2ad5aecc9c8ab33","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 16:02:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"841c811345d54bafafc585cb5968721e468052c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 17:18:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/88/669388/1/check/openstack-tox-pep8/0582d21/ : SUCCESS in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/88/669388/1/check/openstack-tox-py27/ff170e3/ : SUCCESS in 8m 53s\n- openstack-tox-py36 http://logs.openstack.org/88/669388/1/check/openstack-tox-py36/36a116d/ : SUCCESS in 8m 10s\n- openstack-tox-py37 http://logs.openstack.org/88/669388/1/check/openstack-tox-py37/640a739/ : SUCCESS in 8m 03s\n- openstack-tox-docs http://logs.openstack.org/88/669388/1/check/openstack-tox-docs/f643f0f/html/ : SUCCESS in 3m 44s\n- build-openstack-releasenotes http://logs.openstack.org/88/669388/1/check/build-openstack-releasenotes/6911584/html/ : SUCCESS in 4m 20s\n- openstack-tox-cover http://logs.openstack.org/88/669388/1/check/openstack-tox-cover/a14a785/cover/ : SUCCESS in 8m 53s (non-voting)\n- mistral-devstack http://logs.openstack.org/88/669388/1/check/mistral-devstack/403be46/ : SUCCESS in 46m 24s\n- mistral-devstack-non-apache http://logs.openstack.org/88/669388/1/check/mistral-devstack-non-apache/e6583eb/ : SUCCESS in 49m 14s\n- mistral-devstack-kombu http://logs.openstack.org/88/669388/1/check/mistral-devstack-kombu/0da2c3c/ : SUCCESS in 56m 48s\n- mistral-tox-unit-mysql http://logs.openstack.org/88/669388/1/check/mistral-tox-unit-mysql/1de7c4e/ : SUCCESS in 29m 15s\n- mistral-tox-unit-postgresql http://logs.openstack.org/88/669388/1/check/mistral-tox-unit-postgresql/db94695/ : SUCCESS in 27m 29s\n- mistral-rally-task http://logs.openstack.org/88/669388/1/check/mistral-rally-task/8d92936/ : SUCCESS in 54m 51s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/88/669388/1/check/openstack-tox-lower-constraints/4cd1a3b/ : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8a1e80824f94220aa143c5a478194a108a65fb0b","author":{"_account_id":16635,"name":"deray","email":"debayan.ray@gmail.com","username":"deray"},"date":"2019-07-08 05:58:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nI have a little concern here. Please see my inline comment.","accounts_in_message":[],"_revision_number":1},{"id":"e1894246e627e2dcf0b4b9aa0204043300c37f87","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-08 15:22:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a8892a15dbbc54f223857d3f6fb41444c174f9c","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-08-14 15:56:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1576ecea726cefa80ce65a021bf1a993b3b935f6","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-15 04:29:30.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM. I totally agree that it\u0027s worth supporting py37 only.","accounts_in_message":[],"_revision_number":1},{"id":"f8f3d6797fffbf06945c5a6fcd29e8e691d74911","author":{"_account_id":16635,"name":"deray","email":"debayan.ray@gmail.com","username":"deray"},"date":"2019-08-15 19:07:27.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)\n\nIf Mistral is planning to support only py37 only (and not even py36), then I am fine with this patch. Hence, removing my -1. Thanks, Corey!","accounts_in_message":[],"_revision_number":1},{"id":"fd249a87df4bc32121987ad246b56f4fa00e7294","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-16 04:46:12.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e0d129d85aa5a3e2e4bad66a7cf20057dda29377","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 04:46:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"abc92bce5cfeffeff48fd9b51ec8321f403efb1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 05:51:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc7c4d24bdcd458699baff2405883e72 : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9bb77df132294e62aa935fc0d6070b66 : SUCCESS in 9m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82a27792f1c440dfad97f9b881a17e70 : SUCCESS in 9m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d0749cf34684ad58a6b44e5255d7c2e : SUCCESS in 9m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adbaf62252d84e3a9b235c99cfe97258 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2101cfb4a5314e6a90e2805bc78cf483 : SUCCESS in 5m 02s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1418ac5005134c10a27b1704d8726341 : SUCCESS in 45m 51s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/383d072eac004e9e81e3701474238421 : SUCCESS in 50m 36s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f4f6b2c5ad024099a4d3032feb8debad : SUCCESS in 29m 21s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2b94fbcf8ab34e43984f065580ed4c9e : SUCCESS in 28m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fb4d1ee4b1704e48bbad633cfff12230 : SUCCESS in 43m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1abaf0058a7d4802a10c26b083e539cc : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":1},{"id":"006862bf248f0b03029704fbcb26d7c39984381f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 05:51:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"843717a1d576ed8d245ceb3876f57c4afa9f64c3","revisions":{"843717a1d576ed8d245ceb3876f57c4afa9f64c3":{"kind":"REWORK","_number":1,"created":"2019-07-05 16:02:35.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/88/669388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/88/669388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/88/669388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/88/669388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/88/669388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/88/669388/1"}}},"commit":{"parents":[{"commit":"6a9aa4c295b51e9b11ab4a30e5c002bcc2924609","subject":"Merge \"Initialize the lazily loaded execution \"input\" field in API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a9aa4c295b51e9b11ab4a30e5c002bcc2924609"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:41:06.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 16:02:34.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Ieb3834672827521d209c5963f2bf1a28c26155d4\nStory: #2005924\nTask: #34222\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/843717a1d576ed8d245ceb3876f57c4afa9f64c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/843717a1d576ed8d245ceb3876f57c4afa9f64c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
