)]}'
{"id":"openstack%2Fmistral~669558","triplet_id":"openstack%2Fmistral~master~I5436e34dec032b9028221e3e87c8d88a9b207e23","project":"openstack/mistral","branch":"master","topic":"bp/mistral-redesign-scheduler","hashtags":[],"change_id":"I5436e34dec032b9028221e3e87c8d88a9b207e23","subject":"Add a migration to create the scheduled_jobs table","status":"MERGED","created":"2019-07-07 01:45:53.000000000","updated":"2019-09-09 19:26:06.000000000","submitted":"2019-09-09 19:24:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669558-1568057098215-58a3ffe5","meta_rev_id":"e7f7c08504118e9efa46bc66d6fe7e7b8accb73a","_number":669558,"virtual_id_number":669558,"owner":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-09-09 19:24:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-09 14:01:07.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2019-09-09 08:28:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2019-09-09 08:28:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 21:58:06.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-09-06 07:23:27.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-09-09 08:28:43.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-09-09 14:01:07.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-09-09 19:24:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8214a35ebc9e3985330a5ff65306bddc6f52f0ac","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-07-07 01:45:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db5001f48447935023f52060322ba75d241ca4c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-07 04:05:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/58/669558/1/check/openstack-tox-pep8/ebaa15f/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/58/669558/1/check/openstack-tox-py27/1272772/ : SUCCESS in 7m 44s\n- openstack-tox-py36 http://logs.openstack.org/58/669558/1/check/openstack-tox-py36/8dca4b3/ : SUCCESS in 7m 21s\n- openstack-tox-py37 http://logs.openstack.org/58/669558/1/check/openstack-tox-py37/c2e3f22/ : SUCCESS in 9m 03s\n- openstack-tox-docs http://logs.openstack.org/58/669558/1/check/openstack-tox-docs/c301e9d/html/ : SUCCESS in 3m 51s\n- openstack-tox-cover http://logs.openstack.org/58/669558/1/check/openstack-tox-cover/09ad29c/cover/ : SUCCESS in 8m 12s (non-voting)\n- mistral-devstack http://logs.openstack.org/58/669558/1/check/mistral-devstack/5c5203c/ : SUCCESS in 44m 11s\n- mistral-devstack-non-apache http://logs.openstack.org/58/669558/1/check/mistral-devstack-non-apache/2ef0231/ : SUCCESS in 47m 00s\n- mistral-devstack-kombu http://logs.openstack.org/58/669558/1/check/mistral-devstack-kombu/8e3b3f5/ : SUCCESS in 46m 18s\n- mistral-tox-unit-mysql http://logs.openstack.org/58/669558/1/check/mistral-tox-unit-mysql/e2a0429/ : SUCCESS in 26m 22s\n- mistral-tox-unit-postgresql http://logs.openstack.org/58/669558/1/check/mistral-tox-unit-postgresql/c99b121/ : SUCCESS in 25m 35s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/58/669558/1/check/tripleo-ci-centos-7-scenario003-standalone/bc4d85a/ : SUCCESS in 1h 16m 02s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/58/669558/1/check/tripleo-ci-centos-7-containers-multinode/de2d42e/ : SUCCESS in 2h 13m 23s\n- mistral-rally-task http://logs.openstack.org/58/669558/1/check/mistral-rally-task/0b17bf4/ : SUCCESS in 1h 06m 47s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/58/669558/1/check/openstack-tox-lower-constraints/fe392d9/ : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":1},{"id":"6f1254ee715067603c7051a615b3f4452dbd2f95","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-07-07 12:51:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9d53935c6bff93cda32f26228f09ebd1ab576f05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-07 15:17:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/58/669558/2/check/openstack-tox-pep8/a63e2a2/ : SUCCESS in 4m 34s\n- openstack-tox-py27 http://logs.openstack.org/58/669558/2/check/openstack-tox-py27/ee2e821/ : SUCCESS in 8m 20s\n- openstack-tox-py36 http://logs.openstack.org/58/669558/2/check/openstack-tox-py36/db182ad/ : SUCCESS in 7m 39s\n- openstack-tox-py37 http://logs.openstack.org/58/669558/2/check/openstack-tox-py37/70b738c/ : SUCCESS in 7m 20s\n- openstack-tox-docs http://logs.openstack.org/58/669558/2/check/openstack-tox-docs/e58827c/html/ : SUCCESS in 3m 46s\n- openstack-tox-cover http://logs.openstack.org/58/669558/2/check/openstack-tox-cover/605e41a/cover/ : SUCCESS in 8m 22s (non-voting)\n- mistral-devstack http://logs.openstack.org/58/669558/2/check/mistral-devstack/7b967f7/ : SUCCESS in 43m 36s\n- mistral-devstack-non-apache http://logs.openstack.org/58/669558/2/check/mistral-devstack-non-apache/b2752bd/ : SUCCESS in 43m 56s\n- mistral-devstack-kombu http://logs.openstack.org/58/669558/2/check/mistral-devstack-kombu/369d565/ : SUCCESS in 43m 02s\n- mistral-tox-unit-mysql http://logs.openstack.org/58/669558/2/check/mistral-tox-unit-mysql/bbbb65c/ : SUCCESS in 26m 58s\n- mistral-tox-unit-postgresql http://logs.openstack.org/58/669558/2/check/mistral-tox-unit-postgresql/405c872/ : SUCCESS in 25m 36s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/58/669558/2/check/tripleo-ci-centos-7-scenario003-standalone/6c08c6f/ : SUCCESS in 1h 19m 44s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/58/669558/2/check/tripleo-ci-centos-7-containers-multinode/77bd8d5/ : SUCCESS in 2h 19m 58s\n- mistral-rally-task http://logs.openstack.org/58/669558/2/check/mistral-rally-task/8d11a14/ : SUCCESS in 1h 03m 52s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/58/669558/2/check/openstack-tox-lower-constraints/8d7f06f/ : SUCCESS in 9m 01s","accounts_in_message":[],"_revision_number":2},{"id":"3569e9b18f540b1e1d0460cdf3dd52f5bba95b09","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-07-10 21:58:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4ccbef128f05bf2b0d15e5e0bfe185c0981dc5c5","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-17 06:18:17.000000000","message":"Patch Set 2: Code-Review-1 Workflow-1\n\nLet\u0027s not merge this yet please. I don\u0027t want to create any migrations for this functionality because it is not stable yet. Also, what\u0027s the meaning of this \"scheduler_id\" field and why did you remove timestamps?","accounts_in_message":[],"_revision_number":2},{"id":"22d9415366bf39849557c2e7b9592c1c689e6368","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-07-18 17:58:42.000000000","message":"Patch Set 2:\n\nYeah, I agree that we should stabilize the solution first.\nscheduler_id will be used to define what scheduler captured the job, it allows to optimize queries and quickly find jobs that belong to a given scheduler.\nI removed the indices, because I think they are quite useless here - there shouldn\u0027t be a lot of rows in the table, just active jobs. So we can easily scan through them without indices. But there will be a lot of insertions and removals, and indices make these operations slower.","accounts_in_message":[],"_revision_number":2},{"id":"97bdae3b9ec200d75a25cab9f0e04cf229ac70d8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-05 08:07:43.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"efd18019d13939f9f2a1e8c1058d71493e62826c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-05 08:08:16.000000000","message":"Patch Set 3:\n\nI removed \"scheduler_id\" for now. If needed, we\u0027ll add it later.","accounts_in_message":[],"_revision_number":3},{"id":"1a8ceba50ab778eabe6eed95bc505fc128c72a33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 12:26:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb221d7353254e389f7f1a600c77e6bc : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38a0dfa39e784f6cad1baf38ac05eb78 : SUCCESS in 9m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/369052f336e7410ca5d5ccaa7ec6432d : SUCCESS in 9m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c1c178c9f8f49a8afc1c0c372a29daa : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ab31bd2a394bf5805d78f4a0ec937e : SUCCESS in 5m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/694432f483654ec3b851fcf234d59a52 : SUCCESS in 9m 57s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/668d155275784ee6b3a4caa29ac9a4a2 : SUCCESS in 51m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4f91b53df00540f3bfb8ed531c1d71a6 : SUCCESS in 42m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9212cad86628446e9aadf65fafcbc269 : SUCCESS in 46m 13s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9b10c08835e84683b1c881c1ea4ba860 : SUCCESS in 28m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/73a3209419b6428ab3999ef322cc3cb5 : SUCCESS in 27m 19s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/ef1fba888e824298980aa73405657e4d : SUCCESS in 1h 38m 13s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/aad24c3d622745cd9b25e6fbd18dbabd : SUCCESS in 2h 59m 08s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/61fa24c78f114ce88ef32dde2953041d : SUCCESS in 1h 07m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5873b806501f4e798ba5d56d65738df3 : SUCCESS in 10m 13s","accounts_in_message":[],"_revision_number":3},{"id":"85ce13229cded0480531061c52fe875f62cba81e","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-09-06 07:23:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b801c983c7c23f6124709e473bc38ff2fa72c2f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 07:23:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ee395c90a30091cd7d294c0144b0763e480a737f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 09:21:35.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d71dc1c429a4a5a8d569a5c83fc0b08 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/423b41b035e9477581a916df264e9be0 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b1ac960bf4a4a8cb1e429bd913676c9 : SUCCESS in 7m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6dbb96b451743f28928ef4e1f36a688 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e29e4844fd60431ba0b052f582d9e5a8 : SUCCESS in 4m 18s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2fbcd3f5fcd249f2bd97454dc900330d : SUCCESS in 51m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6cf79f23e14548558efe85baf714489e : SUCCESS in 42m 47s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cdcb537bc5474fb2aadccfa1e092e2de : FAILURE in 29m 00s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c28985c89e3042898ecf0fcb28771309 : SUCCESS in 26m 15s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/916bcecb60454dd5abb85e18cdbfded6 : SUCCESS in 1h 30m 14s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6e5f6e76d0fa469f82349bbc2e1a0379 : SUCCESS in 39m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b35aec501324d96968e35fbd187ef5c : SUCCESS in 8m 55s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/db55ddef8bee4b79989597cb9dc53a5d : SUCCESS in 1h 40m 56s","accounts_in_message":[],"_revision_number":3},{"id":"50fe3f895936aa7bd278e52fe6931ea20ba60ae7","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-06 09:24:26.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5c1b1bfb4488f37745f7146a0a583347163e2d27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 12:27:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fdd1fc6cb7c426fba10a12ec8687689 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/616f1fc89c414053b54e7f12b096ea7c : SUCCESS in 8m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec26a10f6da54fc1a6c8369f7fac2b48 : SUCCESS in 8m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96b4e9b430924a449a9bfa9dafb34174 : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e22b8b682044860bb7df02a4e28f27c : SUCCESS in 5m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42ae7c59a65145078ba6c487394c78b9 : SUCCESS in 9m 14s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e831cd35335e4019b23808a35cf1a92d : SUCCESS in 51m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/45f81dd31d2848df877db25b2d193acf : SUCCESS in 35m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8447656934174b29a48f8792b613c3f9 : SUCCESS in 42m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a785168f7d6042fe844a72b6f3363068 : FAILURE in 29m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7efd3266400d46d9bee7e600bb9229c9 : SUCCESS in 28m 10s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/30bfa2c5358b44b6970ebc7a4f42a442 : SUCCESS in 1h 28m 36s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f4a691507dcd4592b876d92a07a6e24d : SUCCESS in 2h 54m 14s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5a8493bc6d3048c0817be591028e1b0b : SUCCESS in 1h 01m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ce46efdf796497ab9c7ff2c4197cc0b : SUCCESS in 11m 29s","accounts_in_message":[],"_revision_number":3},{"id":"d04102dea5590d8543f00a7241bb864ba7048e35","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-09-08 18:07:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2d8a70f52c09b488d3c49b3f91c860e911a77015","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-08 21:02:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46daa874403b44d1850a6e4094edd746 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86c4b93786a34d78917620d03c853d7f : FAILURE in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9125abe3dc5d4d209435fa2ffa7d3ce8 : FAILURE in 7m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a6af391d41c49c18656194493fd63b8 : FAILURE in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95ad408c2b2a42a1a106b84658da2b4e : SUCCESS in 4m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84e9f15ca30b4ade85356d4910c2857c : FAILURE in 8m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d3ac3123b28244208c7cfe85e3cf95c1 : SUCCESS in 44m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3193ee8d8cdb4beb8acf99ea347cee49 : SUCCESS in 40m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3829b32d79c8495aa586719fe5a8e126 : SUCCESS in 40m 39s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d0ef2b94c54743799cd7036024bf9087 : FAILURE in 27m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c3af5675a8ad4c539ca5e3c39ce1b518 : FAILURE in 25m 46s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/3394847969b740bfb79ff888dc5b486a : FAILURE in 1h 06m 12s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1e5adc44662344dcb630e258475f2628 : SUCCESS in 2h 50m 33s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/010f9d3d13cc4ee98b3597a71c505f2c : FAILURE in 31m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2204f50753be4f9590cb9edada6e9bb7 : SUCCESS in 9m 26s","accounts_in_message":[],"_revision_number":3},{"id":"7b909e9faf3abaa3cc7cff2914de788a5fdc7b23","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-09 04:02:03.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"bd4ec6514c059a80a750b66f19291f8e1abba2e1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-09 04:02:07.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"98989e32a006a1d81e6e8b8ce8f80632649d8a60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 06:52:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27713f2858294228862fb804f508293d : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fd0883972c7462e9b2216779c1a3387 : FAILURE in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bf0941ab1d44045a1d3bd308b0914f3 : FAILURE in 8m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35d09968fbb048ef8a9f0a1eb9bbd421 : FAILURE in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff4fe3eb34e34979a618e563e3f4a271 : SUCCESS in 4m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe8f8494388e41c5b5ea300347249842 : FAILURE in 7m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/df9712a000c24fc18e8f036a0d33304d : SUCCESS in 42m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8b9a3bfc03d247138a1f5282327ca977 : SUCCESS in 38m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/631bdb11a86749b8b0b5cdf3a5cc42e9 : SUCCESS in 38m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d46a0bce7aa541aebf42d4e729e54d2f : FAILURE in 31m 45s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/578788ba49e744f6b5fa23f82b5fd46c : FAILURE in 27m 56s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b49c720fb1f64072bb99373c44f8d2e6 : SUCCESS in 1h 27m 12s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/aa08e00a91334fd881a5043559913a32 : SUCCESS in 2h 49m 19s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/46b84341f20b438cb9d369fd97bf3603 : FAILURE in 29m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/849c5de20e4c4a02a53b519a7a111910 : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":4},{"id":"91b558e89830ccd97ca66987fedf91378c58ec27","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-09 08:28:40.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"f9f7b8f4ad599eee9cf9cc189f6797703e4ca62a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-09 08:28:43.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"001846ce4eb5c2afde40603805c85d8428f3262e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 11:26:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eae4e754cce4961a81e065c8b1e9a35 : SUCCESS in 6m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97748b14b74e4d87ade2746426144363 : SUCCESS in 9m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e865c0b464f45b0a0552c9fd505a78d : SUCCESS in 11m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62b89645930b42739e2921b292244ed4 : SUCCESS in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/773458cfdde449ee9122cbcc9d1d59df : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/effe5fe5c2ff468cba5967ae46716db2 : SUCCESS in 10m 15s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3ba107d033b349e693331e2bc02c1ebd : SUCCESS in 50m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6e8a406e918d420d9983cd5c954e89fe : SUCCESS in 47m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/59e422001dcf45b3a105154e05b38e66 : SUCCESS in 44m 42s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fab4a162a4d34ea8b69dd63eabd7b786 : SUCCESS in 28m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c10b2d7f7e314c1ba40e774b6bd4ecb6 : SUCCESS in 27m 07s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b869d8745d364d49acf57d8f1a204238 : SUCCESS in 1h 34m 24s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1715a8e1afe3464c98de2516ac4e5120 : SUCCESS in 2h 48m 55s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/23a2a0461f35439fbf5c5ec74d55738d : FAILURE in 32m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8027dc4f37a427f8cab7801d8dadd02 : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":5},{"id":"c1ae0b5bafe731879d103146c52d3bf4d16e1e73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 11:26:50.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"223cfc0ec717d6b22043c0ba4e7803b70e204d7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 13:59:15.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55cd384af22d4e948739941600eca752 : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de91df271c7b418d8acfe4a9e7eba99f : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4d188d072034e96948d2faf7967136a : SUCCESS in 9m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac3067dd0e054f2f9943969cc165ca83 : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbdeeb36f0d140f3b3efa47859d728d9 : SUCCESS in 4m 28s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f0f9b29d87804562bf6d896e50440f18 : SUCCESS in 42m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e649d0c2043a4019a05b28057808d699 : SUCCESS in 40m 50s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/75d880e8f92a4848baf34858375561d7 : FAILURE in 27m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/11549c1caa3c44c0a8678b4b79a7ab78 : SUCCESS in 27m 03s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/f148fc2352db4c2196ebf48cb0eabb33 : SUCCESS in 1h 32m 06s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/27f61765e88346f7958f3855c0c58a9b : SUCCESS in 41m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ac796a348094782a2c23f50cf49ab2c : SUCCESS in 10m 34s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7a0962b6d2f647afad6cf5aaca468cd7 : SUCCESS in 1h 56m 01s","accounts_in_message":[],"_revision_number":5},{"id":"a227c4a43cdd1b11408c4817eb67c706a0fb4d47","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-09 14:01:07.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"383b83ad14facd5409794a751e8d15b7416a97a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 17:31:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3db018240afa48a78ad9bdc2c32075cc : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/725c02f096cb48ae93953414301edb52 : SUCCESS in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6ec21bd692246d486a942d4d71237c1 : SUCCESS in 11m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d7aa660a7a94618a7e65b343d863d15 : SUCCESS in 10m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b798295e93f4324af44dddc87ecf865 : SUCCESS in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc9d9bffa8014c88af37fb186a7a52aa : SUCCESS in 10m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/54e2e40753634fbfa0a98d4080595187 : SUCCESS in 50m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f85e5299eb874dafa695d38e7e30f9de : SUCCESS in 49m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/808da93ae4cd4a91beebc93a3a76dda2 : SUCCESS in 50m 06s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8c0b22c20dfd4e339f47b0d796288825 : SUCCESS in 28m 16s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/07f24e2b7726478ca290e4a0600e56b9 : SUCCESS in 28m 06s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/62afaba3d4e340aa8ae717b24eb0fbaa : SUCCESS in 1h 45m 22s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ce943a4a42c14e88b3d53f87a96c380a : SUCCESS in 3h 08m 01s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2b3faffaf63f48328df7a9f5704b928a : FAILURE in 45m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e5dcb1e38dd44a2917d463f60091985 : SUCCESS in 9m 16s","accounts_in_message":[],"_revision_number":5},{"id":"82f8f59c8625ad01d323b90e30cc43e240442f1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 17:31:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"62e86327e699b74877ac2158ae0f4a28b8bceb9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 19:24:57.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d280344844d3486e88c96f68409f8c7b : SUCCESS in 6m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/345ae8ddd69c483194a0ca440369a6bb : SUCCESS in 9m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5abd14d67594cac97605d16fe176b6a : SUCCESS in 9m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa28a2b5c6044b7284873b1fd513fe43 : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d1851db3a874c0e9faec6199c8a59ad : SUCCESS in 6m 15s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f23b32576bc5453c8253bb4ec5b64f47 : SUCCESS in 48m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5f2f745785524267a1277a590ba9f819 : SUCCESS in 42m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0cc8ba459a314b80a252d9b1fa479af4 : SUCCESS in 28m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/27b131bc9dc04d77be6c9a51a4477488 : SUCCESS in 28m 48s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/46b3030fa260424a8fbd226cc8e37ed7 : SUCCESS in 1h 37m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/30359c88c27b4ca0acbd4f7dfa6c8b07 : SUCCESS in 42m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0eba8b0dad2145acb34fd1bb00834556 : SUCCESS in 9m 26s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7eef6e92be6047d28e9f6c5d60b3532b : SUCCESS in 1h 47m 35s","accounts_in_message":[],"_revision_number":5},{"id":"c9cd42bce772f46f3c207082d80a4bb0de8b0e17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 19:24:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"7db3dbc1006b0751cd27fe7395940356f76acf50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 19:26:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f944eec066da4b48aca59e927263ee2e : SUCCESS in 48s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"21f95a58732b44a543226ca5a43e58acccab32ab","revisions":{"9b7bdcc1b163fbc45bb15d9a304cd3f24469b3f7":{"kind":"REWORK","_number":1,"created":"2019-07-07 01:45:53.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/58/669558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/58/669558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/669558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/669558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/669558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/58/669558/1"}}},"commit":{"parents":[{"commit":"43d23c0e254e237303c91d91fb0d1330552df610","subject":"Create needed infrastructure to switch scheduler implementations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/43d23c0e254e237303c91d91fb0d1330552df610"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-07-06 21:15:27.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-07-06 22:04:36.000000000","tz":120},"subject":"Add a migration to create the scheduled_jobs table","message":"Add a migration to create the scheduled_jobs table\n\nThis commit adds a required migration and additionally creates a\nnew field in the model called \u0027scheduler_id\u0027.\n\nChange-Id: I5436e34dec032b9028221e3e87c8d88a9b207e23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9b7bdcc1b163fbc45bb15d9a304cd3f24469b3f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9b7bdcc1b163fbc45bb15d9a304cd3f24469b3f7"}]},"branch":"refs/heads/master"},"e9a0fb1ab95780677807df39330fe83dfa00e667":{"kind":"REWORK","_number":2,"created":"2019-07-07 12:51:43.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/58/669558/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/58/669558/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/669558/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/669558/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/669558/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/58/669558/2"}}},"commit":{"parents":[{"commit":"43d23c0e254e237303c91d91fb0d1330552df610","subject":"Create needed infrastructure to switch scheduler implementations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/43d23c0e254e237303c91d91fb0d1330552df610"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-07-06 21:15:27.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-07-07 12:47:46.000000000","tz":120},"subject":"Add a migration to create the scheduled_jobs table","message":"Add a migration to create the scheduled_jobs table\n\nThis commit adds a required migration and additionally creates a\nnew field in the model called \u0027scheduler_id\u0027.\n\nChange-Id: I5436e34dec032b9028221e3e87c8d88a9b207e23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e9a0fb1ab95780677807df39330fe83dfa00e667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e9a0fb1ab95780677807df39330fe83dfa00e667"}]},"branch":"refs/heads/master"},"36cca9ba1505e8bcd924a97a019d1dbb40107e73":{"kind":"REWORK","_number":3,"created":"2019-09-05 08:07:43.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/58/669558/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/58/669558/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/669558/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/669558/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/669558/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/58/669558/3"}}},"commit":{"parents":[{"commit":"43d23c0e254e237303c91d91fb0d1330552df610","subject":"Create needed infrastructure to switch scheduler implementations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/43d23c0e254e237303c91d91fb0d1330552df610"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-07-06 21:15:27.000000000","tz":120},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-05 08:07:27.000000000","tz":0},"subject":"Add a migration to create the scheduled_jobs table","message":"Add a migration to create the scheduled_jobs table\n\nThis commit adds a required migration and additionally creates a\nnew field in the model called \u0027scheduler_id\u0027.\n\nChange-Id: I5436e34dec032b9028221e3e87c8d88a9b207e23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/36cca9ba1505e8bcd924a97a019d1dbb40107e73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/36cca9ba1505e8bcd924a97a019d1dbb40107e73"}]},"branch":"refs/heads/master"},"dea6c2de1cb5c5c31b4eaff89ff2f8b5920e1490":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-09-09 04:02:03.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/58/669558/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/58/669558/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/669558/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/669558/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/669558/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/58/669558/4"}}},"commit":{"parents":[{"commit":"b8a8aad6452f236aec37de04fbbaf93bb3744bbf","subject":"Merge \"Fix misspell word\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b8a8aad6452f236aec37de04fbbaf93bb3744bbf"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-07-06 21:15:27.000000000","tz":120},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-09 04:02:03.000000000","tz":0},"subject":"Add a migration to create the scheduled_jobs table","message":"Add a migration to create the scheduled_jobs table\n\nThis commit adds a required migration and additionally creates a\nnew field in the model called \u0027scheduler_id\u0027.\n\nChange-Id: I5436e34dec032b9028221e3e87c8d88a9b207e23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dea6c2de1cb5c5c31b4eaff89ff2f8b5920e1490"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dea6c2de1cb5c5c31b4eaff89ff2f8b5920e1490"}]},"branch":"refs/heads/master"},"21f95a58732b44a543226ca5a43e58acccab32ab":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-09-09 08:28:40.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/58/669558/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/58/669558/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/669558/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/669558/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/58/669558/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/58/669558/5"}}},"commit":{"parents":[{"commit":"fdfb65a8ffd2fb0a539a880961e2daa197be7dc7","subject":"Fix workflow language tests after updating jsonschema to 3.0.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fdfb65a8ffd2fb0a539a880961e2daa197be7dc7"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-07-06 21:15:27.000000000","tz":120},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-09 08:28:40.000000000","tz":0},"subject":"Add a migration to create the scheduled_jobs table","message":"Add a migration to create the scheduled_jobs table\n\nThis commit adds a required migration and additionally creates a\nnew field in the model called \u0027scheduler_id\u0027.\n\nChange-Id: I5436e34dec032b9028221e3e87c8d88a9b207e23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/21f95a58732b44a543226ca5a43e58acccab32ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/21f95a58732b44a543226ca5a43e58acccab32ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
