)]}'
{"id":"openstack%2Fmistral~669559","triplet_id":"openstack%2Fmistral~master~I803fa8724298d6d79f14cd4d6b67eeba05b9a529","project":"openstack/mistral","branch":"master","topic":"bp/mistral-redesign-scheduler","hashtags":[],"change_id":"I803fa8724298d6d79f14cd4d6b67eeba05b9a529","subject":"Add db api tests for scheduled jobs","status":"MERGED","created":"2019-07-07 01:45:53.000000000","updated":"2019-09-11 06:41:34.000000000","submitted":"2019-09-11 06:40:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669559-1568184001046-032679dc","meta_rev_id":"3d75fa1ad9517e0cdbb02922c6d9e80fac2309e5","_number":669559,"virtual_id_number":669559,"owner":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-09-11 06:40:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":2,"date":"2019-09-10 05:10:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2019-09-11 04:19:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-10 05:10:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2019-09-11 04:19:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-08 10:38:45.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-09-09 09:50:33.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-09-11 04:19:43.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-09-11 06:40:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"222af7b594659150206067c9ea78bea7869460cc","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-07-07 01:45:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d16b9298c9b7edd1e80b540cf21489c7e277247d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-07 03:53:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/669559/1/check/openstack-tox-pep8/8ea2632/ : SUCCESS in 4m 49s\n- openstack-tox-py27 http://logs.openstack.org/59/669559/1/check/openstack-tox-py27/073734c/ : SUCCESS in 8m 47s\n- openstack-tox-py36 http://logs.openstack.org/59/669559/1/check/openstack-tox-py36/522e474/ : SUCCESS in 11m 06s\n- openstack-tox-py37 http://logs.openstack.org/59/669559/1/check/openstack-tox-py37/4dc7b72/ : SUCCESS in 9m 21s\n- openstack-tox-docs http://logs.openstack.org/59/669559/1/check/openstack-tox-docs/aea6932/html/ : SUCCESS in 6m 34s\n- openstack-tox-cover http://logs.openstack.org/59/669559/1/check/openstack-tox-cover/74af492/cover/ : SUCCESS in 9m 20s (non-voting)\n- mistral-devstack http://logs.openstack.org/59/669559/1/check/mistral-devstack/1002c9d/ : SUCCESS in 40m 22s\n- mistral-devstack-non-apache http://logs.openstack.org/59/669559/1/check/mistral-devstack-non-apache/a7cb0cc/ : SUCCESS in 38m 04s\n- mistral-devstack-kombu http://logs.openstack.org/59/669559/1/check/mistral-devstack-kombu/8445777/ : SUCCESS in 38m 00s\n- mistral-tox-unit-mysql http://logs.openstack.org/59/669559/1/check/mistral-tox-unit-mysql/51e8d5d/ : SUCCESS in 27m 04s\n- mistral-tox-unit-postgresql http://logs.openstack.org/59/669559/1/check/mistral-tox-unit-postgresql/8aef300/ : SUCCESS in 27m 12s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/59/669559/1/check/tripleo-ci-centos-7-scenario003-standalone/1965bd4/ : SUCCESS in 1h 04m 24s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/59/669559/1/check/tripleo-ci-centos-7-containers-multinode/0d591b3/ : SUCCESS in 1h 55m 33s\n- mistral-rally-task http://logs.openstack.org/59/669559/1/check/mistral-rally-task/56636a6/ : SUCCESS in 55m 11s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/669559/1/check/openstack-tox-lower-constraints/b63eb11/ : SUCCESS in 10m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7b1cb324a649f6a6cc3f35068191bf0404e2fd99","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-07-07 12:51:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"7d854988bfb177a81d27c9618ce4d9047b9d9a84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-07 15:15:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/669559/2/check/openstack-tox-pep8/692eb5f/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/59/669559/2/check/openstack-tox-py27/9e04cea/ : SUCCESS in 7m 46s\n- openstack-tox-py36 http://logs.openstack.org/59/669559/2/check/openstack-tox-py36/354deef/ : SUCCESS in 8m 06s\n- openstack-tox-py37 http://logs.openstack.org/59/669559/2/check/openstack-tox-py37/9dcf132/ : SUCCESS in 7m 55s\n- openstack-tox-docs http://logs.openstack.org/59/669559/2/check/openstack-tox-docs/374fa3d/html/ : SUCCESS in 3m 35s\n- openstack-tox-cover http://logs.openstack.org/59/669559/2/check/openstack-tox-cover/e8f2386/cover/ : SUCCESS in 7m 51s (non-voting)\n- mistral-devstack http://logs.openstack.org/59/669559/2/check/mistral-devstack/770f1a7/ : SUCCESS in 40m 07s\n- mistral-devstack-non-apache http://logs.openstack.org/59/669559/2/check/mistral-devstack-non-apache/b40e6ab/ : SUCCESS in 43m 14s\n- mistral-devstack-kombu http://logs.openstack.org/59/669559/2/check/mistral-devstack-kombu/cd7e984/ : SUCCESS in 42m 09s\n- mistral-tox-unit-mysql http://logs.openstack.org/59/669559/2/check/mistral-tox-unit-mysql/88922a6/ : SUCCESS in 26m 52s\n- mistral-tox-unit-postgresql http://logs.openstack.org/59/669559/2/check/mistral-tox-unit-postgresql/6b47b9e/ : SUCCESS in 25m 31s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/59/669559/2/check/tripleo-ci-centos-7-scenario003-standalone/c0272e8/ : SUCCESS in 1h 29m 06s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/59/669559/2/check/tripleo-ci-centos-7-containers-multinode/091f9ff/ : SUCCESS in 2h 17m 35s\n- mistral-rally-task http://logs.openstack.org/59/669559/2/check/mistral-rally-task/78fdd91/ : SUCCESS in 1h 00m 59s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/59/669559/2/check/openstack-tox-lower-constraints/65870a4/ : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":2},{"id":"1ee28b9a0c8bd86d714360abbfda71da8b698b52","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-07-10 22:09:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9d3db22d3e1b3ca1d1140c24acd6aab9de0f3969","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-08 10:38:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"89375fdcda5d44869ce4f15ba2d8e8e543ac31e0","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-09-09 09:50:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3ae48d0cfcb8279be5c04149f008c332f0d71595","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-10 05:10:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b1c5627bafc6c612b6a39803e25c8ec2ca13a7b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 08:10:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d218758fdb5a4b2f98280c5e90fd19c8 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61f203c304614c6c96a0fcd3d4de3c7c : SUCCESS in 9m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b32b702c809b4507a25d798e1f249ab8 : SUCCESS in 9m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34b0aa784f4b4af687223d076c136139 : SUCCESS in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3747d4d2d97b4644842f73522f9c3f3b : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2106ea11c704c66ae51f24eaefb2410 : SUCCESS in 8m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c1d56fb2bb3641629f48bbebc62ff733 : SUCCESS in 43m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bfc84e5fab5c4e59a0d66640bb3c8ee6 : SUCCESS in 46m 10s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a98ac833e9094810863fc3a7ecea2319 : SUCCESS in 44m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bbcbb94e4da24844bef3d5560e3389f2 : SUCCESS in 32m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2eff11abf4414f50b7ef236b31318670 : SUCCESS in 28m 29s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/8dc42d35c39b4abeb0ae4b3edb005d6d : SUCCESS in 1h 45m 37s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0b8f734e36db48f98fea4bd549b157de : SUCCESS in 2h 59m 17s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6e0910eaf9cf49f2aaee5552a5bd1eb1 : FAILURE in 40m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1ffa9d25e58477c86fb6112b3b446f2 : SUCCESS in 9m 18s","accounts_in_message":[],"_revision_number":3},{"id":"e0557efb50259e337f090ea253b7a71bcf0a95a0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-09-11 04:19:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a5b3d9e3901b99cbc73d74e677edb56dc1f58637","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 04:19:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ce07251468098950f818e6475fba903ab1985545","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 06:40:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62dfa73491eb429aaa292c01158f9f0c : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/594c0df9c0184091afb72db5c6872919 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90b87fa8a6354ee99c6af0014a0bb5a2 : SUCCESS in 8m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78321439a4034db68c8fb4c0f23364f6 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aac3f9286f14e3981ae88e95e16470c : SUCCESS in 4m 09s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a01c0da9e0ec43f0bcc63f092391adb0 : SUCCESS in 40m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1e52e3f70bd64b74b4c40d8719ee3480 : SUCCESS in 44m 47s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/10c89b41b8504d078bab173107ce6e3e : SUCCESS in 28m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/30828747e04b49768b93d0345a7d06a8 : SUCCESS in 27m 13s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5144125d669240aba4015bab354b3694 : SUCCESS in 1h 32m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d7b3faa4266442d88494f7a8597fecd1 : SUCCESS in 39m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c15389f565d4fcbbf0a7f2895be4a73 : SUCCESS in 10m 12s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/9f1336f3d2e94e59a8ed70b8baac32ea : SUCCESS in 2h 18m 43s","accounts_in_message":[],"_revision_number":3},{"id":"7d3ec8973e58603533f913a827a7c2ad43e2a5bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 06:40:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"bd5fd0dd8e0347f4076dfef1dfc4943e304ceef3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 06:41:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/043aa58d6d39427d89022c0e558b37fb : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4a1be8c4c606466c2f5d9ca747f588d025ca7d1f","revisions":{"cbad64cdc15d60012f7d3a38a65062e08208995e":{"kind":"REWORK","_number":1,"created":"2019-07-07 01:45:53.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/59/669559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/59/669559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/59/669559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/59/669559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/59/669559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/59/669559/1"}}},"commit":{"parents":[{"commit":"9b7bdcc1b163fbc45bb15d9a304cd3f24469b3f7","subject":"Add a migration to create the scheduled_jobs table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9b7bdcc1b163fbc45bb15d9a304cd3f24469b3f7"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-07-06 22:03:19.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-07-06 22:30:44.000000000","tz":120},"subject":"Add db api tests for scheduled jobs","message":"Add db api tests for scheduled jobs\n\nChange-Id: I803fa8724298d6d79f14cd4d6b67eeba05b9a529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cbad64cdc15d60012f7d3a38a65062e08208995e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cbad64cdc15d60012f7d3a38a65062e08208995e"}]},"branch":"refs/heads/master"},"924cad9b60a39168a162a4c8b46bb07ee0b8b9f8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-07 12:51:43.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/59/669559/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/59/669559/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/59/669559/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/59/669559/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/59/669559/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/59/669559/2"}}},"commit":{"parents":[{"commit":"e9a0fb1ab95780677807df39330fe83dfa00e667","subject":"Add a migration to create the scheduled_jobs table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e9a0fb1ab95780677807df39330fe83dfa00e667"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-07-06 22:03:19.000000000","tz":120},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-07-07 12:47:56.000000000","tz":120},"subject":"Add db api tests for scheduled jobs","message":"Add db api tests for scheduled jobs\n\nChange-Id: I803fa8724298d6d79f14cd4d6b67eeba05b9a529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/924cad9b60a39168a162a4c8b46bb07ee0b8b9f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/924cad9b60a39168a162a4c8b46bb07ee0b8b9f8"}]},"branch":"refs/heads/master"},"4a1be8c4c606466c2f5d9ca747f588d025ca7d1f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-10 05:10:01.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/59/669559/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/59/669559/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/59/669559/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/59/669559/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/59/669559/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/59/669559/3"}}},"commit":{"parents":[{"commit":"21f95a58732b44a543226ca5a43e58acccab32ab","subject":"Add a migration to create the scheduled_jobs table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/21f95a58732b44a543226ca5a43e58acccab32ab"}]}],"author":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-07-06 22:03:19.000000000","tz":120},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-09-10 05:10:01.000000000","tz":0},"subject":"Add db api tests for scheduled jobs","message":"Add db api tests for scheduled jobs\n\nChange-Id: I803fa8724298d6d79f14cd4d6b67eeba05b9a529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4a1be8c4c606466c2f5d9ca747f588d025ca7d1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4a1be8c4c606466c2f5d9ca747f588d025ca7d1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
