)]}'
{"id":"openstack%2Fmistral~670211","triplet_id":"openstack%2Fmistral~master~I9b50bb40d2f1fc883f32406f4bc11f03a15f0a70","project":"openstack/mistral","branch":"master","topic":"bp/mistral-dry-run","hashtags":[],"change_id":"I9b50bb40d2f1fc883f32406f4bc11f03a15f0a70","subject":"Dry run feature","status":"ABANDONED","created":"2019-07-10 23:02:50.000000000","updated":"2026-03-28 19:37:16.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"080745dd69addaa46161776e57a5c1cb70fb4478","_number":670211,"virtual_id_number":670211,"owner":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-25 07:46:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":-1,"date":"2019-12-25 05:45:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 23:27:21.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-07-11 23:27:21.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-07-11 23:27:21.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-07-11 23:27:21.000000000","updated_by":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"reviewer":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"state":"REVIEWER"},{"updated":"2019-07-29 18:41:12.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-12-25 05:45:42.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-12-25 07:46:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3f3f1e114904b8549f8449f03ff2dc4097e6e58","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-07-10 23:02:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff7a66bad366ba5d890adc70c2d03c5a27841a95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 00:29:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/11/670211/1/check/openstack-tox-pep8/85df763/ : SUCCESS in 5m 16s\n- openstack-tox-py27 http://logs.openstack.org/11/670211/1/check/openstack-tox-py27/1f2b132/ : FAILURE in 9m 32s\n- openstack-tox-py36 http://logs.openstack.org/11/670211/1/check/openstack-tox-py36/213e6bb/ : FAILURE in 8m 36s\n- openstack-tox-py37 http://logs.openstack.org/11/670211/1/check/openstack-tox-py37/76059f3/ : FAILURE in 10m 43s\n- openstack-tox-docs http://logs.openstack.org/11/670211/1/check/openstack-tox-docs/bb189c8/html/ : SUCCESS in 4m 32s\n- openstack-tox-cover http://logs.openstack.org/11/670211/1/check/openstack-tox-cover/c3d5adf/ : FAILURE in 9m 50s (non-voting)\n- mistral-devstack http://logs.openstack.org/11/670211/1/check/mistral-devstack/51c1d8b/ : FAILURE in 42m 45s\n- mistral-devstack-non-apache http://logs.openstack.org/11/670211/1/check/mistral-devstack-non-apache/7f2ad62/ : FAILURE in 43m 04s\n- mistral-devstack-kombu http://logs.openstack.org/11/670211/1/check/mistral-devstack-kombu/0902751/ : FAILURE in 46m 38s\n- mistral-tox-unit-mysql http://logs.openstack.org/11/670211/1/check/mistral-tox-unit-mysql/d97660a/ : FAILURE in 30m 10s\n- mistral-tox-unit-postgresql http://logs.openstack.org/11/670211/1/check/mistral-tox-unit-postgresql/951f62c/ : FAILURE in 29m 35s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/11/670211/1/check/tripleo-ci-centos-7-scenario003-standalone/46bc2c7/ : SUCCESS in 1h 17m 30s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/11/670211/1/check/tripleo-ci-centos-7-containers-multinode/2024f92/ : FAILURE in 1h 20m 37s\n- mistral-rally-task http://logs.openstack.org/11/670211/1/check/mistral-rally-task/60f2daf/ : SUCCESS in 58m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/11/670211/1/check/openstack-tox-lower-constraints/8496957/ : FAILURE in 10m 21s","accounts_in_message":[],"_revision_number":1},{"id":"8f6e9456426996e69b03bb5c2fa269d12002b4d8","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-07-11 14:47:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2425263542a9c86a95076e8555ea8cfc45096c6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 17:11:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/11/670211/2/check/openstack-tox-pep8/2e10d77/ : SUCCESS in 5m 36s\n- openstack-tox-py27 http://logs.openstack.org/11/670211/2/check/openstack-tox-py27/c2fd614/ : FAILURE in 9m 35s\n- openstack-tox-py36 http://logs.openstack.org/11/670211/2/check/openstack-tox-py36/2ddf417/ : FAILURE in 8m 37s\n- openstack-tox-py37 http://logs.openstack.org/11/670211/2/check/openstack-tox-py37/e5ea5a1/ : FAILURE in 8m 50s\n- openstack-tox-docs http://logs.openstack.org/11/670211/2/check/openstack-tox-docs/8748384/html/ : SUCCESS in 4m 01s\n- openstack-tox-cover http://logs.openstack.org/11/670211/2/check/openstack-tox-cover/ab2a07d/ : FAILURE in 11m 24s (non-voting)\n- mistral-devstack http://logs.openstack.org/11/670211/2/check/mistral-devstack/1d0930c/ : SUCCESS in 52m 07s\n- mistral-devstack-non-apache http://logs.openstack.org/11/670211/2/check/mistral-devstack-non-apache/932856e/ : SUCCESS in 50m 49s\n- mistral-devstack-kombu http://logs.openstack.org/11/670211/2/check/mistral-devstack-kombu/862072b/ : SUCCESS in 45m 04s\n- mistral-tox-unit-mysql http://logs.openstack.org/11/670211/2/check/mistral-tox-unit-mysql/c23fd51/ : FAILURE in 33m 33s\n- mistral-tox-unit-postgresql http://logs.openstack.org/11/670211/2/check/mistral-tox-unit-postgresql/7de08be/ : FAILURE in 32m 31s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/11/670211/2/check/tripleo-ci-centos-7-scenario003-standalone/5a5d3a9/ : SUCCESS in 1h 27m 21s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/11/670211/2/check/tripleo-ci-centos-7-containers-multinode/29e8d0f/ : SUCCESS in 2h 13m 06s\n- mistral-rally-task http://logs.openstack.org/11/670211/2/check/mistral-rally-task/437d4af/ : SUCCESS in 1h 10m 33s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/11/670211/2/check/openstack-tox-lower-constraints/a663332/ : FAILURE in 11m 06s","accounts_in_message":[],"_revision_number":2},{"id":"3c174be2ce350b54271efccea3e4a92dc3285aca","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-07-11 20:08:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a6839146b7c4cf3e8a98a5fcb90ce2307dc6b7dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 22:27:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/11/670211/3/check/openstack-tox-pep8/ec0b703/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/11/670211/3/check/openstack-tox-py27/e851e95/ : FAILURE in 8m 24s\n- openstack-tox-py36 http://logs.openstack.org/11/670211/3/check/openstack-tox-py36/c5b103e/ : FAILURE in 8m 12s\n- openstack-tox-py37 http://logs.openstack.org/11/670211/3/check/openstack-tox-py37/e69244a/ : FAILURE in 8m 20s\n- openstack-tox-docs http://logs.openstack.org/11/670211/3/check/openstack-tox-docs/18718b3/html/ : SUCCESS in 4m 00s\n- openstack-tox-cover http://logs.openstack.org/11/670211/3/check/openstack-tox-cover/3bcc99a/ : FAILURE in 8m 04s (non-voting)\n- mistral-devstack http://logs.openstack.org/11/670211/3/check/mistral-devstack/0930cdf/ : SUCCESS in 47m 38s\n- mistral-devstack-non-apache http://logs.openstack.org/11/670211/3/check/mistral-devstack-non-apache/78ba466/ : SUCCESS in 50m 54s\n- mistral-devstack-kombu http://logs.openstack.org/11/670211/3/check/mistral-devstack-kombu/47f38da/ : SUCCESS in 48m 54s\n- mistral-tox-unit-mysql http://logs.openstack.org/11/670211/3/check/mistral-tox-unit-mysql/25acc0a/ : FAILURE in 26m 51s\n- mistral-tox-unit-postgresql http://logs.openstack.org/11/670211/3/check/mistral-tox-unit-postgresql/77d9e88/ : FAILURE in 25m 43s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/11/670211/3/check/tripleo-ci-centos-7-scenario003-standalone/3b33605/ : FAILURE in 51m 05s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/11/670211/3/check/tripleo-ci-centos-7-containers-multinode/01273d8/ : SUCCESS in 2h 10m 55s\n- mistral-rally-task http://logs.openstack.org/11/670211/3/check/mistral-rally-task/5162222/ : SUCCESS in 1h 06m 01s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/11/670211/3/check/openstack-tox-lower-constraints/acb3590/ : FAILURE in 9m 35s","accounts_in_message":[],"_revision_number":3},{"id":"e0a1bebb57e1eac9a1903650925a4faaa0f0e4d8","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-07-11 23:26:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ad17265fe4cdc82688ec72ea6168f77c9b38139e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 01:42:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/11/670211/4/check/openstack-tox-pep8/0bc3373/ : SUCCESS in 5m 47s\n- openstack-tox-py27 http://logs.openstack.org/11/670211/4/check/openstack-tox-py27/309ddd6/ : SUCCESS in 9m 18s\n- openstack-tox-py36 http://logs.openstack.org/11/670211/4/check/openstack-tox-py36/8dcf59d/ : SUCCESS in 9m 33s\n- openstack-tox-py37 http://logs.openstack.org/11/670211/4/check/openstack-tox-py37/0064a86/ : SUCCESS in 10m 18s\n- openstack-tox-docs http://logs.openstack.org/11/670211/4/check/openstack-tox-docs/f0a1045/html/ : SUCCESS in 5m 32s\n- openstack-tox-cover http://logs.openstack.org/11/670211/4/check/openstack-tox-cover/1722319/cover/ : SUCCESS in 8m 49s (non-voting)\n- mistral-devstack http://logs.openstack.org/11/670211/4/check/mistral-devstack/7e80925/ : SUCCESS in 45m 06s\n- mistral-devstack-non-apache http://logs.openstack.org/11/670211/4/check/mistral-devstack-non-apache/a2255c9/ : SUCCESS in 45m 49s\n- mistral-devstack-kombu http://logs.openstack.org/11/670211/4/check/mistral-devstack-kombu/8ee4fa1/ : SUCCESS in 46m 23s\n- mistral-tox-unit-mysql http://logs.openstack.org/11/670211/4/check/mistral-tox-unit-mysql/2670a4d/ : SUCCESS in 27m 41s\n- mistral-tox-unit-postgresql http://logs.openstack.org/11/670211/4/check/mistral-tox-unit-postgresql/bd10b3d/ : SUCCESS in 26m 22s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/11/670211/4/check/tripleo-ci-centos-7-scenario003-standalone/ecf7912/ : SUCCESS in 1h 20m 10s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/11/670211/4/check/tripleo-ci-centos-7-containers-multinode/ea745b7/ : SUCCESS in 2h 10m 18s\n- mistral-rally-task http://logs.openstack.org/11/670211/4/check/mistral-rally-task/feeb522/ : SUCCESS in 59m 33s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/11/670211/4/check/openstack-tox-lower-constraints/4e1793d/ : SUCCESS in 11m 28s","accounts_in_message":[],"_revision_number":4},{"id":"a9e64223cd5c2d5073d52ee086d2b8f838914969","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-07-17 11:07:15.000000000","message":"Patch Set 4: Code-Review-1\n\nThank you! Although this change seems ok, I would like to ask how useful/valid the base use case is. That is, to check the execution of the WF without executing actions. The problem I feel with the assumption that this can be done in any real workflows is that in most cases the edges from one task to the other will contain conditions. Also, the inputs and publish statements will require data to be present. As much as I would like to have this kind of functionality, it feels to be a blunt statement that this will be applicable in real world scenarios. Also, in most cases, the real challenge stems from the desire to be able to cover the possible branches of execution.\nMy humble opinion is that we shouldn\u0027t try to support this use case as it introduces maintenance cost with complexity and the power yield of this feature is negligible in most cases. So, in fact, I\u0027d vote -2 but I\u0027m open to discussion, of course.","accounts_in_message":[],"_revision_number":4},{"id":"cff22961e015e61adef4a8ced7df297554284255","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-07-18 15:28:40.000000000","message":"Patch Set 4:\n\n\u003e Thank you! Although this change seems ok, I would like to ask how\n \u003e useful/valid the base use case is. That is, to check the execution\n \u003e of the WF without executing actions. The problem I feel with the\n \u003e assumption that this can be done in any real workflows is that in\n \u003e most cases the edges from one task to the other will contain\n \u003e conditions. Also, the inputs and publish statements will require\n \u003e data to be present. As much as I would like to have this kind of\n \u003e functionality, it feels to be a blunt statement that this will be\n \u003e applicable in real world scenarios. Also, in most cases, the real\n \u003e challenge stems from the desire to be able to cover the possible\n \u003e branches of execution.\n \u003e My humble opinion is that we shouldn\u0027t try to support this use case\n \u003e as it introduces maintenance cost with complexity and the power\n \u003e yield of this feature is negligible in most cases. So, in fact, I\u0027d\n \u003e vote -2 but I\u0027m open to discussion, of course.\n\nAndras, thanks for you reply. The main idea is to provide the opportunity to run any tasks without any impact to any external system. If anyone wants to run a task in dry run mode, he should be sure that action under this task supports it.\nSo this patch is only an instrument and people who want to use this feature should implement this behavior in their actions. So, if any action in normal mode should return some value, the dry run mode should return some stub and so on. \nFor now, I just want to provide an instrument for this, maybe later it would be great to extend openstack actions and std (I am not sure if std actions should be changed to support this or not)","accounts_in_message":[],"_revision_number":4},{"id":"b9800f1e8e052a9ba41a4c49b004a86abf2e4714","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-07-29 18:41:12.000000000","message":"Patch Set 4: -Code-Review\n\n\u003e Andras, thanks for you reply. The main idea is to provide the\n \u003e opportunity to run any tasks without any impact to any external\n \u003e system. If anyone wants to run a task in dry run mode, he should be\n \u003e sure that action under this task supports it.\n \u003e So this patch is only an instrument and people who want to use this\n \u003e feature should implement this behavior in their actions. So, if any\n \u003e action in normal mode should return some value, the dry run mode\n \u003e should return some stub and so on.\n \u003e For now, I just want to provide an instrument for this, maybe later\n \u003e it would be great to extend openstack actions and std (I am not\n \u003e sure if std actions should be changed to support this or not)\n\nThanks Oleg, I really appreciate your feedback. Adding \u0027programmatic\u0027 features to the language is good on one hand but it also means code is added to the workflow. That code can fail and is practically a mock so it is up to the developer to recreate the behavior of the real world. I still have doubts about how this can be used in real life workflows. I\u0027m not against adding it and if it works for your use cases, it\u0027s great.","accounts_in_message":[],"_revision_number":4},{"id":"3319e68b90af3730a9c11f1b91cfe0ba7a330479","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-12-07 17:32:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"61d7eab0d037282275e15deff92b5363633a1bdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 19:32:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7fd74b456e64239925a61f301f9b369 : FAILURE in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95ca7f37444a4ed0928156f06a1db277 : TIMED_OUT in 40m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8bcb283256e041b3b879090c0cb09820 : TIMED_OUT in 41m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ffb345d19d4aaaaa6fb3433d9f2db7 : SUCCESS in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7d874572fe64bfbbb5ffb9ea3205dbe : TIMED_OUT in 50m 54s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0ef243672aba4efa9c023737d8f2d8d4 : FAILURE in 1h 04m 53s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b49642da13a6454ca9bd8f9fb3b2149c : FAILURE in 1h 06m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/62fe2d3f74b94684b1e927dbbd08f757 : FAILURE in 1h 02m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/af6b0d31358742ddabfe24b877cb6e7a : TIMED_OUT in 1h 00m 52s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f9a7fdd53ef547c192af15fec705e606 : TIMED_OUT in 1h 00m 53s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/cbd475fde9404ec2bdf9792e63ff30d1 : SUCCESS in 1h 18m 02s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/567147a1fbc34beebf2fb8ded2645d90 : FAILURE in 1h 54m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5623158535e84b5d8bdbf9ba81749d6c : FAILURE in 57m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89eff2925a494fc782f1e86ae4adb6c5 : TIMED_OUT in 40m 57s","accounts_in_message":[],"_revision_number":5},{"id":"776a4a0d598c0070a732b87eb0d103020aa1ba33","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-10 05:30:06.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"9e9573ab54dbc43b05042b5008022512cdb12024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 08:09:41.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89c33c5a39df4b24971847cf1fd6a656 : FAILURE in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23dc6f3300634593a6d3e2709dbdb1f4 : TIMED_OUT in 41m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/255f9875676e47c186e9c1e069929c84 : TIMED_OUT in 41m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb1dc81b292d42e2b857fcfcc321270a : SUCCESS in 6m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32140164c63047bb965a1d14675e4e89 : TIMED_OUT in 51m 13s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1c9527558b0841adb9322344a644dbad : TIMED_OUT in 2h 34m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8c127e28ba6647c29fe7af774acbd26a : FAILURE in 1h 05m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/51d0958bbccd42c28ebf53b5a5421e02 : FAILURE in 1h 05m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9d2b1b5ccb8944c28a6a9cf81ddfbdff : TIMED_OUT in 1h 00m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/168dbaf257074fad96840e53a2565b43 : TIMED_OUT in 1h 01m 01s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a313defcc64b4c1b863a30d4d7cdbf82 : SUCCESS in 1h 32m 37s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4b312740b6cd4a1c944ae1af63c62b8c : FAILURE in 1h 47m 21s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/541cba3a86f24fc180cd507e3c64a364 : FAILURE in 1h 20m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c6e20de5a424c12ace242eb55352807 : TIMED_OUT in 41m 25s","accounts_in_message":[],"_revision_number":6},{"id":"e1935912608f41b6a37013d1b5744fabb1ca8f4f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-25 05:01:15.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"a25bc437bea7a843810daa9d3c6cbc57f25c7af2","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-25 05:45:42.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"d61438fb72e0e6500f6fe9fab2cb4018fc026fc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-25 07:46:53.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d85c89b316f4364bfa163d8417a8e61 : FAILURE in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30c691a2d68748fb91a08c64958375ba : TIMED_OUT in 41m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2edb693c8f084f1bb4859d7a2ebff9b3 : TIMED_OUT in 41m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed99ce112700432796b63caff1dcd919 : SUCCESS in 6m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3d243a91f694b9d872cae59a6b7b9e0 : TIMED_OUT in 51m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/81008d04af754cd194b94288dff70457 : FAILURE in 1h 07m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ee785934a95a4abbac20b0a8840c344c : FAILURE in 1h 09m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7a50554daacd41418dbc6e5cec5fa777 : FAILURE in 1h 04m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ba4abdc2809b4c388d5379a9664f9eea : TIMED_OUT in 1h 01m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f796be833f864b8c95114a9b633f769f : TIMED_OUT in 1h 01m 05s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ef82912c178c4da38f121c551994acd9 : FAILURE in 2h 09m 55s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f627315faf6e4496ab299ea61a699c53 : FAILURE in 53m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfa2945537794523bd058c8eec4841d4 : TIMED_OUT in 40m 57s","accounts_in_message":[],"_revision_number":7},{"id":"080745dd69addaa46161776e57a5c1cb70fb4478","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-03-28 19:37:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"97ed8d17f83cde476805c3c2b978a6e29cedc55f","revisions":{"3fec162447412e4f45895bd9bba3a5b065718dee":{"kind":"REWORK","_number":1,"created":"2019-07-10 23:02:50.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/11/670211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/670211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/670211/1"}}},"commit":{"parents":[{"commit":"6a9aa4c295b51e9b11ab4a30e5c002bcc2924609","subject":"Merge \"Initialize the lazily loaded execution \"input\" field in API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a9aa4c295b51e9b11ab4a30e5c002bcc2924609"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-07-10 23:01:26.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-07-10 23:01:26.000000000","tz":180},"subject":"Dry run feature","message":"Dry run feature\n\nNow it is possible to execute tasks in dry-run mode, so instead of \u0027run\u0027\naction method \u0027test\u0027 will be executed. It could be configured\nfor a single task or for a whole workflow in task-defaults section.\n\nChange-Id: I9b50bb40d2f1fc883f32406f4bc11f03a15f0a70\nImplements: blueprint mistral-dry-run\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3fec162447412e4f45895bd9bba3a5b065718dee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3fec162447412e4f45895bd9bba3a5b065718dee"}]},"branch":"refs/heads/master"},"7e3058c7f926353e06faea82fc696eae53383729":{"kind":"REWORK","_number":2,"created":"2019-07-11 14:47:22.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/11/670211/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/670211/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/670211/2"}}},"commit":{"parents":[{"commit":"6a9aa4c295b51e9b11ab4a30e5c002bcc2924609","subject":"Merge \"Initialize the lazily loaded execution \"input\" field in API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a9aa4c295b51e9b11ab4a30e5c002bcc2924609"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-07-10 23:01:26.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-07-11 14:47:11.000000000","tz":180},"subject":"Dry run feature","message":"Dry run feature\n\nNow it is possible to execute tasks in dry-run mode, so instead of \u0027run\u0027\naction method \u0027test\u0027 will be executed. It could be configured\nfor a single task or for a whole workflow in task-defaults section.\n\nChange-Id: I9b50bb40d2f1fc883f32406f4bc11f03a15f0a70\nImplements: blueprint mistral-dry-run\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7e3058c7f926353e06faea82fc696eae53383729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7e3058c7f926353e06faea82fc696eae53383729"}]},"branch":"refs/heads/master"},"10697374f1097897e061dc73f0a47b66ff3ddf6d":{"kind":"REWORK","_number":3,"created":"2019-07-11 20:08:00.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/11/670211/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/670211/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/670211/3"}}},"commit":{"parents":[{"commit":"6a9aa4c295b51e9b11ab4a30e5c002bcc2924609","subject":"Merge \"Initialize the lazily loaded execution \"input\" field in API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a9aa4c295b51e9b11ab4a30e5c002bcc2924609"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-07-10 23:01:26.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-07-11 20:07:42.000000000","tz":180},"subject":"Dry run feature","message":"Dry run feature\n\nNow it is possible to execute tasks in dry-run mode, so instead of \u0027run\u0027\naction method \u0027test\u0027 will be executed. It could be configured\nfor a single task or for a whole workflow in task-defaults section.\n\nChange-Id: I9b50bb40d2f1fc883f32406f4bc11f03a15f0a70\nImplements: blueprint mistral-dry-run\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/10697374f1097897e061dc73f0a47b66ff3ddf6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/10697374f1097897e061dc73f0a47b66ff3ddf6d"}]},"branch":"refs/heads/master"},"edbf5444c8c37d73aef8dc21236353ea98170058":{"kind":"REWORK","_number":4,"created":"2019-07-11 23:26:49.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/11/670211/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/670211/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/670211/4"}}},"commit":{"parents":[{"commit":"6a9aa4c295b51e9b11ab4a30e5c002bcc2924609","subject":"Merge \"Initialize the lazily loaded execution \"input\" field in API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a9aa4c295b51e9b11ab4a30e5c002bcc2924609"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-07-10 23:01:26.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-07-11 23:25:46.000000000","tz":180},"subject":"Dry run feature","message":"Dry run feature\n\nNow it is possible to execute tasks in dry-run mode, so instead of \u0027run\u0027\naction method \u0027test\u0027 will be executed. It could be configured\nfor a single task or for a whole workflow in task-defaults section.\n\nChange-Id: I9b50bb40d2f1fc883f32406f4bc11f03a15f0a70\nImplements: blueprint mistral-dry-run\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/edbf5444c8c37d73aef8dc21236353ea98170058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/edbf5444c8c37d73aef8dc21236353ea98170058"}]},"branch":"refs/heads/master"},"dc3c4966706493800f5e5517d9c0b28bf3f21d8e":{"kind":"REWORK","_number":5,"created":"2019-12-07 17:32:23.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/11/670211/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/670211/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/670211/5"}}},"commit":{"parents":[{"commit":"7b58b9c2672684abbe2960e321a8c92ed7b94727","subject":"Fix log messages in the action heartbeats mechanism","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7b58b9c2672684abbe2960e321a8c92ed7b94727"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-07-10 23:01:26.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-12-07 17:32:07.000000000","tz":180},"subject":"Dry run feature","message":"Dry run feature\n\nNow it is possible to execute tasks in dry-run mode, so instead of \u0027run\u0027\naction method \u0027test\u0027 will be executed. It could be configured\nfor a single task or for a whole workflow in task-defaults section.\n\nChange-Id: I9b50bb40d2f1fc883f32406f4bc11f03a15f0a70\nImplements: blueprint mistral-dry-run\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dc3c4966706493800f5e5517d9c0b28bf3f21d8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dc3c4966706493800f5e5517d9c0b28bf3f21d8e"}]},"branch":"refs/heads/master"},"ffbfc28434fec550069f770fd743a0ea9623bbfc":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-12-10 05:30:06.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/11/670211/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/670211/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/670211/6"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-07-10 23:01:26.000000000","tz":180},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-10 05:30:06.000000000","tz":0},"subject":"Dry run feature","message":"Dry run feature\n\nNow it is possible to execute tasks in dry-run mode, so instead of \u0027run\u0027\naction method \u0027test\u0027 will be executed. It could be configured\nfor a single task or for a whole workflow in task-defaults section.\n\nChange-Id: I9b50bb40d2f1fc883f32406f4bc11f03a15f0a70\nImplements: blueprint mistral-dry-run\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ffbfc28434fec550069f770fd743a0ea9623bbfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ffbfc28434fec550069f770fd743a0ea9623bbfc"}]},"branch":"refs/heads/master"},"97ed8d17f83cde476805c3c2b978a6e29cedc55f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-12-25 05:01:15.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/11/670211/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/11/670211/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/11/670211/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/11/670211/7"}}},"commit":{"parents":[{"commit":"541aabbfe8a7afd876adcfb209d18aafdef2d5f6","subject":"Merge \"Use MISTRAL_SERVICE_HOST as the host ip for standalone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/541aabbfe8a7afd876adcfb209d18aafdef2d5f6"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-07-10 23:01:26.000000000","tz":180},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-25 05:01:15.000000000","tz":0},"subject":"Dry run feature","message":"Dry run feature\n\nNow it is possible to execute tasks in dry-run mode, so instead of \u0027run\u0027\naction method \u0027test\u0027 will be executed. It could be configured\nfor a single task or for a whole workflow in task-defaults section.\n\nChange-Id: I9b50bb40d2f1fc883f32406f4bc11f03a15f0a70\nImplements: blueprint mistral-dry-run\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/97ed8d17f83cde476805c3c2b978a6e29cedc55f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/97ed8d17f83cde476805c3c2b978a6e29cedc55f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
