)]}'
{"id":"openstack%2Fmistral~671914","triplet_id":"openstack%2Fmistral~master~I69ca615ed446214463f8fa7a3e303ff0bf32b460","project":"openstack/mistral","branch":"master","topic":"eyalb/devstack","hashtags":[],"change_id":"I69ca615ed446214463f8fa7a3e303ff0bf32b460","subject":"When installing devstack enable the bash completion","status":"MERGED","created":"2019-07-21 05:38:52.000000000","updated":"2019-07-31 06:27:45.000000000","submitted":"2019-07-31 06:27:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671914-1564554465582-f6dc5ada","meta_rev_id":"8d04033583b5eec042f744dcebcba6f3f806d0e8","_number":671914,"virtual_id_number":671914,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-07-31 06:27:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2019-07-29 07:21:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":1,"date":"2019-07-30 05:29:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2019-07-31 04:23:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-25 09:37:51.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-07-30 05:29:50.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-07-31 04:23:29.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-07-31 06:27:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a3c263acca6f54a54853daa61c79c9dff80be28","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-21 05:38:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c4df6359f55e8b3c7f13057c9b54887604b5521","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-21 06:41:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e0ff228a70b41f76687a659a8c565cfcfae383d2","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-21 07:09:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"52e0590b3a51f001833e951826df9cca4d27617b","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-21 07:53:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"10bff25e21378f6faac9c30e6d0e6d0db331bdb8","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-21 08:57:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4},{"id":"6ab57bcaa508c17ea81a0e4ce9b7706811563a4a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-21 10:42:56.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"b917a1ce5b2b78ecb1ff28a1406c09a161079dd3","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-21 10:48:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"840359a977852c36d93c9b90d0134bc838d61c9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-21 13:13:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/14/671914/5/check/openstack-tox-pep8/f3c41c7/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/14/671914/5/check/openstack-tox-py27/6b3f197/ : SUCCESS in 7m 19s\n- openstack-tox-py36 http://logs.openstack.org/14/671914/5/check/openstack-tox-py36/0d0bee2/ : SUCCESS in 10m 25s\n- openstack-tox-py37 http://logs.openstack.org/14/671914/5/check/openstack-tox-py37/00c46ce/ : SUCCESS in 7m 31s\n- openstack-tox-docs http://logs.openstack.org/14/671914/5/check/openstack-tox-docs/bc8d4f9/html/ : SUCCESS in 3m 46s\n- openstack-tox-cover http://logs.openstack.org/14/671914/5/check/openstack-tox-cover/6ee7fc3/cover/ : SUCCESS in 8m 02s (non-voting)\n- mistral-devstack http://logs.openstack.org/14/671914/5/check/mistral-devstack/3b30837/ : SUCCESS in 41m 56s\n- mistral-devstack-non-apache http://logs.openstack.org/14/671914/5/check/mistral-devstack-non-apache/d4a1513/ : SUCCESS in 37m 17s\n- mistral-devstack-kombu http://logs.openstack.org/14/671914/5/check/mistral-devstack-kombu/7ae0151/ : SUCCESS in 41m 32s\n- mistral-tox-unit-mysql http://logs.openstack.org/14/671914/5/check/mistral-tox-unit-mysql/3663674/ : SUCCESS in 27m 42s\n- mistral-tox-unit-postgresql http://logs.openstack.org/14/671914/5/check/mistral-tox-unit-postgresql/b1e8f1e/ : SUCCESS in 25m 18s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/14/671914/5/check/tripleo-ci-centos-7-scenario003-standalone/6101098/ : SUCCESS in 1h 09m 22s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/14/671914/5/check/tripleo-ci-centos-7-containers-multinode/b6000c4/ : SUCCESS in 2h 19m 43s\n- mistral-rally-task http://logs.openstack.org/14/671914/5/check/mistral-rally-task/32ed8b9/ : SUCCESS in 54m 15s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/14/671914/5/check/openstack-tox-lower-constraints/686e55c/ : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":5},{"id":"0a951b951ef7d562d137e68177403f4889df6c90","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-22 04:29:00.000000000","message":"Patch Set 5: Code-Review+2\n\nLooks good to me, I\u0027d prefer that someone else take a look.","accounts_in_message":[],"_revision_number":5},{"id":"3c10793c784cb66e88b26a9b56705b96bb99c9cb","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-07-24 09:08:32.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"83ec060e1aabfab68e689cf2a0b0c4ae8d3f0bd2","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-07-25 09:37:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"34ae537c8ea87aaea7a86303f59487ebd227ce2a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-25 11:10:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"aff63a3ca864df25f1ac0a9ce263f76e9ab596fd","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-25 11:11:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"38a56ce870d981d6b510b853ad32a85f53dbda39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 13:37:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/14/671914/6/check/openstack-tox-pep8/cdd34ab/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/14/671914/6/check/openstack-tox-py27/90ce793/ : SUCCESS in 10m 02s\n- openstack-tox-py36 http://logs.openstack.org/14/671914/6/check/openstack-tox-py36/5c44cc3/ : SUCCESS in 9m 19s\n- openstack-tox-py37 http://logs.openstack.org/14/671914/6/check/openstack-tox-py37/8c02760/ : SUCCESS in 8m 37s\n- openstack-tox-docs http://logs.openstack.org/14/671914/6/check/openstack-tox-docs/b58177e/html/ : SUCCESS in 5m 27s\n- openstack-tox-cover http://logs.openstack.org/14/671914/6/check/openstack-tox-cover/f625ee2/cover/ : SUCCESS in 11m 28s (non-voting)\n- mistral-devstack http://logs.openstack.org/14/671914/6/check/mistral-devstack/efbb75c/ : SUCCESS in 45m 49s\n- mistral-devstack-non-apache http://logs.openstack.org/14/671914/6/check/mistral-devstack-non-apache/ce1e2a5/ : SUCCESS in 51m 21s\n- mistral-devstack-kombu http://logs.openstack.org/14/671914/6/check/mistral-devstack-kombu/20b8e20/ : SUCCESS in 42m 46s\n- mistral-tox-unit-mysql http://logs.openstack.org/14/671914/6/check/mistral-tox-unit-mysql/d802a1a/ : SUCCESS in 28m 21s\n- mistral-tox-unit-postgresql http://logs.openstack.org/14/671914/6/check/mistral-tox-unit-postgresql/dcf2c41/ : SUCCESS in 27m 14s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/14/671914/6/check/tripleo-ci-centos-7-scenario003-standalone/b14f3b7/ : SUCCESS in 1h 21m 20s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/14/671914/6/check/tripleo-ci-centos-7-containers-multinode/7b49820/ : SUCCESS in 2h 20m 46s\n- mistral-rally-task http://logs.openstack.org/14/671914/6/check/mistral-rally-task/a0bec94/ : SUCCESS in 1h 08m 46s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/14/671914/6/check/openstack-tox-lower-constraints/8ef7bee/ : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":6},{"id":"d0e6d3275da8ae40938dba9ec3af0ebaddca20cb","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-29 07:21:23.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"3de71636cb6764df3ea504ddde0482f90248aabc","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-07-30 05:29:50.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"a5d33a0386c25e908cb0680e08d0eff4111a5b47","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-31 04:23:29.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"8203a6e85e972b217c56d9d1054ec877e6a5f7ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 04:23:42.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"733ccdc04f05d36fc19d8bd731a9235dd21ba669","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 06:27:45.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/14/671914/6/gate/openstack-tox-pep8/59125de/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/14/671914/6/gate/openstack-tox-py27/0d9fda0/ : SUCCESS in 8m 51s\n- openstack-tox-py36 http://logs.openstack.org/14/671914/6/gate/openstack-tox-py36/700f739/ : SUCCESS in 9m 03s\n- openstack-tox-py37 http://logs.openstack.org/14/671914/6/gate/openstack-tox-py37/5d5c5dd/ : SUCCESS in 8m 14s\n- openstack-tox-docs http://logs.openstack.org/14/671914/6/gate/openstack-tox-docs/839a0ad/html/ : SUCCESS in 4m 47s\n- mistral-devstack http://logs.openstack.org/14/671914/6/gate/mistral-devstack/37910c3/ : SUCCESS in 43m 21s\n- mistral-devstack-non-apache http://logs.openstack.org/14/671914/6/gate/mistral-devstack-non-apache/1dcf221/ : SUCCESS in 43m 07s\n- mistral-tox-unit-mysql http://logs.openstack.org/14/671914/6/gate/mistral-tox-unit-mysql/f4dcee5/ : SUCCESS in 28m 24s\n- mistral-tox-unit-postgresql http://logs.openstack.org/14/671914/6/gate/mistral-tox-unit-postgresql/221ba77/ : SUCCESS in 28m 18s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/14/671914/6/gate/tripleo-ci-centos-7-scenario003-standalone/0a3f164/ : SUCCESS in 1h 12m 17s\n- mistral-devstack-kombu http://logs.openstack.org/14/671914/6/gate/mistral-devstack-kombu/d705e32/ : SUCCESS in 45m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/671914/6/gate/openstack-tox-lower-constraints/c5f1cb4/ : SUCCESS in 9m 37s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/14/671914/6/gate/tripleo-ci-centos-7-undercloud-containers/d3767f5/ : SUCCESS in 1h 34m 02s","accounts_in_message":[],"_revision_number":6},{"id":"504dfac3a98c8dcdcbfe57b7a1ca9eb62d9e3cf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 06:27:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"90a9a48f21cbbf9ccf270799d0fc2aa25895636f","revisions":{"060611d1f87a8e07e8c749e69819a9a5ce89b798":{"kind":"REWORK","_number":1,"created":"2019-07-21 05:38:52.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/14/671914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/671914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/671914/1"}}},"commit":{"parents":[{"commit":"258575c21de3e5049694d7ee9b34b6322179cb11","subject":"Merge \"Add workflow_execution_id to task notification object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/258575c21de3e5049694d7ee9b34b6322179cb11"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-21 05:38:33.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-21 05:38:33.000000000","tz":180},"subject":"When installing devstack enable the bash completion","message":"When installing devstack enable the bash completion\n\nIn devstack if we install the client then copy the bash\ncompletion script to its folder.\nalso use the standart devstack function to install the\nmistral client\n\nChange-Id: I69ca615ed446214463f8fa7a3e303ff0bf32b460\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/060611d1f87a8e07e8c749e69819a9a5ce89b798"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/060611d1f87a8e07e8c749e69819a9a5ce89b798"}]},"branch":"refs/heads/master"},"9f913bb551aae4f0d617b49fbf06d0938e51250b":{"kind":"REWORK","_number":2,"created":"2019-07-21 06:41:15.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/14/671914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/671914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/671914/2"}}},"commit":{"parents":[{"commit":"258575c21de3e5049694d7ee9b34b6322179cb11","subject":"Merge \"Add workflow_execution_id to task notification object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/258575c21de3e5049694d7ee9b34b6322179cb11"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-21 05:38:33.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-21 06:40:32.000000000","tz":180},"subject":"When installing devstack enable the bash completion","message":"When installing devstack enable the bash completion\n\nIn devstack if we install the client then copy the bash\ncompletion script to its folder.\nalso use the standart devstack function to install the\nmistral client\n\nChange-Id: I69ca615ed446214463f8fa7a3e303ff0bf32b460\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f913bb551aae4f0d617b49fbf06d0938e51250b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f913bb551aae4f0d617b49fbf06d0938e51250b"}]},"branch":"refs/heads/master"},"1cbded8acfec031590d8841ad29c2ca09d6b664c":{"kind":"REWORK","_number":3,"created":"2019-07-21 07:09:21.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/14/671914/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/671914/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/671914/3"}}},"commit":{"parents":[{"commit":"258575c21de3e5049694d7ee9b34b6322179cb11","subject":"Merge \"Add workflow_execution_id to task notification object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/258575c21de3e5049694d7ee9b34b6322179cb11"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-21 05:38:33.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-21 07:08:50.000000000","tz":180},"subject":"When installing devstack enable the bash completion","message":"When installing devstack enable the bash completion\n\nIn devstack if we install the client then copy the bash\ncompletion script to its folder.\nalso use the standart devstack function to install the\nmistral client\n\nChange-Id: I69ca615ed446214463f8fa7a3e303ff0bf32b460\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1cbded8acfec031590d8841ad29c2ca09d6b664c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1cbded8acfec031590d8841ad29c2ca09d6b664c"}]},"branch":"refs/heads/master"},"c21bb4520365bcb829fbe9a022d13b595133fa44":{"kind":"REWORK","_number":4,"created":"2019-07-21 07:53:40.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/14/671914/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/671914/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/671914/4"}}},"commit":{"parents":[{"commit":"258575c21de3e5049694d7ee9b34b6322179cb11","subject":"Merge \"Add workflow_execution_id to task notification object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/258575c21de3e5049694d7ee9b34b6322179cb11"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-21 05:38:33.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-21 07:53:22.000000000","tz":180},"subject":"When installing devstack enable the bash completion","message":"When installing devstack enable the bash completion\n\nIn devstack if we install the client then copy the bash\ncompletion script to its folder.\nalso use the standart devstack function to install the\nmistral client\n\nChange-Id: I69ca615ed446214463f8fa7a3e303ff0bf32b460\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c21bb4520365bcb829fbe9a022d13b595133fa44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c21bb4520365bcb829fbe9a022d13b595133fa44"}]},"branch":"refs/heads/master"},"85503d2f72066701e56adc7da55c8bbb999737e4":{"kind":"REWORK","_number":5,"created":"2019-07-21 10:48:49.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/14/671914/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/671914/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/671914/5"}}},"commit":{"parents":[{"commit":"258575c21de3e5049694d7ee9b34b6322179cb11","subject":"Merge \"Add workflow_execution_id to task notification object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/258575c21de3e5049694d7ee9b34b6322179cb11"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-21 05:38:33.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-21 10:48:09.000000000","tz":180},"subject":"When installing devstack enable the bash completion","message":"When installing devstack enable the bash completion\n\nIn devstack if we install the client then copy the bash\ncompletion script to its folder.\nalso use the standart devstack function to install the\nmistral client\n\nChange-Id: I69ca615ed446214463f8fa7a3e303ff0bf32b460\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/85503d2f72066701e56adc7da55c8bbb999737e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/85503d2f72066701e56adc7da55c8bbb999737e4"}]},"branch":"refs/heads/master"},"90a9a48f21cbbf9ccf270799d0fc2aa25895636f":{"kind":"REWORK","_number":6,"created":"2019-07-25 11:10:12.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/14/671914/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/671914/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/671914/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/671914/6"}}},"commit":{"parents":[{"commit":"258575c21de3e5049694d7ee9b34b6322179cb11","subject":"Merge \"Add workflow_execution_id to task notification object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/258575c21de3e5049694d7ee9b34b6322179cb11"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-21 05:38:33.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-25 11:09:06.000000000","tz":180},"subject":"When installing devstack enable the bash completion","message":"When installing devstack enable the bash completion\n\nIn devstack if we install the client then copy the bash\ncompletion script to its folder.\nalso use the standart devstack function to install the\nmistral client\n\nChange-Id: I69ca615ed446214463f8fa7a3e303ff0bf32b460\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/90a9a48f21cbbf9ccf270799d0fc2aa25895636f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/90a9a48f21cbbf9ccf270799d0fc2aa25895636f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
