)]}'
{"id":"openstack%2Fmistral~672185","triplet_id":"openstack%2Fmistral~stable%2Frocky~Ib2c72e89612c9b2b4bccd37cb797804ea40e0b1f","project":"openstack/mistral","branch":"stable/rocky","topic":"bug/1837468","hashtags":[],"change_id":"Ib2c72e89612c9b2b4bccd37cb797804ea40e0b1f","subject":"disable triggering version discovery in test_generator unit tests","status":"MERGED","created":"2019-07-22 22:18:41.000000000","updated":"2019-07-29 15:43:31.000000000","submitted":"2019-07-29 09:23:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672185-1564392202799-8da850ff","meta_rev_id":"b8302d9c74475a2829a1247f1822d0e70755e359","_number":672185,"virtual_id_number":672185,"owner":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-07-29 09:23:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-07-29 05:06:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2019-07-29 05:06:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 22:19:09.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-07-25 07:46:41.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-07-29 05:06:35.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-07-29 09:23:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68b1553f7e3376417866003a360ce650e5e72739","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-07-22 22:18:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"231ad87d569132b6be52a914c26f78d1ba07a328","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 01:06:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/85/672185/1/check/openstack-tox-pep8/a90e26c/ : FAILURE in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/85/672185/1/check/openstack-tox-py27/1e05a20/ : RETRY_LIMIT in 3m 35s\n- openstack-tox-py35 http://logs.openstack.org/85/672185/1/check/openstack-tox-py35/1132a29/ : RETRY_LIMIT in 2m 35s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/672185/1/check/build-openstack-sphinx-docs/6a88f67/html/ : SUCCESS in 5m 39s\n- build-openstack-releasenotes http://logs.openstack.org/85/672185/1/check/build-openstack-releasenotes/c730692/html/ : SUCCESS in 5m 26s\n- openstack-tox-cover http://logs.openstack.org/85/672185/1/check/openstack-tox-cover/2b1108b/ : RETRY_LIMIT in 2m 28s (non-voting)\n- mistral-devstack http://logs.openstack.org/85/672185/1/check/mistral-devstack/dd553e2/ : SUCCESS in 41m 12s\n- mistral-devstack-non-apache http://logs.openstack.org/85/672185/1/check/mistral-devstack-non-apache/6396e49/ : SUCCESS in 43m 55s\n- mistral-devstack-kombu http://logs.openstack.org/85/672185/1/check/mistral-devstack-kombu/fe3bc20/ : SUCCESS in 51m 06s\n- mistral-tox-unit-mysql http://logs.openstack.org/85/672185/1/check/mistral-tox-unit-mysql/ae78e1b/ : RETRY_LIMIT in 2m 50s\n- mistral-tox-unit-postgresql http://logs.openstack.org/85/672185/1/check/mistral-tox-unit-postgresql/1e5cd52/ : RETRY_LIMIT in 2m 31s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/85/672185/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/a86a33a/ : SUCCESS in 2h 45m 37s (non-voting)\n- mistral-rally-task http://logs.openstack.org/85/672185/1/check/mistral-rally-task/ebfb819/ : SUCCESS in 1h 10m 33s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/85/672185/1/check/openstack-tox-lower-constraints/d4c3016/ : RETRY_LIMIT in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f7fb1ed862f4eb8964bf0f9e3a9f1a7b25f862fa","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-07-23 14:49:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5ac502e9c2ae4a6b8d975822558c5cbfbf298161","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:22:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/85/672185/2/check/openstack-tox-pep8/0c07db7/ : SUCCESS in 6m 08s\n- openstack-tox-py27 http://logs.openstack.org/85/672185/2/check/openstack-tox-py27/7bf8434/ : RETRY_LIMIT in 3m 36s\n- openstack-tox-py35 http://logs.openstack.org/85/672185/2/check/openstack-tox-py35/9062b9a/ : RETRY_LIMIT in 3m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/672185/2/check/build-openstack-sphinx-docs/fc6f87f/html/ : SUCCESS in 4m 41s\n- build-openstack-releasenotes http://logs.openstack.org/85/672185/2/check/build-openstack-releasenotes/6f6f022/html/ : SUCCESS in 4m 58s\n- openstack-tox-cover http://logs.openstack.org/85/672185/2/check/openstack-tox-cover/6083b58/ : RETRY_LIMIT in 3m 04s (non-voting)\n- mistral-devstack http://logs.openstack.org/85/672185/2/check/mistral-devstack/bb424d1/ : SUCCESS in 48m 02s\n- mistral-devstack-non-apache http://logs.openstack.org/85/672185/2/check/mistral-devstack-non-apache/16b8641/ : SUCCESS in 48m 35s\n- mistral-devstack-kombu http://logs.openstack.org/85/672185/2/check/mistral-devstack-kombu/cc555de/ : SUCCESS in 42m 06s\n- mistral-tox-unit-mysql http://logs.openstack.org/85/672185/2/check/mistral-tox-unit-mysql/1596cb6/ : RETRY_LIMIT in 2m 09s\n- mistral-tox-unit-postgresql http://logs.openstack.org/85/672185/2/check/mistral-tox-unit-postgresql/f5ef442/ : RETRY_LIMIT in 2m 27s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/85/672185/2/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/d365e29/ : SUCCESS in 2h 24m 58s (non-voting)\n- mistral-rally-task http://logs.openstack.org/85/672185/2/check/mistral-rally-task/09873f5/ : SUCCESS in 1h 00m 50s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/85/672185/2/check/openstack-tox-lower-constraints/8dfbeb0/ : RETRY_LIMIT in 3m 21s","accounts_in_message":[],"_revision_number":2},{"id":"6fb7757e8c5ec6bee73e10ee5cfa8109ff0b5646","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-07-23 19:01:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"46415255cccbaa9dd78ceef94ae2ef080494d430","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 21:52:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/85/672185/2/check/openstack-tox-pep8/315f0fd/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/85/672185/2/check/openstack-tox-py27/d250e2a/ : RETRY_LIMIT in 4m 28s\n- openstack-tox-py35 http://logs.openstack.org/85/672185/2/check/openstack-tox-py35/55aefd3/ : RETRY_LIMIT in 2m 59s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/672185/2/check/build-openstack-sphinx-docs/981eebd/html/ : SUCCESS in 4m 22s\n- build-openstack-releasenotes http://logs.openstack.org/85/672185/2/check/build-openstack-releasenotes/d7a2ff0/html/ : SUCCESS in 3m 54s\n- openstack-tox-cover http://logs.openstack.org/85/672185/2/check/openstack-tox-cover/ecac696/ : RETRY_LIMIT in 3m 00s (non-voting)\n- mistral-devstack http://logs.openstack.org/85/672185/2/check/mistral-devstack/e77c557/ : SUCCESS in 55m 14s\n- mistral-devstack-non-apache http://logs.openstack.org/85/672185/2/check/mistral-devstack-non-apache/2ef1c98/ : SUCCESS in 54m 53s\n- mistral-devstack-kombu http://logs.openstack.org/85/672185/2/check/mistral-devstack-kombu/d2c1c92/ : SUCCESS in 54m 21s\n- mistral-tox-unit-mysql http://logs.openstack.org/85/672185/2/check/mistral-tox-unit-mysql/2018297/ : RETRY_LIMIT in 4m 49s\n- mistral-tox-unit-postgresql http://logs.openstack.org/85/672185/2/check/mistral-tox-unit-postgresql/77a0d20/ : RETRY_LIMIT in 3m 29s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/85/672185/2/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/8f74a01/ : SUCCESS in 2h 26m 25s (non-voting)\n- mistral-rally-task http://logs.openstack.org/85/672185/2/check/mistral-rally-task/b14f6ef/ : SUCCESS in 1h 06m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/85/672185/2/check/openstack-tox-lower-constraints/072185a/ : RETRY_LIMIT in 3m 03s","accounts_in_message":[],"_revision_number":2},{"id":"ad607d497f8da70eec852321279d834ff8d07d5d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-24 15:01:31.000000000","message":"Patch Set 2: Code-Review-1\n\nIn OpenStack any bug usually has to be fixed in master branch first, and if the problem exists on stable branches then backport the fix branch-by-branch. (See process description: [1])\nIn case the bug does not exist in master anymore, then please indicate that in commit message.\n\nAnyway, the \u0027command not found\u0027 errors in the Zuul checks that cause the failures will disappear, whenever the bindep fix [2] lands on stable/rocky.\n\n[1] https://docs.openstack.org/project-team-guide/stable-branches.html#processes\n[2] https://review.opendev.org/#/c/672543/","accounts_in_message":[],"_revision_number":2},{"id":"838504ab8da28ba009f43a5ed2c432cc4fb9bd88","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-07-24 21:00:07.000000000","message":"Patch Set 2:\n\n\u003e In OpenStack any bug usually has to be fixed in master branch\n \u003e first, and if the problem exists on stable branches then backport\n \u003e the fix branch-by-branch. (See process description: [1])\n \u003e In case the bug does not exist in master anymore, then please\n \u003e indicate that in commit message.\n \u003e \n \u003e Anyway, the \u0027command not found\u0027 errors in the Zuul checks that\n \u003e cause the failures will disappear, whenever the bindep fix [2]\n \u003e lands on stable/rocky.\n \u003e \n \u003e [1] https://docs.openstack.org/project-team-guide/stable-branches.html#processes\n \u003e [2] https://review.opendev.org/#/c/672543/\n\nThanks for the review! Yes, the bug is only applicable to the stable/rocky branch and with the latest openstacksdk 0.7.3 release.\n\nhttps://review.opendev.org/#/c/671384/\n\nThe current problem is fixed in stable/stein release by this patch.\n\nhttps://review.opendev.org/#/c/614211/\n\nPrior to the above patch, we used to create the senlinclient.v1.client.Client object via __init__, which causes keystoneauth1 to perform version discovery if clustering_default_microversion is set. Please see the bug report for more details.\n\nI\u0027ll update the commit message to indicate that the fix is for stable/rocky only.","accounts_in_message":[],"_revision_number":2},{"id":"f8969c8ab4d8ca7ddaf8a4d80129868607e49469","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-07-24 21:03:41.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"78604549c06959b4d50483025de77ae6d11da776","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 23:18:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/85/672185/3/check/openstack-tox-pep8/ae30bac/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/85/672185/3/check/openstack-tox-py27/7dcc10b/ : RETRY_LIMIT in 2m 54s\n- openstack-tox-py35 http://logs.openstack.org/85/672185/3/check/openstack-tox-py35/4c85ec8/ : RETRY_LIMIT in 2m 59s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/672185/3/check/build-openstack-sphinx-docs/f7ebfcb/html/ : SUCCESS in 3m 55s\n- build-openstack-releasenotes http://logs.openstack.org/85/672185/3/check/build-openstack-releasenotes/eacd7a1/html/ : SUCCESS in 3m 43s\n- openstack-tox-cover http://logs.openstack.org/85/672185/3/check/openstack-tox-cover/e516be3/ : RETRY_LIMIT in 3m 32s (non-voting)\n- mistral-devstack http://logs.openstack.org/85/672185/3/check/mistral-devstack/08c1c97/ : SUCCESS in 45m 31s\n- mistral-devstack-non-apache http://logs.openstack.org/85/672185/3/check/mistral-devstack-non-apache/406f788/ : SUCCESS in 44m 54s\n- mistral-devstack-kombu http://logs.openstack.org/85/672185/3/check/mistral-devstack-kombu/9f85ab8/ : SUCCESS in 44m 10s\n- mistral-tox-unit-mysql http://logs.openstack.org/85/672185/3/check/mistral-tox-unit-mysql/3a53baf/ : RETRY_LIMIT in 4m 03s\n- mistral-tox-unit-postgresql http://logs.openstack.org/85/672185/3/check/mistral-tox-unit-postgresql/fd0200f/ : RETRY_LIMIT in 3m 00s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/85/672185/3/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/61d57c1/ : SUCCESS in 2h 09m 45s (non-voting)\n- mistral-rally-task http://logs.openstack.org/85/672185/3/check/mistral-rally-task/3dd6575/ : SUCCESS in 1h 00m 05s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/85/672185/3/check/openstack-tox-lower-constraints/766de7a/ : RETRY_LIMIT in 3m 38s","accounts_in_message":[],"_revision_number":3},{"id":"95774db7abedff4cfa3259dcf649ed10c09496e7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-25 07:46:41.000000000","message":"Patch Set 3:\n\nThanks for the clarification, Guang Yee!\n\nAlso, feel free to squash the bindep fix [1] to this patch as I think that both this patch and the bindep fix are necessary to pass Zuul checks.\n\n[1] https://review.opendev.org/#/c/672543/","accounts_in_message":[],"_revision_number":3},{"id":"0922487d7d97dc5935e606fd03fab1d988eb79fd","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-07-25 17:42:06.000000000","message":"Patch Set 3:\n\n\u003e Thanks for the clarification, Guang Yee!\n \u003e \n \u003e Also, feel free to squash the bindep fix [1] to this patch as I\n \u003e think that both this patch and the bindep fix are necessary to pass\n \u003e Zuul checks.\n \u003e \n \u003e [1] https://review.opendev.org/#/c/672543/\n\nThanks! I\u0027ll cherry-pick the changes from 672543.","accounts_in_message":[],"_revision_number":3},{"id":"ae30b6a64424f4cafa0efb1de6b2e9b0ee24ca6e","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-07-25 17:42:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7183e97e22d0bd46de8a26bad22e4aff2530e3e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 20:20:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/85/672185/4/check/openstack-tox-pep8/15a0639/ : SUCCESS in 6m 12s\n- openstack-tox-py27 http://logs.openstack.org/85/672185/4/check/openstack-tox-py27/cdb78ae/ : SUCCESS in 9m 55s\n- openstack-tox-py35 http://logs.openstack.org/85/672185/4/check/openstack-tox-py35/e61b6b3/ : SUCCESS in 9m 46s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/672185/4/check/build-openstack-sphinx-docs/2715d5e/html/ : SUCCESS in 6m 09s\n- build-openstack-releasenotes http://logs.openstack.org/85/672185/4/check/build-openstack-releasenotes/2b54b14/html/ : SUCCESS in 5m 16s\n- openstack-tox-cover http://logs.openstack.org/85/672185/4/check/openstack-tox-cover/6a68220/cover/ : SUCCESS in 9m 34s (non-voting)\n- mistral-devstack http://logs.openstack.org/85/672185/4/check/mistral-devstack/8628013/ : SUCCESS in 43m 48s\n- mistral-devstack-non-apache http://logs.openstack.org/85/672185/4/check/mistral-devstack-non-apache/bb15e13/ : SUCCESS in 43m 09s\n- mistral-devstack-kombu http://logs.openstack.org/85/672185/4/check/mistral-devstack-kombu/85da67c/ : SUCCESS in 48m 29s\n- mistral-tox-unit-mysql http://logs.openstack.org/85/672185/4/check/mistral-tox-unit-mysql/1dfca29/ : SUCCESS in 26m 49s\n- mistral-tox-unit-postgresql http://logs.openstack.org/85/672185/4/check/mistral-tox-unit-postgresql/abcfe75/ : SUCCESS in 28m 56s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/85/672185/4/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/47e00ae/ : SUCCESS in 2h 30m 44s (non-voting)\n- mistral-rally-task http://logs.openstack.org/85/672185/4/check/mistral-rally-task/92b822f/ : SUCCESS in 1h 05m 24s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/85/672185/4/check/openstack-tox-lower-constraints/e68df98/ : FAILURE in 13m 00s","accounts_in_message":[],"_revision_number":4},{"id":"eea175fff45659cbd6819ee14307d15588cf037d","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-07-25 23:30:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fac12b4395da7485ee62888cb6f730499e44aee0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 02:08:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/672185/5/check/openstack-tox-pep8/c215921/ : SUCCESS in 5m 24s\n- openstack-tox-py27 http://logs.openstack.org/85/672185/5/check/openstack-tox-py27/2ee44c5/ : SUCCESS in 8m 10s\n- openstack-tox-py35 http://logs.openstack.org/85/672185/5/check/openstack-tox-py35/4b4120f/ : SUCCESS in 7m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/672185/5/check/build-openstack-sphinx-docs/693ad34/html/ : SUCCESS in 5m 21s\n- build-openstack-releasenotes http://logs.openstack.org/85/672185/5/check/build-openstack-releasenotes/df9d64d/html/ : SUCCESS in 4m 18s\n- openstack-tox-cover http://logs.openstack.org/85/672185/5/check/openstack-tox-cover/172c25c/cover/ : SUCCESS in 8m 25s (non-voting)\n- mistral-devstack http://logs.openstack.org/85/672185/5/check/mistral-devstack/4b4d8d6/ : SUCCESS in 52m 02s\n- mistral-devstack-non-apache http://logs.openstack.org/85/672185/5/check/mistral-devstack-non-apache/45dbb5c/ : SUCCESS in 40m 21s\n- mistral-devstack-kombu http://logs.openstack.org/85/672185/5/check/mistral-devstack-kombu/99c17d9/ : SUCCESS in 46m 53s\n- mistral-tox-unit-mysql http://logs.openstack.org/85/672185/5/check/mistral-tox-unit-mysql/0fe2baf/ : SUCCESS in 26m 06s\n- mistral-tox-unit-postgresql http://logs.openstack.org/85/672185/5/check/mistral-tox-unit-postgresql/3673616/ : SUCCESS in 25m 36s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/85/672185/5/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/c8f2f9b/ : SUCCESS in 2h 35m 01s (non-voting)\n- mistral-rally-task http://logs.openstack.org/85/672185/5/check/mistral-rally-task/76988d7/ : SUCCESS in 56m 24s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/85/672185/5/check/openstack-tox-lower-constraints/1dca7f0/ : SUCCESS in 12m 11s","accounts_in_message":[],"_revision_number":5},{"id":"d4a8d29d05b75f378d330a5a6c366484ba03b0da","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-29 05:06:35.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nLGTM. Please also make a backport for the corresponding release note.","accounts_in_message":[],"_revision_number":5},{"id":"73aa45ec8941c9c83f88737deacaab9ad2a9f63b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 05:06:48.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"386e1f8d239390ef05d8b63e82866886eb25983e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 09:23:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"a8179873010a5e7734be3ae20510853d777e41be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 09:23:22.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/672185/5/gate/openstack-tox-pep8/91270c7/ : SUCCESS in 6m 37s\n- openstack-tox-py27 http://logs.openstack.org/85/672185/5/gate/openstack-tox-py27/27ff407/ : SUCCESS in 9m 31s\n- openstack-tox-py35 http://logs.openstack.org/85/672185/5/gate/openstack-tox-py35/2ebefe4/ : SUCCESS in 10m 45s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/672185/5/gate/build-openstack-sphinx-docs/aa56114/html/ : SUCCESS in 5m 58s\n- build-openstack-releasenotes http://logs.openstack.org/85/672185/5/gate/build-openstack-releasenotes/0ba743e/html/ : SUCCESS in 5m 45s\n- mistral-devstack http://logs.openstack.org/85/672185/5/gate/mistral-devstack/dfdebfb/ : SUCCESS in 55m 55s\n- mistral-devstack-non-apache http://logs.openstack.org/85/672185/5/gate/mistral-devstack-non-apache/685358c/ : SUCCESS in 46m 44s\n- mistral-tox-unit-mysql http://logs.openstack.org/85/672185/5/gate/mistral-tox-unit-mysql/d571787/ : SUCCESS in 27m 28s\n- mistral-tox-unit-postgresql http://logs.openstack.org/85/672185/5/gate/mistral-tox-unit-postgresql/44fa1d9/ : SUCCESS in 24m 34s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/85/672185/5/gate/tripleo-ci-centos-7-scenario003-multinode-oooq-container/9f7e3ef/ : SUCCESS in 2h 36m 18s (non-voting)\n- mistral-devstack-kombu http://logs.openstack.org/85/672185/5/gate/mistral-devstack-kombu/f31b623/ : SUCCESS in 49m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/672185/5/gate/openstack-tox-lower-constraints/d972e0f/ : SUCCESS in 11m 26s","accounts_in_message":[],"_revision_number":5},{"id":"526f0bb9f6a7bc2aa782369c8aaceb15be937626","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-07-29 15:43:31.000000000","message":"Patch Set 5:\n\n\u003e LGTM. Please also make a backport for the corresponding release\n \u003e note.\n\nThanks! I am unclear about your request. Is there a separate branch for the release notes? Please advice.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f2a0b569f767e0f7cea844ed01accccf810ca618","revisions":{"790e9f5b9636699fe6a1e4e6a14912881028d9c6":{"kind":"REWORK","_number":1,"created":"2019-07-22 22:18:41.000000000","uploader":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"ref":"refs/changes/85/672185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/672185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/672185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/672185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/672185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/672185/1"}}},"commit":{"parents":[{"commit":"c9e61b80ad9f762d4acb07466b7fa548addcbd5c","subject":"Add release notes to changes to wflanguage execution object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c9e61b80ad9f762d4acb07466b7fa548addcbd5c"}]}],"author":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2019-07-22 22:15:00.000000000","tz":-420},"committer":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2019-07-22 22:18:40.000000000","tz":-420},"subject":"disable triggering version discovery in test_generator unit tests","message":"disable triggering version discovery in test_generator unit tests\n\nFixed unit.actions.openstack.test_generator.GeneratorTest.test_generator\nfailure by avoid triggering version discovery as it needs to talk to\nKeystone server. Talking to a live Keystone server is not needed for\nthis particular unit test.\n\nChange-Id: Ib2c72e89612c9b2b4bccd37cb797804ea40e0b1f\nCloses-Bug: 1837468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/790e9f5b9636699fe6a1e4e6a14912881028d9c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/790e9f5b9636699fe6a1e4e6a14912881028d9c6"}]},"branch":"refs/heads/stable/rocky"},"87661abc468475cdd6fe2b31050a9eb76ccf66aa":{"kind":"REWORK","_number":2,"created":"2019-07-23 14:49:50.000000000","uploader":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"ref":"refs/changes/85/672185/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/672185/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/672185/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/672185/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/672185/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/672185/2"}}},"commit":{"parents":[{"commit":"c9e61b80ad9f762d4acb07466b7fa548addcbd5c","subject":"Add release notes to changes to wflanguage execution object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c9e61b80ad9f762d4acb07466b7fa548addcbd5c"}]}],"author":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2019-07-22 22:15:00.000000000","tz":-420},"committer":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2019-07-23 14:49:43.000000000","tz":-420},"subject":"disable triggering version discovery in test_generator unit tests","message":"disable triggering version discovery in test_generator unit tests\n\nFixed unit.actions.openstack.test_generator.GeneratorTest.test_generator\nfailure by avoid triggering version discovery as it needs to talk to\nKeystone server. Talking to a live Keystone server is not needed for\nthis particular unit test.\n\nChange-Id: Ib2c72e89612c9b2b4bccd37cb797804ea40e0b1f\nCloses-Bug: 1837468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87661abc468475cdd6fe2b31050a9eb76ccf66aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87661abc468475cdd6fe2b31050a9eb76ccf66aa"}]},"branch":"refs/heads/stable/rocky"},"bbc5d886fe00352c2ed8ec9062ad5d41dac90dcb":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-07-24 21:03:41.000000000","uploader":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"ref":"refs/changes/85/672185/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/672185/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/672185/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/672185/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/672185/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/672185/3"}}},"commit":{"parents":[{"commit":"c9e61b80ad9f762d4acb07466b7fa548addcbd5c","subject":"Add release notes to changes to wflanguage execution object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c9e61b80ad9f762d4acb07466b7fa548addcbd5c"}]}],"author":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2019-07-22 22:15:00.000000000","tz":-420},"committer":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2019-07-24 21:00:22.000000000","tz":-420},"subject":"disable triggering version discovery in test_generator unit tests","message":"disable triggering version discovery in test_generator unit tests\n\nFixed unit.actions.openstack.test_generator.GeneratorTest.test_generator\nfailure by avoid triggering version discovery as it needs to talk to\nKeystone server. Talking to a live Keystone server is not needed for\nthis particular unit test.\n\nNOTE: this patch is only applicable to stable/rocky and with the latest\nopenstacksdk 0.17.3 release.\n\nChange-Id: Ib2c72e89612c9b2b4bccd37cb797804ea40e0b1f\nCloses-Bug: 1837468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bbc5d886fe00352c2ed8ec9062ad5d41dac90dcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bbc5d886fe00352c2ed8ec9062ad5d41dac90dcb"}]},"branch":"refs/heads/stable/rocky"},"190c9cf53c4932d0528dfe0f6a87aa9e15cb52f6":{"kind":"REWORK","_number":4,"created":"2019-07-25 17:42:15.000000000","uploader":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"ref":"refs/changes/85/672185/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/672185/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/672185/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/672185/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/672185/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/672185/4"}}},"commit":{"parents":[{"commit":"c9e61b80ad9f762d4acb07466b7fa548addcbd5c","subject":"Add release notes to changes to wflanguage execution object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c9e61b80ad9f762d4acb07466b7fa548addcbd5c"}]}],"author":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2019-07-22 22:15:00.000000000","tz":-420},"committer":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2019-07-25 17:42:15.000000000","tz":-420},"subject":"disable triggering version discovery in test_generator unit tests","message":"disable triggering version discovery in test_generator unit tests\n\nFixed unit.actions.openstack.test_generator.GeneratorTest.test_generator\nfailure by avoid triggering version discovery as it needs to talk to\nKeystone server. Talking to a live Keystone server is not needed for\nthis particular unit test.\n\nNOTE: this patch is only applicable to stable/rocky and with the latest\nopenstacksdk 0.17.3 release.\n\nNOTE: this patch also includes change I9ab552dd4bf3826f358d73a4bddaf1885b1be130\nfrom https://review.opendev.org/#/c/672543/ as both are needed in order\nto pass Jenkins.\n\nChange-Id: Ib2c72e89612c9b2b4bccd37cb797804ea40e0b1f\nCloses-Bug: 1837468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/190c9cf53c4932d0528dfe0f6a87aa9e15cb52f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/190c9cf53c4932d0528dfe0f6a87aa9e15cb52f6"}]},"branch":"refs/heads/stable/rocky"},"f2a0b569f767e0f7cea844ed01accccf810ca618":{"kind":"REWORK","_number":5,"created":"2019-07-25 23:30:28.000000000","uploader":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"ref":"refs/changes/85/672185/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/672185/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/672185/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/672185/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/672185/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/672185/5"}}},"commit":{"parents":[{"commit":"c9e61b80ad9f762d4acb07466b7fa548addcbd5c","subject":"Add release notes to changes to wflanguage execution object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c9e61b80ad9f762d4acb07466b7fa548addcbd5c"}]}],"author":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2019-07-22 22:15:00.000000000","tz":-420},"committer":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2019-07-25 23:30:22.000000000","tz":-420},"subject":"disable triggering version discovery in test_generator unit tests","message":"disable triggering version discovery in test_generator unit tests\n\nFixed unit.actions.openstack.test_generator.GeneratorTest.test_generator\nfailure by avoid triggering version discovery as it needs to talk to\nKeystone server. Talking to a live Keystone server is not needed for\nthis particular unit test.\n\nNOTE: this patch is only applicable to stable/rocky and with the latest\nopenstacksdk 0.17.3 release.\n\nNOTE: this patch also includes change I9ab552dd4bf3826f358d73a4bddaf1885b1be130\nfrom https://review.opendev.org/#/c/672543/ as both are needed in order\nto pass Jenkins.\n\nChange-Id: Ib2c72e89612c9b2b4bccd37cb797804ea40e0b1f\nCloses-Bug: 1837468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f2a0b569f767e0f7cea844ed01accccf810ca618"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f2a0b569f767e0f7cea844ed01accccf810ca618"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
