)]}'
{"id":"openstack%2Fmistral~672240","triplet_id":"openstack%2Fmistral~stable%2Fstein~Ie5159ba52b6a682e00027ce6bf6d3c5c3f4f1eb9","project":"openstack/mistral","branch":"stable/stein","topic":"remove_deprecated_nova_cmds-stable/stein","hashtags":[],"change_id":"Ie5159ba52b6a682e00027ce6bf6d3c5c3f4f1eb9","subject":"Remove deprecated nova commands","status":"MERGED","created":"2019-07-23 09:17:45.000000000","updated":"2019-07-23 20:33:54.000000000","submitted":"2019-07-23 20:33:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672240-1563914034917-e6e18664","meta_rev_id":"e64d8a8b821ed4cab232f21d99e9ab739e21034a","_number":672240,"virtual_id_number":672240,"owner":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"date":"2019-07-23 14:23:37.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":2,"date":"2019-07-23 20:33:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-07-23 19:35:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-07-23 19:35:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 10:58:11.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-07-23 11:00:32.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-07-23 19:35:55.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-07-23 20:33:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d1be77e673ada1436918936fa080e079bef2a8e","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-23 09:17:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"05cfd2819b54f44714a22222a61702b2aa9fff8b","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-23 10:28:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1599d1694c512fbe7401dd4d07b88a3b4ddcaf1d","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-07-23 10:58:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d60ee9ebaa3a82640d019bff44b37bc963f382f1","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-07-23 11:00:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"219701cc5cd9c38928edb32be3527628bc1e15ea","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-23 11:07:47.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"08be7d7448685abaa8937338590e7153b7e54711","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-23 11:30:17.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"fb79bbfe1e93a2dbade46235b474fce790a5e66b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 14:02:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/40/672240/4/check/openstack-tox-pep8/8f97c17/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/40/672240/4/check/openstack-tox-py27/a66a13a/ : SUCCESS in 8m 15s\n- openstack-tox-py35 http://logs.openstack.org/40/672240/4/check/openstack-tox-py35/904a9d5/ : SUCCESS in 8m 28s\n- openstack-tox-py36 http://logs.openstack.org/40/672240/4/check/openstack-tox-py36/d2b05af/ : SUCCESS in 7m 02s\n- openstack-tox-py37 http://logs.openstack.org/40/672240/4/check/openstack-tox-py37/4f1968d/ : SUCCESS in 9m 24s\n- openstack-tox-docs http://logs.openstack.org/40/672240/4/check/openstack-tox-docs/c540ca7/html/ : SUCCESS in 4m 49s\n- build-openstack-releasenotes http://logs.openstack.org/40/672240/4/check/build-openstack-releasenotes/fa52d5c/html/ : SUCCESS in 3m 52s\n- openstack-tox-cover http://logs.openstack.org/40/672240/4/check/openstack-tox-cover/b7419c7/cover/ : SUCCESS in 10m 25s (non-voting)\n- mistral-devstack http://logs.openstack.org/40/672240/4/check/mistral-devstack/b677ae2/ : SUCCESS in 40m 03s\n- mistral-devstack-non-apache http://logs.openstack.org/40/672240/4/check/mistral-devstack-non-apache/69a3123/ : FAILURE in 18m 50s\n- mistral-devstack-kombu http://logs.openstack.org/40/672240/4/check/mistral-devstack-kombu/b64661e/ : SUCCESS in 43m 58s\n- mistral-tox-unit-mysql http://logs.openstack.org/40/672240/4/check/mistral-tox-unit-mysql/d1b496f/ : SUCCESS in 26m 43s\n- mistral-tox-unit-postgresql http://logs.openstack.org/40/672240/4/check/mistral-tox-unit-postgresql/da58b6b/ : SUCCESS in 25m 51s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/40/672240/4/check/tripleo-ci-centos-7-scenario003-standalone/8879b9e/ : SUCCESS in 1h 23m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/40/672240/4/check/tripleo-ci-centos-7-containers-multinode/ec0d9cb/ : SUCCESS in 2h 16m 57s (non-voting)\n- mistral-rally-task http://logs.openstack.org/40/672240/4/check/mistral-rally-task/8590eb8/ : SUCCESS in 1h 10m 33s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/40/672240/4/check/openstack-tox-lower-constraints/ca0ff97/ : SUCCESS in 11m 10s","accounts_in_message":[],"_revision_number":4},{"id":"fcbd8546e9a72f3384ea79a630f9c2d076ab275e","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-23 14:23:37.000000000","message":"Patch Set 4:\n\nrecheck (tls error while downloading stuff)","accounts_in_message":[],"_revision_number":4},{"id":"e98e7d216d53089ee36e36a164b05db0860b2bae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:13:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/40/672240/4/check/openstack-tox-pep8/b1b6961/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/40/672240/4/check/openstack-tox-py27/996b32e/ : SUCCESS in 7m 37s\n- openstack-tox-py35 http://logs.openstack.org/40/672240/4/check/openstack-tox-py35/842cb70/ : SUCCESS in 10m 08s\n- openstack-tox-py36 http://logs.openstack.org/40/672240/4/check/openstack-tox-py36/1356f96/ : SUCCESS in 9m 34s\n- openstack-tox-py37 http://logs.openstack.org/40/672240/4/check/openstack-tox-py37/24bee67/ : SUCCESS in 9m 19s\n- openstack-tox-docs http://logs.openstack.org/40/672240/4/check/openstack-tox-docs/ce2132b/html/ : SUCCESS in 5m 45s\n- build-openstack-releasenotes http://logs.openstack.org/40/672240/4/check/build-openstack-releasenotes/33bb32f/html/ : SUCCESS in 4m 32s\n- openstack-tox-cover http://logs.openstack.org/40/672240/4/check/openstack-tox-cover/d64072d/cover/ : SUCCESS in 8m 27s (non-voting)\n- mistral-devstack http://logs.openstack.org/40/672240/4/check/mistral-devstack/0766940/ : SUCCESS in 45m 02s\n- mistral-devstack-non-apache http://logs.openstack.org/40/672240/4/check/mistral-devstack-non-apache/863fda6/ : SUCCESS in 45m 48s\n- mistral-devstack-kombu http://logs.openstack.org/40/672240/4/check/mistral-devstack-kombu/8a61a99/ : SUCCESS in 49m 18s\n- mistral-tox-unit-mysql http://logs.openstack.org/40/672240/4/check/mistral-tox-unit-mysql/00e6b46/ : SUCCESS in 30m 53s\n- mistral-tox-unit-postgresql http://logs.openstack.org/40/672240/4/check/mistral-tox-unit-postgresql/b13b489/ : SUCCESS in 26m 30s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/40/672240/4/check/tripleo-ci-centos-7-scenario003-standalone/dd79c3f/ : SUCCESS in 1h 28m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/40/672240/4/check/tripleo-ci-centos-7-containers-multinode/0fd9e93/ : SUCCESS in 2h 38m 15s (non-voting)\n- mistral-rally-task http://logs.openstack.org/40/672240/4/check/mistral-rally-task/41ef4e8/ : SUCCESS in 1h 08m 28s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/40/672240/4/check/openstack-tox-lower-constraints/097906a/ : SUCCESS in 10m 11s","accounts_in_message":[],"_revision_number":4},{"id":"5f328633cbcca9819c7bc725cf8ca19ed4a0e020","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-23 19:35:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"dcfb761ea6f8b6c267b67718e8025f32ee292ff1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 19:36:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7d002dddedc9af609c77725a0495eb2911f6b638","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 20:33:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/40/672240/4/gate/openstack-tox-pep8/5f8affd/ : SUCCESS in 5m 56s\n- openstack-tox-py27 http://logs.openstack.org/40/672240/4/gate/openstack-tox-py27/ec59f7d/ : SUCCESS in 9m 35s\n- openstack-tox-py35 http://logs.openstack.org/40/672240/4/gate/openstack-tox-py35/2b4bc9b/ : SUCCESS in 11m 02s\n- openstack-tox-py36 http://logs.openstack.org/40/672240/4/gate/openstack-tox-py36/2f7061e/ : SUCCESS in 8m 08s\n- openstack-tox-py37 http://logs.openstack.org/40/672240/4/gate/openstack-tox-py37/64fae2e/ : SUCCESS in 11m 15s\n- openstack-tox-docs http://logs.openstack.org/40/672240/4/gate/openstack-tox-docs/d32908c/html/ : SUCCESS in 5m 38s\n- build-openstack-releasenotes http://logs.openstack.org/40/672240/4/gate/build-openstack-releasenotes/1e4f4f4/html/ : SUCCESS in 5m 19s\n- mistral-devstack http://logs.openstack.org/40/672240/4/gate/mistral-devstack/3ffa2bd/ : SUCCESS in 50m 56s\n- mistral-devstack-non-apache http://logs.openstack.org/40/672240/4/gate/mistral-devstack-non-apache/d80ea76/ : SUCCESS in 49m 42s\n- mistral-tox-unit-mysql http://logs.openstack.org/40/672240/4/gate/mistral-tox-unit-mysql/1e66460/ : SUCCESS in 29m 52s\n- mistral-tox-unit-postgresql http://logs.openstack.org/40/672240/4/gate/mistral-tox-unit-postgresql/deffd84/ : SUCCESS in 26m 07s\n- mistral-devstack-kombu http://logs.openstack.org/40/672240/4/gate/mistral-devstack-kombu/8ba81b9/ : SUCCESS in 45m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/672240/4/gate/openstack-tox-lower-constraints/06e1f14/ : SUCCESS in 11m 11s","accounts_in_message":[],"_revision_number":4},{"id":"b45fa4cc59b7085109996aad05be6481be91aa0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 20:33:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"83ce5574253bf233b430562c19a415236d004495","revisions":{"3812e9c897950dc0b97af5650ab0ebb992c63d32":{"kind":"REWORK","_number":1,"created":"2019-07-23 09:17:45.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/40/672240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/40/672240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/672240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/672240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/672240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/40/672240/1"}}},"commit":{"parents":[{"commit":"6ff82c316e9b2464eedd5a8df215101918ebff9b","subject":"Add release notes to changes to wflanguage execution object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6ff82c316e9b2464eedd5a8df215101918ebff9b"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-05-16 14:25:37.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-23 09:17:45.000000000","tz":0},"subject":"Remove deprecated nova commands","message":"Remove deprecated nova commands\n\nnovaclient.cert was removed 18 months ago\nhttps://github.com/openstack/python-novaclient/commit/4bc4078fcb4f2a47269882cc83e1d637db1f37c2\nthis commit removes those calls.\n\nChange-Id: Ie5159ba52b6a682e00027ce6bf6d3c5c3f4f1eb9\n(cherry picked from commit dbc4c9115009a0a989b18d0f7887537ef35241c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3812e9c897950dc0b97af5650ab0ebb992c63d32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3812e9c897950dc0b97af5650ab0ebb992c63d32"}]},"branch":"refs/heads/stable/stein"},"e79f470c76591f799f4c8ae0f0adcdda9f17c46e":{"kind":"REWORK","_number":2,"created":"2019-07-23 10:28:35.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/40/672240/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/40/672240/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/672240/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/672240/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/672240/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/40/672240/2"}}},"commit":{"parents":[{"commit":"6ff82c316e9b2464eedd5a8df215101918ebff9b","subject":"Add release notes to changes to wflanguage execution object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6ff82c316e9b2464eedd5a8df215101918ebff9b"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-05-16 14:25:37.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-23 10:28:30.000000000","tz":0},"subject":"Remove deprecated nova commands","message":"Remove deprecated nova commands\n\nnovaclient.cert was removed 18 months ago\nhttps://github.com/openstack/python-novaclient/commit/4bc4078fcb4f2a47269882cc83e1d637db1f37c2\nthis commit removes those calls.\n\nChange-Id: Ie5159ba52b6a682e00027ce6bf6d3c5c3f4f1eb9\n(cherry picked from commit dbc4c9115009a0a989b18d0f7887537ef35241c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e79f470c76591f799f4c8ae0f0adcdda9f17c46e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e79f470c76591f799f4c8ae0f0adcdda9f17c46e"}]},"branch":"refs/heads/stable/stein"},"346c2a0137b612493e05d167e6037a4bdfe79960":{"kind":"REWORK","_number":3,"created":"2019-07-23 11:07:47.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/40/672240/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/40/672240/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/672240/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/672240/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/672240/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/40/672240/3"}}},"commit":{"parents":[{"commit":"6ff82c316e9b2464eedd5a8df215101918ebff9b","subject":"Add release notes to changes to wflanguage execution object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6ff82c316e9b2464eedd5a8df215101918ebff9b"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-05-16 14:25:37.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-23 11:07:39.000000000","tz":0},"subject":"Remove deprecated nova commands","message":"Remove deprecated nova commands\n\nnovaclient.cert was removed 18 months ago\nhttps://github.com/openstack/python-novaclient/commit/4bc4078fcb4f2a47269882cc83e1d637db1f37c2\nthis commit removes those calls.\n\nChange-Id: Ie5159ba52b6a682e00027ce6bf6d3c5c3f4f1eb9\n(cherry picked from commit dbc4c9115009a0a989b18d0f7887537ef35241c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/346c2a0137b612493e05d167e6037a4bdfe79960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/346c2a0137b612493e05d167e6037a4bdfe79960"}]},"branch":"refs/heads/stable/stein"},"83ce5574253bf233b430562c19a415236d004495":{"kind":"REWORK","_number":4,"created":"2019-07-23 11:30:17.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/40/672240/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/40/672240/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/672240/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/672240/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/672240/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/40/672240/4"}}},"commit":{"parents":[{"commit":"6ff82c316e9b2464eedd5a8df215101918ebff9b","subject":"Add release notes to changes to wflanguage execution object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6ff82c316e9b2464eedd5a8df215101918ebff9b"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-05-16 14:25:37.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-23 11:30:13.000000000","tz":0},"subject":"Remove deprecated nova commands","message":"Remove deprecated nova commands\n\nnovaclient.cert was removed 18 months ago\nhttps://github.com/openstack/python-novaclient/commit/4bc4078fcb4f2a47269882cc83e1d637db1f37c2\nthis commit removes those calls.\n\nChange-Id: Ie5159ba52b6a682e00027ce6bf6d3c5c3f4f1eb9\n(cherry picked from commit dbc4c9115009a0a989b18d0f7887537ef35241c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/83ce5574253bf233b430562c19a415236d004495"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/83ce5574253bf233b430562c19a415236d004495"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
