)]}'
{"id":"openstack%2Fmistral~672245","triplet_id":"openstack%2Fmistral~master~I4fa15994a7359a5f90a0a4671d47b19fe928cf33","project":"openstack/mistral","branch":"master","topic":"bug/1837532","hashtags":[],"change_id":"I4fa15994a7359a5f90a0a4671d47b19fe928cf33","subject":"Retry a DB transaction on \"Too many connections\" error","status":"MERGED","created":"2019-07-23 10:00:42.000000000","updated":"2020-02-27 02:09:34.000000000","submitted":"2019-07-24 17:19:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672245-1563988782346-82382829","meta_rev_id":"44b335685fc5737d5083fba1fca6049ff1a76cf7","_number":672245,"virtual_id_number":672245,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":15353,"name":"Noa Koffman","email":"noa.koffman@nokia.com","username":"noakoffman"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-07-24 17:19:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-24 12:31:52.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":2,"date":"2019-07-24 09:57:45.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":15353,"name":"Noa Koffman","email":"noa.koffman@nokia.com","username":"noakoffman"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":15353,"name":"Noa Koffman","email":"noa.koffman@nokia.com","username":"noakoffman"},{"value":1,"date":"2019-07-24 10:39:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15353,"name":"Noa Koffman","email":"noa.koffman@nokia.com","username":"noakoffman"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 10:12:51.000000000","updated_by":{"_account_id":15353,"name":"Noa Koffman","email":"noa.koffman@nokia.com","username":"noakoffman"},"reviewer":{"_account_id":15353,"name":"Noa Koffman","email":"noa.koffman@nokia.com","username":"noakoffman"},"state":"REVIEWER"},{"updated":"2019-07-24 09:57:45.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-07-24 12:31:52.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-07-24 17:19:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c1b3c64044a4618393c60c7428de266b7167dffa","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-23 10:00:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54a1f10adcf7bfb5ffb6956b15b72985a0cb61b1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-23 10:11:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a4daa30435f8a5db2776aaa91a5331db83a1bef2","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-23 11:06:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"22ddc579dfc4143e8f4981e9b5b51be27a2063b6","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-23 11:31:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"954f82518699260a18e415d56245d9588dfe925c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-23 13:22:29.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"1b80b836068d21a48dd7c7e00c8f82e9a898b91d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:59:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/45/672245/5/check/openstack-tox-pep8/f37e7bb/ : SUCCESS in 7m 10s\n- openstack-tox-py27 http://logs.openstack.org/45/672245/5/check/openstack-tox-py27/c38099f/ : SUCCESS in 10m 55s\n- openstack-tox-py36 http://logs.openstack.org/45/672245/5/check/openstack-tox-py36/a9afa35/ : SUCCESS in 10m 29s\n- openstack-tox-py37 http://logs.openstack.org/45/672245/5/check/openstack-tox-py37/d2f7f66/ : SUCCESS in 9m 26s\n- openstack-tox-docs http://logs.openstack.org/45/672245/5/check/openstack-tox-docs/121adcd/html/ : SUCCESS in 4m 52s\n- requirements-check http://logs.openstack.org/45/672245/5/check/requirements-check/81ed524/ : SUCCESS in 2m 33s\n- openstack-tox-cover http://logs.openstack.org/45/672245/5/check/openstack-tox-cover/5048c50/cover/ : SUCCESS in 9m 04s (non-voting)\n- mistral-devstack http://logs.openstack.org/45/672245/5/check/mistral-devstack/16df435/ : SUCCESS in 46m 11s\n- mistral-devstack-non-apache http://logs.openstack.org/45/672245/5/check/mistral-devstack-non-apache/7bd4b41/ : FAILURE in 45m 01s\n- mistral-devstack-kombu http://logs.openstack.org/45/672245/5/check/mistral-devstack-kombu/586c02b/ : SUCCESS in 45m 41s\n- mistral-tox-unit-mysql http://logs.openstack.org/45/672245/5/check/mistral-tox-unit-mysql/b28cefc/ : SUCCESS in 27m 21s\n- mistral-tox-unit-postgresql http://logs.openstack.org/45/672245/5/check/mistral-tox-unit-postgresql/7f703de/ : SUCCESS in 27m 13s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/45/672245/5/check/tripleo-ci-centos-7-scenario003-standalone/5bacdef/ : SUCCESS in 1h 04m 53s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/45/672245/5/check/tripleo-ci-centos-7-containers-multinode/b675151/ : SUCCESS in 2h 20m 10s\n- mistral-rally-task http://logs.openstack.org/45/672245/5/check/mistral-rally-task/021e07f/ : SUCCESS in 59m 58s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/45/672245/5/check/openstack-tox-lower-constraints/e62fd23/ : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":5},{"id":"ec4afc313c221863129debf9acb2ffa3a053812e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-23 19:34:36.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c353b8356e60fc536f21292bfa1508af1b5bc727","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 22:11:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/45/672245/5/check/openstack-tox-pep8/f6f38dc/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/45/672245/5/check/openstack-tox-py27/1895743/ : SUCCESS in 10m 01s\n- openstack-tox-py36 http://logs.openstack.org/45/672245/5/check/openstack-tox-py36/3d02d30/ : SUCCESS in 8m 39s\n- openstack-tox-py37 http://logs.openstack.org/45/672245/5/check/openstack-tox-py37/54b93a0/ : SUCCESS in 9m 03s\n- openstack-tox-docs http://logs.openstack.org/45/672245/5/check/openstack-tox-docs/c82fa30/html/ : SUCCESS in 3m 45s\n- requirements-check http://logs.openstack.org/45/672245/5/check/requirements-check/20275d3/ : SUCCESS in 2m 47s\n- openstack-tox-cover http://logs.openstack.org/45/672245/5/check/openstack-tox-cover/c3168d2/cover/ : SUCCESS in 10m 06s (non-voting)\n- mistral-devstack http://logs.openstack.org/45/672245/5/check/mistral-devstack/d3f20a1/ : SUCCESS in 41m 21s\n- mistral-devstack-non-apache http://logs.openstack.org/45/672245/5/check/mistral-devstack-non-apache/21a0c2d/ : SUCCESS in 47m 16s\n- mistral-devstack-kombu http://logs.openstack.org/45/672245/5/check/mistral-devstack-kombu/e30c73d/ : SUCCESS in 49m 42s\n- mistral-tox-unit-mysql http://logs.openstack.org/45/672245/5/check/mistral-tox-unit-mysql/87d059d/ : SUCCESS in 33m 17s\n- mistral-tox-unit-postgresql http://logs.openstack.org/45/672245/5/check/mistral-tox-unit-postgresql/168b709/ : SUCCESS in 26m 22s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/45/672245/5/check/tripleo-ci-centos-7-scenario003-standalone/d60cbd0/ : SUCCESS in 1h 18m 52s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/45/672245/5/check/tripleo-ci-centos-7-containers-multinode/3813d82/ : SUCCESS in 2h 29m 44s\n- mistral-rally-task http://logs.openstack.org/45/672245/5/check/mistral-rally-task/be96893/ : SUCCESS in 56m 50s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/45/672245/5/check/openstack-tox-lower-constraints/12e1e05/ : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":5},{"id":"33438ab169845dd52f1a5cfb0b1d10e4c49f26c4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-24 01:02:37.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 06181d5e5a738f1b4134f4cd31c962ec7baead95","accounts_in_message":[],"_revision_number":5},{"id":"b7b4e2aea35da8f6e8e3df15c162910f33331d7e","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-07-24 09:57:45.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"da085a245f0e898b459a2a5827e5f9745696e38e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-24 10:39:28.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"3451c7fac9b42817a627031c0a22befac0cfae4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 10:39:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e11036c7651339d1c791568e4d77267184b79e5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 12:19:43.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/45/672245/5/gate/openstack-tox-pep8/2ca951a/ : SUCCESS in 5m 12s\n- openstack-tox-py27 http://logs.openstack.org/45/672245/5/gate/openstack-tox-py27/afdb929/ : SUCCESS in 9m 34s\n- openstack-tox-py36 http://logs.openstack.org/45/672245/5/gate/openstack-tox-py36/b3566e9/ : SUCCESS in 9m 34s\n- openstack-tox-py37 http://logs.openstack.org/45/672245/5/gate/openstack-tox-py37/55aff61/ : SUCCESS in 12m 00s\n- openstack-tox-docs http://logs.openstack.org/45/672245/5/gate/openstack-tox-docs/e746c43/html/ : SUCCESS in 5m 33s\n- requirements-check http://logs.openstack.org/45/672245/5/gate/requirements-check/bea67ea/ : SUCCESS in 3m 18s\n- mistral-devstack http://logs.openstack.org/45/672245/5/gate/mistral-devstack/aa18522/ : SUCCESS in 45m 11s\n- mistral-devstack-non-apache http://logs.openstack.org/45/672245/5/gate/mistral-devstack-non-apache/2aeeed4/ : SUCCESS in 48m 10s\n- mistral-tox-unit-mysql http://logs.openstack.org/45/672245/5/gate/mistral-tox-unit-mysql/0447791/ : SUCCESS in 27m 43s\n- mistral-tox-unit-postgresql http://logs.openstack.org/45/672245/5/gate/mistral-tox-unit-postgresql/8ffbe40/ : SUCCESS in 27m 22s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/45/672245/5/gate/tripleo-ci-centos-7-scenario003-standalone/cf5b466/ : FAILURE in 1h 12m 38s\n- mistral-devstack-kombu http://logs.openstack.org/45/672245/5/gate/mistral-devstack-kombu/ec2e97c/ : SUCCESS in 52m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/672245/5/gate/openstack-tox-lower-constraints/e1a2002/ : SUCCESS in 9m 59s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/45/672245/5/gate/tripleo-ci-centos-7-undercloud-containers/635ccf6/ : SUCCESS in 1h 38m 44s","accounts_in_message":[],"_revision_number":5},{"id":"6a851e047611c5b0d9e2638f35386b1d5697b407","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-24 12:31:52.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c7e52cfc8ce298a284a7bd2da9e5b7762d6205b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 15:31:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/45/672245/5/check/openstack-tox-pep8/1da2a52/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/45/672245/5/check/openstack-tox-py27/88d644d/ : SUCCESS in 8m 46s\n- openstack-tox-py36 http://logs.openstack.org/45/672245/5/check/openstack-tox-py36/83f8bf7/ : SUCCESS in 8m 51s\n- openstack-tox-py37 http://logs.openstack.org/45/672245/5/check/openstack-tox-py37/8640285/ : SUCCESS in 9m 44s\n- openstack-tox-docs http://logs.openstack.org/45/672245/5/check/openstack-tox-docs/0df9e6e/html/ : SUCCESS in 6m 10s\n- requirements-check http://logs.openstack.org/45/672245/5/check/requirements-check/53dedda/ : SUCCESS in 4m 06s\n- openstack-tox-cover http://logs.openstack.org/45/672245/5/check/openstack-tox-cover/a707e32/cover/ : SUCCESS in 10m 17s (non-voting)\n- mistral-devstack http://logs.openstack.org/45/672245/5/check/mistral-devstack/131ba6c/ : SUCCESS in 47m 36s\n- mistral-devstack-non-apache http://logs.openstack.org/45/672245/5/check/mistral-devstack-non-apache/b622977/ : SUCCESS in 51m 47s\n- mistral-devstack-kombu http://logs.openstack.org/45/672245/5/check/mistral-devstack-kombu/766d306/ : SUCCESS in 43m 46s\n- mistral-tox-unit-mysql http://logs.openstack.org/45/672245/5/check/mistral-tox-unit-mysql/1dc859b/ : SUCCESS in 28m 24s\n- mistral-tox-unit-postgresql http://logs.openstack.org/45/672245/5/check/mistral-tox-unit-postgresql/7493ef9/ : SUCCESS in 27m 25s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/45/672245/5/check/tripleo-ci-centos-7-scenario003-standalone/02f7474/ : SUCCESS in 1h 15m 31s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/45/672245/5/check/tripleo-ci-centos-7-containers-multinode/0664657/ : SUCCESS in 2h 58m 15s\n- mistral-rally-task http://logs.openstack.org/45/672245/5/check/mistral-rally-task/9aa3a2b/ : SUCCESS in 1h 00m 21s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/45/672245/5/check/openstack-tox-lower-constraints/48b3482/ : SUCCESS in 13m 19s","accounts_in_message":[],"_revision_number":5},{"id":"5e142467e2d58a2be3ddefe4af3176187577b92c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 15:31:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3c6a2335f0e5c368b66dc7ce5042b83cb8f8566d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 17:19:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"d7dcf0ce8a542635c10ef7b0a8d50fb69d97f632","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 17:19:42.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/45/672245/5/gate/openstack-tox-pep8/87d0c96/ : SUCCESS in 6m 53s\n- openstack-tox-py27 http://logs.openstack.org/45/672245/5/gate/openstack-tox-py27/20144e8/ : SUCCESS in 7m 38s\n- openstack-tox-py36 http://logs.openstack.org/45/672245/5/gate/openstack-tox-py36/5d7bc6a/ : SUCCESS in 10m 59s\n- openstack-tox-py37 http://logs.openstack.org/45/672245/5/gate/openstack-tox-py37/21597fb/ : SUCCESS in 8m 35s\n- openstack-tox-docs http://logs.openstack.org/45/672245/5/gate/openstack-tox-docs/e6b0d11/html/ : SUCCESS in 5m 56s\n- requirements-check http://logs.openstack.org/45/672245/5/gate/requirements-check/87696aa/ : SUCCESS in 3m 11s\n- mistral-devstack http://logs.openstack.org/45/672245/5/gate/mistral-devstack/1a4fec8/ : SUCCESS in 53m 28s\n- mistral-devstack-non-apache http://logs.openstack.org/45/672245/5/gate/mistral-devstack-non-apache/4c13a1a/ : SUCCESS in 45m 27s\n- mistral-tox-unit-mysql http://logs.openstack.org/45/672245/5/gate/mistral-tox-unit-mysql/42fb831/ : SUCCESS in 27m 33s\n- mistral-tox-unit-postgresql http://logs.openstack.org/45/672245/5/gate/mistral-tox-unit-postgresql/7954300/ : SUCCESS in 27m 55s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/45/672245/5/gate/tripleo-ci-centos-7-scenario003-standalone/177c5a4/ : SUCCESS in 1h 17m 13s\n- mistral-devstack-kombu http://logs.openstack.org/45/672245/5/gate/mistral-devstack-kombu/8dce31d/ : SUCCESS in 45m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/672245/5/gate/openstack-tox-lower-constraints/4ef42ea/ : SUCCESS in 10m 55s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/45/672245/5/gate/tripleo-ci-centos-7-undercloud-containers/da7f4c8/ : SUCCESS in 1h 33m 17s","accounts_in_message":[],"_revision_number":5},{"id":"baebf53c33a8eef5401a6d6c5379a3d01426b27a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-27 02:09:34.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit f8fb0922c928ff68d3cc004706e3c930d58dbffb","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d","revisions":{"d60d1d0345ee0751d0bf1d5037cbc7cf43071ad5":{"kind":"REWORK","_number":1,"created":"2019-07-23 10:00:42.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/45/672245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/672245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/672245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/672245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/672245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/672245/1"}}},"commit":{"parents":[{"commit":"f2e8d614d932c615d068d3cb9ad01ac99350fcc5","subject":"Merge \"remove mistral-tempest-plugin from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f2e8d614d932c615d068d3cb9ad01ac99350fcc5"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-07-23 09:59:57.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-07-23 09:59:57.000000000","tz":420},"subject":"WIP: Retry a DB transaction on \"Too many connections\" error","message":"WIP: Retry a DB transaction on \"Too many connections\" error\n\nChange-Id: I4fa15994a7359a5f90a0a4671d47b19fe928cf33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d60d1d0345ee0751d0bf1d5037cbc7cf43071ad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d60d1d0345ee0751d0bf1d5037cbc7cf43071ad5"}]},"branch":"refs/heads/master"},"1519562de27f80d94af491060b222fd08af5eff8":{"kind":"REWORK","_number":2,"created":"2019-07-23 10:11:41.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/45/672245/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/672245/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/672245/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/672245/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/672245/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/672245/2"}}},"commit":{"parents":[{"commit":"f2e8d614d932c615d068d3cb9ad01ac99350fcc5","subject":"Merge \"remove mistral-tempest-plugin from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f2e8d614d932c615d068d3cb9ad01ac99350fcc5"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-07-23 09:59:57.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-07-23 10:11:18.000000000","tz":420},"subject":"Retry a DB transaction on \"Too many connections\" error","message":"Retry a DB transaction on \"Too many connections\" error\n\n* Writing a unit test is very problematic but the fix has been\n  tested manually.\n\nCloses-Bug: #1837532\n\nChange-Id: I4fa15994a7359a5f90a0a4671d47b19fe928cf33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1519562de27f80d94af491060b222fd08af5eff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1519562de27f80d94af491060b222fd08af5eff8"}]},"branch":"refs/heads/master"},"6dec3bc3216eaf29c3abe62ba90c85c59aee96d6":{"kind":"REWORK","_number":3,"created":"2019-07-23 11:06:22.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/45/672245/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/672245/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/672245/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/672245/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/672245/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/672245/3"}}},"commit":{"parents":[{"commit":"f2e8d614d932c615d068d3cb9ad01ac99350fcc5","subject":"Merge \"remove mistral-tempest-plugin from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f2e8d614d932c615d068d3cb9ad01ac99350fcc5"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-07-23 09:59:57.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-07-23 11:05:57.000000000","tz":420},"subject":"Retry a DB transaction on \"Too many connections\" error","message":"Retry a DB transaction on \"Too many connections\" error\n\n* Writing a unit test is very problematic but the fix has been\n  tested manually.\n\nCloses-Bug: #1837532\n\nChange-Id: I4fa15994a7359a5f90a0a4671d47b19fe928cf33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6dec3bc3216eaf29c3abe62ba90c85c59aee96d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6dec3bc3216eaf29c3abe62ba90c85c59aee96d6"}]},"branch":"refs/heads/master"},"41f8369f29250f0c4d89b72e80a6395066193d61":{"kind":"REWORK","_number":4,"created":"2019-07-23 11:31:56.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/45/672245/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/672245/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/672245/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/672245/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/672245/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/672245/4"}}},"commit":{"parents":[{"commit":"f2e8d614d932c615d068d3cb9ad01ac99350fcc5","subject":"Merge \"remove mistral-tempest-plugin from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f2e8d614d932c615d068d3cb9ad01ac99350fcc5"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-07-23 09:59:57.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-07-23 11:31:34.000000000","tz":420},"subject":"Retry a DB transaction on \"Too many connections\" error","message":"Retry a DB transaction on \"Too many connections\" error\n\n* Writing a unit test is very problematic but the fix has been\n  tested manually.\n\nCloses-Bug: #1837532\n\nChange-Id: I4fa15994a7359a5f90a0a4671d47b19fe928cf33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/41f8369f29250f0c4d89b72e80a6395066193d61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/41f8369f29250f0c4d89b72e80a6395066193d61"}]},"branch":"refs/heads/master"},"4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d":{"kind":"REWORK","_number":5,"created":"2019-07-23 13:22:29.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/45/672245/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/672245/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/672245/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/672245/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/672245/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/672245/5"}}},"commit":{"parents":[{"commit":"f2e8d614d932c615d068d3cb9ad01ac99350fcc5","subject":"Merge \"remove mistral-tempest-plugin from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f2e8d614d932c615d068d3cb9ad01ac99350fcc5"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-07-23 09:59:57.000000000","tz":420},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-23 13:22:25.000000000","tz":0},"subject":"Retry a DB transaction on \"Too many connections\" error","message":"Retry a DB transaction on \"Too many connections\" error\n\n* Writing a unit test is very problematic but the fix has been\n  tested manually.\n\nCloses-Bug: #1837532\n\nChange-Id: I4fa15994a7359a5f90a0a4671d47b19fe928cf33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
