)]}'
{"id":"openstack%2Fmistral~672543","triplet_id":"openstack%2Fmistral~stable%2Frocky~I9ab552dd4bf3826f358d73a4bddaf1885b1be130","project":"openstack/mistral","branch":"stable/rocky","topic":"bindep-fallback","hashtags":[],"change_id":"I9ab552dd4bf3826f358d73a4bddaf1885b1be130","subject":"Add bindep.txt file for binary dependencies used in unit tests","status":"ABANDONED","created":"2019-07-24 14:51:31.000000000","updated":"2019-07-29 14:23:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"77d8caad347f99b3c41cbddbd599b7672add559e","_number":672543,"virtual_id_number":672543,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-07-25 17:43:20.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":-1,"date":"2019-07-24 17:33:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 17:33:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-25 17:43:20.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"}],"messages":[{"id":"37dd3916be63a82f21dce42b8c840449743f3af2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-24 14:51:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fe382b2f59a238a341f134242c74354ad3bf72b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 17:33:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/43/672543/1/check/openstack-tox-pep8/e1aaca5/ : SUCCESS in 5m 48s\n- openstack-tox-py27 http://logs.openstack.org/43/672543/1/check/openstack-tox-py27/a068ad3/ : FAILURE in 9m 49s\n- openstack-tox-py35 http://logs.openstack.org/43/672543/1/check/openstack-tox-py35/60d6b2b/ : FAILURE in 12m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/43/672543/1/check/build-openstack-sphinx-docs/b203b58/html/ : SUCCESS in 6m 37s\n- build-openstack-releasenotes http://logs.openstack.org/43/672543/1/check/build-openstack-releasenotes/502b95f/html/ : SUCCESS in 5m 03s\n- openstack-tox-cover http://logs.openstack.org/43/672543/1/check/openstack-tox-cover/c24f598/ : FAILURE in 8m 52s (non-voting)\n- mistral-devstack http://logs.openstack.org/43/672543/1/check/mistral-devstack/78131d9/ : SUCCESS in 37m 25s\n- mistral-devstack-non-apache http://logs.openstack.org/43/672543/1/check/mistral-devstack-non-apache/5e517b0/ : SUCCESS in 56m 11s\n- mistral-devstack-kombu http://logs.openstack.org/43/672543/1/check/mistral-devstack-kombu/5c90640/ : SUCCESS in 1h 02m 40s\n- mistral-tox-unit-mysql http://logs.openstack.org/43/672543/1/check/mistral-tox-unit-mysql/d0fabc8/ : FAILURE in 31m 14s\n- mistral-tox-unit-postgresql http://logs.openstack.org/43/672543/1/check/mistral-tox-unit-postgresql/95ed7fe/ : FAILURE in 26m 29s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/43/672543/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/16c087a/ : SUCCESS in 2h 29m 10s (non-voting)\n- mistral-rally-task http://logs.openstack.org/43/672543/1/check/mistral-rally-task/21d6977/ : SUCCESS in 1h 12m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/43/672543/1/check/openstack-tox-lower-constraints/29added/ : SUCCESS in 11m 56s","accounts_in_message":[],"_revision_number":1},{"id":"fad58b2a00a925b0a715b0a6e7c3129b599224c3","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-07-25 17:43:20.000000000","message":"Patch Set 1:\n\nthis patch had been cherry-picked into https://review.opendev.org/#/c/672185/ as both are needed in order to pass jenkins","accounts_in_message":[],"_revision_number":1},{"id":"64a211fbdb8ed372d319ad1065593bbb2a8f370e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-29 14:23:25.000000000","message":"Abandoned\n\nCovered by https://review.opendev.org/#/c/672185/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"282ce55a4a61b92ca8f1996bc52183ce38d95fcb","revisions":{"282ce55a4a61b92ca8f1996bc52183ce38d95fcb":{"kind":"REWORK","_number":1,"created":"2019-07-24 14:51:31.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/43/672543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/43/672543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/43/672543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/43/672543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/43/672543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/43/672543/1"}}},"commit":{"parents":[{"commit":"c9e61b80ad9f762d4acb07466b7fa548addcbd5c","subject":"Add release notes to changes to wflanguage execution object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c9e61b80ad9f762d4acb07466b7fa548addcbd5c"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-06-27 09:54:49.000000000","tz":180},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-07-24 14:51:29.000000000","tz":120},"subject":"Add bindep.txt file for binary dependencies used in unit tests","message":"Add bindep.txt file for binary dependencies used in unit tests\n\nNote(elod.illes): this patch is cherry-picked from master, however the\nfile itself exists on stable/stein, too, squashed into a different\npatch.\n\nChange-Id: I9ab552dd4bf3826f358d73a4bddaf1885b1be130\n(cherry picked from commit 26f6e19299e43c443db4002fe1d3a98084adf4fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/282ce55a4a61b92ca8f1996bc52183ce38d95fcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/282ce55a4a61b92ca8f1996bc52183ce38d95fcb"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
