)]}'
{"id":"openstack%2Fmistral~673298","triplet_id":"openstack%2Fmistral~master~I69c32f1a6fc34cd3c0c843e847cb190c2fa53010","project":"openstack/mistral","branch":"master","topic":"bug/1721571","hashtags":[],"change_id":"I69c32f1a6fc34cd3c0c843e847cb190c2fa53010","subject":"Improve error message when sync actions timeout","status":"MERGED","created":"2019-07-29 14:22:02.000000000","updated":"2019-08-06 09:29:20.000000000","submitted":"2019-08-06 09:29:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673298-1565083760859-3342d8ab","meta_rev_id":"361add3cb6a1c9844c110714b19c427d3a80314b","_number":673298,"virtual_id_number":673298,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-08-06 09:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":2,"date":"2019-08-06 07:38:41.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2019-08-06 07:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-08-06 04:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":1,"date":"2019-08-06 07:38:41.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2019-08-06 07:38:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 11:07:10.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-07-31 09:27:24.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-08-06 04:14:06.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-08-06 07:38:41.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-08-06 07:38:56.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-08-06 09:29:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5823c551d4a8ca85127e1175761f567111f9cead","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-29 14:22:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71b15bbd81fdd23ba058ec129bbcf55230af5199","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 17:00:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/98/673298/1/check/openstack-tox-pep8/12a54d6/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/98/673298/1/check/openstack-tox-py27/87fbcb2/ : SUCCESS in 10m 13s\n- openstack-tox-py36 http://logs.openstack.org/98/673298/1/check/openstack-tox-py36/109284d/ : SUCCESS in 8m 56s\n- openstack-tox-py37 http://logs.openstack.org/98/673298/1/check/openstack-tox-py37/28fb9da/ : SUCCESS in 10m 09s\n- openstack-tox-docs http://logs.openstack.org/98/673298/1/check/openstack-tox-docs/825b696/html/ : SUCCESS in 6m 51s\n- openstack-tox-cover http://logs.openstack.org/98/673298/1/check/openstack-tox-cover/71625ef/cover/ : SUCCESS in 11m 33s (non-voting)\n- mistral-devstack http://logs.openstack.org/98/673298/1/check/mistral-devstack/eff40d5/ : SUCCESS in 52m 58s\n- mistral-devstack-non-apache http://logs.openstack.org/98/673298/1/check/mistral-devstack-non-apache/0af7bc4/ : SUCCESS in 46m 58s\n- mistral-devstack-kombu http://logs.openstack.org/98/673298/1/check/mistral-devstack-kombu/adbd5a4/ : SUCCESS in 48m 40s\n- mistral-tox-unit-mysql http://logs.openstack.org/98/673298/1/check/mistral-tox-unit-mysql/4d1dab6/ : FAILURE in 26m 34s\n- mistral-tox-unit-postgresql http://logs.openstack.org/98/673298/1/check/mistral-tox-unit-postgresql/6216a55/ : SUCCESS in 27m 03s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/98/673298/1/check/tripleo-ci-centos-7-scenario003-standalone/b9c73f5/ : SUCCESS in 1h 29m 02s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/98/673298/1/check/tripleo-ci-centos-7-containers-multinode/5181153/ : SUCCESS in 2h 26m 13s\n- mistral-rally-task http://logs.openstack.org/98/673298/1/check/mistral-rally-task/773082a/ : SUCCESS in 1h 00m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/98/673298/1/check/openstack-tox-lower-constraints/274e582/ : SUCCESS in 14m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3d3625498d4ed16b0a506b369b655076616c9eb1","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-29 17:10:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"92730345be659d46fb96d231f25b2c7a4fa84b25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 19:44:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/98/673298/1/check/openstack-tox-pep8/3550c40/ : SUCCESS in 6m 13s\n- openstack-tox-py27 http://logs.openstack.org/98/673298/1/check/openstack-tox-py27/2c49487/ : SUCCESS in 9m 52s\n- openstack-tox-py36 http://logs.openstack.org/98/673298/1/check/openstack-tox-py36/98d55c2/ : SUCCESS in 11m 38s\n- openstack-tox-py37 http://logs.openstack.org/98/673298/1/check/openstack-tox-py37/071fccd/ : SUCCESS in 11m 50s\n- openstack-tox-docs http://logs.openstack.org/98/673298/1/check/openstack-tox-docs/7445509/html/ : SUCCESS in 5m 24s\n- openstack-tox-cover http://logs.openstack.org/98/673298/1/check/openstack-tox-cover/54306f0/cover/ : SUCCESS in 9m 45s (non-voting)\n- mistral-devstack http://logs.openstack.org/98/673298/1/check/mistral-devstack/5b2c792/ : SUCCESS in 1h 00m 17s\n- mistral-devstack-non-apache http://logs.openstack.org/98/673298/1/check/mistral-devstack-non-apache/31090b4/ : SUCCESS in 47m 15s\n- mistral-devstack-kombu http://logs.openstack.org/98/673298/1/check/mistral-devstack-kombu/95da899/ : SUCCESS in 41m 39s\n- mistral-tox-unit-mysql http://logs.openstack.org/98/673298/1/check/mistral-tox-unit-mysql/1bf64f8/ : FAILURE in 26m 52s\n- mistral-tox-unit-postgresql http://logs.openstack.org/98/673298/1/check/mistral-tox-unit-postgresql/fdba393/ : SUCCESS in 27m 57s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/98/673298/1/check/tripleo-ci-centos-7-scenario003-standalone/303a037/ : SUCCESS in 1h 17m 57s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/98/673298/1/check/tripleo-ci-centos-7-containers-multinode/34f3d5b/ : SUCCESS in 2h 28m 18s\n- mistral-rally-task http://logs.openstack.org/98/673298/1/check/mistral-rally-task/20c20f2/ : SUCCESS in 1h 01m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/98/673298/1/check/openstack-tox-lower-constraints/962e812/ : SUCCESS in 12m 52s","accounts_in_message":[],"_revision_number":1},{"id":"07f25002d63dcabc029982a04b3ac4af009503e8","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-30 04:03:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e8cad6d81b3027c68a6e6e6cf5d27c7de5f9d71e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 06:29:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/98/673298/1/check/openstack-tox-pep8/294810b/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/98/673298/1/check/openstack-tox-py27/6a6fdaa/ : SUCCESS in 7m 25s\n- openstack-tox-py36 http://logs.openstack.org/98/673298/1/check/openstack-tox-py36/7075e5f/ : SUCCESS in 7m 33s\n- openstack-tox-py37 http://logs.openstack.org/98/673298/1/check/openstack-tox-py37/ad39b0e/ : SUCCESS in 11m 16s\n- openstack-tox-docs http://logs.openstack.org/98/673298/1/check/openstack-tox-docs/83e872a/html/ : SUCCESS in 5m 31s\n- openstack-tox-cover http://logs.openstack.org/98/673298/1/check/openstack-tox-cover/73b1c0d/cover/ : SUCCESS in 7m 52s (non-voting)\n- mistral-devstack http://logs.openstack.org/98/673298/1/check/mistral-devstack/828a49c/ : SUCCESS in 43m 04s\n- mistral-devstack-non-apache http://logs.openstack.org/98/673298/1/check/mistral-devstack-non-apache/83a3333/ : SUCCESS in 43m 36s\n- mistral-devstack-kombu http://logs.openstack.org/98/673298/1/check/mistral-devstack-kombu/b37dd46/ : SUCCESS in 45m 26s\n- mistral-tox-unit-mysql http://logs.openstack.org/98/673298/1/check/mistral-tox-unit-mysql/63426de/ : SUCCESS in 26m 34s\n- mistral-tox-unit-postgresql http://logs.openstack.org/98/673298/1/check/mistral-tox-unit-postgresql/5b5c047/ : SUCCESS in 28m 54s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/98/673298/1/check/tripleo-ci-centos-7-scenario003-standalone/b7bd330/ : SUCCESS in 1h 18m 53s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/98/673298/1/check/tripleo-ci-centos-7-containers-multinode/39d7aed/ : SUCCESS in 2h 20m 47s\n- mistral-rally-task http://logs.openstack.org/98/673298/1/check/mistral-rally-task/7304605/ : SUCCESS in 52m 39s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/98/673298/1/check/openstack-tox-lower-constraints/ab9155a/ : SUCCESS in 10m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7989e0b8d798414f5dab2f22730e3e24cc5d86ef","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-07-30 11:07:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b2557a0f5c175f31327b226e26a998a4ab09685c","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-07-31 09:27:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6364d2fe7c3d78c621c1c1386d1286472b159df5","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-01 07:51:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f6e945e9a79135755518aae40bf871c702769135","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-01 12:59:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dc8fd05ec579417dee0f06a8c2de690729d3455e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-01 13:01:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"828f1ec1c7fe0e25224d31774096d338009be280","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-01 13:33:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"178d46feb460fc77e8162f05bb5bb87b8921baba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 17:54:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/98/673298/4/check/openstack-tox-pep8/bdde02e/ : SUCCESS in 6m 36s\n- openstack-tox-py27 https://logs.opendev.org/98/673298/4/check/openstack-tox-py27/19c54aa/ : SUCCESS in 10m 39s\n- openstack-tox-py36 https://logs.opendev.org/98/673298/4/check/openstack-tox-py36/c13e476/ : SUCCESS in 10m 19s\n- openstack-tox-py37 https://logs.opendev.org/98/673298/4/check/openstack-tox-py37/f60e1ae/ : SUCCESS in 10m 11s\n- openstack-tox-docs https://logs.opendev.org/98/673298/4/check/openstack-tox-docs/30b43ad/html/ : SUCCESS in 6m 46s\n- openstack-tox-cover https://logs.opendev.org/98/673298/4/check/openstack-tox-cover/3ee6f35/cover/ : SUCCESS in 10m 24s (non-voting)\n- mistral-devstack https://logs.opendev.org/98/673298/4/check/mistral-devstack/16b2425/ : SUCCESS in 53m 28s\n- mistral-devstack-non-apache https://logs.opendev.org/98/673298/4/check/mistral-devstack-non-apache/06d1d39/ : SUCCESS in 55m 03s\n- mistral-devstack-kombu https://logs.opendev.org/98/673298/4/check/mistral-devstack-kombu/a8506f6/ : SUCCESS in 52m 43s\n- mistral-tox-unit-mysql https://logs.opendev.org/98/673298/4/check/mistral-tox-unit-mysql/07f327d/ : SUCCESS in 29m 00s\n- mistral-tox-unit-postgresql https://logs.opendev.org/98/673298/4/check/mistral-tox-unit-postgresql/54eb7ff/ : SUCCESS in 26m 47s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/98/673298/4/check/tripleo-ci-centos-7-scenario003-standalone/4e4f493/ : SUCCESS in 1h 16m 43s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/98/673298/4/check/tripleo-ci-centos-7-containers-multinode/de7d0e9/ : SUCCESS in 2h 36m 55s\n- mistral-rally-task https://logs.opendev.org/98/673298/4/check/mistral-rally-task/cab2dca/ : SUCCESS in 1h 08m 32s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/673298/4/check/openstack-tox-lower-constraints/e52b26d/ : SUCCESS in 11m 18s","accounts_in_message":[],"_revision_number":4},{"id":"7a5a51db8ac9f25d520990f956539668796e270c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-02 08:56:32.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nJust one thing related to mocking that I\u0027d prefer to fix.","accounts_in_message":[],"_revision_number":4},{"id":"58be65840cc24da8c6153c5fac6593b80955710b","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-04 11:16:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d4f368fd0716f8139c60cbd47f99caa3ca3d75ec","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-04 11:24:07.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4731a785985476b4e552f8ab225ada3d243623fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 13:53:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/98/673298/5/check/openstack-tox-pep8/f3d9a3c/ : SUCCESS in 3m 59s\n- openstack-tox-py27 https://logs.opendev.org/98/673298/5/check/openstack-tox-py27/2c20d0a/ : SUCCESS in 8m 11s\n- openstack-tox-py36 https://logs.opendev.org/98/673298/5/check/openstack-tox-py36/b701e89/ : SUCCESS in 8m 14s\n- openstack-tox-py37 https://logs.opendev.org/98/673298/5/check/openstack-tox-py37/4746332/ : SUCCESS in 8m 15s\n- openstack-tox-docs https://logs.opendev.org/98/673298/5/check/openstack-tox-docs/17ebd27/html/ : SUCCESS in 4m 21s\n- openstack-tox-cover https://logs.opendev.org/98/673298/5/check/openstack-tox-cover/aee05ed/cover/ : SUCCESS in 7m 58s (non-voting)\n- mistral-devstack https://logs.opendev.org/98/673298/5/check/mistral-devstack/61a67f4/ : SUCCESS in 48m 02s\n- mistral-devstack-non-apache https://logs.opendev.org/98/673298/5/check/mistral-devstack-non-apache/2835975/ : SUCCESS in 37m 03s\n- mistral-devstack-kombu https://logs.opendev.org/98/673298/5/check/mistral-devstack-kombu/c5f77a4/ : SUCCESS in 35m 17s\n- mistral-tox-unit-mysql https://logs.opendev.org/98/673298/5/check/mistral-tox-unit-mysql/a4dc544/ : SUCCESS in 29m 31s\n- mistral-tox-unit-postgresql https://logs.opendev.org/98/673298/5/check/mistral-tox-unit-postgresql/040f8d0/ : SUCCESS in 25m 32s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/98/673298/5/check/tripleo-ci-centos-7-scenario003-standalone/da4d7d0/ : SUCCESS in 1h 23m 06s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/98/673298/5/check/tripleo-ci-centos-7-containers-multinode/872a159/ : SUCCESS in 2h 35m 40s\n- mistral-rally-task https://logs.opendev.org/98/673298/5/check/mistral-rally-task/842cff1/ : SUCCESS in 53m 53s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/673298/5/check/openstack-tox-lower-constraints/dee0c92/ : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":5},{"id":"fcf9b38f0755426fbbde40656743257d0e003156","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-05 10:15:44.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"6be511364277a45402fcceedfef19402e41933e6","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-05 10:26:28.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e06b236df48903ad52355c54ec2d70165fa423e3","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-05 11:15:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0066644ed46b6f677642e438fa195e8e809f330f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 13:49:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/98/673298/6/check/openstack-tox-pep8/3ca1f90/ : SUCCESS in 4m 51s\n- openstack-tox-py27 https://logs.opendev.org/98/673298/6/check/openstack-tox-py27/11756d7/ : SUCCESS in 12m 28s\n- openstack-tox-py36 https://logs.opendev.org/98/673298/6/check/openstack-tox-py36/d2e2844/ : SUCCESS in 12m 56s\n- openstack-tox-py37 https://logs.opendev.org/98/673298/6/check/openstack-tox-py37/93d1f9b/ : SUCCESS in 15m 10s\n- openstack-tox-docs https://logs.opendev.org/98/673298/6/check/openstack-tox-docs/26e1090/html/ : SUCCESS in 4m 22s\n- openstack-tox-cover https://logs.opendev.org/98/673298/6/check/openstack-tox-cover/5001182/cover/ : SUCCESS in 13m 13s (non-voting)\n- mistral-devstack https://logs.opendev.org/98/673298/6/check/mistral-devstack/cf1ecf0/ : SUCCESS in 50m 51s\n- mistral-devstack-non-apache https://logs.opendev.org/98/673298/6/check/mistral-devstack-non-apache/543b4b4/ : SUCCESS in 50m 16s\n- mistral-devstack-kombu https://logs.opendev.org/98/673298/6/check/mistral-devstack-kombu/175a7df/ : SUCCESS in 46m 42s\n- mistral-tox-unit-mysql https://logs.opendev.org/98/673298/6/check/mistral-tox-unit-mysql/7978913/ : SUCCESS in 29m 02s\n- mistral-tox-unit-postgresql https://logs.opendev.org/98/673298/6/check/mistral-tox-unit-postgresql/1b55057/ : SUCCESS in 31m 13s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/98/673298/6/check/tripleo-ci-centos-7-scenario003-standalone/4fb4fae/ : SUCCESS in 1h 30m 32s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/98/673298/6/check/tripleo-ci-centos-7-containers-multinode/e23700a/ : SUCCESS in 2h 27m 57s\n- mistral-rally-task https://logs.opendev.org/98/673298/6/check/mistral-rally-task/890b0ff/ : SUCCESS in 1h 06m 32s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/98/673298/6/check/openstack-tox-lower-constraints/dca8ab2/ : SUCCESS in 14m 42s","accounts_in_message":[],"_revision_number":6},{"id":"b701286ab0d99322385a854b6724102d7da3c12c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-06 04:14:06.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"6e335bc0d6b2a499dda4b6cb14b2c2bec54e99b7","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-06 07:38:41.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d59d1c7f3ab72c60a3ef0ea3fa1ffee81023b915","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 07:38:54.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"13bbd8926db00827e34c95c4b335c3776de2cc96","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-08-06 07:38:56.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0baf5e9a877aa742bb2cdf6622296df5e38e0db3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 09:29:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"21b104ffc3a62e1512c1f2ba72b8fbb9923a4c84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 09:29:20.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/98/673298/6/gate/openstack-tox-pep8/08c288f/ : SUCCESS in 6m 05s\n- openstack-tox-py27 https://logs.opendev.org/98/673298/6/gate/openstack-tox-py27/8de6502/ : SUCCESS in 9m 26s\n- openstack-tox-py36 https://logs.opendev.org/98/673298/6/gate/openstack-tox-py36/b5b1b85/ : SUCCESS in 9m 14s\n- openstack-tox-py37 https://logs.opendev.org/98/673298/6/gate/openstack-tox-py37/d9f061a/ : SUCCESS in 11m 10s\n- openstack-tox-docs https://logs.opendev.org/98/673298/6/gate/openstack-tox-docs/675511a/html/ : SUCCESS in 5m 35s\n- mistral-devstack https://logs.opendev.org/98/673298/6/gate/mistral-devstack/c924e72/ : SUCCESS in 54m 54s\n- mistral-devstack-non-apache https://logs.opendev.org/98/673298/6/gate/mistral-devstack-non-apache/222692e/ : SUCCESS in 44m 54s\n- mistral-tox-unit-mysql https://logs.opendev.org/98/673298/6/gate/mistral-tox-unit-mysql/951e5a7/ : SUCCESS in 27m 43s\n- mistral-tox-unit-postgresql https://logs.opendev.org/98/673298/6/gate/mistral-tox-unit-postgresql/47f992f/ : SUCCESS in 27m 10s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/98/673298/6/gate/tripleo-ci-centos-7-scenario003-standalone/b9cde70/ : SUCCESS in 1h 14m 32s\n- mistral-devstack-kombu https://logs.opendev.org/98/673298/6/gate/mistral-devstack-kombu/f78ee9d/ : SUCCESS in 47m 02s\n- openstack-tox-lower-constraints https://logs.opendev.org/98/673298/6/gate/openstack-tox-lower-constraints/6cc5120/ : SUCCESS in 12m 07s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/98/673298/6/gate/tripleo-ci-centos-7-undercloud-containers/d6961b9/ : SUCCESS in 1h 47m 42s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"babcd0d6831b4596ce3f392990aecefc60ae7193","revisions":{"4c3742b0425ab96515601ae4bd8279d4d7d1e05f":{"kind":"REWORK","_number":1,"created":"2019-07-29 14:22:02.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/98/673298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/98/673298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/98/673298/1"}}},"commit":{"parents":[{"commit":"4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d","subject":"Retry a DB transaction on \"Too many connections\" error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-29 14:20:08.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-29 14:20:08.000000000","tz":180},"subject":"Improve error message when sync actions timeout","message":"Improve error message when sync actions timeout\n\nWrap the exception in case of a MessageTimeout to\nsomething more clear\n\nChange-Id: I69c32f1a6fc34cd3c0c843e847cb190c2fa53010\nCloses-bug: #1721571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4c3742b0425ab96515601ae4bd8279d4d7d1e05f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4c3742b0425ab96515601ae4bd8279d4d7d1e05f"}]},"branch":"refs/heads/master"},"8267718affa6b8f55dbc70a0c0c6f519bf43eb12":{"kind":"REWORK","_number":2,"created":"2019-08-01 12:59:30.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/98/673298/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/98/673298/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/98/673298/2"}}},"commit":{"parents":[{"commit":"4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d","subject":"Retry a DB transaction on \"Too many connections\" error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-29 14:20:08.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-01 12:58:23.000000000","tz":180},"subject":"Improve error message when sync actions timeout","message":"Improve error message when sync actions timeout\n\nMake the exception in case of a MessageTimeout to\nbe something more clear\n\nChange-Id: I69c32f1a6fc34cd3c0c843e847cb190c2fa53010\nCloses-bug: #1721571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8267718affa6b8f55dbc70a0c0c6f519bf43eb12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8267718affa6b8f55dbc70a0c0c6f519bf43eb12"}]},"branch":"refs/heads/master"},"6a7154e04a3c1e674458066195695f24d0ee8aef":{"kind":"REWORK","_number":3,"created":"2019-08-01 13:01:40.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/98/673298/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/98/673298/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/98/673298/3"}}},"commit":{"parents":[{"commit":"4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d","subject":"Retry a DB transaction on \"Too many connections\" error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-29 14:20:08.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-01 13:01:25.000000000","tz":180},"subject":"Improve error message when sync actions timeout","message":"Improve error message when sync actions timeout\n\nMake the exception in case of a MessageTimeout to\nbe something more clear\n\nChange-Id: I69c32f1a6fc34cd3c0c843e847cb190c2fa53010\nCloses-bug: #1721571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a7154e04a3c1e674458066195695f24d0ee8aef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a7154e04a3c1e674458066195695f24d0ee8aef"}]},"branch":"refs/heads/master"},"51b8b07ab1cbca3d631da8d7ba16e03a495dc8ae":{"kind":"REWORK","_number":4,"created":"2019-08-01 13:33:52.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/98/673298/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/98/673298/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/98/673298/4"}}},"commit":{"parents":[{"commit":"4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d","subject":"Retry a DB transaction on \"Too many connections\" error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-29 14:20:08.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-01 13:33:36.000000000","tz":180},"subject":"Improve error message when sync actions timeout","message":"Improve error message when sync actions timeout\n\nMake the exception in case of a MessageTimeout to\nbe something more clear\n\nChange-Id: I69c32f1a6fc34cd3c0c843e847cb190c2fa53010\nCloses-bug: #1721571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/51b8b07ab1cbca3d631da8d7ba16e03a495dc8ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/51b8b07ab1cbca3d631da8d7ba16e03a495dc8ae"}]},"branch":"refs/heads/master"},"f1803fa39cb6bc7eb972a105bd94b4e7cdef4d28":{"kind":"REWORK","_number":5,"created":"2019-08-04 11:16:24.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/98/673298/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/98/673298/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/98/673298/5"}}},"commit":{"parents":[{"commit":"4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d","subject":"Retry a DB transaction on \"Too many connections\" error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-29 14:20:08.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-04 11:15:57.000000000","tz":180},"subject":"Improve error message when sync actions timeout","message":"Improve error message when sync actions timeout\n\nMake the exception in case of a MessageTimeout to\nbe something more clear\n\nChange-Id: I69c32f1a6fc34cd3c0c843e847cb190c2fa53010\nCloses-bug: #1721571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f1803fa39cb6bc7eb972a105bd94b4e7cdef4d28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f1803fa39cb6bc7eb972a105bd94b4e7cdef4d28"}]},"branch":"refs/heads/master"},"babcd0d6831b4596ce3f392990aecefc60ae7193":{"kind":"REWORK","_number":6,"created":"2019-08-05 11:15:25.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/98/673298/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/98/673298/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/673298/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/98/673298/6"}}},"commit":{"parents":[{"commit":"4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d","subject":"Retry a DB transaction on \"Too many connections\" error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-29 14:20:08.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-05 11:15:06.000000000","tz":180},"subject":"Improve error message when sync actions timeout","message":"Improve error message when sync actions timeout\n\nMake the exception in case of a MessageTimeout to\nbe something more clear\n\nChange-Id: I69c32f1a6fc34cd3c0c843e847cb190c2fa53010\nCloses-bug: #1721571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/babcd0d6831b4596ce3f392990aecefc60ae7193"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/babcd0d6831b4596ce3f392990aecefc60ae7193"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
