)]}'
{"id":"openstack%2Fmistral~673544","triplet_id":"openstack%2Fmistral~master~I85b9cc76ad7c10a9ffac7ae61232460381723b8f","project":"openstack/mistral","branch":"master","topic":"eyalb/test_names","hashtags":[],"change_id":"I85b9cc76ad7c10a9ffac7ae61232460381723b8f","subject":"Rename the test class so it will be PyCharm friendly","status":"MERGED","created":"2019-07-30 14:44:01.000000000","updated":"2019-07-31 05:48:17.000000000","submitted":"2019-07-31 05:48:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673544-1564552097995-8da850ff","meta_rev_id":"c14a398f1816046a9cce4df6382b6962adb8e350","_number":673544,"virtual_id_number":673544,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-07-31 05:48:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-07-31 04:16:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-30 22:00:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-07-31 04:16:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 22:00:07.000000000","updated_by":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"reviewer":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"state":"REVIEWER"},{"updated":"2019-07-31 04:16:16.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-07-31 05:48:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41dfc16a360aeea40b1b3a4647d468a87bedf104","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-30 14:44:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5753223a9d50025ee0f9ca948713190dbe01aca4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 17:16:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/44/673544/1/check/openstack-tox-pep8/8be8d5f/ : SUCCESS in 5m 24s\n- openstack-tox-py27 http://logs.openstack.org/44/673544/1/check/openstack-tox-py27/e5dcdeb/ : SUCCESS in 9m 59s\n- openstack-tox-py36 http://logs.openstack.org/44/673544/1/check/openstack-tox-py36/cfbb9f6/ : SUCCESS in 12m 29s\n- openstack-tox-py37 http://logs.openstack.org/44/673544/1/check/openstack-tox-py37/03b2c94/ : SUCCESS in 9m 25s\n- openstack-tox-docs http://logs.openstack.org/44/673544/1/check/openstack-tox-docs/d1d6ae3/html/ : SUCCESS in 4m 39s\n- openstack-tox-cover http://logs.openstack.org/44/673544/1/check/openstack-tox-cover/31a42b7/cover/ : SUCCESS in 9m 07s (non-voting)\n- mistral-devstack http://logs.openstack.org/44/673544/1/check/mistral-devstack/a5e5772/ : SUCCESS in 46m 13s\n- mistral-devstack-non-apache http://logs.openstack.org/44/673544/1/check/mistral-devstack-non-apache/147a480/ : SUCCESS in 48m 07s\n- mistral-devstack-kombu http://logs.openstack.org/44/673544/1/check/mistral-devstack-kombu/721c4fc/ : SUCCESS in 47m 01s\n- mistral-tox-unit-mysql http://logs.openstack.org/44/673544/1/check/mistral-tox-unit-mysql/de115d8/ : SUCCESS in 29m 00s\n- mistral-tox-unit-postgresql http://logs.openstack.org/44/673544/1/check/mistral-tox-unit-postgresql/d11e667/ : SUCCESS in 27m 38s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/44/673544/1/check/tripleo-ci-centos-7-scenario003-standalone/84b6535/ : SUCCESS in 1h 06m 07s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/44/673544/1/check/tripleo-ci-centos-7-containers-multinode/7a0d670/ : SUCCESS in 2h 24m 41s\n- mistral-rally-task http://logs.openstack.org/44/673544/1/check/mistral-rally-task/ae7d99d/ : SUCCESS in 1h 08m 52s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/44/673544/1/check/openstack-tox-lower-constraints/4a73ae2/ : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":1},{"id":"cd460b33670c7c38d0e9e0001b883572b6907e95","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-07-30 22:00:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6f44ed5176cef8862854708f1e55b88575a25792","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-31 04:16:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"06c9b980e65214eb068fa5f4004f1a287c3a62db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 04:16:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d624b1b2d0449b0f2f36e36d91efd09058a2b97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 05:48:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"88b63d241cd476e89c20936a8f436fba2ceb15cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 05:48:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/44/673544/1/gate/openstack-tox-pep8/ab8cd91/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/44/673544/1/gate/openstack-tox-py27/374606c/ : SUCCESS in 8m 11s\n- openstack-tox-py36 http://logs.openstack.org/44/673544/1/gate/openstack-tox-py36/78494dc/ : SUCCESS in 8m 43s\n- openstack-tox-py37 http://logs.openstack.org/44/673544/1/gate/openstack-tox-py37/3dca504/ : SUCCESS in 9m 17s\n- openstack-tox-docs http://logs.openstack.org/44/673544/1/gate/openstack-tox-docs/bf9fb63/html/ : SUCCESS in 4m 57s\n- mistral-devstack http://logs.openstack.org/44/673544/1/gate/mistral-devstack/7519b9b/ : SUCCESS in 52m 16s\n- mistral-devstack-non-apache http://logs.openstack.org/44/673544/1/gate/mistral-devstack-non-apache/a33b9de/ : SUCCESS in 47m 04s\n- mistral-tox-unit-mysql http://logs.openstack.org/44/673544/1/gate/mistral-tox-unit-mysql/b91fa49/ : SUCCESS in 27m 37s\n- mistral-tox-unit-postgresql http://logs.openstack.org/44/673544/1/gate/mistral-tox-unit-postgresql/e24d902/ : SUCCESS in 26m 18s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/44/673544/1/gate/tripleo-ci-centos-7-scenario003-standalone/46eb828/ : SUCCESS in 1h 14m 57s\n- mistral-devstack-kombu http://logs.openstack.org/44/673544/1/gate/mistral-devstack-kombu/e4d033e/ : SUCCESS in 45m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/44/673544/1/gate/openstack-tox-lower-constraints/b1dffd0/ : SUCCESS in 9m 08s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/44/673544/1/gate/tripleo-ci-centos-7-undercloud-containers/79624b8/ : SUCCESS in 1h 31m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"77118d3bb90fd7a27f4eafd1da569a73e7b6acb1","revisions":{"77118d3bb90fd7a27f4eafd1da569a73e7b6acb1":{"kind":"REWORK","_number":1,"created":"2019-07-30 14:44:01.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/44/673544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/673544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/673544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/673544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/673544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/673544/1"}}},"commit":{"parents":[{"commit":"c77eeb2d14193a815fa93d6db4f04ac872696ade","subject":"Merge \"Bulk delete of stored delayed calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c77eeb2d14193a815fa93d6db4f04ac872696ade"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-30 14:43:45.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-07-30 14:43:45.000000000","tz":180},"subject":"Rename the test class so it will be PyCharm friendly","message":"Rename the test class so it will be PyCharm friendly\n\nIn order to run from PyCharm test classed need to end or start\nwith the word \u0027Test\u0027\n\nChange-Id: I85b9cc76ad7c10a9ffac7ae61232460381723b8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/77118d3bb90fd7a27f4eafd1da569a73e7b6acb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/77118d3bb90fd7a27f4eafd1da569a73e7b6acb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
