)]}'
{"id":"openstack%2Fmistral~673595","triplet_id":"openstack%2Fmistral~master~Id5f629085e0ecfb4fc75ff82c1ed0283f3d2c490","project":"openstack/mistral","branch":"master","topic":"retire-api-site","hashtags":[],"change_id":"Id5f629085e0ecfb4fc75ff82c1ed0283f3d2c490","subject":"Update api-ref location","status":"MERGED","created":"2019-07-30 17:50:44.000000000","updated":"2019-07-31 08:04:26.000000000","submitted":"2019-07-31 08:04:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673595-1564560266121-f6dc5ada","meta_rev_id":"f5aa1c9417078f2573cd5982152a68177c1ee770","_number":673595,"virtual_id_number":673595,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-31 05:53:50.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-07-31 08:04:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-07-31 04:17:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-07-31 04:17:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-31 04:17:06.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-07-31 08:04:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e67f9969b2beb2d3e5ea11cdb02c3a9d2d32707","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-30 17:50:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61f3529395d7edef4f146220063604750c760af3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 19:02:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/95/673595/1/check/openstack-tox-pep8/0898409/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/95/673595/1/check/openstack-tox-py27/b6ef529/ : SUCCESS in 11m 28s\n- openstack-tox-py36 http://logs.openstack.org/95/673595/1/check/openstack-tox-py36/ae61c48/ : SUCCESS in 9m 25s\n- openstack-tox-py37 http://logs.openstack.org/95/673595/1/check/openstack-tox-py37/871035b/ : SUCCESS in 9m 28s\n- openstack-tox-docs http://logs.openstack.org/95/673595/1/check/openstack-tox-docs/0873fd5/html/ : SUCCESS in 4m 18s\n- build-openstack-releasenotes http://logs.openstack.org/95/673595/1/check/build-openstack-releasenotes/d327dd3/html/ : SUCCESS in 4m 29s\n- openstack-tox-cover http://logs.openstack.org/95/673595/1/check/openstack-tox-cover/048b476/cover/ : SUCCESS in 9m 32s (non-voting)\n- mistral-devstack http://logs.openstack.org/95/673595/1/check/mistral-devstack/0c57263/ : SUCCESS in 49m 33s\n- mistral-devstack-non-apache http://logs.openstack.org/95/673595/1/check/mistral-devstack-non-apache/8ec78da/ : SUCCESS in 45m 51s\n- mistral-devstack-kombu http://logs.openstack.org/95/673595/1/check/mistral-devstack-kombu/591123c/ : SUCCESS in 48m 21s\n- mistral-tox-unit-mysql http://logs.openstack.org/95/673595/1/check/mistral-tox-unit-mysql/8d15c57/ : SUCCESS in 30m 20s\n- mistral-tox-unit-postgresql http://logs.openstack.org/95/673595/1/check/mistral-tox-unit-postgresql/94cace5/ : SUCCESS in 27m 03s\n- mistral-rally-task http://logs.openstack.org/95/673595/1/check/mistral-rally-task/7179d2f/ : SUCCESS in 1h 04m 17s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/95/673595/1/check/openstack-tox-lower-constraints/52dd44c/ : SUCCESS in 9m 57s","accounts_in_message":[],"_revision_number":1},{"id":"33eaabf5e9dd4aa5034c55b6d6fa4291b54c0697","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-31 04:17:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f54b1758783cb9fd9507ad31074f9e1f0af3e9c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 04:17:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"042757d28375eaf6d15fdf76d4fb28ab6829bc4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 05:48:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/95/673595/1/gate/openstack-tox-pep8/67966ad/ : SUCCESS in 5m 36s\n- openstack-tox-py27 http://logs.openstack.org/95/673595/1/gate/openstack-tox-py27/bb3391f/ : SUCCESS in 7m 20s\n- openstack-tox-py36 http://logs.openstack.org/95/673595/1/gate/openstack-tox-py36/5ac4cb9/ : SUCCESS in 7m 34s\n- openstack-tox-py37 http://logs.openstack.org/95/673595/1/gate/openstack-tox-py37/dbb4994/ : SUCCESS in 8m 34s\n- openstack-tox-docs http://logs.openstack.org/95/673595/1/gate/openstack-tox-docs/c600a13/html/ : SUCCESS in 3m 52s\n- build-openstack-releasenotes http://logs.openstack.org/95/673595/1/gate/build-openstack-releasenotes/5b97b96/html/ : SUCCESS in 3m 26s\n- mistral-devstack http://logs.openstack.org/95/673595/1/gate/mistral-devstack/e44edcf/ : SUCCESS in 44m 35s\n- mistral-devstack-non-apache http://logs.openstack.org/95/673595/1/gate/mistral-devstack-non-apache/84bad3b/ : SUCCESS in 44m 45s\n- mistral-tox-unit-mysql http://logs.openstack.org/95/673595/1/gate/mistral-tox-unit-mysql/2a32ea2/ : FAILURE in 26m 44s\n- mistral-tox-unit-postgresql http://logs.openstack.org/95/673595/1/gate/mistral-tox-unit-postgresql/98cef52/ : SUCCESS in 26m 41s\n- mistral-devstack-kombu http://logs.openstack.org/95/673595/1/gate/mistral-devstack-kombu/966cfc6/ : SUCCESS in 42m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/673595/1/gate/openstack-tox-lower-constraints/6faae7c/ : SUCCESS in 8m 35s","accounts_in_message":[],"_revision_number":1},{"id":"64f49238f19ab3b5c5dc28c6ed6ed5432073306b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-31 05:53:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4900ccafa6ab4b3efabce1a0309ad9ea16d886e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 07:12:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/95/673595/1/check/openstack-tox-pep8/03067eb/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/95/673595/1/check/openstack-tox-py27/4ba3951/ : SUCCESS in 8m 08s\n- openstack-tox-py36 http://logs.openstack.org/95/673595/1/check/openstack-tox-py36/19feb9e/ : SUCCESS in 9m 51s\n- openstack-tox-py37 http://logs.openstack.org/95/673595/1/check/openstack-tox-py37/f5915cf/ : SUCCESS in 12m 07s\n- openstack-tox-docs http://logs.openstack.org/95/673595/1/check/openstack-tox-docs/410f565/html/ : SUCCESS in 6m 10s\n- build-openstack-releasenotes http://logs.openstack.org/95/673595/1/check/build-openstack-releasenotes/d493b50/html/ : SUCCESS in 4m 58s\n- openstack-tox-cover http://logs.openstack.org/95/673595/1/check/openstack-tox-cover/ace5bd8/cover/ : SUCCESS in 11m 32s (non-voting)\n- mistral-devstack http://logs.openstack.org/95/673595/1/check/mistral-devstack/55e26df/ : SUCCESS in 56m 27s\n- mistral-devstack-non-apache http://logs.openstack.org/95/673595/1/check/mistral-devstack-non-apache/3f0b7a5/ : SUCCESS in 56m 39s\n- mistral-devstack-kombu http://logs.openstack.org/95/673595/1/check/mistral-devstack-kombu/7908bf2/ : SUCCESS in 57m 02s\n- mistral-tox-unit-mysql http://logs.openstack.org/95/673595/1/check/mistral-tox-unit-mysql/91bc8ac/ : SUCCESS in 29m 59s\n- mistral-tox-unit-postgresql http://logs.openstack.org/95/673595/1/check/mistral-tox-unit-postgresql/22779f9/ : SUCCESS in 28m 31s\n- mistral-rally-task http://logs.openstack.org/95/673595/1/check/mistral-rally-task/8c0fb96/ : SUCCESS in 1h 11m 40s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/95/673595/1/check/openstack-tox-lower-constraints/e6f4af9/ : SUCCESS in 12m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b45c76a6cd585beb82d43fa77ebd562d8fc59000","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 07:12:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0bf88888d7c7a5e41c22155c457ccab8938d12a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 08:04:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/95/673595/1/gate/openstack-tox-pep8/43083bd/ : SUCCESS in 7m 31s\n- openstack-tox-py27 http://logs.openstack.org/95/673595/1/gate/openstack-tox-py27/f81ebfb/ : SUCCESS in 10m 10s\n- openstack-tox-py36 http://logs.openstack.org/95/673595/1/gate/openstack-tox-py36/6f2f5df/ : SUCCESS in 10m 16s\n- openstack-tox-py37 http://logs.openstack.org/95/673595/1/gate/openstack-tox-py37/1c80f22/ : SUCCESS in 9m 45s\n- openstack-tox-docs http://logs.openstack.org/95/673595/1/gate/openstack-tox-docs/8084de3/html/ : SUCCESS in 5m 12s\n- build-openstack-releasenotes http://logs.openstack.org/95/673595/1/gate/build-openstack-releasenotes/e934c49/html/ : SUCCESS in 3m 56s\n- mistral-devstack http://logs.openstack.org/95/673595/1/gate/mistral-devstack/555a3df/ : SUCCESS in 44m 01s\n- mistral-devstack-non-apache http://logs.openstack.org/95/673595/1/gate/mistral-devstack-non-apache/c9c46af/ : SUCCESS in 47m 59s\n- mistral-tox-unit-mysql http://logs.openstack.org/95/673595/1/gate/mistral-tox-unit-mysql/1469a99/ : SUCCESS in 30m 24s\n- mistral-tox-unit-postgresql http://logs.openstack.org/95/673595/1/gate/mistral-tox-unit-postgresql/3f1033f/ : SUCCESS in 26m 37s\n- mistral-devstack-kombu http://logs.openstack.org/95/673595/1/gate/mistral-devstack-kombu/8f4e19a/ : SUCCESS in 45m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/673595/1/gate/openstack-tox-lower-constraints/9685b9e/ : SUCCESS in 10m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d1a26d8c4a763385a6e9e8a01184fba504fa4b3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 08:04:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c3c3b1b117c3d198ca0af2c25d8fab6d27b36c0","revisions":{"2c3c3b1b117c3d198ca0af2c25d8fab6d27b36c0":{"kind":"REWORK","_number":1,"created":"2019-07-30 17:50:44.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/95/673595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/673595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/673595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/673595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/673595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/673595/1"}}},"commit":{"parents":[{"commit":"c77eeb2d14193a815fa93d6db4f04ac872696ade","subject":"Merge \"Bulk delete of stored delayed calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c77eeb2d14193a815fa93d6db4f04ac872696ade"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-30 17:50:34.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-30 17:50:34.000000000","tz":120},"subject":"Update api-ref location","message":"Update api-ref location\n\nThe api documentation is now published on docs.openstack.org instead\nof developer.openstack.org. Update all links that are changed to the\nnew location.\n\nNote that redirects will be set up as well but let\u0027s point now to the\nnew location.\n\nFor details, see:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-July/007828.html\n\nChange-Id: Id5f629085e0ecfb4fc75ff82c1ed0283f3d2c490\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2c3c3b1b117c3d198ca0af2c25d8fab6d27b36c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2c3c3b1b117c3d198ca0af2c25d8fab6d27b36c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
