)]}'
{"id":"openstack%2Fmistral~674033","triplet_id":"openstack%2Fmistral~master~If8ff2a19068fd3aebb3f9d3a55d9ed6dfaefdb14","project":"openstack/mistral","branch":"master","topic":"bug/manual","hashtags":[],"change_id":"If8ff2a19068fd3aebb3f9d3a55d9ed6dfaefdb14","subject":"A1 B should be tasks rather than workflow levels","status":"MERGED","created":"2019-08-01 13:38:28.000000000","updated":"2019-08-02 07:46:00.000000000","submitted":"2019-08-02 07:46:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674033-1564731960795-520616fe","meta_rev_id":"7060af0dfe19d61967b254f0852da3975cbf9d94","_number":674033,"virtual_id_number":674033,"owner":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-02 06:48:14.000000000","_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-08-02 07:46:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2019-08-02 06:36:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2019-08-02 06:36:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 13:49:49.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-08-02 05:00:43.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-08-02 07:46:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82831dbd7ec73039571c5c5ca4ff660512330c8b","author":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"date":"2019-08-01 13:38:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f13618b66bdaabaef1cafd7b2d77622bd18d738","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-08-01 13:49:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"eda4d74015547d297014ebc01937caa54422e641","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 14:29:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/33/674033/1/check/openstack-tox-pep8/cf6d2cd/ : FAILURE in 5m 05s\n- openstack-tox-docs https://logs.opendev.org/33/674033/1/check/openstack-tox-docs/3a87dce/html/ : SUCCESS in 4m 30s\n- openstack-tox-cover https://logs.opendev.org/33/674033/1/check/openstack-tox-cover/0686b1f/cover/ : SUCCESS in 13m 59s (non-voting)\n- mistral-tox-unit-mysql https://logs.opendev.org/33/674033/1/check/mistral-tox-unit-mysql/95203d9/ : FAILURE in 28m 53s\n- mistral-tox-unit-postgresql https://logs.opendev.org/33/674033/1/check/mistral-tox-unit-postgresql/7c7d5b9/ : SUCCESS in 26m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d3802f8fcc63e58a962f6f15cc38dfa746df9e3f","author":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"date":"2019-08-02 02:51:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cecf43a55dc7f7b5f8c4da55793549ef49b65065","author":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"date":"2019-08-02 03:12:34.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bcfbbc887ef44478f84151b50c638daa5cc8eaee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 03:42:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/33/674033/3/check/openstack-tox-pep8/2104371/ : SUCCESS in 5m 38s\n- openstack-tox-docs https://logs.opendev.org/33/674033/3/check/openstack-tox-docs/ed98de8/html/ : SUCCESS in 4m 56s\n- openstack-tox-cover https://logs.opendev.org/33/674033/3/check/openstack-tox-cover/d32ee4d/cover/ : SUCCESS in 8m 49s (non-voting)\n- mistral-tox-unit-mysql https://logs.opendev.org/33/674033/3/check/mistral-tox-unit-mysql/df952ce/ : SUCCESS in 27m 14s\n- mistral-tox-unit-postgresql https://logs.opendev.org/33/674033/3/check/mistral-tox-unit-postgresql/ad77ae7/ : FAILURE in 28m 21s","accounts_in_message":[],"_revision_number":3},{"id":"c6b167f24faf97734ae69887a379420656589fb8","author":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"date":"2019-08-02 04:42:57.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"63bc54bbe9e0df535425ce060504925fc17c99bc","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-02 05:00:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b213af8c9501dfd543103660df4da6f5ef63f1b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 05:11:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/33/674033/4/check/openstack-tox-pep8/897f015/ : SUCCESS in 4m 28s\n- openstack-tox-docs https://logs.opendev.org/33/674033/4/check/openstack-tox-docs/8580961/html/ : SUCCESS in 4m 06s\n- openstack-tox-cover https://logs.opendev.org/33/674033/4/check/openstack-tox-cover/78bbd66/cover/ : SUCCESS in 8m 23s (non-voting)\n- mistral-tox-unit-mysql https://logs.opendev.org/33/674033/4/check/mistral-tox-unit-mysql/6a28c15/ : FAILURE in 27m 39s\n- mistral-tox-unit-postgresql https://logs.opendev.org/33/674033/4/check/mistral-tox-unit-postgresql/b66d7b7/ : SUCCESS in 27m 04s","accounts_in_message":[],"_revision_number":4},{"id":"1a525e7d0aaee70cb6e0f68a5a85b061c3ab4701","author":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"date":"2019-08-02 06:36:57.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"71e16c07002eb3637f83d06f6cde18e8fee3c3d2","author":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"date":"2019-08-02 06:48:14.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"8bae67f1a6a6be9de6ba4ae058566def993e7083","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 07:10:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/33/674033/5/check/openstack-tox-pep8/efcda1c/ : SUCCESS in 7m 32s\n- openstack-tox-docs https://logs.opendev.org/33/674033/5/check/openstack-tox-docs/0fa02b0/html/ : SUCCESS in 6m 35s\n- openstack-tox-cover https://logs.opendev.org/33/674033/5/check/openstack-tox-cover/d44fc73/cover/ : SUCCESS in 9m 08s (non-voting)\n- mistral-tox-unit-mysql https://logs.opendev.org/33/674033/5/check/mistral-tox-unit-mysql/0b1b4cd/ : SUCCESS in 28m 14s\n- mistral-tox-unit-postgresql https://logs.opendev.org/33/674033/5/check/mistral-tox-unit-postgresql/0402dd0/ : SUCCESS in 30m 12s","accounts_in_message":[],"_revision_number":5},{"id":"090efe44db80fcdbb2c1da9a9f271e7c2364bb29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 07:10:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3f9d0b51a4c9952f8db0ce8cb08176270964d59a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 07:46:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"8348fbf5780ff2e5bf66c59af9922a02b69753e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 07:46:00.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/33/674033/5/gate/openstack-tox-pep8/cc0265d/ : SUCCESS in 6m 58s\n- openstack-tox-docs https://logs.opendev.org/33/674033/5/gate/openstack-tox-docs/492cee0/html/ : SUCCESS in 3m 48s\n- mistral-tox-unit-mysql https://logs.opendev.org/33/674033/5/gate/mistral-tox-unit-mysql/cff614a/ : SUCCESS in 31m 25s\n- mistral-tox-unit-postgresql https://logs.opendev.org/33/674033/5/gate/mistral-tox-unit-postgresql/8f9de81/ : SUCCESS in 29m 18s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1a2ff63733e53ee39b59f877c56065afb443343a","revisions":{"ce65ecda2c183104c3272ec5f549e34c3590155a":{"kind":"REWORK","_number":1,"created":"2019-08-01 13:38:28.000000000","uploader":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"ref":"refs/changes/33/674033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/33/674033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/674033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/674033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/674033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/33/674033/1"}}},"commit":{"parents":[{"commit":"276d088b0957c1d3b5924934b47ca40d54697577","subject":"Merge \"Update api-ref location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/276d088b0957c1d3b5924934b47ca40d54697577"}]}],"author":{"name":"TitanLi","email":"lisheng0706@gmail.com","date":"2019-08-01 13:32:38.000000000","tz":480},"committer":{"name":"TitanLi","email":"lisheng0706@gmail.com","date":"2019-08-01 13:32:38.000000000","tz":480},"subject":"A1, B grade should be a task rather than a workflow level","message":"A1, B grade should be a task rather than a workflow level\n\nChange-Id: If8ff2a19068fd3aebb3f9d3a55d9ed6dfaefdb14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ce65ecda2c183104c3272ec5f549e34c3590155a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ce65ecda2c183104c3272ec5f549e34c3590155a"}]},"branch":"refs/heads/master"},"53214acc20e02892df8a16aa244ba856006eb931":{"kind":"REWORK","_number":2,"created":"2019-08-02 02:51:05.000000000","uploader":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"ref":"refs/changes/33/674033/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/33/674033/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/674033/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/674033/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/674033/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/33/674033/2"}}},"commit":{"parents":[{"commit":"276d088b0957c1d3b5924934b47ca40d54697577","subject":"Merge \"Update api-ref location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/276d088b0957c1d3b5924934b47ca40d54697577"}]}],"author":{"name":"TitanLi","email":"lisheng0706@gmail.com","date":"2019-08-01 13:32:38.000000000","tz":480},"committer":{"name":"TitanLi","email":"lisheng0706@gmail.com","date":"2019-08-02 02:38:00.000000000","tz":480},"subject":"A1,B should be tasks rather than workflow levels","message":"A1,B should be tasks rather than workflow levels\n\nrecheck no bug\nIn Mistral Workflow Language v2 specification document\nTask A1 and Task B should be in the task area otherwise there will be an error\n\nChange-Id: If8ff2a19068fd3aebb3f9d3a55d9ed6dfaefdb14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/53214acc20e02892df8a16aa244ba856006eb931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/53214acc20e02892df8a16aa244ba856006eb931"}]},"branch":"refs/heads/master"},"a13396e488fca6713ad1c66b1d73c3bde96904f8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-02 03:12:34.000000000","uploader":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"ref":"refs/changes/33/674033/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/33/674033/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/674033/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/674033/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/674033/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/33/674033/3"}}},"commit":{"parents":[{"commit":"276d088b0957c1d3b5924934b47ca40d54697577","subject":"Merge \"Update api-ref location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/276d088b0957c1d3b5924934b47ca40d54697577"}]}],"author":{"name":"TitanLi","email":"lisheng0706@gmail.com","date":"2019-08-01 13:32:38.000000000","tz":480},"committer":{"name":"TitanLi","email":"lisheng0706@gmail.com","date":"2019-08-02 03:11:35.000000000","tz":480},"subject":"A1,B should be tasks rather than workflow levels","message":"A1,B should be tasks rather than workflow levels\n\nrecheck no bug\nIn Mistral Workflow Language v2 specification document\nTask A1 and Task B should be in the task area\notherwise there will be an error\n\nChange-Id: If8ff2a19068fd3aebb3f9d3a55d9ed6dfaefdb14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a13396e488fca6713ad1c66b1d73c3bde96904f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a13396e488fca6713ad1c66b1d73c3bde96904f8"}]},"branch":"refs/heads/master"},"c5806aabf2636678c35a6a4771b6c6e005c3914f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-08-02 04:42:57.000000000","uploader":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"ref":"refs/changes/33/674033/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/33/674033/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/674033/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/674033/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/674033/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/33/674033/4"}}},"commit":{"parents":[{"commit":"276d088b0957c1d3b5924934b47ca40d54697577","subject":"Merge \"Update api-ref location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/276d088b0957c1d3b5924934b47ca40d54697577"}]}],"author":{"name":"TitanLi","email":"lisheng0706@gmail.com","date":"2019-08-01 13:32:38.000000000","tz":480},"committer":{"name":"TitanLi","email":"lisheng0706@gmail.com","date":"2019-08-02 04:41:22.000000000","tz":480},"subject":"A1 B should be tasks rather than workflow levels","message":"A1 B should be tasks rather than workflow levels\n\nrecheck no bug A1 and B should be in the task area\n\nChange-Id: If8ff2a19068fd3aebb3f9d3a55d9ed6dfaefdb14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c5806aabf2636678c35a6a4771b6c6e005c3914f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c5806aabf2636678c35a6a4771b6c6e005c3914f"}]},"branch":"refs/heads/master"},"1a2ff63733e53ee39b59f877c56065afb443343a":{"kind":"NO_CHANGE","_number":5,"created":"2019-08-02 06:36:57.000000000","uploader":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"ref":"refs/changes/33/674033/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/33/674033/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/674033/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/674033/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/674033/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/33/674033/5"}}},"commit":{"parents":[{"commit":"276d088b0957c1d3b5924934b47ca40d54697577","subject":"Merge \"Update api-ref location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/276d088b0957c1d3b5924934b47ca40d54697577"}]}],"author":{"name":"TitanLi","email":"lisheng0706@gmail.com","date":"2019-08-01 13:32:38.000000000","tz":480},"committer":{"name":"TitanLi","email":"lisheng0706@gmail.com","date":"2019-08-02 04:43:19.000000000","tz":480},"subject":"A1 B should be tasks rather than workflow levels","message":"A1 B should be tasks rather than workflow levels\n\nrecheck no bug A1 and B should be in the task area\n\nChange-Id: If8ff2a19068fd3aebb3f9d3a55d9ed6dfaefdb14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1a2ff63733e53ee39b59f877c56065afb443343a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1a2ff63733e53ee39b59f877c56065afb443343a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
