)]}'
{"id":"openstack%2Fmistral~674053","triplet_id":"openstack%2Fmistral~master~I8f2054c820f7774eb84155f1843cc871c87ca823","project":"openstack/mistral","branch":"master","topic":"bug/manual","hashtags":[],"change_id":"I8f2054c820f7774eb84155f1843cc871c87ca823","subject":"ref:https://specs.openstack.org/openstack/mistral-specs/specs/pike/approved/advanced_publishing.html","status":"ABANDONED","created":"2019-08-01 14:35:39.000000000","updated":"2019-08-02 02:10:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f46f19df4a98d7adf36dc3491429e6f249330939","_number":674053,"virtual_id_number":674053,"owner":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-08-01 18:23:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 18:23:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b232edb01847121ffad6bb390b75a47c23b2e23c","author":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"date":"2019-08-01 14:35:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f48d452c1d207f30611d26fa725c3c3cfe78d12a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 18:23:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/53/674053/1/check/openstack-tox-pep8/6ed925d/ : FAILURE in 5m 24s\n- openstack-tox-py27 https://logs.opendev.org/53/674053/1/check/openstack-tox-py27/e5cde88/ : SUCCESS in 9m 52s\n- openstack-tox-py36 https://logs.opendev.org/53/674053/1/check/openstack-tox-py36/3ad9aba/ : SUCCESS in 9m 13s\n- openstack-tox-py37 https://logs.opendev.org/53/674053/1/check/openstack-tox-py37/58b93bf/ : SUCCESS in 9m 05s\n- openstack-tox-docs https://logs.opendev.org/53/674053/1/check/openstack-tox-docs/2146aa0/html/ : SUCCESS in 4m 50s\n- openstack-tox-cover https://logs.opendev.org/53/674053/1/check/openstack-tox-cover/1040023/cover/ : SUCCESS in 10m 06s (non-voting)\n- mistral-devstack https://logs.opendev.org/53/674053/1/check/mistral-devstack/49145e1/ : SUCCESS in 54m 14s\n- mistral-devstack-non-apache https://logs.opendev.org/53/674053/1/check/mistral-devstack-non-apache/5fbad72/ : SUCCESS in 52m 26s\n- mistral-devstack-kombu https://logs.opendev.org/53/674053/1/check/mistral-devstack-kombu/0210452/ : SUCCESS in 53m 18s\n- mistral-tox-unit-mysql https://logs.opendev.org/53/674053/1/check/mistral-tox-unit-mysql/d75b6c1/ : SUCCESS in 28m 25s\n- mistral-tox-unit-postgresql https://logs.opendev.org/53/674053/1/check/mistral-tox-unit-postgresql/a026e91/ : SUCCESS in 27m 20s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/53/674053/1/check/tripleo-ci-centos-7-scenario003-standalone/dc7a9fe/ : SUCCESS in 1h 24m 02s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/53/674053/1/check/tripleo-ci-centos-7-containers-multinode/3ecab38/ : SUCCESS in 2h 31m 49s\n- mistral-rally-task https://logs.opendev.org/53/674053/1/check/mistral-rally-task/48fad72/ : SUCCESS in 1h 11m 00s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/53/674053/1/check/openstack-tox-lower-constraints/6fba0a2/ : SUCCESS in 15m 43s","accounts_in_message":[],"_revision_number":1},{"id":"56a3faab28cb2d5fb9ce4941ef68a06af6ba634c","author":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"date":"2019-08-02 02:10:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cbf6356b869a6e96c598785176950fc306c6ff25","revisions":{"cbf6356b869a6e96c598785176950fc306c6ff25":{"kind":"REWORK","_number":1,"created":"2019-08-01 14:35:39.000000000","uploader":{"_account_id":29811,"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","username":"TitanLi"},"ref":"refs/changes/53/674053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/53/674053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/674053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/674053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/53/674053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/53/674053/1"}}},"commit":{"parents":[{"commit":"ce65ecda2c183104c3272ec5f549e34c3590155a","subject":"A1, B grade should be a task rather than a workflow level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ce65ecda2c183104c3272ec5f549e34c3590155a"}]}],"author":{"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","date":"2019-08-01 14:35:39.000000000","tz":0},"committer":{"name":"WEN-SHENG,LI","email":"lisheng0706@gmail.com","date":"2019-08-01 14:35:39.000000000","tz":0},"subject":"ref:https://specs.openstack.org/openstack/mistral-specs/specs/pike/approved/advanced_publishing.html","message":"ref:https://specs.openstack.org/openstack/mistral-specs/specs/pike/approved/advanced_publishing.html\n\nChange-Id: I8f2054c820f7774eb84155f1843cc871c87ca823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cbf6356b869a6e96c598785176950fc306c6ff25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cbf6356b869a6e96c598785176950fc306c6ff25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
